)]}'
{"id":"openstack%2Fneutron~891086","triplet_id":"openstack%2Fneutron~master~Ia144503a1a9417558019e961986f16b6fc5228b8","project":"openstack/neutron","branch":"master","topic":"dnm/sql_master_b","attention_set":{},"removed_from_attention_set":{"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-08-10 20:12:49.000000000","reason":"\u003cGERRIT_ACCOUNT_34271\u003e replied on the change","reason_account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}}},"hashtags":[],"change_id":"Ia144503a1a9417558019e961986f16b6fc5228b8","subject":"[DNM] test with sql master branches","status":"ABANDONED","created":"2023-08-10 17:27:19.000000000","updated":"2023-08-15 17:36:25.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bdd2b82ea89a4db9d504e8bbcff1c08e1c74e07f","_number":891086,"virtual_id_number":891086,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-10 21:23:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-10 17:27:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 17:27:57.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"dbd5f5fb16f43241afc4c941c0a9241a3cad2519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-10 17:27:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f2b282e1c0c133279bc1a21cb2c06ee611aace8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 17:27:50.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 891086,1","accounts_in_message":[],"_revision_number":1},{"id":"cf2aa653fb920c4174c19141ec9bda0a78c4de02","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-10 17:27:57.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 891086,1","accounts_in_message":[],"_revision_number":1},{"id":"f0c05c5e64c5b2b2419860fb4068bf2caddd6bce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 17:27:58.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 891086,1","accounts_in_message":[],"_revision_number":1},{"id":"7cfaa7ec51360a1b3e76487ade53d74a83c86084","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-10 17:29:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"413f0d439de716093e89fbf9df48439d845ce140","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 17:29:58.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/neutron on branch master.  The error was:\n  \n    found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 18, column 31","accounts_in_message":[],"_revision_number":1},{"id":"9aa9c2be84b3217f0fe4f22795e3a49982eb6b32","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-10 17:30:01.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/neutron/zuul.d/project.yaml@master\", line 18, column 31","accounts_in_message":[],"_revision_number":1},{"id":"cc7756baaa44adc3847f529855f0e2634b5ac3e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 17:30:02.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/neutron on branch master.  The error was:\n  \n    found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 18, column 31","accounts_in_message":[],"_revision_number":1},{"id":"eacbdba6294101b26ae03cf71316e87f484391d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-10 18:54:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"743e607a39811a5a1e126fa9de9e07ae37a1e16d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 20:00:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4164b556b8ac41be88a0ec38c72d577b\n\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2f4fb10661104993bd287958de203994 : FAILURE in 1h 01m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e526ce0639c833241f339ef94508555041b24296","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-10 20:12:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd792d91d76278b74bb4b6e3a36ae375247545d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 21:23:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdacb3d9a8004c78afd901084c6167a6\n\n- neutron-functional-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/300868b5f5434b6fa4df4a86f4e3d64c : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":2},{"id":"bdd2b82ea89a4db9d504e8bbcff1c08e1c74e07f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-08-15 17:36:25.000000000","message":"Abandoned\n\nonly used for testing","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ca4bde28b6eb57b21a1aefa6668dbbcb1ff6f988","revisions":{"5a3b12133a13a68c7d778327bd12d0bfad2515d8":{"kind":"REWORK","_number":1,"created":"2023-08-10 17:27:19.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/891086/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/891086/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/891086/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/891086/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/891086/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/891086/1"}}},"commit":{"parents":[{"commit":"5813a83487d3d6d545b8a2ef019731029da46b56","subject":"Merge \"ovs-agent: React to DB down just like to server down\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5813a83487d3d6d545b8a2ef019731029da46b56"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-10 17:24:11.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-10 17:24:11.000000000","tz":-240},"subject":"[DNM] test with sql master branches","message":"[DNM] test with sql master branches\n\nDepends-On: I0abc9c0dfc65e3e031246f514ff6e1e3593a00f4\nChange-Id: Ia144503a1a9417558019e961986f16b6fc5228b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a3b12133a13a68c7d778327bd12d0bfad2515d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5a3b12133a13a68c7d778327bd12d0bfad2515d8"}]},"branch":"refs/heads/master"},"ca4bde28b6eb57b21a1aefa6668dbbcb1ff6f988":{"kind":"REWORK","_number":2,"created":"2023-08-10 18:54:18.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/86/891086/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/86/891086/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/891086/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/891086/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/86/891086/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/86/891086/2"}}},"commit":{"parents":[{"commit":"5813a83487d3d6d545b8a2ef019731029da46b56","subject":"Merge \"ovs-agent: React to DB down just like to server down\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5813a83487d3d6d545b8a2ef019731029da46b56"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-10 17:24:11.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-10 18:54:06.000000000","tz":-240},"subject":"[DNM] test with sql master branches","message":"[DNM] test with sql master branches\n\nDepends-On: I0abc9c0dfc65e3e031246f514ff6e1e3593a00f4\nChange-Id: Ia144503a1a9417558019e961986f16b6fc5228b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca4bde28b6eb57b21a1aefa6668dbbcb1ff6f988"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca4bde28b6eb57b21a1aefa6668dbbcb1ff6f988"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
