)]}'
{"id":"openstack%2Fneutron~891276","triplet_id":"openstack%2Fneutron~stable%2Fxena~Id5bc24b76b7aa510d4066fbc5f7b8037cc8c740d","project":"openstack/neutron","branch":"stable/xena","topic":"bug/2029297-stable/2023.1-stable/zed-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-08-18 23:52:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5bc24b76b7aa510d4066fbc5f7b8037cc8c740d","subject":"hash-ring: Retry all DB operations if inactive","status":"MERGED","created":"2023-08-14 12:53:27.000000000","updated":"2023-08-18 23:53:39.000000000","submitted":"2023-08-18 23:52:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891276-bug/2029297-stable/2023.1-stable/zed-stable/yoga-stable/xena","cherry_pick_of_change":891275,"cherry_pick_of_patch_set":1,"meta_rev_id":"3564be2f08b5720e157ccb7f6048d6114bcc47fd","_number":891276,"virtual_id_number":891276,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-18 23:52:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-08-17 14:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-08-18 14:34:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-08-18 22:24:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-14 13:32:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-14 15:52:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-17 14:36:08.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-08-18 14:34:05.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"801a03b22e841913a1a37cc816c3b04bae58572f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-08-14 12:53:27.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"e20de13f3ed1d23a3a2f37097246daa2deece765","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 13:32:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60678e00402446f1a3e152627f69bbb8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/96f5f40d13d04e52bb93cded32da114f : SUCCESS in 19m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92c41517254c46f9992bf444c4d5adc2 : SUCCESS in 37m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a379e89ef84af4ef4f55c115f157b040d61ca9d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 15:52:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f584114bb51474ea0ae5f7fe10096ba\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/c84403c9987544a989879d49b389eacb : SUCCESS in 45m 00s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/70e002bad3f3426bb5042ff8872135e1 : SUCCESS in 1h 52m 47s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/245b344d9388432f987da215866fba25 : TIMED_OUT in 2h 49m 43s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/5c5cdd11faf24880a7b15eebb8a4aa7d : SUCCESS in 1h 30m 48s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/776d4385f81049ef9d736b0dfd76034c : SUCCESS in 1h 42m 28s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/4075cf432cb54b64a1c3fdde97c2badb : SUCCESS in 38m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f1ce3894d9b498c9e829654a064c054 : SUCCESS in 33m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7ff12e43344a25a8935468377dd39b : SUCCESS in 12m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23a0b070bdfb44939ba678c80be4e92d : SUCCESS in 30m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0337e085d99c46b788b2a4ad5b47b10e : SUCCESS in 19m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdeac2f353014966b0a835a4c51dce20 : SUCCESS in 22m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a70bed50e37c4db28c88bc507b6a6c44 : SUCCESS in 9m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ec92710d2dc4bbab4abe2f481678151 : SUCCESS in 44m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d12517b6885f44b7a8e9eec64274e5e0 : FAILURE in 1h 05m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d28d0bb526834b1b9ed5c283fa8f6f5c : SUCCESS in 59m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9bc41bbc99c34b23b9dff2c06dff4d9b : SUCCESS in 53m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/69ace55a72d74385bd6143d892a1df40 : SUCCESS in 49m 35s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/17c1d7aa6c0e417bb13c40027bd0f304 : FAILURE in 58m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1e094e7008c444ad899dd89e4c9ea9a3 : SUCCESS in 1h 13m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2eb924baf9944ea1be5bcd77d3ccb9db : SUCCESS in 53m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f271e0f4627a49cab643aa1487feaa84 : SUCCESS in 1h 05m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da5da4cccbb538a7a2c3ecf34d909f305e6003b2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-17 14:36:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8abc0a2b2c10b07a8b78ab89865efa9ecb1dfba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 15:25:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e4600ab20e54d8485450e03f1589631\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccb2a6d3b609417a8ece7ea8395daf35 : SUCCESS in 17m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e5d63a4781b546a69702081d1b445acf : SUCCESS in 47m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e48da38acf6c4805f5e32c38734c2e2375ae7091","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 17:25:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61ebcb37524048f0bc649c600f296d5a\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/579c82a0cfba4ab5aacda7df38aad761 : SUCCESS in 47m 48s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/16f1694a586d4819bd32e31580676613 : SUCCESS in 2h 21m 40s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/d5547a40b929474db594124296a0c097 : SUCCESS in 2h 46m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/9aa5e4ac99ba4ef896cf43f806513277 : SUCCESS in 1h 46m 42s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/5ce662bbcc134a3aa05b9eb99475656e : SUCCESS in 1h 46m 41s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/21cf7806169343efaf607a401c5be12c : TIMED_OUT in 1h 02m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85484fd212504daca011ee39b49e57a5 : SUCCESS in 35m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f35ca555aa34a73b9fe4ba69ad84507 : SUCCESS in 14m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a3e8e6150824aa5975427f99d353011 : SUCCESS in 25m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/136e149d11db4dbdafa98eeeae45c238 : SUCCESS in 25m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e26eac34ec94080a6d0e1d76c896aa0 : SUCCESS in 26m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/318a246e0aa1456e9a42c5bb44795248 : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b9c8cec901074f6193e54d90df18143e : SUCCESS in 57m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad8b7d4b82194c4a9c840b020aa7d60b : SUCCESS in 1h 24m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f0d7ee601e1f4ac5af15c9aac751cdb3 : SUCCESS in 1h 16m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/57e8e54dd64149e4ab19c38c67e7ca28 : SUCCESS in 1h 01m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/86ac4f1c9b6b48d0badb466ecca95626 : SUCCESS in 1h 03m 08s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0d1030f69d304519a8ada1939204bbf6 : SUCCESS in 1h 27m 57s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e444ef1caa7a4fcba30d4cd2d7477767 : SUCCESS in 1h 21m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d559008f38f641d7ad0c29ebcd6aed9c : SUCCESS in 1h 05m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ca56258307e3426782742ef63ddd8d17 : SUCCESS in 1h 03m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ead1f59df36db423067f8702474eb26babd31133","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-18 14:34:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b3fbf5509516265d0f287681bc9f0ae52a666c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 14:59:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bd7905026c64720858f0eb518d3ade2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6f55ce225f5476f9205bfc8339001aa : SUCCESS in 18m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b83cc7b34d98417398047a7eefe7af64 : SUCCESS in 18m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e614ef953f4e749a685c393f0ed7c549ab97c4ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 16:27:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69111e32835f475ca45180b42ed8cdae\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/e44f9a5b1bd74e8cb214c0d4624d7c13 : SUCCESS in 51m 20s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/8e3f3e69bd994fbe949405f8928fb83a : SUCCESS in 1h 41m 10s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/4aa2ebabd94e40b38773b299ea1226bc : SUCCESS in 1h 52m 37s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/6de96a5b9a2543f88c8b4ec62d233f6f : SUCCESS in 1h 21m 04s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/2de39d48c9d5438badc94cb833e6a356 : SUCCESS in 1h 36m 25s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/50cb78980d62446d9efe22e409149ab1 : SUCCESS in 36m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de49707a4ba04c58a1521cfa5a1e2d59 : SUCCESS in 38m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab6b345a604c4be1b44eb8fae346330a : SUCCESS in 15m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3638a5d1cd3f49ce8dfc153215bac6e1 : SUCCESS in 29m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb36743790944fd0b077ba6fe466c445 : SUCCESS in 26m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a7504f289d5453e90d3d7f7d7b64065 : SUCCESS in 24m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a702605491fc4c4795b34abac696b572 : SUCCESS in 10m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6c144b1e10a4d988eee98408534d08e : FAILURE in 44m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/561c4c11d6474404b7eede124a059c47 : SUCCESS in 1h 26m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bfd4aa1cb78e4b788cf88fc44b52cfe6 : SUCCESS in 1h 18m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1dc7638b0b040eb997c927febe4814f : SUCCESS in 54m 56s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ef121571a11748fdb3f3651ff9116267 : SUCCESS in 1h 12m 19s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ce74c9e43f0942dba01dc247b98fb253 : SUCCESS in 1h 16m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/767e5b7b4bd044e193ba6b20d71952da : SUCCESS in 1h 22m 39s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/811c03909ac64ff29f40653badcf5627 : SUCCESS in 44m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4a08b7ec88c64ee498e639989e83f857 : SUCCESS in 52m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22bd1c9edb7493a4329b5d2af21a29aa7dbd08f0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-18 18:41:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e569d0a0c0fcd1c4b3eb5728c5b1c214ee8c2310","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 19:30:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/742dd8c6487d46909e550be266265a3c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c772cf317c246c8bc4cb6f639d55855 : SUCCESS in 47m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5be9a388593043d0bbd56d29813d00da : SUCCESS in 17m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b721867ffac8bb2c8d2a8ebd4bac26eaa68d86c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 20:53:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee34337e72cb4eea8f00fa95f3e963bc\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/d422cf2f2879485d86ec12b47cf475ad : SUCCESS in 39m 19s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/33e1383ed5d2442990e76e60a6c2b74c : SUCCESS in 1h 31m 39s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/2a6acce50d5e4c56a8160444a59d3740 : SUCCESS in 1h 34m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/8beae1901632431a93268d7898a6c886 : SUCCESS in 1h 32m 30s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/318a479e4dc44d3fbf7f47bcdf9de0e0 : SUCCESS in 2h 05m 36s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/af01d73cf558423cb43249abed8a2fe9 : SUCCESS in 29m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d635fef0487441bf9ae79ed9149d3a71 : SUCCESS in 27m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dde8aec41b54789a55d70587901c411 : SUCCESS in 11m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80f6ba1fd7444c99a122d98622dd454c : SUCCESS in 23m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/362ba4cbad26438e8372fc504e59297e : SUCCESS in 20m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a704ade4d26a409a90cbdc69db9b3c0c : SUCCESS in 21m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b2f47b247f4173a7e589608f7fa2e4 : SUCCESS in 10m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4bd2aeec47de444fa7ba0b303cb07690 : SUCCESS in 49m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a48658fdfb014e44b7936655834bd5ee : SUCCESS in 1h 28m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b1be655ba2cc45978ca831e933bec495 : SUCCESS in 1h 14m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/35f83481db094c9791e909ccb837e4d6 : SUCCESS in 1h 05m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/daa030dfbc124523a82d977d67c97cd7 : SUCCESS in 52m 04s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eb7bd23cfd0f45afba95a30834b333aa : SUCCESS in 1h 09m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c5a90b9c324f4a78a1d9476bf05504a4 : SUCCESS in 1h 19m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcbdcfd3250b44fca0b5e9ee98057041 : SUCCESS in 49m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/877f788308064bc28de3d0e1294fafd6 : SUCCESS in 56m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4e891e35a559aceedbb85b3e5f52d255a454b57","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-18 22:24:00.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cbad3aa5dce7bbdc2cdd7d82f77ba29dca73fcf5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 22:24:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f33c4c074af563808ff0d67caffe0099cad34d07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 23:52:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bc21da05c3f468984dce0cf9fd4339d\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/5121faae7b3c4537be34a497913ad66e : SUCCESS in 57m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45cd097caf9347b09dd9a6929ac61a28 : SUCCESS in 13m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46c46ce9a61d4279b30ed020d58a6437 : SUCCESS in 27m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c27ce17339243b4a9318040f15b4e14 : SUCCESS in 23m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a882f55d4336426490d47f792f75cb93 : SUCCESS in 9m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7187178c95c54ef3a8901cdca0ffabeb : SUCCESS in 50m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e6ff3412c5374e7282a25bb64ceb599f : SUCCESS in 1h 27m 08s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/32ae1d14946648c39ec4798bb16170e8 : SUCCESS in 57m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/23feb0a6156d4b72b230ac9e564ce8c1 : SUCCESS in 51m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2415762569144c1bb6009a07e5f041b3 : SUCCESS in 49m 19s","accounts_in_message":[],"_revision_number":1},{"id":"342dffc9f1a3d41041a05223200cf48764d5aa86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 23:52:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3564be2f08b5720e157ccb7f6048d6114bcc47fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 23:53:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd3e50ed97594d7aa3d8d9bb35072a9a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d122e9a365784ce89dfb37ed427122a8 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20337b5f8a6b5033c99e7a12351befaac3b43b35","revisions":{"20337b5f8a6b5033c99e7a12351befaac3b43b35":{"kind":"REWORK","_number":1,"created":"2023-08-14 12:53:27.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/76/891276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/891276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/891276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/891276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/891276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/891276/1"}}},"commit":{"parents":[{"commit":"51d83673c2aedaeee6f87f42d7e8749a5b4ffc46","subject":"Merge \"[OVN] Hash Ring: Better handle Neutron worker failures\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51d83673c2aedaeee6f87f42d7e8749a5b4ffc46"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-07-28 01:38:42.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-08-14 12:53:27.000000000","tz":0},"subject":"hash-ring: Retry all DB operations if inactive","message":"hash-ring: Retry all DB operations if inactive\n\nNeutron should be resilient to scenarios where connectivity to a service\nNeutron depends on is lost. SQL database is one of it and hash ring DB\noperations is used very early in the Neutron startup process.\n\nIf the connection is lost right after new API worker was spawned then\nOVN IDLs are not instantiated and Neutron silently hangs on each request\nthat uses OVN DB because the event signaling OVN is ready never happens.\n\nCloses-bug: #2029297\nChange-Id: Id5bc24b76b7aa510d4066fbc5f7b8037cc8c740d\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n(cherry picked from commit a505ff7dbb4f0b7f5487fa8389842dd3ee583d50)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20337b5f8a6b5033c99e7a12351befaac3b43b35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/20337b5f8a6b5033c99e7a12351befaac3b43b35"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
