)]}'
{"id":"openstack%2Fneutron~891295","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I52ec4b346271889ebaa7b7f84981eae5503d02d3","project":"openstack/neutron","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-08-22 13:49:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I52ec4b346271889ebaa7b7f84981eae5503d02d3","subject":"[OVN] Retry retrieving LSP hosting information","status":"MERGED","created":"2023-08-14 12:28:59.000000000","updated":"2023-08-22 13:51:11.000000000","submitted":"2023-08-22 13:49:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891295","meta_rev_id":"5bf712428c91049cddf95eeeedad9a143386d285","_number":891295,"virtual_id_number":891295,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 13:49:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-08-22 12:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-08-14 15:10:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-21 13:27:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-08-22 12:00:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-14 13:14:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-14 14:15:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-14 15:10:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-08-21 13:27:16.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2023-08-22 12:00:09.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"e2c81cf8c3ae7cbc35c45cff2c8dbc6421e1de7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-14 12:28:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e73c366ec65bf6827629535fef43ea25ec021f3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 13:14:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1861b434ba95443db6deb85274aebef5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/728b1527702047d7a11858923816119b : SUCCESS in 20m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fecb9b2e0174e77ab98d9acb8a45a17 : SUCCESS in 43m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb7bbe7df8ba8ed47a08a505227d5d3b1f4a8eb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 14:15:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20551088a76d4227810929b9915e5960\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/9b632e03dae94cbeb5481454504bba39 : SUCCESS in 49m 53s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/ba70a9b3bbba4d3499a63059af855b48 : SUCCESS in 1h 10m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbec0ada90a14b53a6bd8ba3afee78b8 : SUCCESS in 38m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56868927ac3747038bb1f7ed3b05f6bb : SUCCESS in 14m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dc2edaf52a94329a98b008e93cd5fd3 : SUCCESS in 28m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5ccd424a95740d6ba2bad9b79481fcf : SUCCESS in 26m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b92041f75ea40a6be42b25271942569 : SUCCESS in 11m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d23614f6c46e476ba464785f01c095dd : SUCCESS in 1h 11m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f74d3b560bb48e482364804a306d8ad : SUCCESS in 1h 03m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d3a8fceae3774d08a71a41acec345b63 : SUCCESS in 1h 39m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f34d6d38ebd4ac7998cbd856666b5da004d4850a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-14 15:10:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9b6dc0615cb251c476f8e3b5d86c4872c8eb55a7","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-08-21 13:27:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aa310d31803d07d21c3a23a81ff2dd614f0bd6e9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-08-22 12:00:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d8302bf8221bc263662d75c4c58f98290bfc4ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 12:00:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1d5b0cb1e60b0d7c25b1a45fa752bf24110ea4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 13:49:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/116163cb7ab64b0399bb319ef4120934\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/0ca4710b7b3a4bcb829e728ad6b92cbe : SUCCESS in 53m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2adf5611ed0b48d0b4f3e4db6dc527d5 : SUCCESS in 11m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41026979f70b4b3f84363dd56fc30ede : SUCCESS in 28m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63482bd69f3e48a99e17277ee8c0847a : SUCCESS in 35m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f581bc0e354895b57bcfce053c7ae9 : SUCCESS in 10m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1259c3538a44291b9bf87629e5bd1eb : SUCCESS in 1h 10m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96e70915bda64c5984d03f44beb45a5a : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":1},{"id":"686a8921e2c0db2ba8060ad1f00775ccfb62de15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 13:49:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5bf712428c91049cddf95eeeedad9a143386d285","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 13:51:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1d9cb88a8d64cd485b647a89f7e698b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c9395aba1a0488fae677998ae36b3a2 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10a981a1589a82f739980807475cea8151fa9a70","revisions":{"10a981a1589a82f739980807475cea8151fa9a70":{"kind":"REWORK","_number":1,"created":"2023-08-14 12:28:59.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/95/891295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/95/891295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/891295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/891295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/95/891295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/95/891295/1"}}},"commit":{"parents":[{"commit":"3944dff994dc8827bb2405dc11b3e42230520e3e","subject":"Merge \"[OVN] Prevent Trunk creation/deletion with parent port bound\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3944dff994dc8827bb2405dc11b3e42230520e3e"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-07-20 09:25:23.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-14 12:22:42.000000000","tz":60},"subject":"[OVN] Retry retrieving LSP hosting information","message":"[OVN] Retry retrieving LSP hosting information\n\nThere\u0027s a sync issue while trying to fetch the hosting information for\nthe LSP before we write it to the OVN database, sometimes the\ninformation is not yet present and we end up with an empty string (\"\")\nfor the host attribute of portbindings. This patch adds a retry\nmechanism to solve this sync issue.\n\nConflicts:\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n  neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/ovsdb/test_ovn_client.py\n\nChange-Id: I52ec4b346271889ebaa7b7f84981eae5503d02d3\nRelated-Bug: #2020058\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 3044b938b9fa7dfef126db9acc18782978238dc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10a981a1589a82f739980807475cea8151fa9a70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10a981a1589a82f739980807475cea8151fa9a70"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
