)]}'
{"id":"openstack%2Fneutron~891502","triplet_id":"openstack%2Fneutron~stable%2Fzed~Id7bf5a6aa547795ba78724eed59ba9d4fb74f758","project":"openstack/neutron","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-08-21 19:51:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7bf5a6aa547795ba78724eed59ba9d4fb74f758","subject":"[OVN] ovn-db-sync check for router port differences","status":"MERGED","created":"2023-08-15 14:39:16.000000000","updated":"2023-08-21 19:53:15.000000000","submitted":"2023-08-21 19:51:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891502","meta_rev_id":"c9e2cef4f564bfbeacdf009f5e69c6c3ccf75dbe","_number":891502,"virtual_id_number":891502,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-21 19:51:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-08-18 09:27:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-08-21 17:58:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-08-21 17:58:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-15 15:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-15 19:24:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-18 09:27:03.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-08-21 17:58:29.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"c80810a8c6d09bc7c61ac6e94c736911e8c475b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-15 14:39:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1098e22c6db865b7337396b397468bba99b1240","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 15:22:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e75ca590484d4568999f78b080a64a9d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bad50b7cd5c340dea9ccbc024a624997 : SUCCESS in 20m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b62bb8af29e949caa4782605afbe7fa3 : SUCCESS in 29m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0798a11881d73f6833be5baff2b7d9d1843779de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 19:24:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/089e7fae63114d539a74f04ed13c755c\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/a3c265d8f23b4761b72f1411baa1ad37 : SUCCESS in 1h 17m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b51d42ee44cf434baefc1119f854fcd7 : SUCCESS in 36m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8e9dc09741146e586796fd907785682 : SUCCESS in 15m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0c87b053efa4a91887f7edd907696e9 : SUCCESS in 26m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f768ca149e6143338dbf7cb9e5f311a0 : SUCCESS in 31m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/366beab5833147f98f13ce1651af183b : SUCCESS in 27m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74539b9d054d4709b9254c41b21ebfa2 : SUCCESS in 12m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/08698d446efd4744bbc31cd8b0150b2d : SUCCESS in 55m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3988c53569b4d588efff01870f36a4b : SUCCESS in 1h 05m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2f3d134b18594192bda29140b291214f : SUCCESS in 1h 38m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71d597a0bae762bf256410e5fe5610f04a3bdd45","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-18 09:27:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b173973fd61a21c01cdb48f414403cb382816a8f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-21 17:58:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"66a32136002faaf7d7d2ba147fc6bd82f0342c64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 17:59:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56c7ad573d7fe376eb68485469612abf0bfc4aeb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 19:51:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c3a4d3204484228bafc231df3e9de17\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/24f18710d0754ea98fca5753a8a5d0cb : SUCCESS in 1h 41m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27ffb35c456b4a32814ec6e367fea95a : SUCCESS in 11m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9675ca3e0f214834b05e72c1042f080a : SUCCESS in 27m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/489c33618ed14fc0986c4b3f4f8040e9 : SUCCESS in 26m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87a8f0e47d2a4383a4a889a8916b0f2e : SUCCESS in 8m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/01260a398eaa448b9219c2641527f158 : SUCCESS in 1h 02m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/714b9216026949318f355dbbcbb3451f : SUCCESS in 51m 31s","accounts_in_message":[],"_revision_number":1},{"id":"266fdebcef31b9d2eec30e52baea18105e45c413","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 19:51:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c9e2cef4f564bfbeacdf009f5e69c6c3ccf75dbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 19:53:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1e8a944f4104d0dabed907cae32c6d1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82027fb1687c4bf294424ce4f937be5f : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc71377ba7f8c99910565969a462c3bd5b8b4379","revisions":{"bc71377ba7f8c99910565969a462c3bd5b8b4379":{"kind":"REWORK","_number":1,"created":"2023-08-15 14:39:16.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/02/891502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/891502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/891502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/891502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/891502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/891502/1"}}},"commit":{"parents":[{"commit":"8c4a603c6091202769294d6bea378bc5d87fc759","subject":"Merge \"dvr: Avoid installing non-dvr openflow rule on startup\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c4a603c6091202769294d6bea378bc5d87fc759"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 14:17:40.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-15 14:36:36.000000000","tz":60},"subject":"[OVN] ovn-db-sync check for router port differences","message":"[OVN] ovn-db-sync check for router port differences\n\nPrior to this patch the ovn-db-sync script did not check if the router\nports were actually out-of-sync before marking them to be updated. This\nbehavior introduced irrelevant information in the sync report (specially\nwhen ran in \"log\" mode) making the user think that the databases were\nout-of-sync even when they were not.\n\nThis patch adds the code checking for differences in the Neutron Router\nPorts and OVN Logical Router Port entries prior to updating them.\n\nConflicts:\n  neutron/tests/unit/fake_resources.py\n\nChange-Id: Id7bf5a6aa547795ba78724eed59ba9d4fb74f758\nCloses-Bug: #2030773\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 4693836a1b58b477298e51cb47622222e3556752)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc71377ba7f8c99910565969a462c3bd5b8b4379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc71377ba7f8c99910565969a462c3bd5b8b4379"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
