)]}'
{"id":"openstack%2Fneutron~891503","triplet_id":"openstack%2Fneutron~stable%2Fyoga~Id7bf5a6aa547795ba78724eed59ba9d4fb74f758","project":"openstack/neutron","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-08-21 19:51:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7bf5a6aa547795ba78724eed59ba9d4fb74f758","subject":"[OVN] ovn-db-sync check for router port differences","status":"MERGED","created":"2023-08-15 14:40:45.000000000","updated":"2023-08-21 19:53:23.000000000","submitted":"2023-08-21 19:51:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891503","meta_rev_id":"d6ecda15a12d0ad82c5893e6944888c4767893cb","_number":891503,"virtual_id_number":891503,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-21 19:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-08-18 09:27:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-08-21 17:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-08-21 17:58:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-15 15:17:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-15 19:24:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-18 09:27:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-08-21 17:58:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"084aa3916b933d3220255b46c881c0040fb41ca2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-15 14:40:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6f8f3fb3c4f8212c0a70e4020675488883cfda3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 15:17:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf214a64f45c404585b9c0b494b466d1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/45819842173e43f384de4e1e8df75774 : SUCCESS in 21m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28af37da69f245e1b5d9d48d78bb8689 : SUCCESS in 20m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32468e641adfc91fcc813d8d44d012b03a9a1834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 19:24:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c49bab126904d4d8340802456e76c56\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/1dd93e406219463dbd0775b1485f81c1 : SUCCESS in 47m 34s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/810460884cad4c95b0fc483f1c090631 : SUCCESS in 1h 13m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dfa7907cac149bf832204afadcbe3ba : SUCCESS in 38m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c967995ac97a4a1eb5c319d235769fa7 : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69eb6c53e8b44262b42768bc1b9f2154 : SUCCESS in 30m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/113413d58f5b484485c66262a059ca85 : SUCCESS in 25m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55b2f7af3d3e46bf9a0152ab472e56ce : SUCCESS in 9m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c57875081fe74156ae8554a680317268 : SUCCESS in 1h 01m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/214d13ce88304936be6ab1cc2a694144 : SUCCESS in 1h 02m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/25c5e24292ba4610b5ce684a1a1b60e3 : SUCCESS in 1h 25m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a8ae4d8ceebb0617e199e8f1d72dd1ef08fe459","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-18 09:27:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a1d5591f0b7ad1f12d417aa3430d817fcd09a1c9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-21 17:58:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1680b667965a9b166885989f4bf4278a2f7cbdc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 17:59:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d72c6d6ae14d43804e35a6797a06b2d62850b6a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 19:51:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/681f270181d14a6ba60d39988bfd0559\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/d00aac2e1d844b14baa1a46ad9b18cd1 : SUCCESS in 48m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d610382787644dbaad536d4807430e9e : SUCCESS in 11m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28252aa604c44b7890b208399548c997 : SUCCESS in 29m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/111bb041736e46beb1cdc9558032b16b : SUCCESS in 24m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ccacdb4c14f4926836512711cea949d : SUCCESS in 9m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/67b7632806d3425a9cafa1b08b387f56 : SUCCESS in 53m 02s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9f90cc2b6b34f3c833b43204bde67e7 : SUCCESS in 48m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f93c656eaec17cf0b249789b27bba3702e302e0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 19:51:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d6ecda15a12d0ad82c5893e6944888c4767893cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 19:53:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/367fff0333824578ae316d64b79419fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2600069706b94b48a0bfb876718e7073 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b288a1264a07d9590f2280d3cd7476b8117b94b2","revisions":{"b288a1264a07d9590f2280d3cd7476b8117b94b2":{"kind":"REWORK","_number":1,"created":"2023-08-15 14:40:45.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/03/891503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/891503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/891503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/891503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/891503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/891503/1"}}},"commit":{"parents":[{"commit":"3944dff994dc8827bb2405dc11b3e42230520e3e","subject":"Merge \"[OVN] Prevent Trunk creation/deletion with parent port bound\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3944dff994dc8827bb2405dc11b3e42230520e3e"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 14:17:40.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-15 14:35:45.000000000","tz":60},"subject":"[OVN] ovn-db-sync check for router port differences","message":"[OVN] ovn-db-sync check for router port differences\n\nPrior to this patch the ovn-db-sync script did not check if the router\nports were actually out-of-sync before marking them to be updated. This\nbehavior introduced irrelevant information in the sync report (specially\nwhen ran in \"log\" mode) making the user think that the databases were\nout-of-sync even when they were not.\n\nThis patch adds the code checking for differences in the Neutron Router\nPorts and OVN Logical Router Port entries prior to updating them.\n\nConflicts:\n  neutron/tests/unit/fake_resources.py\n\nChange-Id: Id7bf5a6aa547795ba78724eed59ba9d4fb74f758\nCloses-Bug: #2030773\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 4693836a1b58b477298e51cb47622222e3556752)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b288a1264a07d9590f2280d3cd7476b8117b94b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b288a1264a07d9590f2280d3cd7476b8117b94b2"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
