)]}'
{"id":"openstack%2Fneutron~892053","triplet_id":"openstack%2Fneutron~master~Ief49981ef318ad53fef362ca9970cfc135415306","project":"openstack/neutron","branch":"master","topic":"bug/2031971","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-08-21 13:47:06.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"Ief49981ef318ad53fef362ca9970cfc135415306","subject":"DNM: Testing bug 2031971","status":"ABANDONED","created":"2023-08-19 16:09:11.000000000","updated":"2023-08-21 16:18:05.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"309d936514fa643cafbcbbaf9fbefb730a29497f","_number":892053,"virtual_id_number":892053,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-21 16:02:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-19 16:33:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-19 18:11:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79b5d87bf6d5862e0a4b3a6b0787c0b855706078","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-19 16:09:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cef8896c3c302fb730bdf1b145e5d66324a0b589","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 16:33:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5994260c2ae4a9fb97ad724bf6a9eda\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2fede50fd314e04acccdc90940d1f17 : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4993043a4d941d0ad8c59dd774d101d : SUCCESS in 22m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f9091dbfb1d840afa780c6e73e605aff : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ded1f12e1a264815af6699621e763701 : RETRY_LIMIT in 1m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1111707f173bcdbda1261b84a51e3d05d35b726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 18:11:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c5210ffee8743dabc57c1ab20365473\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ec63105e4c4047e2a2cc815ed72fca47 : SUCCESS in 1h 35m 22s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fe3c964b351140058dc6895e6ed997e7 : SUCCESS in 1h 50m 59s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a7aebd1e09d14f6f88b7c45e6c31d97e : SUCCESS in 1h 33m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/28ceb3dbf1c44d938a70704c3ad5f0a6 : SUCCESS in 1h 54m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/af01efa79bb142cb9c7fde9a9f850ffc : SUCCESS in 34m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/813b6efd026548aca4c9b4e98196dfe8 : SUCCESS in 32m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbbe7924d2134a72a30faf4e182c04b4 : SUCCESS in 12m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6011cbf00c294f828bdcf1dc2ac0e12e : SUCCESS in 21m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f7a31765daf4102bee651c84d922ded : SUCCESS in 28m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8f05650062a4a36a1c4ceec5bec038d : SUCCESS in 22m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9af21e767584a09a64b3f0db84ad2f5 : RETRY_LIMIT in 1m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b06a15423c8343b3a931627e04ae0580 : SUCCESS in 9m 12s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/e4469747dfb04ba2b4dcc4b2e3349b69 : SUCCESS in 25m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8215f6aad55f4683aaaa194b8445eecb : FAILURE in 1h 01m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/913202673e7a4be7bcde9f2aedae5c56 : SUCCESS in 1h 55m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6cfa2742cb9b438695c45de8cdf8fa9e : SUCCESS in 36m 43s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fd7c5ab1fb834834b0c3366ec70b6721 : SUCCESS in 37m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e4f74942ed4a4e19aecc885d9edc66c2 : SUCCESS in 1h 34m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/50faa3dad2cd4820a75297614188350c : SUCCESS in 1h 35m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bc2e810bf7ae4ab49aabd81be2829c61 : SUCCESS in 57m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/511e90520d254bf0ac0b0b5a43a03526 : SUCCESS in 1h 31m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/767ff50760ce41fbb80fa1c1d16005eb : SUCCESS in 1h 36m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a14ef11afcf84faf950b928b146b0044 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8e90afd16875de58aa2cebfe8bd6d8cf472b1354","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-21 13:47:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3afcb0009434efc8c0bb06ae8130901f68f1f1ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 14:15:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba467b369798465c9900942f26b37229\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0334f6436c44f2392ab732359597e6e : SUCCESS in 24m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5463aa20f14e4f5aba4d4f7f5e5c18e2 : SUCCESS in 27m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea4788fccd554adfa7ff2f370732d6a4 : SUCCESS in 24m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/433bf807ff334a75ad4aa6c8acc70b3b : RETRY_LIMIT in 2m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7bb1d2bf38cf0fd2903323e6e4d3a6b640f60691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 16:02:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07d9d357ec3d45c09e2acdfdea4769cf\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/136ab7a0d6904ad6a03d58a9e8d6ad54 : SUCCESS in 2h 03m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b926ed3db24d4b68aba8a1452a9e232c : SUCCESS in 2h 13m 36s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1221ea718c7940ad8fcf2e395af9121a : SUCCESS in 1h 54m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fb2e2be234c94d8a8b593528d3b160fc : SUCCESS in 2h 13m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a7e1fc88922a4490b38d00737f831797 : SUCCESS in 47m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf5b41f42cb74889b2dae790885a1b30 : SUCCESS in 43m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd96f21a8ba14d1da8fb9dcf6990f186 : SUCCESS in 11m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/696413e85b624487a31d8736a295e733 : SUCCESS in 32m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/602f4a8256ca42f79cebb3197f73507e : SUCCESS in 30m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f29c94c753045febe247381caea6e10 : SUCCESS in 26m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e2fe2e183364238a7465604d068de00 : RETRY_LIMIT in 1m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47ff46f48a96403991f702d97538083c : SUCCESS in 11m 59s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/830a6d00af3c4c438ef61944ac7d3493 : SUCCESS in 24m 32s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f94062bba1154c48afb6ede60d7a6bbb : SUCCESS in 1h 00m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e590b9c8f1240c9a17ce0ca706b58b4 : SUCCESS in 2h 07m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/39dd5910d0524bcbad072daa346a93a1 : SUCCESS in 1h 10m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/996daf7aed19414fa12ea967569db81a : SUCCESS in 50m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/868aed07f7e34c869bdf75be0bfbd3df : SUCCESS in 1h 29m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e5438ac120044aedb3bf53c7aaa99bdb : SUCCESS in 1h 23m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7fb6f30ae152467a96c9004ef100c755 : SUCCESS in 1h 01m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/934848f43402463896130547d8eb8985 : SUCCESS in 1h 40m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dd0446e939cc410f9a5ebd21150c2ada : SUCCESS in 1h 40m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/42e494b86af64c8695b74da6e9319e10 : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":1},{"id":"309d936514fa643cafbcbbaf9fbefb730a29497f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-21 16:18:05.000000000","message":"Abandoned\n\nRally jobs seem stable now","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6efe40ec61391231b4ca00cdfb81b61d92ffd5e5","revisions":{"6efe40ec61391231b4ca00cdfb81b61d92ffd5e5":{"kind":"REWORK","_number":1,"created":"2023-08-19 16:09:11.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/53/892053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/892053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/892053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/892053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/892053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/892053/1"}}},"commit":{"parents":[{"commit":"be9c4d585db1833b410062f51a6d11a41818b9c1","subject":"Merge \"Force DB migration script to be run before some fullstack tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be9c4d585db1833b410062f51a6d11a41818b9c1"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-19 16:04:48.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-19 16:04:48.000000000","tz":-240},"subject":"DNM: Testing bug 2031971","message":"DNM: Testing bug 2031971\n\nUsing this to test rally jobs failures.\n\nRelated-bug: #2031971\nDepends-on: https://review.opendev.org/c/openstack/rally-openstack/+/892052\nChange-Id: Ief49981ef318ad53fef362ca9970cfc135415306\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6efe40ec61391231b4ca00cdfb81b61d92ffd5e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6efe40ec61391231b4ca00cdfb81b61d92ffd5e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
