)]}'
{"id":"openstack%2Fneutron~892116","triplet_id":"openstack%2Fneutron~master~I32102d7b8f200892fbb45cd6f69263d63dcb1325","project":"openstack/neutron","branch":"master","topic":"bug/2012332","attention_set":{},"removed_from_attention_set":{"34125":{"account":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"last_update":"2023-10-04 07:57:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I32102d7b8f200892fbb45cd6f69263d63dcb1325","subject":"Add allowed_address_pairs add/remove atomic operations","status":"ABANDONED","created":"2023-08-21 04:00:19.000000000","updated":"2023-10-04 07:57:10.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ae884a1ab142ebae0f9099b384d468f3993f138e","_number":892116,"virtual_id_number":892116,"owner":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-21 06:15:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-21 04:30:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-21 06:15:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 12:34:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-04 07:57:10.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"705e9e253e8e81776459ed642fa6caa1e95902df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"date":"2023-08-21 04:00:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4d2fd89ed4ecc8df3335f48def752d8ccd47b2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 04:30:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4cbf013287145c5a82eab1033d8b7b7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e73afadfda14b829b739c964e9da646 : SUCCESS in 20m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d3749495f3f4b6691f0afbbd3b7e37f : SUCCESS in 19m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9a00adfe967b4f44938c206ddda12c7b : SUCCESS in 29m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7db0d5a1ed824befbd5911849521f52c : RETRY_LIMIT in 3m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03850252616226109f058a3883f9257afaf103f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 06:15:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee331870e45b4338ad43abb4b6242ea7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/618f5f19ce6a4b5ca980b5f47c7aaf42 : FAILURE in 1h 17m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0542c64518344ccaaf2a90d902777162 : FAILURE in 1h 31m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/04bd122990b24eba9277eb4c5368abed : FAILURE in 1h 19m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/baed1bda599b4a10868a683389fcb007 : SUCCESS in 1h 29m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3eae9fb5b2934af0b5b1c0a64263d4ad : SUCCESS in 34m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3329437e4a714ccd8f8042cf92916dfd : SUCCESS in 27m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e6852804025492690e3698c000cdd21 : SUCCESS in 10m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/220bd562253b4c23ae7ec1bd0972b743 : SUCCESS in 27m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fdce462310740628b072c7ecb46e4fc : SUCCESS in 30m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b58a314c6154fb8bd5696d955d77228 : SUCCESS in 26m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/00cd71de8d7646808fa65c4a8b223fe1 : RETRY_LIMIT in 2m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16fde6e987264a318d55d51f56ebdfdf : SUCCESS in 8m 37s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/11e7f8aaa7c043bdb4e5a4d373fde42b : SUCCESS in 25m 13s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/46a2a46441174356a3acb60dc3fb1ccd : FAILURE in 52m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d256694e0d424816a9567630855da94c : SUCCESS in 2h 02m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fb13af3c5a784c26bb4260184a4d8416 : FAILURE in 36m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/822e17b2e1054ad780202350e9cb9d85 : FAILURE in 1h 02m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9c1e9e5ab74e4c4e904eb11195552955 : FAILURE in 1h 07m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/29bd8a771ff7449aa628f39e2331ef7a : FAILURE in 48m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f141022b9683424f93bea2bc64fa8e41 : FAILURE in 42m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/301ec58952fc490c9b9dbc270f929a89 : SUCCESS in 1h 17m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c1c70ff7006f4899a02b5e33365b209f : SUCCESS in 2h 08m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/502907dd25e34c13b0bbc222e3cad922 : FAILURE in 40m 40s","accounts_in_message":[],"_revision_number":1},{"id":"f3173cd36fcc045d068dee59c318693bd84023e1","author":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"date":"2023-08-21 06:20:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"78119e26ab66d129ef61f1f19f2c22931fd80768","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 06:44:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9c349774bdb45e39891a2e3c954a7a2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/712c1621ab1546dd8b8d00ee96005655 : SUCCESS in 22m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ea1597d869a4a43a54375f93a5a9c94 : SUCCESS in 21m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f6a8676871042848c628702dea11026 : SUCCESS in 20m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/45c4d6a2279b49d3a0dfb90d3f1022c6 : RETRY_LIMIT in 3m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fc8bc9959bc5ad8b1e5f5122c1fed063a646575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 08:34:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af344d927cca4696a4650e1a28b74312\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0921cb70965a422094059ac79d22a0d0 : FAILURE in 1h 27m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/137989df3be543ebafd462c7f9c0630a : FAILURE in 1h 38m 54s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1080b113aa5b4a4d920327bb10be40e2 : FAILURE in 1h 30m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2f6d5b6ecd0648abbc8b7d50e781ece3 : SUCCESS in 2h 09m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea2c93ef434b489d9519b7a3d7126a26 : SUCCESS in 31m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7da1f650f4c438295a6702a20aee187 : SUCCESS in 35m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/186cc28c8c994ff1ab6a128e7f39ee66 : SUCCESS in 8m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dcbe94299cc4f2b8aea0df48e3fc93c : SUCCESS in 28m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43042d83774b4d3f8962335a8244e41c : SUCCESS in 27m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4aa67629400f48ada84b2f8c109e4609 : SUCCESS in 27m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48b7019e6d7f475f9205e119d3eb5ca9 : RETRY_LIMIT in 1m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0417b67d29fe4bd3b874dfe74fb92dce : SUCCESS in 10m 53s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7334b54fb6fd4fd4bad1ed85f27bdb5d : SUCCESS in 27m 23s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/03c0a8399e3a43a2beee201364490c99 : FAILURE in 54m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/21bc69b0e20744e39b16319eca60332a : FAILURE in 1h 47m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/85a4738dfe954ed48adff529b5e79ec6 : FAILURE in 56m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/525959fd44a04349a2b0d79b2b9887c0 : FAILURE in 57m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d1bf3a1c50954519b9d609c266a77f23 : FAILURE in 44m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cd1f0674a1c344fa9bdf7ea7dbfce60c : FAILURE in 59m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b25838f270534efb871767c6a5a45203 : FAILURE in 55m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a698674bf99443e4a0ee99e9a4ab1f48 : SUCCESS in 1h 18m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/16359d8ed89347e59f84989fac4e84de : SUCCESS in 1h 45m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d1a08fb4b9c14701872c13796b98e855 : FAILURE in 1h 02m 20s","accounts_in_message":[],"_revision_number":1},{"id":"e5ada04d14289b3f6a026db8baf9b1dc22258359","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-30 12:34:28.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/16/892116/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ae884a1ab142ebae0f9099b384d468f3993f138e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-04 07:57:10.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5da59fe787a8a48b178b755a7a681192c3c0342","revisions":{"d5da59fe787a8a48b178b755a7a681192c3c0342":{"kind":"REWORK","_number":1,"created":"2023-08-21 04:00:19.000000000","uploader":{"_account_id":34125,"name":"huanghailun","display_name":"huanghailun","email":"hailun.huang@easystack.cn","username":"lnun"},"ref":"refs/changes/16/892116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/892116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/892116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/892116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/892116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/892116/1"}}},"commit":{"parents":[{"commit":"be9c4d585db1833b410062f51a6d11a41818b9c1","subject":"Merge \"Force DB migration script to be run before some fullstack tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be9c4d585db1833b410062f51a6d11a41818b9c1"}]}],"author":{"name":"hailun.huang","email":"hailun.huang@easystack.cn","date":"2023-08-21 03:45:21.000000000","tz":0},"committer":{"name":"hailun.huang","email":"hailun.huang@easystack.cn","date":"2023-08-21 03:45:21.000000000","tz":0},"subject":"Add allowed_address_pairs add/remove atomic operations","message":"Add allowed_address_pairs add/remove atomic operations\n\nThe old method for operation allowed_address_pairs is \u0027update_port\u0027 and\n\u0027create_port\u0027.\nThe new method of updating allowed_address_pairs is to atomically modify it\nby \u0027add_allowed_address_pairs\u0027 and \u0027remove_allowed_address_pairs\u0027.\n\nCloses-Bug: #2012332\n\nCo-Authored-By: Rodolfo Alonso Hernandez ralonsoh@redhat.com\n\nDepends-On: https://review.opendev.org/880615\nChange-Id: I32102d7b8f200892fbb45cd6f69263d63dcb1325\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5da59fe787a8a48b178b755a7a681192c3c0342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d5da59fe787a8a48b178b755a7a681192c3c0342"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
