)]}'
{"id":"openstack%2Fneutron~892134","triplet_id":"openstack%2Fneutron~master~I53c339d8a55b6f1cebb411821478bf19da4b5cba","project":"openstack/neutron","branch":"master","topic":"bug/2031646","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-08-22 16:25:29.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-08-22 15:51:11.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-08-22 16:25:29.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-08-22 13:06:39.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I53c339d8a55b6f1cebb411821478bf19da4b5cba","subject":"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue","status":"MERGED","created":"2023-08-21 11:05:57.000000000","updated":"2023-08-22 16:27:58.000000000","submitted":"2023-08-22 16:25:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892134-bug/2031646","meta_rev_id":"53cd174869e6cd098c324eb5717e26cbca1640ed","_number":892134,"virtual_id_number":892134,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 16:25:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2023-08-22 15:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-08-22 13:06:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2023-08-22 15:51:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-21 11:16:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-08-21 12:06:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-21 14:01:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-08-22 05:40:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-08-22 13:06:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"2b787f098cff33f095196b1617dac14057a6d802","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-21 11:05:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b91f4766f3d3995e26e19d4fa30dbed0eed3ecb7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-21 11:16:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4326956a3005e86a253943529b8b3dc895a3febb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-21 11:22:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"638ef3e3da48f77f9ee8e9c4e2589a1c9bd81cee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-21 11:22:59.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"2463eed5332b54071f7ab7b09b0757fa010f1829","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 12:06:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/22d93b54620a405e90f4273c98e573f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b83b8c87a8e40038d93c505e80bbba2 : SUCCESS in 38m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/172b7766ba254034bada61d1dc5cc2e3 : SUCCESS in 10m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aecb730b83bb4ef2837ec7481b9953e9 : SUCCESS in 25m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e7670f6e1484070ae897188f1d1b0fa : SUCCESS in 27m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b687a2e1a301405c8514bb7c5c31ed36 : SUCCESS in 27m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9dc3fe603904b28984ab7152418ea99 : RETRY_LIMIT in 1m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f9b40fb164c450190ec2d818717faca : SUCCESS in 10m 45s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/706c8061893e48dd9dea5b419a1442af : FAILURE in 33m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0fe312e72e5335e94d8110a15d6e8cc8b8d2a0e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 12:19:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8dd9d0a5be944a9bb33ab63688c71ea\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23429b997cba40b48c852cd7daa539ba : SUCCESS in 23m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec4a40b5a5fb471683894478be9323d0 : SUCCESS in 54m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7e4ce1d1a64c4277a68108eee290beeb : SUCCESS in 19m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c9f4a4643177455ab9c1cfb2f27647e4 : RETRY_LIMIT in 1m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1711cbe581f5bd5ebf9c7de31f84621405c0755d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-21 13:02:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"deb400c854bc97c3c2adf1d791f4ddc076daf70d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-21 13:18:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae603024c9b66230b59f59d59cc7a9430aab0a48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 13:18:26.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/892132 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"c3763789f8c653ba615dca9fad8dc6c06f2e78bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 13:18:28.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/892132 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"171267107b116e367399c19f8c9fb452cad2817a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-21 13:57:39.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"65a41e06792b0575758b117ba77813eeca22d8c4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-21 14:01:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"832aec819ea545c92081d3ea4833edf61215dd9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 14:44:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc04b337aba74e94ba6816016f4e07ff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e75eb7736a94cebb00a31db67eeb189 : SUCCESS in 44m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dec9816db9f4a5994b529083fd549da : SUCCESS in 11m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/185803b672e84a4faf6324a44b6d2deb : SUCCESS in 32m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/436719e16c3d472f99650e8c152b99d3 : SUCCESS in 30m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/771e1b1b68b84a6f9ef0ffc920176e6d : SUCCESS in 28m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8597f3192e6d4cdc8276bd38595de6c9 : RETRY_LIMIT in 2m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f43342078d64eaba96ce998c3c823eb : SUCCESS in 11m 38s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/796e954a8850435b96facec897050ec7 : FAILURE in 31m 08s","accounts_in_message":[],"_revision_number":3},{"id":"3edfd0f7a6d802370a961e73fea9d18c91995cdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 14:45:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe3f7e63365e4a488945f181664e8597\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c59b8ffa081d4928a090c07c656a69f0 : SUCCESS in 45m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2afb270ce9934792b6bfe94ef0a69b44 : SUCCESS in 23m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4465681d722d4102ac0bef332c8486d3 : SUCCESS in 44m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b6a79ee531904e9197faa4028ad600cf : RETRY_LIMIT in 1m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ca7c037687d4318722049ce224072c1296384f0d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-21 15:57:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7505212dae146b592e58d7c9a2b98b8d19a6b848","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-21 16:51:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ffe347d83fd7a4d5cf88c170a9416414c084c6a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-08-22 05:40:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ababf5dae9c51b791d4a06e4e99e3da275f1936c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-22 11:15:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"088ae223313b3321ca2baacd906f6e97acb269a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-22 11:15:48.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c5e511778f7b5aabd5234d22da26209f3f838132","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 12:11:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/558f0e7b23954f4fbf3cf018c476acb5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/16242f002ff54eb3a10ab345c9622ab7 : SUCCESS in 52m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0431fcfecab439f8d701b4c1621e075 : SUCCESS in 19m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6bb76fdf59554c078ec03db0765a59c7 : SUCCESS in 52m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/73b67302d35247eead3ba82421b91d3a : SUCCESS in 19m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3fc8bbaa856f3cded483c728836f1f8152b4a027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 12:17:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c161eaece584005a59ffea42447c1fa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/921ea2bbcb4745faa4df141650e8b8f7 : SUCCESS in 44m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67291d6bb7d74c72a0bc22c99928e695 : SUCCESS in 10m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d63ddc3efc964e788a5722a1b4601ab0 : SUCCESS in 30m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0fd5d6f6d7e45a7a020aaa18f150f41 : SUCCESS in 39m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/daebe587667e4488a05387fe5766637d : SUCCESS in 22m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6861233761a4f5194040155eeefe54f : SUCCESS in 27m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd5bf10ecb8457eba9d54b8e8b9c89e : SUCCESS in 10m 40s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/48a93f53f3f642c5ab4624b7302498ef : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":4},{"id":"2ca1386e4ee069b5233b6930bd7a4b61ebb3f002","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-22 13:06:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4c2d3e6a4899a9865498c957a3f81e7f4fc09253","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-08-22 15:51:11.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4e71ed54376a158494238553efefdf18c1d71171","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 15:51:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"824e380d4be05135bff8409cfe4a8f8153c4f40c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 16:25:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8bd3631d46f4b64a2a152cc943c6786\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2845b28a999b4702976bf0ae88cff1a3 : SUCCESS in 9m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91147a01645c4246a955c958049fbb84 : SUCCESS in 29m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c4cbf19f1114d8f8948b4d08e180754 : SUCCESS in 29m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c238995558047c5be9151f4d338937e : SUCCESS in 26m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99fe5e7deae34e83b11f4c71b15a0ee3 : SUCCESS in 13m 46s","accounts_in_message":[],"_revision_number":4},{"id":"8e6475f1b2d64c1b51a4801729a6f3bcd33dc352","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 16:25:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"53cd174869e6cd098c324eb5717e26cbca1640ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 16:27:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e6575d1d02a4635a4537fa2386f78af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c149d11d94a34ed9a6d2dfb91b07fe3c : SUCCESS in 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7848eb0bf925ae872012e08012e15fed9ab32af1","revisions":{"949c82d20eb0637cdd54b1b410232b3d221fb6d7":{"kind":"REWORK","_number":1,"created":"2023-08-21 11:05:57.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/34/892134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/892134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/892134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/892134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/892134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/892134/1"}}},"commit":{"parents":[{"commit":"b7eab20af15e28c00efa88c053886d052ddb8218","subject":"Remove local CI job \"tox-py311\" superseded by \"openstack-tox-py311\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7eab20af15e28c00efa88c053886d052ddb8218"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-21 11:04:59.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-21 11:04:59.000000000","tz":0},"subject":"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue","message":"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue\n\nThis job will be non voting but will help us during the SQLAlchemy\n2.0 migration.\n\nRelated-Bug: #2031646\nChange-Id: I53c339d8a55b6f1cebb411821478bf19da4b5cba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/949c82d20eb0637cdd54b1b410232b3d221fb6d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/949c82d20eb0637cdd54b1b410232b3d221fb6d7"}]},"branch":"refs/heads/master"},"39f16b0541034eb22fcdaafe091030cc758ddc4b":{"kind":"REWORK","_number":2,"created":"2023-08-21 11:22:59.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/34/892134/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/892134/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/892134/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/892134/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/892134/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/892134/2"}}},"commit":{"parents":[{"commit":"b7eab20af15e28c00efa88c053886d052ddb8218","subject":"Remove local CI job \"tox-py311\" superseded by \"openstack-tox-py311\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7eab20af15e28c00efa88c053886d052ddb8218"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-21 11:04:59.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-08-21 11:22:59.000000000","tz":0},"subject":"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue","message":"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue\n\nThis job will help us during the SQLAlchemy 2.0 migration.\n\nRelated-Bug: #2031646\nChange-Id: I53c339d8a55b6f1cebb411821478bf19da4b5cba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39f16b0541034eb22fcdaafe091030cc758ddc4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39f16b0541034eb22fcdaafe091030cc758ddc4b"}]},"branch":"refs/heads/master"},"e1dcff2e09057f1d18319c87cff00481ce227568":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-21 13:57:39.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/34/892134/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/892134/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/892134/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/892134/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/892134/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/892134/3"}}},"commit":{"parents":[{"commit":"72b5120ac2c5d1723ea62b87c51c1eec75b42c95","subject":"Remove local CI job \"tox-py311\" superseded by \"openstack-tox-py311\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72b5120ac2c5d1723ea62b87c51c1eec75b42c95"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-21 11:04:59.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-08-21 13:57:39.000000000","tz":0},"subject":"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue","message":"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue\n\nThis job will help us during the SQLAlchemy 2.0 migration.\n\nRelated-Bug: #2031646\nChange-Id: I53c339d8a55b6f1cebb411821478bf19da4b5cba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1dcff2e09057f1d18319c87cff00481ce227568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e1dcff2e09057f1d18319c87cff00481ce227568"}]},"branch":"refs/heads/master","description":"Rebase"},"7848eb0bf925ae872012e08012e15fed9ab32af1":{"kind":"REWORK","_number":4,"created":"2023-08-22 11:15:48.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/34/892134/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/34/892134/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/892134/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/892134/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/34/892134/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/34/892134/4"}}},"commit":{"parents":[{"commit":"72b5120ac2c5d1723ea62b87c51c1eec75b42c95","subject":"Remove local CI job \"tox-py311\" superseded by \"openstack-tox-py311\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72b5120ac2c5d1723ea62b87c51c1eec75b42c95"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-21 11:04:59.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-08-22 11:15:47.000000000","tz":0},"subject":"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue","message":"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue\n\nThis job will help us during the SQLAlchemy 2.0 migration.\n\nRelated-Bug: #2031646\nChange-Id: I53c339d8a55b6f1cebb411821478bf19da4b5cba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7848eb0bf925ae872012e08012e15fed9ab32af1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7848eb0bf925ae872012e08012e15fed9ab32af1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
