)]}'
{"id":"openstack%2Fneutron~892182","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~Id7bf5a6aa547795ba78724eed59ba9d4fb74f758","project":"openstack/neutron","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-08-22 13:57:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7bf5a6aa547795ba78724eed59ba9d4fb74f758","subject":"[OVN] ovn-db-sync check for router port differences","status":"MERGED","created":"2023-08-21 18:01:18.000000000","updated":"2023-08-22 13:58:49.000000000","submitted":"2023-08-22 13:57:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892182","cherry_pick_of_change":891506,"cherry_pick_of_patch_set":2,"meta_rev_id":"bf3fcf9e2c368575412df3af94324aaa7780c91a","_number":892182,"virtual_id_number":892182,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 13:57:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2023-08-22 12:36:36.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2023-08-22 11:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2023-08-22 12:36:36.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-21 18:01:18.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"CC"},{"updated":"2023-08-21 19:39:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-22 08:03:40.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2023-08-22 11:58:02.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"d6589b59d153c07e4040e4837f0200ac2e9f6fd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-21 18:01:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"9336e94beb30f1a6f8be76e473a0b5910b433dd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 19:39:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4ab58c4e67a4ca7a2a4982735182837\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/8637254f659d468198ec9e763218b2e4 : SUCCESS in 45m 00s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/00b1e0dff41c4867a52283e0e7338ed8 : SUCCESS in 1h 36m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cc745ed35e4420e891ddbed45e921d3 : SUCCESS in 35m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a36b1df3ff04dd89528c58136ef6810 : SUCCESS in 13m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94ba05019f374957978843f0f56e203d : SUCCESS in 26m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6e5714c14af48ef81a721854012a6af : SUCCESS in 24m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87282a9c73034b0296db589b6d9a01f1 : SUCCESS in 10m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f6eab6f1f77a49389a92ba27349bd43e : SUCCESS in 1h 07m 42s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/bff423e6bc4644fc80a9d1eafe7bbde3 : SUCCESS in 1h 16m 49s (non-voting)\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/b92148cbceac48a799eb33d38700af11 : SUCCESS in 51m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d7b53165ceab46ec97ffda728e86b7fd : SUCCESS in 53m 27s (non-voting)\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/0f28f171c881408fa00f0fcdf31ee2ab : SUCCESS in 46m 22s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9d5f1409620b4a13aab99f549a02a463 : FAILURE in 34m 46s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/877f1a521d034b9aab12a800dcf3ebe2 : SUCCESS in 1h 19m 07s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5a50185d984148c3864fea4c0ffe3178 : SUCCESS in 1h 20m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/21be1b24d8114b43a1e121da5b6511e9 : FAILURE in 31m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59f5601d547f76ae39ccdd6e56424a56445e1092","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-22 08:02:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f09d094822bebf38bd2cae06af299e58aa4b90eb","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-22 08:03:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d826578d556f8753bd7b404c7ca306875671eec8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-08-22 11:58:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a2e958421402cb3c5f7cfd236cc79c3b5bb10b9","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-22 12:36:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1beaf0ed53e15ab66d4890756cfce1f5402d30b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 12:37:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5eb0a12e634cf237f4d72174e5dd50f781af831a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 13:57:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a72b2bb1ab7411788ff0ba06de676db\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/45a2dd031c7b45ed92822e7a6dca5787 : SUCCESS in 54m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b8fa7141e9479cb91e92ca26ee2bda : SUCCESS in 13m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2370ab9cc5142179d1e8be2fe5c40b9 : SUCCESS in 29m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46a8109a1b8c4f80a372794be34c0dde : SUCCESS in 27m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/513d8d8dcfcf452a8b2b0e707fda8957 : SUCCESS in 11m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/82824b65dbcd4ff29106bbe9ebebd7bc : SUCCESS in 1h 13m 57s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/42a1b99c69ee4423a8bc9e6ae65c20c4 : SUCCESS in 1h 07m 23s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/c8499b5f64ea48f496c8d3f12787f148 : SUCCESS in 1h 15m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ad7464bfec39b0734934e8080352a814e7d6d90a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 13:57:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bf3fcf9e2c368575412df3af94324aaa7780c91a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 13:58:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bac55dca24b47ccb320097ddd010be5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adc5c374fd264777afd38ce3f5636a74 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d45ee7f34f339c683c03f5391800b9848aaf11e","revisions":{"8d45ee7f34f339c683c03f5391800b9848aaf11e":{"kind":"REWORK","_number":1,"created":"2023-08-21 18:01:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/82/892182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/892182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/892182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/892182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/892182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/892182/1"}}},"commit":{"parents":[{"commit":"0f3557555fa6270abca8669267aeade4c123e514","subject":"Merge \"Fix ACL sync when default sg group is created\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f3557555fa6270abca8669267aeade4c123e514"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 14:17:40.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-21 18:01:18.000000000","tz":0},"subject":"[OVN] ovn-db-sync check for router port differences","message":"[OVN] ovn-db-sync check for router port differences\n\nPrior to this patch the ovn-db-sync script did not check if the router\nports were actually out-of-sync before marking them to be updated. This\nbehavior introduced irrelevant information in the sync report (specially\nwhen ran in \"log\" mode) making the user think that the databases were\nout-of-sync even when they were not.\n\nThis patch adds the code checking for differences in the Neutron Router\nPorts and OVN Logical Router Port entries prior to updating them.\n\nNOTE FOR THIS BACKPORT:\n\nThe lrp_get() method from ovsdbapp is not present for this stable\nbranch, so this patch was adapted to use the get_lrouter_port()\ninstead.\n\nConflicts:\n  neutron/tests/unit/fake_resources.py\n\nChange-Id: Id7bf5a6aa547795ba78724eed59ba9d4fb74f758\nCloses-Bug: #2030773\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 4693836a1b58b477298e51cb47622222e3556752)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d45ee7f34f339c683c03f5391800b9848aaf11e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d45ee7f34f339c683c03f5391800b9848aaf11e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
