)]}'
{"id":"openstack%2Fneutron~892183","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Id7bf5a6aa547795ba78724eed59ba9d4fb74f758","project":"openstack/neutron","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-08-22 08:02:03.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-08-22 14:26:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7bf5a6aa547795ba78724eed59ba9d4fb74f758","subject":"[OVN] ovn-db-sync check for router port differences","status":"MERGED","created":"2023-08-21 18:01:42.000000000","updated":"2023-08-22 14:27:43.000000000","submitted":"2023-08-22 14:26:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892183","cherry_pick_of_change":892182,"cherry_pick_of_patch_set":1,"meta_rev_id":"79555bcdf240f2f7adb885ba5acd7ecf7c1b49e5","_number":892183,"virtual_id_number":892183,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 14:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2023-08-22 12:36:41.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2023-08-22 11:58:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":1,"date":"2023-08-22 12:36:41.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-21 18:01:42.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"CC"},{"updated":"2023-08-21 18:04:43.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2023-08-21 20:00:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-22 11:58:14.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"7f9ee2f74d5f638238c35992c2df95736d37c97b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-21 18:01:42.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.\n\nThe following files contain Git conflicts:\n* neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_db_sync.py","accounts_in_message":[],"_revision_number":1},{"id":"73dd1b6c078b677e5d1f0246b80303cc7b6093ab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-21 18:04:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ed0cbfe638a2fba3648ae5cf931f50d1df3a7393","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-21 18:04:55.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"09c637812db8d8de43062dc98b7292bb1e3bbfb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 20:00:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6de1a7e26cd54d1eb9621c15957d4ec2\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/f4d8b6d61b724f019cf0e443045b5b81 : SUCCESS in 48m 24s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/6b0d8cd6eb4b4f33b11fde6ad3e76df9 : SUCCESS in 1h 54m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daadc6e64de24c3f862b1090245bbdda : SUCCESS in 27m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1bb3160aa55428892ce08e338c13d5c : SUCCESS in 14m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/614be862caf24aca9012036db1a0077f : SUCCESS in 27m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9c07fc0cc0f47b4a9e033da504a46a2 : SUCCESS in 31m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e35d26df547448bbe298984e8e4653b : SUCCESS in 32m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3837fc35ae9a4b96a5d48fe9e91e1193 : SUCCESS in 10m 45s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/de1aaa5cd7be4b7d90febced7b9f6c87 : SUCCESS in 49m 54s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/9882317fb2e540e6820fa27622bdfaa6 : SUCCESS in 1h 17m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/807efbca26ae4efca92a3d55724ce8e0 : SUCCESS in 1h 15m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/2d1235f093114fb5923260fc9dc8fc19 : SUCCESS in 1h 13m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6d1c78394ff4b909387def8d316161e : SUCCESS in 1h 03m 01s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8cb7509e694b476d92021d6929fcac84 : SUCCESS in 1h 07m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/250a7fde90254453b5a75adb1a269617 : SUCCESS in 1h 01m 21s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/14a247be41374305a45a7fdf3ac89972 : SUCCESS in 1h 09m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6e8e844a3a9e49e29b379ceb1059d6e9 : SUCCESS in 1h 27m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e77a9bfeec124c828d3f577ff9e7c088 : FAILURE in 49m 07s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fb37cd15581245e0adc7515b3ada9544 : SUCCESS in 1h 09m 45s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5abb096842e540cd981804143f3fade6 : SUCCESS in 52m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f46ae0d49cfe4431bc6c5a495b0c4699 : SUCCESS in 42m 18s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b23abf727a3047f1aa205567a281a2c1 : SUCCESS in 1h 38m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eecc8eec56e94b06b36ceee28d428dc8 : SUCCESS in 59m 57s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/05865c0777e94381a24f9acff39b91ad : SUCCESS in 1h 17m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e8ab09dbd4df4c53b195314dc761f1f7 : SUCCESS in 51m 33s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/2992a68dcdb54b83a2a0e7123d868010 : SUCCESS in 1h 10m 21s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/590efb5a800741868ddca3f7a8195d73 : SUCCESS in 50m 56s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ee232397df4a4ce098e7741b3675e5c6 : SUCCESS in 1h 27m 08s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/9ed5faaf192d40ca840f6e8cd8d25d63 : FAILURE in 1h 27m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"00b61e62bbaef7b4f5c4aadaa2c335b40e62e12f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-22 08:02:03.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c399262d37087b500f29299e1668c36e059b2b68","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-08-22 11:58:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a168f8a55ce5d5a5ebf8d99203411db6a9522693","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-22 12:36:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0988fdd373bff57861ea35d367e24d27934d79ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 12:37:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"165e1a0326cb77bce5760beb69bc4c329834f74a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:26:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fc25efdecef4cfb8f93012a622e2cb0\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/d7519cd9469d4549841dc9abf4b692dc : SUCCESS in 57m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f12f7102bfb44bb1a028c6b19450639f : SUCCESS in 14m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6362c64b5b474f8a8b5ccaec3aeeb328 : SUCCESS in 31m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3f01cf7cf11243429ca3701cb49a4a08 : SUCCESS in 30m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f504d92a2a914c7c911519703d828c3d : SUCCESS in 10m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9abf4dc32b964a7b854ab9383d68ec05 : SUCCESS in 48m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/504be21783c840d0bf990b18a81fec57 : SUCCESS in 1h 26m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aab9b5ad18444b9eb7aa3bf6b3139a5a : SUCCESS in 1h 09m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/72cf7e8f84b84c17803c3e41a024826d : SUCCESS in 1h 20m 10s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/305a8c17391747d1941f360072a4d390 : SUCCESS in 1h 02m 45s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2541335b7147491b976ecf97fb1c24aa : SUCCESS in 1h 27m 29s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bf4e2262ba0644bc8f478ea98862a69c : SUCCESS in 1h 43m 18s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7cd217b11a1e4d1a8aee283f1abaa9c7 : SUCCESS in 1h 43m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/545ae17b6de54555abc707471ac043d9 : SUCCESS in 59m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b07fbe13ffa442ebaf4fd789510fd793 : SUCCESS in 1h 16m 05s","accounts_in_message":[],"_revision_number":2},{"id":"0f4ca62cb66b24ac32f7da34765e04ffb61e26e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:26:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"79555bcdf240f2f7adb885ba5acd7ecf7c1b49e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:27:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43cc64c7d4dd45b592abcb1ef60d9940\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce8634a2e6f2428b9c930f73f913e369 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6af6771deb66b75f0189733771e9cc288d6139ed","revisions":{"85278e5fed4b3b9b0052e4abe34a36634d7b9e97":{"kind":"REWORK","_number":1,"created":"2023-08-21 18:01:42.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/83/892183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/892183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/892183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/892183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/892183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/892183/1"}}},"commit":{"parents":[{"commit":"a6f5160b6c20144db5293b615f3c9e35c8fc59c7","subject":"Fix ACL sync when default sg group is created","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6f5160b6c20144db5293b615f3c9e35c8fc59c7"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 14:17:40.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-21 18:01:42.000000000","tz":0},"subject":"[OVN] ovn-db-sync check for router port differences","message":"[OVN] ovn-db-sync check for router port differences\n\nPrior to this patch the ovn-db-sync script did not check if the router\nports were actually out-of-sync before marking them to be updated. This\nbehavior introduced irrelevant information in the sync report (specially\nwhen ran in \"log\" mode) making the user think that the databases were\nout-of-sync even when they were not.\n\nThis patch adds the code checking for differences in the Neutron Router\nPorts and OVN Logical Router Port entries prior to updating them.\n\nNOTE FOR THIS BACKPORT:\n\nThe lrp_get() method from ovsdbapp is not present for this stable\nbranch, so this patch was adapted to use the get_lrouter_port()\ninstead.\n\nConflicts:\n  neutron/tests/unit/fake_resources.py\n\nChange-Id: Id7bf5a6aa547795ba78724eed59ba9d4fb74f758\nCloses-Bug: #2030773\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 4693836a1b58b477298e51cb47622222e3556752)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85278e5fed4b3b9b0052e4abe34a36634d7b9e97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/85278e5fed4b3b9b0052e4abe34a36634d7b9e97"}]},"branch":"refs/heads/stable/ussuri"},"6af6771deb66b75f0189733771e9cc288d6139ed":{"kind":"REWORK","_number":2,"created":"2023-08-21 18:04:43.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/83/892183/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/892183/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/892183/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/892183/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/892183/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/892183/2"}}},"commit":{"parents":[{"commit":"a6f5160b6c20144db5293b615f3c9e35c8fc59c7","subject":"Fix ACL sync when default sg group is created","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6f5160b6c20144db5293b615f3c9e35c8fc59c7"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-08 14:17:40.000000000","tz":60},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-08-21 18:04:09.000000000","tz":-240},"subject":"[OVN] ovn-db-sync check for router port differences","message":"[OVN] ovn-db-sync check for router port differences\n\nPrior to this patch the ovn-db-sync script did not check if the router\nports were actually out-of-sync before marking them to be updated. This\nbehavior introduced irrelevant information in the sync report (specially\nwhen ran in \"log\" mode) making the user think that the databases were\nout-of-sync even when they were not.\n\nThis patch adds the code checking for differences in the Neutron Router\nPorts and OVN Logical Router Port entries prior to updating them.\n\nNOTE FOR THIS BACKPORT:\n\nThe lrp_get() method from ovsdbapp is not present for this stable\nbranch, so this patch was adapted to use the get_lrouter_port()\ninstead.\n\nConflicts:\n  neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_db_sync.py\n  neutron/tests/unit/fake_resources.py\n\nChange-Id: Id7bf5a6aa547795ba78724eed59ba9d4fb74f758\nCloses-Bug: #2030773\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 4693836a1b58b477298e51cb47622222e3556752)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6af6771deb66b75f0189733771e9cc288d6139ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6af6771deb66b75f0189733771e9cc288d6139ed"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
