)]}'
{"id":"openstack%2Fneutron~892604","triplet_id":"openstack%2Fneutron~master~I72973abbb8b0f9cc5848fd3b4f6463c38c6595f8","project":"openstack/neutron","branch":"master","topic":"bug/2030741","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-01-26 15:48:12.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"23763":{"account":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"last_update":"2024-01-29 21:23:25.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-12-04 11:02:17.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-12-04 11:02:17.000000000","reason":"removed on reply"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-12-04 11:02:17.000000000","reason":"Change was abandoned"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-01-23 22:41:46.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I72973abbb8b0f9cc5848fd3b4f6463c38c6595f8","subject":"[ovn] AZs distribution in L3 port scheduler","status":"MERGED","created":"2023-08-24 13:59:13.000000000","updated":"2024-01-29 21:24:41.000000000","submitted":"2024-01-29 21:23:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":88,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892604-bug/2030741","meta_rev_id":"6397f47e43b25cb372d95fa8249314f79bcc50e7","_number":892604,"virtual_id_number":892604,"owner":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-01-26 14:57:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-29 21:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-01-26 15:48:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-29 15:41:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-28 05:15:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-29 19:19:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-24 14:24:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-24 16:16:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-08-24 16:16:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-08-24 16:16:57.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-08-24 16:17:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-08-24 16:18:19.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-08-24 16:36:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 08:13:50.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-11 19:07:23.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-01-28 05:15:38.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"082a2fb739bce431f586cc6f235a1b7c7b1daa1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-08-24 13:59:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e410673cbe667208b9f17b748aa33244f656b0ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-08-24 14:02:20.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6c419b68103e8038f38fdd2a9807840bd124bc6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 14:24:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9512010b3c9944279073aaa3b41a9235\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b26685d57ae84cef88d782e76e0bd5e5 : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/497fcdc1b5034974be07749100368927 : SUCCESS in 21m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5787ea3f62e64511817cd6f550b61d05 : SUCCESS in 21m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d76145033d10457d9dbd0b576eb9d0f0 : SUCCESS in 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5992f6c00714e5a3518b0e87c848a72f64cc828e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-24 16:18:19.000000000","message":"Patch Set 2:\n\n(11 comments)","accounts_in_message":[],"_revision_number":2},{"id":"78a305317ceea739bb908d8402e4b15e991e5173","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 16:36:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e43929b93c844aff9dc4c1d4a9d7ed2d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/52a90f22b00c4e118008fd20170d2bb9 : SUCCESS in 1h 20m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a777e8c7183348dab7efa04512cb5052 : FAILURE in 1h 55m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d66986560ed24e61a1102afe3f2a95a2 : SUCCESS in 1h 41m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ecae6e605fbf434e898d63ed917d1b7c : SUCCESS in 2h 16m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c552a68c2d6545fd9c0d102c1ce69bba : SUCCESS in 41m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a7d0dcf847d4024a6a137ac3f54e090 : SUCCESS in 44m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9835a095841c4de2889abafa4640cf0c : SUCCESS in 11m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7295f76e8c7e41358eafa3a580d8fcec : SUCCESS in 31m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb352754856a413cbfbc8585bc22e37c : SUCCESS in 27m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68e9be21472e4b48a7d991f13e1d603d : SUCCESS in 27m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/830f9af927a74e478bcc5799fb98f18c : SUCCESS in 26m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb4713f61d004b6bb270f3ad75b1c4d2 : SUCCESS in 11m 56s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4c6d1bb4b67c46bc8441c6b878e01ec3 : SUCCESS in 34m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2655322ea7d4f109246bb0aceaed5c5 : SUCCESS in 1h 06m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/76e6607d67bd413bb81035c02f36e67f : TIMED_OUT in 2h 32m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/30b8ea709c094ca9930ad1588fca866b : SUCCESS in 43m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ae06bef05354410d8e1271997c034346 : SUCCESS in 1h 03m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/937a46707cf14ba985c23579ce139149 : SUCCESS in 1h 40m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c41374d35c264a9ca36763fb921c1256 : SUCCESS in 1h 25m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8fb042cd3ca04d84b79b522b79eb1986 : SUCCESS in 55m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e81733b9b571493fa4ad11cb5e49bec3 : SUCCESS in 1h 37m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a696d2fe68ed4d51b79b67b8f1a579d8 : SUCCESS in 1h 42m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ae45ded47f7e4b5f92aa688feb669c76 : SUCCESS in 1h 13m 56s","accounts_in_message":[],"_revision_number":2},{"id":"ce347ccd22a4635ebf52ff7a65727a1e025c89ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-08-25 07:28:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"72e0a2589370441c3e198014dfcd6c3d7828daf1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 08:00:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c4f8680b6d5428d858593a01a45f43f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f325044b4a33469ca367c91f4728c60d : SUCCESS in 23m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c38fae35157e4eebbff63eb0fd229b58 : SUCCESS in 25m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b93d1a8b772d4690b328b5fa6d12eccf : SUCCESS in 23m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/26d67d03d44749d799a92448bd72191e : SUCCESS in 30m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b9d2137e43ad8ec21524e9282db8ae7f2c2e82af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-08-25 08:18:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ccb792d08642765e9b1437af4cb4c6dd835a3227","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 08:41:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/879e67454d72481e8a663283925fe8f1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc2909ee3b9a46ac944c6d37c1677fc6 : SUCCESS in 21m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7c038bafe714a12a6e6119d1a9c4c9e : SUCCESS in 22m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b525f5e28e1a430492656d9989054cbb : SUCCESS in 20m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e8b74daed30e4a2dbcd30e5d4cfc634a : SUCCESS in 18m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"77230ac0556628f3c81db3745b7156f7f874e95c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 10:54:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c998e4ee00b54804b55a0b304d0ea675\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/28c4ff3b19924e308ddebb3406078d95 : SUCCESS in 1h 34m 31s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3e1dd43404464aadbe96411efad953cd : SUCCESS in 2h 34m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a639a2f205cb4e14a83cbab50ecd7728 : SUCCESS in 2h 01m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/16388320c13b4da2810a72a8cd2b49e8 : SUCCESS in 2h 01m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4f94fad179554cd7af0a13d263e0d9ee : SUCCESS in 40m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2ab1b8703a54b5d88150d4a38042ddb : SUCCESS in 37m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3990db71950a4b55bf9a00efa9dfa605 : SUCCESS in 9m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e61eb414ef24d0982a90ddd3b9e3971 : SUCCESS in 28m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51c368443cff41c0af6f006126f93312 : SUCCESS in 28m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec9c10c2cc5d49aaa5b059340eaf4463 : SUCCESS in 28m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed688790bd4a4efea5dd809030f8b0e9 : SUCCESS in 26m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d32b6c601f4d9982fedd4601c9d0ff : SUCCESS in 11m 30s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/df8f93ca139c444c820ec80779194aef : SUCCESS in 33m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/749d6dc006d14715982a0d6059f695ba : FAILURE in 43m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cb5fa4fd903421a9504f2cf41f162e8 : SUCCESS in 1h 57m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b7939c3336141a3a39cb2908f551568 : SUCCESS in 1h 02m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/19c1854d50224f0d84b66a0d0595e7aa : SUCCESS in 1h 03m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fa39be01165d4fbc9f054fcacd34aa8d : SUCCESS in 1h 32m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1a3e02b256d54b56acddb953447c3e4e : SUCCESS in 1h 29m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/63f7eaa144d547f880a175c19d94e8bb : SUCCESS in 1h 03m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fea77f9f650d4fe0ab9d91ab4f4da94c : SUCCESS in 1h 39m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2ee9fad8e3434f7684790c623317eeb4 : SUCCESS in 1h 45m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7b4cf150c3dd4c9b9f493ac52cd9490c : FAILURE in 1h 02m 01s","accounts_in_message":[],"_revision_number":4},{"id":"1d5ff1e7d986c1d03869e33183c63451abd0b01e","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-08-25 11:33:23.000000000","message":"Patch Set 4:\n\n(12 comments)","accounts_in_message":[],"_revision_number":4},{"id":"abb031a3cbaabc3dcbaec5b88a5c6dd57f21adc7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-28 21:51:42.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"eac06bc47b289681fc898d757190e7492c2a1876","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-08-29 06:37:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"635c8d7842a391f2ff884646e45337c130f80c50","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-08-29 06:42:41.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d1f17a191ec20ff4d8f5dd819eb490a81f66d781","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 07:30:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ec7152641c24be4b4e0e6e42173aa6a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7968eed990dc4b3d90b4eb53a14b6807 : SUCCESS in 47m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c73d6b72d1e4f4faff3ddbf9d07542f : SUCCESS in 27m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/17745fe82a1246cda3b2e4d2c98c5a54 : SUCCESS in 22m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5eb23b7f437e43caad0f4ffd2c069a6a : SUCCESS in 24m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"144e36a034c788b5c34c7cfa6fbf2f387eb1656c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 08:49:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19c90237e7db4beb8d22b55885b5ca88\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f60224721f624494a44f392c7ff065ec : SUCCESS in 1h 17m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4b118006ce044bcaccb93c0bdeedbb3 : SUCCESS in 1h 47m 40s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/edea95d0f7d24ddaa1d780e723d17aad : SUCCESS in 1h 15m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/060bce13f81347ee92f83e9d68f01671 : SUCCESS in 1h 32m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2699c4c4c8fc4b8386c92eef9293fec7 : SUCCESS in 33m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/164a3c8c3923446b85cfd19e8bf956ee : SUCCESS in 31m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0174d740dab4903b9ce56726bb2536e : SUCCESS in 10m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a35142cbfbbe466a9380f5597114ac3d : SUCCESS in 29m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e364b83d4f948e3886e86e2e8d67e4f : SUCCESS in 22m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64af03d0f82549698c1bb8c90220fbf8 : SUCCESS in 25m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15d0a5c79c8d45988c93d29f6f932f91 : SUCCESS in 26m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba6016a9da59469da9ca58d0d5b0690a : SUCCESS in 9m 05s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8475d928be694565a9c9dfcc25e6fe60 : SUCCESS in 23m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/30569441badc481990401b20b4db083e : SUCCESS in 1h 04m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b7350acc904d4c83a8716d5b0b10386a : SUCCESS in 2h 06m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7b72c2635ac429aa560e4a3bd205cd1 : SUCCESS in 1h 01m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a322bb5aa3a9441392b3ddbe7baa4b0d : SUCCESS in 54m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ea233ef9641f4d9c942ee2b4b2fcd2ae : SUCCESS in 1h 28m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b1a10b043d4542d7b82a4007ed64464b : SUCCESS in 1h 20m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/eb7583263c6a4302ad4d3d52bf7cae17 : SUCCESS in 46m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/279b8250f62a47e79a2aa15ab8fc17f2 : SUCCESS in 1h 19m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8a2815cba08146db957c3aa38cd2c566 : SUCCESS in 1h 24m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2d550dfd845e45ab9e706c326c9cda29 : SUCCESS in 1h 00m 23s","accounts_in_message":[],"_revision_number":5},{"id":"2aa6c583cf5c45d1ff9038174cb6e9f2575dc602","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-29 14:12:27.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a1b05df6c5bf658192307f9d0e836f872c2c907b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-29 15:07:02.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"60dba140398e51cef224d348467232e842d10eef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-08-30 06:40:58.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"0d34b7380d0abe271b475891cdc1f269680dbc21","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-08-30 06:42:15.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"39da4a8a476b2496f060c406f44b5e6219560292","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 07:37:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3996d7f4043f40b89aae4733f3b7d793\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/916d01b84026479fb436db50af805cc0 : SUCCESS in 49m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6dd9c527120c4cfbbd4514d355539494 : SUCCESS in 43m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f249ddaafd714c1e88578ede860858db : SUCCESS in 50m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/10bad9e6303f4a7a8bf0de740901cb92 : SUCCESS in 21m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"da7d11560cb5fc0336510e5e591c16d3365f342b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-30 07:52:09.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f941e0a47f78af379493ae91ff37acdc91349c3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 09:05:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a0edaa9f89643bd980c9b253e856359\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e0b92b79530b4e55b59dc1028d313168 : SUCCESS in 1h 26m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/45f1e582fbee458e87bf375e6da180fd : SUCCESS in 1h 46m 00s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e1fa1674ff8f4f6e866afab4f8b2288b : SUCCESS in 1h 41m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ba821e06fac44cb9b2af0a811210a007 : SUCCESS in 1h 37m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ea9d9924c4247c9aa9e80f480195079 : SUCCESS in 38m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d16888fd024849f8885fcf642a8a9bc6 : SUCCESS in 34m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e34bd4dc80814fbaa3a7eba68e024706 : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19d34efd377b4296a802f159306aee36 : SUCCESS in 23m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/076334d64ca64982a04ebfdbf5c4d671 : SUCCESS in 26m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/451a2f3f962f4f71b86468006fe6ac6c : SUCCESS in 26m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/896cee49a78742bda883986e56c94092 : SUCCESS in 27m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a33c42ef9fe04cb7a8960f79b6ff9997 : SUCCESS in 10m 40s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/884bc699e6f243a2bff17820d35853f2 : SUCCESS in 30m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/432d1584a4874a07a9396eb24289918d : SUCCESS in 55m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/686f5e2593be468a921e035ec47567ac : SUCCESS in 2h 01m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1692d1ee4f1a48adac219ee5285a0d2b : SUCCESS in 43m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a12510f0ba53479b9e513b6c217fe519 : SUCCESS in 1h 04m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/507a25263d274ebd9651d172fc54e1f4 : SUCCESS in 1h 33m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/beb2091bb276444d88e074a4a43cefe1 : SUCCESS in 1h 31m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/71bfecd37f4b4f43841c2b65284eded8 : SUCCESS in 1h 05m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/be3efdd3e9aa480892261b0f3da917f8 : SUCCESS in 1h 48m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e6aeebca6bd04b2d9d9281504ddf6d22 : SUCCESS in 2h 18m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f3b0cff34b3b4257bce604f41a8b09ee : SUCCESS in 51m 59s","accounts_in_message":[],"_revision_number":6},{"id":"e1cf200a1cb7e7298b2e9a36bfb9af81752fc027","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-30 20:47:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f01ad434561699987395edfb4f4af908d2bd18ed","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-31 08:13:50.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/04/892604/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e51106f89ba7648c7e32731b514f18f108fe8443","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-31 08:49:14.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/04/892604/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"838e5e929055c54125385d41af106c5e86226d72","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-31 12:22:47.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/04/892604/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"39b3d212cc9dc8a0bfbc61508bf5fe6754f65e5d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-08-31 13:00:18.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2023/Arista_Tempest_Testrefs/changes/04/892604/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"3603a9bdf6d939ee2d94d5c4873d185e2af12db1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-01 00:43:02.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/04/892604/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"2257616fd0de78a2ea9cdd9023e66a95cc4310c3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-01 13:56:15.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/04/892604/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"2dd9b13c221410aa04d3a6943acdc15245cd82db","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-04 14:36:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"76ed76180abc9d5b336f0047179b1387e05566fb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-12 13:16:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4c53f5c3d0875e4aa7a876243ac06c451100db72","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-09-25 07:00:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d3ca157d2c9db3c688fe53720e250658e77c3ac4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-09-26 07:04:01.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"a11f809c7de0fa8f532f03f410a43fb5d9d87824","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-26 07:38:10.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/04/892604/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"1548f43e06d4917d12f811d4ecf7360f96f4bfb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 08:02:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f3ee6262b784ce6bfd8a35111278a33\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77a9d865194349d489991bd4a5ed0800 : SUCCESS in 53m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a3eafc2934a49c0831577e394aa0277 : SUCCESS in 52m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ab43afe98c9a44c0ae93acc324e36ce1 : FAILURE in 55m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6951d90c040a4b09b44a596821691b07 : SUCCESS in 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7edf20d91c96929503152d7c1e85c3c63931451b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 09:56:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8afc56d91604896a03ec3ddc52b9ec4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fb39611b9d3249c793c0640a4f41311c : TIMED_OUT in 2h 50m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/03e737fb08cf4bc28f9a09002bf8a726 : FAILURE in 1h 45m 16s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/daae17204b014cff800f3b43a770bcb7 : FAILURE in 1h 55m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7c1dc903314545bfa633cadb33a2a03b : SUCCESS in 1h 53m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a57aa2604564cb09b4e67613678f447 : SUCCESS in 33m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47c9ad2f81614e3192b4c28024c6851d : SUCCESS in 10m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab767b1b3f42469190605dacf87a0ab3 : SUCCESS in 25m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac955da9930e4e3e8abba10647d1630b : SUCCESS in 28m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2dcdbb040e3b45318436c59d4a153c91 : SUCCESS in 23m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/062bd40bb35648119775dd03ad53e108 : SUCCESS in 21m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/231b6aeb5e4f447aa955d62e683ecf02 : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38b8275384f04ab18763ac5112cf149c : SUCCESS in 18m 35s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/431217e12aee4a2b86b89c09fa637db3 : SUCCESS in 25m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e897bf6f9fbc47d0881114012c4553be : SUCCESS in 57m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/179bd4d7610e4ee59594bb179d5757dc : SUCCESS in 2h 35m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4256d8cdd65b48fe83914d3d90b1f948 : SUCCESS in 37m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fdf9d31e358b427fb3211182a61d80b0 : SUCCESS in 39m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/10c0e294273b4e51b3ede0e0456e3f6d : SUCCESS in 1h 14m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a80d3b5f0fa54c88aa9649947cefafd7 : SUCCESS in 1h 17m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/53797a49663145dbbb2d3717860b96cb : SUCCESS in 35m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/73eb13e95f9d493097430c1a53e0eb85 : SUCCESS in 1h 19m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/517947e8d19343f68c47900719657ca0 : SUCCESS in 2h 03m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ec8aacebe19b4c1b835367528d85de5f : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":7},{"id":"ade38f417f1801c50d1e7ef23ac0eb410cef935e","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-09-28 11:41:53.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"39908ba02a6fd89596f0f8fd831b95bb13325a1e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-04 02:11:30.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6c0e878a58f074d0c9d0cee3f2aaebf4c07ad702","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-10-05 06:17:37.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d2e73412f5f26937c10bfdb8dc48d2bd8380d329","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-10-05 06:19:49.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"811000aa956658a1f9e522fadae2f32c872f1e05","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-05 06:50:37.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/04/892604/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"a55733fbf1c9ac4bfc16be59ec4e71a15abd1740","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 07:20:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b25ae55bf594d45bb9269dfa62ec9af\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26f65f8e3f7649fd9825b479b5a07d35 : SUCCESS in 59m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a3d61771f554318a3f11f26112f85b1 : SUCCESS in 59m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e093a845cc98454aa5203d0d86ea685d : SUCCESS in 58m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a40c2bee5fbf4287ba40df3535a713a2 : SUCCESS in 52m 32s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"943a3242bd1faa55bae64603952f808cf458f769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 08:39:38.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/173d27032cd6467ea5bbfbde1f92155d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1a79c859f8ef437da846c65026d37659 : SUCCESS in 1h 32m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8a08e4459a0e47178fcb57d36d09e072 : SUCCESS in 2h 04m 24s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b10a3051eabd4e2d879475d448888511 : SUCCESS in 1h 41m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b0623576f4404dd3a79622e00b9e8714 : SUCCESS in 1h 46m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba5f3164c1dc4f338230d516c51a137c : SUCCESS in 30m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/661d35a8eab94cafad2a04553af20fb3 : SUCCESS in 11m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7485165e02b74c749009b78af977b561 : SUCCESS in 22m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/493e0d984c75446bab7a61b4f2222271 : SUCCESS in 19m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b36003c33004669a3f97432f6eda8a8 : SUCCESS in 27m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03afc2b9298c472490ff7738cc51ca99 : SUCCESS in 22m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2f2e5fc8479421eb68deb5c5d2d3ab1 : SUCCESS in 10m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ea0e44eb5a541d8ae14719fe32959aa : SUCCESS in 16m 10s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b745efda0a354026a3117493f19156bb : SUCCESS in 29m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6212e1b81e3d44ee8c323a1d2abb5bcb : TIMED_OUT in 2h 12m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fca02aabbfb40b694e368869709c268 : FAILURE in 2h 16m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3f5c65c6b464580bffed94c361afd00 : SUCCESS in 42m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/10065f199e804cefb2cfdb7a5c44ceb2 : SUCCESS in 1h 02m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/86ddac72f16746e797298e27f165a8e8 : SUCCESS in 1h 21m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/50cd1ebb52db4197be7a3b3c7fa611a8 : SUCCESS in 1h 00m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6968dc3dbb9a40bd90ba296c96ce4e5b : SUCCESS in 1h 06m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2f5cba9f0d0443c3b04ac884468d3cec : SUCCESS in 1h 54m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/290cf71c3b1a4620b98e23404cb0d44f : SUCCESS in 1h 25m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/34ef1aed846c4f3cb0782f3434421529 : SUCCESS in 1h 02m 04s","accounts_in_message":[],"_revision_number":8},{"id":"429cd0124320d471bed4b921fbf30800234f60a5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 19:27:27.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"53159825dd2103ac31c1fc3d326ce1af652d9ac1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 20:23:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76f641cdf8fe4ed6b6deebdd21c93571\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/109b84a0ecdc4d92bcac893b830f8dcc : SUCCESS in 53m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3db7b5e9f59d4c04b0d6eecce2e3b9f2 : SUCCESS in 22m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/68aaa398f2614016b4b729d8182b0701 : SUCCESS in 49m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"87df46d372da03c0b6c0960915e306c61fc22f19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 21:38:41.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8b151c3c6d9481c85d9a63d220582a0\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3e7f37b83fba442a848beeb5e63e515b : FAILURE in 1h 40m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0d4bcf35e172457dae506c7b64d4c90f : SUCCESS in 1h 40m 35s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/8f43ffde1f654ab79a66a1e510a2415f : SUCCESS in 1h 55m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8423f2177c7341baaef4ce4740b48580 : SUCCESS in 1h 57m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02470b3491dd444eb0b26e2ed594a197 : SUCCESS in 30m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/453c0eb3ad7842dfbb207f4537607f31 : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c52b9e08ea0a4d449af6de9a3b7a8583 : SUCCESS in 21m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b6984a50cea41108984cb41705bc565 : SUCCESS in 18m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd1ad5f1ec7746d8b2a0e5099f22f614 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75549073796843adbbc90a93e350359d : SUCCESS in 12m 39s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/10ae5974165f47da85e6d55800e9638e : SUCCESS in 28m 45s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2afad945dd5f4813b423a6262d151758 : SUCCESS in 47m 34s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1b701662d9864d14b28d4d63aeb1d797 : SUCCESS in 53m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6ae55cfc7844319966fc71f1f1aa4bd : SUCCESS in 55m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/632ac956a5234d92817ee333c68cfd2d : SUCCESS in 2h 07m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/deef1eb29c154b15b061863a679a00a0 : SUCCESS in 51m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e50c88f8e02742bb9c0a86b4b1bed111 : SUCCESS in 53m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/03c27f91923b4ee49d7931a60626d842 : SUCCESS in 1h 25m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/48cbb77808cc438489cfce338f5f4a36 : SUCCESS in 1h 22m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/30ad59c9090b479583993976ede7dda2 : SUCCESS in 54m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d8eb61c2a95c4ad1a946e70759ae08a0 : SUCCESS in 1h 39m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5cd07bcc466145b3be0ba48c6fca82a4 : SUCCESS in 1h 38m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d62b1afa1cc64151b2e0144487a424bb : SUCCESS in 1h 11m 57s","accounts_in_message":[],"_revision_number":8},{"id":"55345a15c03baa7262e097d964a6fde2b2a9dcd5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-12-04 11:02:17.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":8},{"id":"775f4998b1a920bb2d7751543c47f0f4e9985e70","tag":"autogenerated:gerrit:restore","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-12-04 12:21:41.000000000","message":"Restored","accounts_in_message":[],"_revision_number":8},{"id":"b16f77eeb61315c52774d70d28cbde3e5afa9af5","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-12-04 12:22:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2fcc7c90e2028c42cfb3e7cfe910509bb7667529","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 13:04:32.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0fbc50adfea49278c0f1f69112459dd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a7e403ec2d54a61998b6088101e3aed : SUCCESS in 16m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b2e40498dc364d23a277eb8cd10f9693 : SUCCESS in 40m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/17aba212109546049ef5528cdb7d9a2e : SUCCESS in 16m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0ef2ad90027b5ddbd41929942e00eee4d636cbe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 15:25:40.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/91b03a19176a4248a8a2293c3dd42ffd\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/971d9d2d389f49dfb76664f6b6dd1e2d : SUCCESS in 1h 23m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a58887f75283410fa0cefa04e48348f0 : SUCCESS in 1h 36m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/3d0960243e7a4a82b43b83233329139e : SUCCESS in 1h 33m 05s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0cd6a99a2cbb4a10b45ef6bd04f911a1 : SUCCESS in 1h 47m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a50df580fc6c4604a767a063055a85e0 : SUCCESS in 32m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b0abc11736e4cc6ae2706c0a8e94f6c : SUCCESS in 12m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80ed93605f6443e5959652d30a72600f : SUCCESS in 24m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/799eab51d0c04b6f815ba86fed2e6af1 : SUCCESS in 21m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa5ff90528914b1c8498c2c42020339c : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76e9204f2b6c403e82ecc408d4396d9f : SUCCESS in 16m 39s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/73047c4ab2e5478c9c8d72e2e1015d31 : SUCCESS in 21m 54s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/af12d20107574f39ba3ea2230ee197b5 : SUCCESS in 46m 28s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f8dbbe48937640e999fe4b5c02486691 : SUCCESS in 49m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1543a8454604cafa40e59f7967b7c26 : SUCCESS in 1h 06m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d658f1a68dd740e6968543b2daf7148a : TIMED_OUT in 3h 02m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3cdee4b4c3fc4362a8a2e46f732ab3b8 : SUCCESS in 47m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fd0e78a852c5488092e17624e4a712a7 : SUCCESS in 1h 02m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2110774bd70441e386a36fd7178da60f : SUCCESS in 1h 27m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/377769ed340848f4849e5ffc33474f2e : SUCCESS in 1h 24m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9a240be591a34bf1a7caf4f2f37b8859 : SUCCESS in 44m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6697d479ec46478a97422238f608f5af : SUCCESS in 1h 28m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/40c198cd18a64758ab85cccd0eada798 : SUCCESS in 2h 13m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bd9a4d4ddac74c0280ff03b865ce15b1 : SUCCESS in 1h 17m 53s","accounts_in_message":[],"_revision_number":8},{"id":"a0534df6ace0286c404a276224b9f0cddce61a79","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-12-05 06:53:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4ffb88470afe1526da74844d667668227ff86ee1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 07:31:10.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33495e0d23ed4dcb80266f20a93e684a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d564bbd3da804212bf30fad45282c02e : SUCCESS in 20m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/95ec5eba7103441e8ed32a6dd3dc0a7b : SUCCESS in 21m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1135339984574ce7a77f1fb0f53ce4f5 : SUCCESS in 36m 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"961540741f0cf6cf9aff5e52730dffd4523dfc38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 09:51:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/123191dd88bc4ec0bc9a112afbd1508d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/102dc5568abc4af881c3a912dfb46863 : SUCCESS in 1h 29m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c606e970da7746ed9587e7e2dc1c210a : SUCCESS in 1h 38m 59s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/862b0e8722ee45aa897843ab1a482c93 : SUCCESS in 1h 21m 07s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/13842c1aaebd4745a4ef0313fea7472e : SUCCESS in 1h 26m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4f94d02772d4c0da0555932b897c9e4 : SUCCESS in 39m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/930a47771f5c4060bfc652ec76c7678f : SUCCESS in 12m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f79159154a547e89ec8fdf0095a5242 : SUCCESS in 27m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ffb7682457f048e5a33222a40b766be7 : SUCCESS in 23m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03edee3f9b96405eb5c71a856dea41aa : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65e9eeb62ddd49bc9e2a6b9f77729baf : SUCCESS in 15m 04s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1ac4848924ef41758b9abf77e7e04280 : SUCCESS in 30m 00s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c6e71ab03f6d4829a0793d52424e53de : SUCCESS in 45m 59s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1d7c2167e98f40a5a29e54adfba43e45 : SUCCESS in 53m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/26a49e40bfb04394a5ad1f902845b6d6 : SUCCESS in 55m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4773a1e5fc047779bc439b7a274c1b6 : SUCCESS in 2h 57m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/74473b4cc989416f86bddde882e28de6 : SUCCESS in 48m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fd8946efdc9b42efa5ab8270e6112f29 : SUCCESS in 50m 36s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a72afebeff1a42bfb3234247615287dd : SUCCESS in 1h 17m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bb86756ae21e4705ac7fb42bae68993a : SUCCESS in 1h 26m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/740b897e0f9d477f8c2724171cd101da : SUCCESS in 1h 01m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/41a54e6723e44547ba717829775d28c1 : SUCCESS in 1h 56m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/53370fedce794a5a91cabd1dfff17754 : SUCCESS in 1h 56m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0ccbc34b13cc4d35a62eda0683ea4b46 : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":8},{"id":"de76273fa9a0a44588a60fee1c7a86e2c2fa7d28","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2023-12-05 12:33:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2357aeebe73306284b405146d75019d4146214cb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-11 19:07:23.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"81108ffcd21720350df321dc29b1b260426f6193","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-21 23:47:01.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"efe316c4c0f4af3442640a057fa52954be5d9ee3","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2024-01-22 14:58:27.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a7baff7d97e88037b614bc9ef64df9b26f5dd1cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2024-01-23 12:54:34.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"7870db0ede00455b6be0b940eed800cfc2d0882c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-01-23 13:00:43.000000000","message":"Patch Set 9:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/January-2024/Arista_Tempest_Testrefs/changes/04/892604/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"8c36a89d46770e264949d2983a5c08719f5cf545","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2024-01-23 13:02:39.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5acd8b497b2b840a5e46a59e3d81cd352bc25848","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 13:24:18.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/360951afe59e41c890e2a17f8a39d1c7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3ba3856a7c44ed9b9b32e090f5253dc : SUCCESS in 23m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b4b4d2dab74426cadcaa20aa7bfed7d : SUCCESS in 21m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dd803ddf4fb84839bfc29987010fba48 : SUCCESS in 21m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"eeeb5aad0ed5e2f19a957bf925fa9331575a4651","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 15:25:53.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8624deabe8649da99cd22d034a7f977\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/31a2289bcfa147d387bce489c991bb1f : SUCCESS in 1h 57m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0032c88ebc1b4f0e81b536b18d158884 : SUCCESS in 2h 11m 26s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b5f8b8c3a9cb473ba82e99028ba85750 : SUCCESS in 1h 14m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fc18df89feea47138a5723c270e453f9 : SUCCESS in 2h 05m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4cd86486b8a49138574e1baefacfa2b : SUCCESS in 43m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8dc2dbf29094afbbdd51f534e9fc097 : SUCCESS in 11m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0569750c13a43f9915e78397afc744e : SUCCESS in 26m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11f777f7ded34685b1e13f95141a36e9 : SUCCESS in 23m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e69224c72e24efb973448d98fadd0b3 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1510ec40df845a88e41d276c11767e7 : SUCCESS in 13m 43s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/70fbd684fd42411680bf05522cfd55e1 : SUCCESS in 29m 25s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f2a50da338ad4ba08019dbe9561c6e48 : SUCCESS in 57m 37s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e77063844b7d4288869b55c9b45301ca : SUCCESS in 1h 03m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7be41dddffb94a2c902fa11c9e10980c : FAILURE in 47m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebcb8ee8fd8c46abbc2438cec0729509 : SUCCESS in 1h 47m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e626748bced94c41ad760be6d615e1f8 : SUCCESS in 1h 07m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/13af627c1287497082d4ea593f3722e2 : SUCCESS in 1h 11m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fa217aea12d4413998e0abb2409cdc6d : SUCCESS in 1h 26m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ab4df806cc804df1b5691a64a44cfe52 : SUCCESS in 1h 26m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1b53c183eb31449986ecf286285ff52a : SUCCESS in 1h 15m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/afa83a6036a5493d849acee800c3e422 : SUCCESS in 2h 01m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d52f185ba8b046cba8c3f87281ac7090 : SUCCESS in 2h 09m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4a1c5c7d8fa44ebfb28c9edcb051901e : SUCCESS in 1h 19m 36s","accounts_in_message":[],"_revision_number":9},{"id":"45adb5022453d764932efca9ea9e2a4875058c62","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-23 22:41:46.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9f27af3de60a25e12915ddb41e1cc627c5c1c8d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 23:31:30.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/746c0960e9f7462f84250f93d6eae903\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3f3215dbddb412faa3c6c7830e940ad : SUCCESS in 43m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e9c8dab9213d4dbd9b51b13331008169 : SUCCESS in 44m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6a7acabad25e48deb361ce33804960fc : SUCCESS in 20m 26s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6177b9a9697e33aa1a45c17a1fb5f20be7ab67b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 00:57:48.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d594d1418d694d87b74ca17a7c714f90\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c23bd8dc18654db194955dfe3339079b : SUCCESS in 1h 39m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0f94811f7a014a9eae441074b3ba8e79 : SUCCESS in 1h 38m 27s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/f50e8e7bed4d4fd2836c05e5f85dbd5e : SUCCESS in 1h 31m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ac0401dcb02648108776508a01b48399 : SUCCESS in 1h 33m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac9324c433e045e1a3d1aeb5caeb1005 : SUCCESS in 36m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/928b880f499f4cc49140057066469b31 : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/469599830acf4bc697c90eb1021b4757 : SUCCESS in 25m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/327b292b27db4725bba7d88d8301d99c : SUCCESS in 23m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a510a6340349abaa65584b25acc9c5 : SUCCESS in 10m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f739934a0894a6384b5a07aaf4e92ff : SUCCESS in 13m 04s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f869bafd3d4843718c360f87500065d1 : SUCCESS in 29m 03s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/81e548e49f17451d85286c4314e46837 : SUCCESS in 59m 26s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/71a44b71954645afbec83eb29573b21c : SUCCESS in 1h 12m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdae3ae808764d67bf7f7e68a76b5871 : SUCCESS in 1h 06m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a74dbedea2a8485e925a3ac995cfc37a : SUCCESS in 1h 49m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6283a14446ba47869165e83debe6763d : SUCCESS in 1h 09m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/98e3e4f89d584924ad3aa3cda41cc4c4 : SUCCESS in 1h 07m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c73d6c7f2f7b41548ed826ea5ae3cb94 : FAILURE in 1h 17m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/46740146321045b4b612fceea5409fb8 : SUCCESS in 1h 28m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8c94fe3de86d45e7b85c36fc98f74239 : SUCCESS in 1h 03m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/41998db38cfc4cb586b3953f8299a16e : SUCCESS in 2h 00m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cdcf2501e96b44b6b711f8ebc016702d : SUCCESS in 2h 05m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/94448ee80d3f4d568030f24796b95bce : SUCCESS in 1h 19m 46s","accounts_in_message":[],"_revision_number":9},{"id":"e5f8b0ed57abc512c50ae985c7b6fb76aa3e8c08","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-24 20:46:12.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c6015144ce43d6e510e65eee72c53184833da85d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 21:40:55.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a5723e479ff411abc6b39b62a7d285a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/954444f50a88498b95a1c15aa160021a : SUCCESS in 19m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/189fbbe72a934c12845344b58d4f9e43 : SUCCESS in 52m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c3c75c270baa419a952df51fbd23ec63 : SUCCESS in 19m 56s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5763840a62a985698b821e1f9698a6629c624ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 22:54:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8084dde27ae24f19ab5cefd906c4063b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ad624b3b813a417ea8649abbeed445c0 : SUCCESS in 1h 50m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/11b2611a35a641388918f499e2abdae0 : SUCCESS in 1h 58m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/95469fe7bf9c4211bfef9750cc7f006c : SUCCESS in 1h 46m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7825cbc037e54a2da73de7f6c8dc7589 : SUCCESS in 2h 04m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dcc4835fc77428085ad2a33f77f2524 : SUCCESS in 38m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c57af89e25450788e70558e2f5f6ee : SUCCESS in 11m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d702ecdf43464480ac6c19ebbe16f0a7 : SUCCESS in 26m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84b14ae9334f4aafb3eca7c39fa66c19 : SUCCESS in 24m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ccb7bbaf8ca404390520212facf3102 : SUCCESS in 13m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a130186475143f798834403a215cbdd : SUCCESS in 12m 51s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2ce7f16dee3b42259685c1a12b2ba6bc : SUCCESS in 30m 26s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2671a6fe25714b8c8fe97c12ec77f6a3 : SUCCESS in 56m 51s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/869f3d5c117f4d03bebef2671ec97342 : SUCCESS in 1h 17m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a657ca966124e2b9c35036085f1bfe4 : SUCCESS in 1h 05m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/006544f471e1481db6300dcc686a5d16 : SUCCESS in 1h 55m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6a7e73b11e744bda6effbf84c60d5bb : SUCCESS in 1h 44m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bc01dd33a89d491e8bde7422ad3d21fb : SUCCESS in 51m 43s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/89f34f888f8a4132ab45a9a3d9c7808c : SUCCESS in 1h 39m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/883a73084acf4934a69d87da949ac35e : SUCCESS in 1h 16m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a2d7f88a4b4647ac94d5c62732907fde : SUCCESS in 1h 03m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/98682116b2514a7ead551e559ec67a19 : SUCCESS in 1h 51m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8a32831fc1804424b120456487c44c75 : SUCCESS in 1h 32m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e9eb2608178a4b76aceb83df11a349ba : SUCCESS in 1h 31m 02s","accounts_in_message":[],"_revision_number":9},{"id":"a17858a637fb8b8182336af4af5a9a1f1857ce3f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-25 06:44:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3f740599eae5341f1df321dec5a453cd7f2ef8b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 07:24:51.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45b42667f1b0405d9b218eb946195b32\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f54d32fcc904296948ad9ee2ec7161a : SUCCESS in 18m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2bbf8e0fdbf646d9872625b261a82a4d : SUCCESS in 19m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/da9f91aa7a5c4f83a0bffbdd1d02ed6c : SUCCESS in 38m 11s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"86f11582ac07b0fa4dd0adc60a10315815ad3efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 09:02:46.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44a1c83367ee45b38dc86db7c2a2808f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0d666c92d4b543eebb3275b44018f8bd : SUCCESS in 1h 34m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/89ddcc4a395949668855cc18d8e88396 : SUCCESS in 2h 07m 12s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/2d78e14571604f29a84afb5f6894d95c : SUCCESS in 1h 58m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b82689ade6eb4faa915ca70d6fd1d4dd : SUCCESS in 2h 10m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb5178a4224e4efd81a759676f95d26d : SUCCESS in 38m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/562a4325a692434a99e0d3f7d65cad3e : SUCCESS in 13m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acd98e47d0204d6ab8b6e89037ab83e8 : SUCCESS in 25m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d014567bfda4460c944c848c2ddd582e : SUCCESS in 27m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b80b352f216404ba5124de6a426c211 : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e056f6442c4f487ea0c3e63d0c0c7de7 : SUCCESS in 12m 29s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9f9a3c112003436698d809dfcedbe78e : SUCCESS in 28m 37s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/ab47caa484e94c6ea3c418c67a7d5fa0 : SUCCESS in 50m 35s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1da68e35769e4eb4ac530755aea71471 : SUCCESS in 1h 09m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/073bf811e4004b259dbfab05e5f58102 : SUCCESS in 1h 00m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/405e4a359fbd4d579a392caa3eeb2ff0 : SUCCESS in 1h 36m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/79bfd97ea44e4811a909eeb5542d93fa : SUCCESS in 1h 02m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e9da112998f047f39212aa119abdf044 : SUCCESS in 1h 07m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1d4d42620cae47f7abea83ca630ef08d : SUCCESS in 1h 29m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6aba47c4e4614c5f97268ab787dbf383 : SUCCESS in 1h 15m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1a830ff03e62493e8da6be5ee926f222 : SUCCESS in 1h 08m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/39a4b29e02374278969dd8a7193f7b13 : SUCCESS in 1h 49m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2f521994617a4e259805ae2663bc4299 : SUCCESS in 2h 03m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/77ebbd386dbf4d03b960eb5a7fcc3c82 : SUCCESS in 1h 16m 35s","accounts_in_message":[],"_revision_number":9},{"id":"a017503ecb2e48b5731b3ab82c73afdf2531528a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-25 11:01:33.000000000","message":"Patch Set 9: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a5fcc67eecd60b5192f39284cf0ede02451b77ea","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-25 15:38:39.000000000","message":"Patch Set 9: -Code-Review","accounts_in_message":[],"_revision_number":9},{"id":"3d1323d95e9fd5f574b3e2d5f41d9dd050145687","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2024-01-26 14:22:23.000000000","message":"Patch Set 9:\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"537150b64030b8d824490b484d86ee6113432e15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2024-01-26 14:23:10.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"514a537724b91672f6838a7e5bfb35fbd67a8e33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 14:44:24.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8d08aa02d644a43b8f1116d1627b933\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/381669dbac424599a71cd0b2d32d3c90 : SUCCESS in 20m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0d124f4169b54244b708e1f66751b481 : SUCCESS in 18m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/949e14ac58974566ba3a7e6012c1a0b3 : SUCCESS in 16m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"68a5969b30672ff4599d177170cc2a0971cefc17","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2024-01-26 14:57:34.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/January-2024/Arista_Tempest_Testrefs/changes/04/892604/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"aa307ba3501fd609b08a1961705ee043467e6efd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-26 15:48:12.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"709928caa3d286afadd31e9cd40b408076a78db9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 16:14:36.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/615043d8e39f490ebc39485f9dd95324\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f6c85584aa6849cd9889c211b8e44f66 : SUCCESS in 1h 18m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cb839710c5784a8290edf66b0931b792 : SUCCESS in 1h 25m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/9fd1ad3603044f1fbc3513dc49090152 : SUCCESS in 1h 49m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ffd3ead91d0f4d3b9f49e36edb913730 : SUCCESS in 1h 31m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9dd019c96424485ab153ee1feb4b55ea : SUCCESS in 41m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c5c0c03cd5747cdb903567009ba29de : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4a8512f1c474f968dd25e48fa8195fd : SUCCESS in 27m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f6d175e8c99495384a68d80b3a103e6 : SUCCESS in 24m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bbcb6c3d811413bb3878598ab897e4d : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acf300835d204682be6d4d68e34aa4a4 : SUCCESS in 13m 58s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c48468391e6140858f35519a44ddffed : FAILURE in 31m 40s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9968d6d1d0e94c16a2b6e17c87f1cfb7 : SUCCESS in 1h 01m 06s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8e418a5bae20428ca51afa949754d5d0 : SUCCESS in 1h 13m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5199b3d93d2040ce9555863f810f5b5d : SUCCESS in 59m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8f86867c82b45a3a7a2c3a4b0ca2c55 : SUCCESS in 1h 44m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4c5ed5b861b434d8f736fdaf1d27c04 : SUCCESS in 1h 06m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7177b5f0cfcf43fca39cb35b4dc50451 : SUCCESS in 1h 04m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a51b90907b0b4e3d8757d35b31ff649e : SUCCESS in 1h 31m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6c1e207b7779453f8f42bf39b7497848 : SUCCESS in 1h 27m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b682ce5a493c446f8bb2643e52c5abe2 : SUCCESS in 1h 02m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f2b12789ab484afb8aab71eab8afb52f : SUCCESS in 55m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e6b52e66b5c2439d8a308a9d9adbb17b : SUCCESS in 1h 25m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2059bd7285d549a08a01dcefe0bb2fe7 : SUCCESS in 1h 24m 35s","accounts_in_message":[],"_revision_number":10},{"id":"1e43ebdd97ed55f4c9ff5bba3120d17c16bee854","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2024-01-28 05:15:38.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"432a51b5d5ce8382f99e3c0499d39590051f657a","author":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"date":"2024-01-29 12:09:29.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4da46fd7764500978d70958038e74953cb7e2362","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-29 15:41:29.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"42856036038c7937c36b937159705759b2790376","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 16:16:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ace8118f9c584e48b808f726fd3ac002\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c52e7d3c989c4f94a194c7501c444ee9 : SUCCESS in 30m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/859b23a08bad433f9e356831379cd8a9 : SUCCESS in 23m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5a24ddd21c2a43438750c585ce1380bf : SUCCESS in 32m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9a450eeb3c10c61923c961e5e27d04d74f2db884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 17:46:12.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe60c3ea516149f5880dfc0ee7a9afa8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b8965725899a4bde80b54227ef063f52 : SUCCESS in 1h 08m 39s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dad201780edc480dad69bc55e1357098 : SUCCESS in 1h 36m 23s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/08de22f025d643829bf2a2ca19dad79f : SUCCESS in 1h 35m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/966a41ede0ec4c7cb202b900f8b43acd : SUCCESS in 1h 56m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6aea548b9f34ae783008a3d164ddcf0 : SUCCESS in 39m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b9a159f7816488484cefe4c66014375 : SUCCESS in 11m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/338f872d5d1a43d8abf0416054262193 : SUCCESS in 28m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8df9f55b6b0947a4b71d36de7657ebd5 : SUCCESS in 25m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc13aa625a7045d5a61de7d8acb8cab7 : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a04af055e13546dcab6c3ec328f15647 : SUCCESS in 15m 11s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/644bdbf342624b9a87f7e9f758a62f0a : SUCCESS in 30m 18s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/52149854ce3e473e9717d30bf819fc55 : SUCCESS in 1h 06m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/df06448ef14c48b7b424f02c974d735c : SUCCESS in 56m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/154aec3e573d4bc284cbf8fe86af1e85 : SUCCESS in 42m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c75fccbb82542cb96f8614f58eb17b3 : SUCCESS in 1h 46m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a811bfff292c4aafaec6e1e3dc399e1a : SUCCESS in 54m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a7329f3ad2db49cbabb83299e4f2acfb : SUCCESS in 1h 05m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/99d4c719789146f1ba72a5be740672d5 : SUCCESS in 1h 17m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/985d8ff026f64fa8a499e3a68aa7c37a : SUCCESS in 1h 12m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5f37644b654c47c99b4b2daed35aee8b : SUCCESS in 1h 00m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fbcd72434e6a4efda9f5eeba8da3dc13 : SUCCESS in 1h 34m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/64c699de3e344c35b074eab22466771f : SUCCESS in 1h 27m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/941e3a6361c54c8492cca4c4a50d341c : SUCCESS in 1h 20m 22s","accounts_in_message":[],"_revision_number":10},{"id":"246f9d55454103aa60249a80eba7d3f0f1004681","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-01-29 19:19:45.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"d5f180bfc911af4d84b56483320c3c2119c00e08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 19:20:18.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"eb64b341b0796aa028fbc9437324d8498dd835ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 21:23:24.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fead3722eb6467f964cb9fee1675ba2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b34fc465ddd744049a6953878bed9a1d : SUCCESS in 1h 36m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/93e0d1b5f5c84f4b95d6bb8787e7b16b : SUCCESS in 1h 22m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7e2d9f84fdaa4db082601329a0760a35 : SUCCESS in 2h 01m 54s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/80c970d958314608a8c52362abc73b06 : SUCCESS in 1h 33m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2213ac30f82446fa9bf90b98242cf9a : SUCCESS in 11m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ac3c3742223439ea2e2ef818bd2b64a : SUCCESS in 27m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e711c980f5b4811b5bc1926cf9ff49b : SUCCESS in 24m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e936526f61d4599bfdd6635e51daa69 : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8eaf56044e41435f95589af8acbe8c39 : SUCCESS in 15m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ce1da92eaff54c1a8f84d2b64ea357cf : SUCCESS in 46m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e9e67d967584fa997a1640a8469d88f : SUCCESS in 1h 22m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c21254ca5aed48fc9b327ee0ef17ca26 : SUCCESS in 1h 13m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/304059acdedb443a9b3ae02f3b1fe016 : SUCCESS in 1h 05m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/924fe6be627d400bb93696098fd60678 : SUCCESS in 1h 09m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ca55c0c049c240d5bf33267e45c3838d : SUCCESS in 52m 00s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/547a462e5acf45b993e2050077996d75 : SUCCESS in 1h 10m 21s","accounts_in_message":[],"_revision_number":10},{"id":"966f0b36d82fecc6ae912e0156f6b57304a6ccd3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 21:23:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"6397f47e43b25cb372d95fa8249314f79bcc50e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 21:24:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00e4b12928c6489e9f5f374b0ac641b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79dfb5e517e94be983862758bbac0961 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0623ac5f08847beb10dd48c01d5fadf : SUCCESS in 48s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"a29ea3724e1f6bb54b76d1b9915c13014272fdcd","revisions":{"79dfcde0822a04e469e35001e76ba1591ab1d4ce":{"kind":"REWORK","_number":1,"created":"2023-08-24 13:59:13.000000000","uploader":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"ref":"refs/changes/04/892604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/892604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/892604/1"}}},"commit":{"parents":[{"commit":"d1bfc3d70ada49c9d764565a67e41491e36d44e0","subject":"Merge \"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1bfc3d70ada49c9d764565a67e41491e36d44e0"}]}],"author":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-24 13:56:48.000000000","tz":120},"committer":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-24 13:56:48.000000000","tz":120},"subject":"Add two new l3 ovn schedulers (azchance, azleastloaded) that ensure LRP gateways are distributed over chassis in the different eligible AZs.","message":"Add two new l3 ovn schedulers (azchance, azleastloaded) that ensure LRP gateways are distributed over chassis in the different eligible AZs.\n\nExisting schedulers (chance, leastloaded) already ensure that LRP gateways are scheduled over chassis in eligible AZs. But, depending on the deployment characteristics, all these chassis can be in the same AZ. In some use-cases, it could be needed to have LRP gateways in different AZs to be resilient on failures.\n\nThese two new schedulers (azchance, azleastloaded) re-order the list of standard previous ones to add a priority on selecting chassis in different AZs.\n\nThis should provide a solution for users who need to have their router gateways scheduled on chassis from different AZs.\n\nCloses-Bug: #2030741\nChange-Id: I72973abbb8b0f9cc5848fd3b4f6463c38c6595f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79dfcde0822a04e469e35001e76ba1591ab1d4ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79dfcde0822a04e469e35001e76ba1591ab1d4ce"}]},"branch":"refs/heads/master"},"8205822e135df3adabf40f1b354966cf44e89ef1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-08-24 14:02:20.000000000","uploader":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"ref":"refs/changes/04/892604/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/892604/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/892604/2"}}},"commit":{"parents":[{"commit":"d1bfc3d70ada49c9d764565a67e41491e36d44e0","subject":"Merge \"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1bfc3d70ada49c9d764565a67e41491e36d44e0"}]}],"author":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-24 13:56:48.000000000","tz":120},"committer":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-24 14:02:20.000000000","tz":0},"subject":"[ovn] AZs distribution in L3 port scheduler","message":"[ovn] AZs distribution in L3 port scheduler\n\nAdd two new l3 ovn schedulers (azchance, azleastloaded) that ensure LRP gateways are distributed over chassis in the different eligible AZs.\n\nExisting schedulers (chance, leastloaded) already ensure that LRP gateways are scheduled over chassis in eligible AZs. But, depending on the deployment characteristics, all these chassis can be in the same AZ. In some use-cases, it could be needed to have LRP gateways in different AZs to be resilient on failures.\n\nThese two new schedulers (azchance, azleastloaded) re-order the list of standard previous ones to add a priority on selecting chassis in different AZs.\n\nThis should provide a solution for users who need to have their router gateways scheduled on chassis from different AZs.\n\nCloses-Bug: #2030741\nChange-Id: I72973abbb8b0f9cc5848fd3b4f6463c38c6595f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8205822e135df3adabf40f1b354966cf44e89ef1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8205822e135df3adabf40f1b354966cf44e89ef1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1183188cb122a890a6c8eff6d82023d96a4d47c8":{"kind":"REWORK","_number":3,"created":"2023-08-25 07:28:48.000000000","uploader":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"ref":"refs/changes/04/892604/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/892604/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/892604/3"}}},"commit":{"parents":[{"commit":"d1bfc3d70ada49c9d764565a67e41491e36d44e0","subject":"Merge \"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1bfc3d70ada49c9d764565a67e41491e36d44e0"}]}],"author":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-24 13:56:48.000000000","tz":120},"committer":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-25 07:25:04.000000000","tz":120},"subject":"[ovn] AZs distribution in L3 port scheduler","message":"[ovn] AZs distribution in L3 port scheduler\n\nAdd two new l3 ovn schedulers (azchance, azleastloaded) that ensure LRP gateways are distributed over chassis in the different eligible AZs.\n\nExisting schedulers (chance, leastloaded) already ensure that LRP gateways are scheduled over chassis in eligible AZs. But, depending on the deployment characteristics, all these chassis can be in the same AZ. In some use-cases, it could be needed to have LRP gateways in different AZs to be resilient on failures.\n\nThese two new schedulers (azchance, azleastloaded) re-order the list of standard previous ones to add a priority on selecting chassis in different AZs.\n\nThis should provide a solution for users who need to have their router gateways scheduled on chassis from different AZs.\n\nCloses-Bug: #2030741\nChange-Id: I72973abbb8b0f9cc5848fd3b4f6463c38c6595f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1183188cb122a890a6c8eff6d82023d96a4d47c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1183188cb122a890a6c8eff6d82023d96a4d47c8"}]},"branch":"refs/heads/master"},"28e6a0ad54b1a5c6c7162e4065787715b1daef03":{"kind":"REWORK","_number":4,"created":"2023-08-25 08:18:15.000000000","uploader":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"ref":"refs/changes/04/892604/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/892604/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/892604/4"}}},"commit":{"parents":[{"commit":"d1bfc3d70ada49c9d764565a67e41491e36d44e0","subject":"Merge \"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1bfc3d70ada49c9d764565a67e41491e36d44e0"}]}],"author":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-24 13:56:48.000000000","tz":120},"committer":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-25 08:17:07.000000000","tz":120},"subject":"[ovn] AZs distribution in L3 port scheduler","message":"[ovn] AZs distribution in L3 port scheduler\n\nAdd two new l3 ovn schedulers (azchance, azleastloaded) that ensure LRP gateways are distributed over chassis in the different eligible AZs.\n\nExisting schedulers (chance, leastloaded) already ensure that LRP gateways are scheduled over chassis in eligible AZs. But, depending on the deployment characteristics, all these chassis can be in the same AZ. In some use-cases, it could be needed to have LRP gateways in different AZs to be resilient on failures.\n\nThese two new schedulers (azchance, azleastloaded) re-order the list of standard previous ones to add a priority on selecting chassis in different AZs.\n\nThis should provide a solution for users who need to have their router gateways scheduled on chassis from different AZs.\n\nCloses-Bug: #2030741\nChange-Id: I72973abbb8b0f9cc5848fd3b4f6463c38c6595f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28e6a0ad54b1a5c6c7162e4065787715b1daef03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/28e6a0ad54b1a5c6c7162e4065787715b1daef03"}]},"branch":"refs/heads/master"},"5fdf086a916b85eb877b08bda3f6b7bc740a51a2":{"kind":"REWORK","_number":5,"created":"2023-08-29 06:37:25.000000000","uploader":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"ref":"refs/changes/04/892604/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/892604/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/892604/5"}}},"commit":{"parents":[{"commit":"d1bfc3d70ada49c9d764565a67e41491e36d44e0","subject":"Merge \"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1bfc3d70ada49c9d764565a67e41491e36d44e0"}]}],"author":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-24 13:56:48.000000000","tz":120},"committer":{"name":"B1972","email":"yann.morice@arkea.com","date":"2023-08-29 06:28:23.000000000","tz":120},"subject":"[ovn] AZs distribution in L3 port scheduler","message":"[ovn] AZs distribution in L3 port scheduler\n\nAdd two new l3 ovn schedulers (azchance, azleastloaded) that ensure LRP gateways are distributed over chassis in the different eligible AZs.\n\nExisting schedulers (chance, leastloaded) already ensure that LRP gateways are scheduled over chassis in eligible AZs. But, depending on the deployment characteristics, all these chassis can be in the same AZ. In some use-cases, it could be needed to have LRP gateways in different AZs to be resilient on failures.\n\nThese two new schedulers (azchance, azleastloaded) re-order the list of standard previous ones to add a priority on selecting chassis in different AZs.\n\nThis should provide a solution for users who need to have their router gateways scheduled on chassis from different AZs.\n\nCloses-Bug: #2030741\nChange-Id: I72973abbb8b0f9cc5848fd3b4f6463c38c6595f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fdf086a916b85eb877b08bda3f6b7bc740a51a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fdf086a916b85eb877b08bda3f6b7bc740a51a2"}]},"branch":"refs/heads/master"},"01f7fe8e6d6e6a4d033c3e6f46c7622b2e4813fa":{"kind":"REWORK","_number":6,"created":"2023-08-30 06:40:58.000000000","uploader":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"ref":"refs/changes/04/892604/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/892604/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/892604/6"}}},"commit":{"parents":[{"commit":"d1bfc3d70ada49c9d764565a67e41491e36d44e0","subject":"Merge \"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1bfc3d70ada49c9d764565a67e41491e36d44e0"}]}],"author":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-24 13:56:48.000000000","tz":120},"committer":{"name":"B1972","email":"yann.morice@arkea.com","date":"2023-08-30 06:38:21.000000000","tz":120},"subject":"[ovn] AZs distribution in L3 port scheduler","message":"[ovn] AZs distribution in L3 port scheduler\n\nAdd two new l3 ovn schedulers (azchance, azleastloaded) that ensure LRP gateways are distributed over chassis in the different eligible AZs.\n\nExisting schedulers (chance, leastloaded) already ensure that LRP gateways are scheduled over chassis in eligible AZs. But, depending on the deployment characteristics, all these chassis can be in the same AZ. In some use-cases, it could be needed to have LRP gateways in different AZs to be resilient on failures.\n\nThese two new schedulers (azchance, azleastloaded) re-order the list of standard previous ones to add a priority on selecting chassis in different AZs.\n\nThis should provide a solution for users who need to have their router gateways scheduled on chassis from different AZs.\n\nCloses-Bug: #2030741\nChange-Id: I72973abbb8b0f9cc5848fd3b4f6463c38c6595f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01f7fe8e6d6e6a4d033c3e6f46c7622b2e4813fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01f7fe8e6d6e6a4d033c3e6f46c7622b2e4813fa"}]},"branch":"refs/heads/master"},"4c52fdd1438a87e35c19e4e2a555c3cb9ae8fc95":{"kind":"REWORK","_number":7,"created":"2023-09-26 07:04:01.000000000","uploader":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"ref":"refs/changes/04/892604/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/892604/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/892604/7"}}},"commit":{"parents":[{"commit":"d1bfc3d70ada49c9d764565a67e41491e36d44e0","subject":"Merge \"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1bfc3d70ada49c9d764565a67e41491e36d44e0"}]}],"author":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-24 13:56:48.000000000","tz":120},"committer":{"name":"B1972","email":"yann.morice@arkea.com","date":"2023-09-26 07:00:33.000000000","tz":120},"subject":"[ovn] AZs distribution in L3 port scheduler","message":"[ovn] AZs distribution in L3 port scheduler\n\nUpdate l3 ovn schedulers (chance, leastloaded) to ensure that LRP gateways are distributed over chassis in the different eligible AZs.\n\nPrevious version already ensure that LRP gateways were scheduled over chassis in eligible AZs. But, depending on the deployment characteristics, all these chassis could be in the same AZ. In some use-cases, it could be needed to have LRP gateways in different AZs to be resilient on failures.\n\nThis patch re-order the list of eligible chassis to add a priority on selecting chassis in different AZs.\n\nThis should provide a solution for users who need to have their router gateways scheduled on chassis from different AZs.\n\nCloses-Bug: #2030741\nChange-Id: I72973abbb8b0f9cc5848fd3b4f6463c38c6595f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c52fdd1438a87e35c19e4e2a555c3cb9ae8fc95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c52fdd1438a87e35c19e4e2a555c3cb9ae8fc95"}]},"branch":"refs/heads/master"},"8754965d9d5e0344d50eb479ddccc1ed754e0caa":{"kind":"REWORK","_number":8,"created":"2023-10-05 06:17:37.000000000","uploader":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"ref":"refs/changes/04/892604/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/892604/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/892604/8"}}},"commit":{"parents":[{"commit":"d1bfc3d70ada49c9d764565a67e41491e36d44e0","subject":"Merge \"Add \"openstack-tox-py310-with-sqlalchemy-master\" to check queue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1bfc3d70ada49c9d764565a67e41491e36d44e0"}]}],"author":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-24 13:56:48.000000000","tz":120},"committer":{"name":"B1972","email":"yann.morice@arkea.com","date":"2023-10-05 06:16:34.000000000","tz":120},"subject":"[ovn] AZs distribution in L3 port scheduler","message":"[ovn] AZs distribution in L3 port scheduler\n\nUpdate l3 ovn schedulers (chance, leastloaded) to ensure that LRP gateways are distributed over chassis in the different eligible AZs.\n\nPrevious version already ensure that LRP gateways were scheduled over chassis in eligible AZs. But, depending on the deployment characteristics, all these chassis could be in the same AZ. In some use-cases, it could be needed to have LRP gateways in different AZs to be resilient on failures.\n\nThis patch re-order the list of eligible chassis to add a priority on selecting chassis in different AZs.\n\nThis should provide a solution for users who need to have their router gateways scheduled on chassis from different AZs.\n\nCloses-Bug: #2030741\nChange-Id: I72973abbb8b0f9cc5848fd3b4f6463c38c6595f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8754965d9d5e0344d50eb479ddccc1ed754e0caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8754965d9d5e0344d50eb479ddccc1ed754e0caa"}]},"branch":"refs/heads/master"},"f3922a8d89e7ff30c374f0255917989ed004e01a":{"kind":"REWORK","_number":9,"created":"2024-01-23 12:54:34.000000000","uploader":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"ref":"refs/changes/04/892604/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/892604/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/892604/9"}}},"commit":{"parents":[{"commit":"5ce17647c60448c9cae00ffc886d4e17a9fe0890","subject":"Merge \"Router flavors and service type for OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ce17647c60448c9cae00ffc886d4e17a9fe0890"}]}],"author":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-24 13:56:48.000000000","tz":120},"committer":{"name":"B1972","email":"yann.morice@arkea.com","date":"2024-01-23 07:04:41.000000000","tz":60},"subject":"[ovn] AZs distribution in L3 port scheduler","message":"[ovn] AZs distribution in L3 port scheduler\n\nUpdate l3 ovn schedulers (chance, leastloaded) to ensure that LRP gateways are distributed over chassis in the different eligible AZs.\n\nPrevious version already ensure that LRP gateways were scheduled over chassis in eligible AZs. But, depending on the deployment characteristics, all these chassis could be in the same AZ. In some use-cases, it could be needed to have LRP gateways in different AZs to be resilient on failures.\n\nThis patch re-order the list of eligible chassis to add a priority on selecting chassis in different AZs.\n\nThis should provide a solution for users who need to have their router gateways scheduled on chassis from different AZs.\n\nCloses-Bug: #2030741\nChange-Id: I72973abbb8b0f9cc5848fd3b4f6463c38c6595f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3922a8d89e7ff30c374f0255917989ed004e01a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3922a8d89e7ff30c374f0255917989ed004e01a"}]},"branch":"refs/heads/master"},"a29ea3724e1f6bb54b76d1b9915c13014272fdcd":{"kind":"REWORK","_number":10,"created":"2024-01-26 14:23:10.000000000","uploader":{"_account_id":23763,"name":"morice","email":"yann.morice@arkea.com","username":"ymorice"},"ref":"refs/changes/04/892604/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/892604/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/892604/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/892604/10"}}},"commit":{"parents":[{"commit":"5ce17647c60448c9cae00ffc886d4e17a9fe0890","subject":"Merge \"Router flavors and service type for OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ce17647c60448c9cae00ffc886d4e17a9fe0890"}]}],"author":{"name":"Yann Morice","email":"yann.morice@arkea.com","date":"2023-08-24 13:56:48.000000000","tz":120},"committer":{"name":"B1972","email":"yann.morice@arkea.com","date":"2024-01-26 14:22:34.000000000","tz":60},"subject":"[ovn] AZs distribution in L3 port scheduler","message":"[ovn] AZs distribution in L3 port scheduler\n\nUpdate l3 ovn schedulers (chance, leastloaded) to ensure that LRP gateways are distributed over chassis in the different eligible AZs.\n\nPrevious version already ensure that LRP gateways were scheduled over chassis in eligible AZs. But, depending on the deployment characteristics, all these chassis could be in the same AZ. In some use-cases, it could be needed to have LRP gateways in different AZs to be resilient on failures.\n\nThis patch re-order the list of eligible chassis to add a priority on selecting chassis in different AZs.\n\nThis should provide a solution for users who need to have their router gateways scheduled on chassis from different AZs.\n\nCloses-Bug: #2030741\nChange-Id: I72973abbb8b0f9cc5848fd3b4f6463c38c6595f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a29ea3724e1f6bb54b76d1b9915c13014272fdcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a29ea3724e1f6bb54b76d1b9915c13014272fdcd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
