)]}'
{"id":"openstack%2Fneutron~892773","triplet_id":"openstack%2Fneutron~stable%2Fzed~I8f3f0c2d516e37145eb298b8f51d92fe9905158a","project":"openstack/neutron","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-08-29 18:08:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8f3f0c2d516e37145eb298b8f51d92fe9905158a","subject":"[OVN] Disable the mcast_flood_reports option for LSPs","status":"MERGED","created":"2023-08-25 10:41:55.000000000","updated":"2023-08-29 18:09:43.000000000","submitted":"2023-08-29 18:08:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892773","meta_rev_id":"9ad0c9bd698edf7b3687b35716a4c43e8acdd80c","_number":892773,"virtual_id_number":892773,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-29 18:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-08-28 08:48:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-08-29 13:37:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-08-29 13:37:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-25 11:34:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-25 12:35:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 08:48:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-08-29 13:37:31.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"69f77aeb1c3e4245f49c75637281a4e5489a5126","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-25 10:41:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"634882080075c05cfafd5c0664794d363ae8838a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 11:34:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5ac8a6678de4a04903b5f4f79b1f19c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc5cc1c3edde454e9c3e80a52c4c53d5 : SUCCESS in 22m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4165c159d70540608ea3e3fdeb403596 : SUCCESS in 51m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d81a5cc439db38513b00b20d728879f17912e1fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:35:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7b269466d7046a7b644153572a87ebc\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/3b427bcb0ad44def946aedfb29a9d8db : FAILURE in 1h 49m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39b0093aa72f4315a6d22fe49b0e8c62 : SUCCESS in 34m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26af91ea45dd4197ac0f402a2af724b6 : SUCCESS in 13m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e9f01c7f32342c89bb27dffced08cc9 : SUCCESS in 22m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5174639907e94478b8847fc87baf51dc : SUCCESS in 27m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7df0967a3ccc4dd1bd3f2b1d10b1cb4d : SUCCESS in 27m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b4387f726940efa9f5b2e038934a64 : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75515b11f5ca4fde9cce60d7aab7f3cf : SUCCESS in 9m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/13653d04c3d9474bb56a3e01038a50d4 : SUCCESS in 46m 18s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4e5f892b51443e8abc4c6ab44b16262 : SUCCESS in 1h 00m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/54c8053ed02046278070477be563eb16 : SUCCESS in 1h 08m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a28203bbeb41d22c6b0ae35da0125e880285b1d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-25 12:36:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe824ecd90ff5551ce4355d17ddebe06031356de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 13:10:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bc3e966f09c4945a388812873c2fa3b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b288914675ad47619511d8ee176f127e : SUCCESS in 33m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2bc729f97759414f917bc421ccdb9fe2 : SUCCESS in 20m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8713f46507c67c673625ad565070a8f7ccd7d88e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 14:20:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac1e854c28144c9692c7302cb922d21c\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/4b8ce190e5ce483283c6532d0055f16b : SUCCESS in 1h 42m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1a1e5abe59140439b4f0d1ba4ba43d0 : SUCCESS in 31m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de61810c6555476fa1543ee38ed99d8d : SUCCESS in 13m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b74e44aebf243359a25f120f021fa29 : SUCCESS in 34m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d9a8ec5f65b42adad9bce9d448108b8 : SUCCESS in 25m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72012670bcff4ba0925afa61d8e79b7a : SUCCESS in 25m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0de0130310b4fbcbc40639482bbddf0 : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbf63df090ac4bb089ff7159e7c83a5a : SUCCESS in 9m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/202bb3f7c7fa4fa5aeef7964e2fdf2b3 : SUCCESS in 54m 09s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3df1d64d39e145a79fb965bc159cda76 : SUCCESS in 1h 01m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4a16e58bd0204c2dbb5db1a603822e2f : SUCCESS in 58m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4d29cdd831a8991a259652c9d88bfad13d18e6d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-28 08:48:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"64bdf549c070c3afb342bc71c6faf8aff9b6b1b0","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-29 13:37:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d47e60c4dac959ddfde291f3b43beedfc894186","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 13:38:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6ea6a59860c7c1d9dba56c7356d229589884110","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:08:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca6f08f495fb4fd6a2912dfffa3158aa\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/350d6d027cb74013a44807774e52bed6 : SUCCESS in 1h 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6637f2ac92641b0945f2a183ea186a6 : SUCCESS in 12m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f414349c97849128d1f4467ccb21eee : SUCCESS in 28m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/696a636a70ec402a903754ac49e0ead0 : SUCCESS in 25m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/733ca38c427f4d5798d21bda1f597eca : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3e5833271d7467ea3cbe01c062ad6e5 : SUCCESS in 8m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3793cdc192ad4131aa2d77f0bf04d5c5 : SUCCESS in 44m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1749456c1b0e4687ba26f4942df54635 : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":1},{"id":"04b2ab622366fdf76e74f3a14371a82801b1226b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:08:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9ad0c9bd698edf7b3687b35716a4c43e8acdd80c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:09:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e3821ca35a04db7adaa1c4eb75da909\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18a139b6c55a44de92ea04bb6f190770 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c8a68a4dc6647718ee0d13a3d09fbb8 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7da91baa25c295e0fda9b4a0ffe84c99e60970cc","revisions":{"7da91baa25c295e0fda9b4a0ffe84c99e60970cc":{"kind":"REWORK","_number":1,"created":"2023-08-25 10:41:55.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/73/892773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/892773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/892773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/892773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/892773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/892773/1"}}},"commit":{"parents":[{"commit":"d6ee668cc32725cb7d15d2e08fdb50a761f91fe4","subject":"Merge \"[OVN] Retry retrieving LSP hosting information\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d6ee668cc32725cb7d15d2e08fdb50a761f91fe4"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-07-11 10:29:13.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-25 10:29:31.000000000","tz":60},"subject":"[OVN] Disable the mcast_flood_reports option for LSPs","message":"[OVN] Disable the mcast_flood_reports option for LSPs\n\nThe mcast_flood_reports option was being enabled on LSPs as a workaround\nfor a problem in core OVN. The issue in core OVN has been fixed and this\nworkaround is now causing an increase in the number of actions on the\ntable 38 of OVN (at the risk of hitting a size limit).\n\nThis patch disables the mcast_flood_reports option on newer versions of\nOVN while keeping the backward compatibility with the old ones.\n\nSince the fix in core OVN does not expose any information to the CMS to\ntell us that the issue is fixed this patch uses the NB DB schema version\nto determine if this is an old or a new OVN version.\n\nChange-Id: I8f3f0c2d516e37145eb298b8f51d92fe9905158a\nCloses-Bug: #2026825\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 06dbc5227b9208887bab9bc6623098c80156f36d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7da91baa25c295e0fda9b4a0ffe84c99e60970cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7da91baa25c295e0fda9b4a0ffe84c99e60970cc"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
