)]}'
{"id":"openstack%2Fneutron~892883","triplet_id":"openstack%2Fneutron~stable%2Fzed~I81390af2ea2fc384423518b84de3acf7adaf9193","project":"openstack/neutron","branch":"stable/zed","topic":"bug/2027605","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-08-29 14:46:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I81390af2ea2fc384423518b84de3acf7adaf9193","subject":"[OVN] Skip the port status UP update during a live migration","status":"MERGED","created":"2023-08-28 08:53:26.000000000","updated":"2023-08-29 14:47:38.000000000","submitted":"2023-08-29 14:46:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892883-bug/2027605","meta_rev_id":"59ed93e51c25e47516c3aae4856b5f7cdaafc7eb","_number":892883,"virtual_id_number":892883,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-29 14:46:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-08-29 11:33:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-08-28 16:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2023-08-29 11:33:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 09:43:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-28 10:40:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:25:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-08-29 11:33:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"9457656df72cf15a043255307c2fc26985e5613c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-28 08:53:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76e0ff73e2a96bd8874193a274e768829a6df708","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 09:43:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96f08bdad8a740f7bf14a014c451c88a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e05538c52b349669c6af76bfabd3383 : SUCCESS in 48m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/edaf4c3abf014f2e8ab8d6c33577df2b : SUCCESS in 23m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17af7d066721f2d827c2b7ea8063d1d020f00834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 10:40:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee880e32bb57457c99fc120559d6738f\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/850e7cec8c3a44ef9710d5c0ca7b03ca : SUCCESS in 1h 21m 02s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/f52eba8ef4104d9fb7b33b5da58f05d7 : SUCCESS in 28m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5612987eb1f545df8fa228963555073d : SUCCESS in 33m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a0d96e2a25347daba12a2adb8ab685c : SUCCESS in 12m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73dbcbd7bc8c4e0ebc3d4781a38d4849 : SUCCESS in 25m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcabf8d2c8e3407ba678bdd51cfbdeaf : SUCCESS in 27m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5414eda2308c4fc8a4739c32a81cf512 : SUCCESS in 24m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc7038cc1404e60a550b72269408f84 : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c610766e60f04477a1b654fabcf4431c : SUCCESS in 45m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a0f9c0e4b6cc44ba826e76f561a69249 : SUCCESS in 1h 37m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b7d4d8d1dd94555b795d145ab80f3a3 : SUCCESS in 1h 05m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/88772df4d9224da6974b8e5fb3dbd2d6 : SUCCESS in 56m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24260b8e36f65f41e4027e901a1cc3a29ec40965","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-28 16:25:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cae5aad76a589a0adc30737413ea5eb3681847e4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-29 11:33:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2634aef0adbd0c53613571448c394a6af441553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 11:33:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8842a4bd439ed2e4ad6cb55c3d34cab4ad16e006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:46:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84c20a7605f24bc8aa7b4c108552e9aa\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/502200275cdd4482bc7e041e57247716 : SUCCESS in 1h 23m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac7552d43b184b06a4ecc91c8b61bf82 : SUCCESS in 13m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8372e212169b4d6dac4600ed1b7d3ebf : SUCCESS in 24m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be18e7f1ed514e1a9b6cbe7e2435dffa : SUCCESS in 24m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6bac59c85ef4a21af3aee39e975e239 : SUCCESS in 11m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/17f402e239504128809b677a8e921294 : SUCCESS in 1h 01m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9094b0a5ca5b4273a1d3680c98f67b56 : SUCCESS in 1h 48m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8f44f447d914fb39fcc2071ca1b757b : SUCCESS in 47m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ee0f6636c34274ccc2b664c66eb2f5c73d07bbff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:46:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"59ed93e51c25e47516c3aae4856b5f7cdaafc7eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:47:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db3455809699404baddf72ae2a8b259b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9200ece34a4485bf024d5f34543bad : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a9990dba1777dc5510d4fc2b249d8adc98e3e12","revisions":{"6a9990dba1777dc5510d4fc2b249d8adc98e3e12":{"kind":"REWORK","_number":1,"created":"2023-08-28 08:53:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/83/892883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/892883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/892883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/892883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/892883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/892883/1"}}},"commit":{"parents":[{"commit":"9a537fdf6cdd38cf910acb41a5cb6d27f7f68a73","subject":"Fix ovn-metadata agent sync of unused namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a537fdf6cdd38cf910acb41a5cb6d27f7f68a73"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-17 23:27:58.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-28 08:52:35.000000000","tz":0},"subject":"[OVN] Skip the port status UP update during a live migration","message":"[OVN] Skip the port status UP update during a live migration\n\nSkip the port status UP update during a live migration due to a\n``PortBindingChassisUpdateEvent`` event.\n\nThe port status will be set by  the ``LogicalSwitchPortCreateUpEvent``\nand ``LogicalSwitchPortCreateDownEvent`` events, that will be issued\nwhen the port is deleted from the source host and created in the\ndestination host. This is a planned operation and controlled by Nova,\nnot a reactive event due to an unplanned host down event.\n\nConflicts:\n    neutron/common/ovn/utils.py\n\nRelated-Bug: #2027605\nChange-Id: I81390af2ea2fc384423518b84de3acf7adaf9193\n(cherry picked from commit e1f887ca9ffa73dde8ab41590f2131f3297aa3cd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a9990dba1777dc5510d4fc2b249d8adc98e3e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a9990dba1777dc5510d4fc2b249d8adc98e3e12"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
