)]}'
{"id":"openstack%2Fneutron~892884","triplet_id":"openstack%2Fneutron~stable%2Fzed~I2370ea2f96e2e31dbd43bf232a63394388e6945f","project":"openstack/neutron","branch":"stable/zed","topic":"bug/2027605","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-08-29 18:10:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2370ea2f96e2e31dbd43bf232a63394388e6945f","subject":"[OVN][Trunk] Set the subports correct host during live migration","status":"MERGED","created":"2023-08-28 08:53:26.000000000","updated":"2023-09-05 10:52:26.000000000","submitted":"2023-08-29 18:10:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892884-bug/2027605","meta_rev_id":"f7432d52037a8ae29f2d84c5f40b9f64cbcca378","_number":892884,"virtual_id_number":892884,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-29 18:10:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-08-29 11:36:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-08-28 16:27:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2023-08-29 14:46:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 09:57:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-28 11:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:27:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-08-29 11:36:47.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"635ed14f053d70b46918e443d94b98495be8e626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-28 08:53:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7beffdbf0fd73fcd7f221b86f66fa2cb9e49bb73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 09:57:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fb0fa7c09df401386ada50dfdf249d8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89fb4fb632c6472b893b4ed4f0a40cb4 : SUCCESS in 1h 01m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71f64103f889487c8d8307b8aa528755 : SUCCESS in 1h 00m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e79e9960beb1809a27ea8e8ca58b2079c7497a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 11:15:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7307ec17354c49a98ce567ea91c44639\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/df9031c73fad41f480bf96e01b5f4728 : SUCCESS in 1h 06m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1c7ee1f30424a48a69a3871b4db92b1 : SUCCESS in 33m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d37eaffbba6d4139a8a1277d2c7c163d : SUCCESS in 10m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/feaeb4031fd04d9d970c272843403893 : SUCCESS in 27m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df935578d5a54ee9a0c65ee87d5d9c26 : SUCCESS in 23m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cda2849db0f34a9f8708917328d1bf05 : SUCCESS in 24m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1879a5be9349bc8ac796dce49558af : SUCCESS in 10m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5da6e1ba6b6401ebea49955d1fa395f : SUCCESS in 53m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8484fdcac8564875a3f6a32c5564f1ca : SUCCESS in 1h 50m 46s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04d8f0f202ae476fb3273ae10a68eba6 : SUCCESS in 54m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7a5e9fc834bd4ef7baa7f9a30e2573c7 : SUCCESS in 1h 09m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66e59c0c5edf38c70b4eb9e10fb46839441767a2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-28 16:27:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46c8b58244d4fa6fcbd3e8e67fb6812bb4faf971","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-29 11:36:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5ab6c4bc32511b34d8c6b3f07c283bfaac7e281","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-29 14:46:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"18ff8428a88e5f9cdc61d0a9cc619c424308d6c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:47:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a19285e9042b9cb991308c4eb58f21984986ad40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:10:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a8fd7619fc46c3a3916f8f153aaef1\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/b8567a3ad59442d0a8606cc08895cfca : SUCCESS in 1h 40m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb3e6be20d304e3e9d4932093f4f7d03 : SUCCESS in 12m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e5542fb16134f1d97d2ddb485c01299 : SUCCESS in 25m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2f7c652e4dc40f38bc3de97191d3f7f : SUCCESS in 23m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/898a1926d650424cad094c462d76658a : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6fec86da432048ae9de33ad883d05792 : SUCCESS in 59m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e74510a945a344a7b8ec6b542cd1351e : SUCCESS in 1h 47m 52s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80f50642e5ac40ec99e1b74fea280be8 : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5297952a6f5f49dd47b48afd84ed8ac05947c5da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:10:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"69bd565418b05282b7ced0e1c647431f5432e4a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:11:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/245b314c5e69498bac211260eb39ee2f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312de325c99b47e89d3304fcaf45dbab : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f7432d52037a8ae29f2d84c5f40b9f64cbcca378","tag":"autogenerated:gerrit:revert","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-05 10:52:26.000000000","message":"Created a revert of this change as I779b13f10f20f95c4963e0e1667566e4589beb76","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ce5ef7f83776512487980e3d1b0fd1364ffb0f1","revisions":{"1ce5ef7f83776512487980e3d1b0fd1364ffb0f1":{"kind":"REWORK","_number":1,"created":"2023-08-28 08:53:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/892884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/892884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/892884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/892884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/892884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/892884/1"}}},"commit":{"parents":[{"commit":"6a9990dba1777dc5510d4fc2b249d8adc98e3e12","subject":"[OVN] Skip the port status UP update during a live migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a9990dba1777dc5510d4fc2b249d8adc98e3e12"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-18 00:42:45.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-28 08:53:10.000000000","tz":0},"subject":"[OVN][Trunk] Set the subports correct host during live migration","message":"[OVN][Trunk] Set the subports correct host during live migration\n\nDuring the trunk migration, the parent port multiple port binding\nwill contain the destination host. Because this update is done\nbefore the migration is done (in other words, the parent port\nstill has two port binding registers), the method setting the\nbinding profile of the subport will use the destination host\nin advance. At the end of the live migration, the subports\nhost will point to the correct hostname.\n\nRelated-Bug: #2027605\nChange-Id: I2370ea2f96e2e31dbd43bf232a63394388e6945f\n(cherry picked from commit 7ed79c1f7890456488a4b44769ed84fea23c9a39)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ce5ef7f83776512487980e3d1b0fd1364ffb0f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ce5ef7f83776512487980e3d1b0fd1364ffb0f1"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
