)]}'
{"id":"openstack%2Fneutron~892885","triplet_id":"openstack%2Fneutron~stable%2Fxena~I81390af2ea2fc384423518b84de3acf7adaf9193","project":"openstack/neutron","branch":"stable/xena","topic":"bug/2027605","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-08-31 19:06:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I81390af2ea2fc384423518b84de3acf7adaf9193","subject":"[OVN] Skip the port status UP update during a live migration","status":"MERGED","created":"2023-08-28 08:54:10.000000000","updated":"2023-08-31 19:08:00.000000000","submitted":"2023-08-31 19:06:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892885-bug/2027605","meta_rev_id":"d5c9a24fb738eaafdf4f5594502f089d3c8c5eb6","_number":892885,"virtual_id_number":892885,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-31 19:06:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-08-31 16:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-08-29 11:33:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2023-08-31 16:48:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 09:18:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-28 11:14:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 11:33:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-08-31 16:48:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"83b0e66cc5ccf67ae32e8e579e8be21137f5065b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-28 08:54:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ec7d804a2529b2aab0c1bcba17f099fca69a48a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 09:18:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3420558f1a2745ed9c513ba2367fd7a5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e30cf65cd21426faf83a649b49d4f65 : FAILURE in 21m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2389edf3960a427aba4b861275573403 : FAILURE in 23m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d16e0ca53b7492569965b4253cb8100796fdc30b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 11:14:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fba23a75dec64476a454beba52d1e6c1\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/8d62fd736d174d05b39bb508a1addf3d : SUCCESS in 57m 33s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/6232d79f27114395b97aae3e4980e60d : SUCCESS in 2h 10m 52s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/82aad51c966a426c9392003d65bad731 : SUCCESS in 56m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1acacd35249470ea1692646cbfbc54c : FAILURE in 35m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7fbdeedaeda408bbb47b6fc49f62937 : SUCCESS in 13m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/024a033f89d14623947c188f1cbb780c : FAILURE in 25m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2340e6b012cf49118d2f638ca9352729 : FAILURE in 32m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/556fecd52f694cf981a2791bd1c40997 : FAILURE in 32m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d8ea2b9eca4d13bd22a82a1d186832 : SUCCESS in 9m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c5f72b6653a549f1a5388d8424500ff7 : SUCCESS in 52m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ddc6a04a5b4e4669a204d026dad8208e : SUCCESS in 1h 22m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11da2c55243c4f7c9566241149e2dbcc : SUCCESS in 51m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0a8c08d2259b4c6795dbfb736cdea521 : SUCCESS in 1h 07m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7525dbd5684055866e03ea2c261fa3b32dacd55d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-29 08:12:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f309e4ed652ea3ffc88344b8b975b32ae825853b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 08:33:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14a98fb214d84eb5932b4dfaa94c8502\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5acfc8eaa16b4b19b5a7fc04702394a0 : SUCCESS in 19m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5537ef44b7c4491b978411b2323f6b0b : SUCCESS in 20m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d062962fcba28249c9933f14523a8df87ccb9d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 09:56:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55baaa792b3d423eb0f6c4aadc8460d2\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/69068f62f3ae43eab350a0a0c6ff09ca : SUCCESS in 1h 00m 38s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/70b7c398ecb7437a8935affa3d3ca21d : SUCCESS in 1h 39m 20s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/53b9dc2699c046988395178d73fe9cb6 : SUCCESS in 31m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6dccc15cb1c14e1d9f48014372bfdece : SUCCESS in 27m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5550d82d07db4ada986131df2a647805 : SUCCESS in 11m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbe0c8ac6cc44e768bebe0d51120b27c : SUCCESS in 20m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc67ea41a71a4011b9dea23945c8b1a9 : SUCCESS in 19m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffbe5245f0a948aeb16bc3769eaa7d88 : SUCCESS in 19m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/287b416dce69496e88f9f931240deeff : SUCCESS in 8m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8c1471182f44a4ba898404c6eaeb523 : SUCCESS in 41m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8bca6c3b55154088b2e63a5cb9e2acd3 : SUCCESS in 1h 29m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b3d01b3d7bf42aa831708f431b93e65 : SUCCESS in 45m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/24cf96acb6da47f4a4574aad5c822821 : SUCCESS in 1h 02m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7209d0c0984d9f90cdd14c7afedae3dbf9c041c9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-29 11:33:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"21a7d65cbfe00585a02ef697a7c3c13e8da578d0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-31 16:48:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8db4b9963f47b95ecc488b673bef6e6dc151e676","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 16:49:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"74eae7c84faae82649f34a8dd343cdbce08cccb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 19:06:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5c627612ba64777822813b6720fda7d\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/d873316178e5402f984ec37dadfe5cac : SUCCESS in 55m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7abfda8e629245378e158dcacb6925fe : SUCCESS in 11m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ba2d183eab948f0ae64fcdb005454d3 : SUCCESS in 26m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/040c31c2e7e141679bc5851b2f08a35d : SUCCESS in 24m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56c85a26419b415caf90a715162e8163 : SUCCESS in 11m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ec640bed4a24a058d7c2f940960a4af : SUCCESS in 51m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/687523516b71432bad6f1b7f5ec27a91 : SUCCESS in 1h 38m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/460ce253842e43db8127bf465c52d89b : SUCCESS in 1h 12m 03s","accounts_in_message":[],"_revision_number":2},{"id":"2dfcb73d2cf7478e4d49937f00d64b14d3fd2214","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 19:06:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d5c9a24fb738eaafdf4f5594502f089d3c8c5eb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 19:08:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fdc02aef8d24e2ababac9326d0f6d0e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/365a4d5fc40647538c7a1ebec3063581 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2a74b8c4cc5326ee78fc1aada2e624f9a48ddfc","revisions":{"1ee35c78b0e2db63c8e6f748ba2848ffa4075271":{"kind":"REWORK","_number":1,"created":"2023-08-28 08:54:10.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/892885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/892885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/892885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/892885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/892885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/892885/1"}}},"commit":{"parents":[{"commit":"41df9488a8672e4502f553bbd5c65ac4f4a9e6eb","subject":"Fix ovn-metadata agent sync of unused namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41df9488a8672e4502f553bbd5c65ac4f4a9e6eb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-17 23:27:58.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-28 08:53:51.000000000","tz":0},"subject":"[OVN] Skip the port status UP update during a live migration","message":"[OVN] Skip the port status UP update during a live migration\n\nSkip the port status UP update during a live migration due to a\n``PortBindingChassisUpdateEvent`` event.\n\nThe port status will be set by  the ``LogicalSwitchPortCreateUpEvent``\nand ``LogicalSwitchPortCreateDownEvent`` events, that will be issued\nwhen the port is deleted from the source host and created in the\ndestination host. This is a planned operation and controlled by Nova,\nnot a reactive event due to an unplanned host down event.\n\nConflicts:\n    neutron/common/ovn/utils.py\n\nRelated-Bug: #2027605\nChange-Id: I81390af2ea2fc384423518b84de3acf7adaf9193\n(cherry picked from commit e1f887ca9ffa73dde8ab41590f2131f3297aa3cd)\n(cherry picked from commit 6a9990dba1777dc5510d4fc2b249d8adc98e3e12)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ee35c78b0e2db63c8e6f748ba2848ffa4075271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ee35c78b0e2db63c8e6f748ba2848ffa4075271"}]},"branch":"refs/heads/stable/xena"},"f2a74b8c4cc5326ee78fc1aada2e624f9a48ddfc":{"kind":"REWORK","_number":2,"created":"2023-08-29 08:12:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/85/892885/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/892885/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/892885/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/892885/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/892885/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/892885/2"}}},"commit":{"parents":[{"commit":"41df9488a8672e4502f553bbd5c65ac4f4a9e6eb","subject":"Fix ovn-metadata agent sync of unused namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/41df9488a8672e4502f553bbd5c65ac4f4a9e6eb"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-17 23:27:58.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-29 08:11:48.000000000","tz":0},"subject":"[OVN] Skip the port status UP update during a live migration","message":"[OVN] Skip the port status UP update during a live migration\n\nSkip the port status UP update during a live migration due to a\n``PortBindingChassisUpdateEvent`` event.\n\nThe port status will be set by  the ``LogicalSwitchPortCreateUpEvent``\nand ``LogicalSwitchPortCreateDownEvent`` events, that will be issued\nwhen the port is deleted from the source host and created in the\ndestination host. This is a planned operation and controlled by Nova,\nnot a reactive event due to an unplanned host down event.\n\nConflicts:\n    neutron/common/ovn/utils.py\n    neutron/common/ovn/constants.py\n\nRelated-Bug: #2027605\nChange-Id: I81390af2ea2fc384423518b84de3acf7adaf9193\n(cherry picked from commit e1f887ca9ffa73dde8ab41590f2131f3297aa3cd)\n(cherry picked from commit 6a9990dba1777dc5510d4fc2b249d8adc98e3e12)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2a74b8c4cc5326ee78fc1aada2e624f9a48ddfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2a74b8c4cc5326ee78fc1aada2e624f9a48ddfc"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
