)]}'
{"id":"openstack%2Fneutron~892887","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I81390af2ea2fc384423518b84de3acf7adaf9193","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/2027605","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-08-31 19:06:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I81390af2ea2fc384423518b84de3acf7adaf9193","subject":"[OVN] Skip the port status UP update during a live migration","status":"MERGED","created":"2023-08-28 08:54:43.000000000","updated":"2023-08-31 19:08:00.000000000","submitted":"2023-08-31 19:06:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892887-bug/2027605","meta_rev_id":"9cfa986deee838a78fd97f19b2fa9636c04cb40c","_number":892887,"virtual_id_number":892887,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-31 19:06:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-08-29 11:33:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-08-28 16:26:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-08-31 16:41:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 09:44:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-28 11:13:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 16:26:04.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-08-29 11:33:08.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"7481632b45ff4cfe583e65781bb4861bf89f9e34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-28 08:54:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4488c98d021cf61dee71ce9318f34eae1b47d77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 09:44:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e1d6faba3ca4934b58c2b0820249b1e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f80dc84c60f74f56a9d23478cc8e4aba : SUCCESS in 48m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4833d73ed67448dac1766406a11243a : SUCCESS in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca6266e559dc59f1fee33547d7818ce35a0cf49f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 11:13:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a66eb13c38c4a399012c2290d5bea8e\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/7e639a4b627a47a9aae28184ab431903 : SUCCESS in 1h 00m 56s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/644c1d75929848dc8a620e91a2713eba : SUCCESS in 1h 10m 57s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/8102fc6673b147c98ffa857b9fba1e0a : SUCCESS in 30m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19740a74a5cf48a699ffce2d61efc57c : SUCCESS in 37m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc81146e51d345fdb3242881e88d4396 : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cec947d3f91d403b96a8df9957d76952 : SUCCESS in 28m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b73833fc2404ebbb22dcf9e2297901f : SUCCESS in 20m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54deb8567475491e8c30459e220280fb : SUCCESS in 10m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ddbcfc6162df487580e36654146cdb8a : SUCCESS in 58m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0080bffeb2174e5f97eae570f758f666 : SUCCESS in 1h 57m 06s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bcb45ed89d34bb191cd8fbca240f4a8 : SUCCESS in 51m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/87b6cc5ca6864eaf945a8b54a3203f97 : SUCCESS in 1h 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"003ff82f9d5674cd56e832ee7fd1b6ad34315e27","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-28 16:26:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d3e80e08a5516b28b11c5f1fc8b3123d568af307","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-29 11:33:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"40f9203aed0a5ed3be5893cd3ae5082318d2b227","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-31 16:41:38.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2052e6fe6a0e2ba8b3bbb3cbea1865b1ce9ee5a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 16:42:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af0dabfeb0af13c27fc0f801ed1b2a77adf85d54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 19:06:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4f79025ad79473aa578c43dab4e5a27\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/fac02f7b2be04ee585157af35c7afdf8 : SUCCESS in 1h 09m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f31c3e9d57d5482c870c3e1dd36c3a95 : SUCCESS in 14m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a5436f4b2f3423cba231dfb0bc322a3 : SUCCESS in 28m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9db33a08116244199c248313b703d3a9 : SUCCESS in 31m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8469a567bbc54d0cb16f8f9164e600a9 : SUCCESS in 11m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ad8680f8a684b149bfc2817893a0791 : SUCCESS in 51m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b84a3be30c1342ddb791e231a4baf971 : SUCCESS in 1h 48m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22a2860374434255acba25981f480769 : SUCCESS in 51m 33s","accounts_in_message":[],"_revision_number":1},{"id":"4c3ce87468ae6832c4c5628d694b3d55d89ea5d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 19:06:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9cfa986deee838a78fd97f19b2fa9636c04cb40c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 19:08:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0538fd697a04491c988bdfed49a2e8cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d220387119a494395982c6807a9f143 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9186d511a510bd39215581918ffb298ae6c2c4f3","revisions":{"9186d511a510bd39215581918ffb298ae6c2c4f3":{"kind":"REWORK","_number":1,"created":"2023-08-28 08:54:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/87/892887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/892887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/892887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/892887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/892887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/892887/1"}}},"commit":{"parents":[{"commit":"7034f4cf73723d129b950d370b167c1c576d559c","subject":"Fix ovn-metadata agent sync of unused namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7034f4cf73723d129b950d370b167c1c576d559c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-17 23:27:58.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-28 08:54:25.000000000","tz":0},"subject":"[OVN] Skip the port status UP update during a live migration","message":"[OVN] Skip the port status UP update during a live migration\n\nSkip the port status UP update during a live migration due to a\n``PortBindingChassisUpdateEvent`` event.\n\nThe port status will be set by  the ``LogicalSwitchPortCreateUpEvent``\nand ``LogicalSwitchPortCreateDownEvent`` events, that will be issued\nwhen the port is deleted from the source host and created in the\ndestination host. This is a planned operation and controlled by Nova,\nnot a reactive event due to an unplanned host down event.\n\nConflicts:\n    neutron/common/ovn/utils.py\n\nRelated-Bug: #2027605\nChange-Id: I81390af2ea2fc384423518b84de3acf7adaf9193\n(cherry picked from commit e1f887ca9ffa73dde8ab41590f2131f3297aa3cd)\n(cherry picked from commit 6a9990dba1777dc5510d4fc2b249d8adc98e3e12)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9186d511a510bd39215581918ffb298ae6c2c4f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9186d511a510bd39215581918ffb298ae6c2c4f3"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
