)]}'
{"id":"openstack%2Fneutron~892896","triplet_id":"openstack%2Fneutron~stable%2F2023.1~I5f11b568bab4518617409b7c3a02c7222df80a92","project":"openstack/neutron","branch":"stable/2023.1","topic":"bug/2032770","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-08-31 19:05:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f11b568bab4518617409b7c3a02c7222df80a92","subject":"[OVN] Add the \u0027uplink-status-propagation\u0027 extension to ML2/OVN","status":"MERGED","created":"2023-08-28 10:00:21.000000000","updated":"2023-08-31 19:07:29.000000000","submitted":"2023-08-31 19:05:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892896-bug/2032770","meta_rev_id":"805fdd6b4bd8a2b1110d2afa2eff17acd892ed53","_number":892896,"virtual_id_number":892896,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-31 19:05:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-08-31 09:26:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-08-31 09:32:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-08-31 09:32:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 10:25:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-28 12:05:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 09:26:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-08-31 09:32:10.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"a9d51fa0601a0572d155b1f498cb91c32e991854","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-28 10:00:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d70da0effa8a79b91e79f67533a16b8cf45451d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 10:25:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9aff2e52e814b32a5b785c1cafcd41c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/868b388a388840b4848f3928136664fb : SUCCESS in 24m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a2959d36d86e4435b4647c871ac99cf7 : SUCCESS in 24m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3e751835a0f14d99095f2d3ddf602f0d3ebe85e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 12:05:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea3f9646aa4d45bba6cf9790945aaa02\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/d650b59992db41858a89d3ab8b6200c8 : SUCCESS in 1h 51m 00s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/e8f779ac812c484b82e6f11f2c58686a : SUCCESS in 32m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d4a2591c2e744108aa7097c32c0bc7a : SUCCESS in 41m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e25faec3cd54dabb2d5287f0e67e2f9 : SUCCESS in 12m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/898bdf7836784093898041299a2ec345 : SUCCESS in 25m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc6eecdfcf734b72968529d8bec270e8 : SUCCESS in 24m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30548832581a4f8b980ff9d40dc4d4ab : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccccee9114014a34a83dc895ff8a45a6 : SUCCESS in 10m 50s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/1a82e491575f44eca3851b817cf6acf1 : SUCCESS in 23m 10s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ee3adce9b83476d8c214e6c1c4b2d87 : SUCCESS in 56m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b20aaf001f8410e8e11d391f445768e : SUCCESS in 1h 57m 54s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/90b2681de89547cd8203987966047f40 : SUCCESS in 56m 15s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a3e6afa7a44e423d9a77c550923977fc : SUCCESS in 1h 21m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/496180082d5e49ecb78d2f639fe6ba9d : SUCCESS in 1h 41m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0e4ff4e187724e95a97b9fdee407b2d5 : SUCCESS in 57m 06s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/48ae3f9f0cb74c9da5168b90192a6818 : SUCCESS in 1h 17m 34s","accounts_in_message":[],"_revision_number":1},{"id":"f3862107e0b01c6e4dbb15a6b9872e77c75c8981","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-31 09:26:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7cbc127e3fa4d55ac8f07b26d74e50e79084e3cc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-31 09:32:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e4234ac9bb7d18f8ef8068c6c63b83ea9281d46c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 09:32:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e106f1678bfe4478718dcd5762a3b63d7306da76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 11:37:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30b1ff9c1ccb421f84b8e519698b6c27\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/c783f640de2445d38d2bf264113f0446 : SUCCESS in 1h 30m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2bafbf3d52643ee899e93df42e5b1b1 : SUCCESS in 11m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e367474cf0e45878e67ed9058afb87e : SUCCESS in 23m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/537fdbed45a748a286dfcc4f58d79fb4 : SUCCESS in 21m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c25fb5327c504d68855c61dea5ab44ca : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f01ca2426fcb4307a29d45e8fc6ab7c3 : SUCCESS in 8m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0acdecda89cf49c7892bfb50b5efe24f : FAILURE in 43m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/60644fa95a8b45359f18f09e5a270d6e : SUCCESS in 1h 50m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/71ee2669f6d4459996068998fa625627 : SUCCESS in 1h 12m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/11d40c62a83841cd8c3d47d290d94e38 : SUCCESS in 1h 09m 46s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/48e6a0d13206438681bd6ee518ddee2d : SUCCESS in 1h 14m 34s","accounts_in_message":[],"_revision_number":1},{"id":"71d28d057aa603bb09fd32576a73371b00261662","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-31 12:24:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0bacb236b637343902d380bf493c87892d679b2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 12:44:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed2e7d869464b13ad26f8523914a3f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/140c0ee9a6c942d2aec1c399a2046fe8 : SUCCESS in 18m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/421ada4844e14618b6e274dae6a723bc : SUCCESS in 19m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3a49ae28bc974ed42fa67cad32d8db45d72046b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 14:54:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/308ed2ff5b6c4757a51cb5bcd7ae5692\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/fc654fff59924aa1bd0442f276783ed4 : SUCCESS in 2h 28m 45s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/13421eca5d8047e4b40d86f5adb18121 : SUCCESS in 46m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f53270de0e534dfd90761e80834c5b7f : SUCCESS in 27m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b517eec2f8e84424804f3acaecdefa12 : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5643bcf9b1843ac9483cd1bb623f998 : SUCCESS in 20m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb83ea3b6b9e4d809aaf7d198a7ea6c1 : SUCCESS in 22m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d86902bf6a9e4061a1912f9e3478ad97 : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b04505ffce1422fbaf77de080f6f9f6 : SUCCESS in 8m 41s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/994a6d92e6084d0e862b3d07eaad311b : SUCCESS in 23m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/96aacda4fb7641aeab8a099f61cb2460 : SUCCESS in 57m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a9da83f5c3e048b89c6982d4522f761d : SUCCESS in 2h 01m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f3e05aec8b2142e3b703caa23f465421 : SUCCESS in 1h 02m 11s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/24b1a42e79844d5cb4d81f276406298e : SUCCESS in 1h 03m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cf9c4c94d73441b685be962babe72355 : SUCCESS in 1h 58m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6dca692daf8848869f7cce6ca371cb83 : SUCCESS in 1h 24m 34s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/8374763cc45a468fbc79a23c4f45da06 : SUCCESS in 51m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7f079caa982f1b8ebaa1a4d0e93c8624fafe035c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 14:55:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26b6a7f67d97c16171e9b1f26a2d23d2d602e6e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 19:05:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aabb53e15d44435919aebee27e12851\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/9701eeade59b4292b2ddd1a30168ebd3 : SUCCESS in 2h 07m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db5076621f8441a49135a4840453f19e : SUCCESS in 14m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75a850e1a6294cffb50a63cd6d777fdb : SUCCESS in 28m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/faf5b7c0f87d408583efbc1b12dce52e : SUCCESS in 30m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f9a369f6e2446f6a49bfc0735d974bf : SUCCESS in 12m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bcdc613d02c48b799df4dc891f7159f : SUCCESS in 11m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ca898d1040149c4998868041fcb7b75 : SUCCESS in 54m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/178831d0a7c94b3d9fce990b524b6b16 : SUCCESS in 1h 50m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b26978ef64d54bdea8cae9a4ded6be9d : SUCCESS in 57m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5089506872414977bd284db0f52c2190 : SUCCESS in 1h 30m 06s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/645bb6140b154545820c1778f74e147a : SUCCESS in 1h 08m 13s","accounts_in_message":[],"_revision_number":1},{"id":"01ad5c3c418c7f1973317f50d8771c6821c86714","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 19:05:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"805fdd6b4bd8a2b1110d2afa2eff17acd892ed53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 19:07:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df280abdd44a4606bc3286af2dbc57ef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0590b3d1f0d848a29867277e45021448 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4dba779ca0d41b8a1c9f69571db7ce5 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f707101d965e41f4211ccf19a3bbdfcd04c9fe7","revisions":{"4f707101d965e41f4211ccf19a3bbdfcd04c9fe7":{"kind":"REWORK","_number":1,"created":"2023-08-28 10:00:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/96/892896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/96/892896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/892896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/892896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/96/892896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/96/892896/1"}}},"commit":{"parents":[{"commit":"be0cb0690e252e3ef8e9ab0698fb2707d781b548","subject":"Fix ovn-metadata agent sync of unused namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be0cb0690e252e3ef8e9ab0698fb2707d781b548"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-28 09:55:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-28 10:00:12.000000000","tz":0},"subject":"[OVN] Add the \u0027uplink-status-propagation\u0027 extension to ML2/OVN","message":"[OVN] Add the \u0027uplink-status-propagation\u0027 extension to ML2/OVN\n\nCloses-Bug: #2032770\nChange-Id: I5f11b568bab4518617409b7c3a02c7222df80a92\n(cherry picked from commit 0545f40a0401629e68561a54e4a576a9fb15f7fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f707101d965e41f4211ccf19a3bbdfcd04c9fe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f707101d965e41f4211ccf19a3bbdfcd04c9fe7"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
