)]}'
{"id":"openstack%2Fneutron~892898","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I5f11b568bab4518617409b7c3a02c7222df80a92","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/2032770","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-08-31 13:27:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f11b568bab4518617409b7c3a02c7222df80a92","subject":"[OVN] Add the \u0027uplink-status-propagation\u0027 extension to ML2/OVN","status":"MERGED","created":"2023-08-28 10:01:40.000000000","updated":"2023-08-31 13:29:02.000000000","submitted":"2023-08-31 13:27:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892898-bug/2032770","meta_rev_id":"040498109841d0e5430bdab3429cc9305fbda4bb","_number":892898,"virtual_id_number":892898,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-31 13:27:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2023-08-31 09:26:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-08-31 11:36:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-08-31 11:36:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-28 10:27:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-28 11:38:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 09:26:46.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-08-31 11:36:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"8922d63354f6e040e52834931aca44a4ebee58b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-28 10:01:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bc6fd7f155ef20a2fe7965f60ef406601c2ca5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 10:27:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be54b65f2be645a7b49c694e5d41e845\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9561172a177b465b8cb96ad118011aae : SUCCESS in 24m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d485f5a3424844e6955f58d0cbfc5274 : SUCCESS in 24m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"912f0d463ce82d49a008396b7525241b4ce099f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 11:38:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84c0b38c409943c4a6bde3d506694c3e\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/0274a36e62314694b25780df7331a000 : SUCCESS in 57m 04s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/20f2ff5e04af4f0ba0f600a2c92e4287 : SUCCESS in 1h 17m 57s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/ee9eb6cc32e942b1be310fc33666d4ca : SUCCESS in 43m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd71a94900124c6fb4dd948a41f0c239 : SUCCESS in 29m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a81dd056886142dfb93c9b1b24b5ccf8 : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92eb9903f7354434a67b813a1766440f : SUCCESS in 29m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78097d0db4e4459d9245d40a6ffdba4b : SUCCESS in 22m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a616fa36fd1841a1be7bcfaf0f64d09c : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a8276184a3044afb22b2a927d1b18ce : SUCCESS in 12m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f082bcab6d4c4617bfd5c51a78895a47 : SUCCESS in 56m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1899e32c7bdd4335bfa4d1053f5d6fd7 : SUCCESS in 1h 27m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf0df41175164d7d8fbc1c1ca79ba407 : SUCCESS in 1h 03m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f1ead7ff623b404c960608b86c2fe4f8 : SUCCESS in 1h 08m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb9702cfd5ac76abb7c78cbef339644623fc083c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-08-31 09:26:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"527874335ff97b15218024b39a7152b9e8c64f05","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-31 11:36:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a9d48e0c33d8110cbd96a7a26bde31116623c3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 11:37:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b060d1c9d20d6af694f8f1b8ec7adffa019b657","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 13:27:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4926019958541c6a3aad69d2d68da36\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/3b1e2663f6384d0597d3df18c43aa8c6 : SUCCESS in 49m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f34db570578c4e609625df8e0d8184c1 : SUCCESS in 14m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc009c37f60b4f89bbc52319354afe89 : SUCCESS in 28m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63f69b82d8e34ccea0fc111a2e2ccd72 : SUCCESS in 25m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d5ed4fab659430398a3157debfc50b9 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1170d4f4ecd439aa02ea2dcfa449366 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3752b3f8ea4344aab631df91a1b0e70d : SUCCESS in 47m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d338d0f2aef94d5fbcbb955f54403cca : SUCCESS in 1h 45m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a79c76bfe9140cb8c26898bff4c6b89 : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":1},{"id":"1e8b8a86c379f93158c5edcd81d9e740218bc870","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 13:27:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"040498109841d0e5430bdab3429cc9305fbda4bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 13:29:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d3b061d2fbe40e9a627129dade6d855\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6582dffb65fb4a48bfca3a2bf01176ed : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58c274f594514dcab80dc533d9a2e3b7 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62ed750a1596c875bb38e0094a553f11a33d29ec","revisions":{"62ed750a1596c875bb38e0094a553f11a33d29ec":{"kind":"REWORK","_number":1,"created":"2023-08-28 10:01:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/98/892898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/892898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/892898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/892898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/892898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/892898/1"}}},"commit":{"parents":[{"commit":"7034f4cf73723d129b950d370b167c1c576d559c","subject":"Fix ovn-metadata agent sync of unused namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7034f4cf73723d129b950d370b167c1c576d559c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-28 09:55:05.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-08-28 10:01:20.000000000","tz":0},"subject":"[OVN] Add the \u0027uplink-status-propagation\u0027 extension to ML2/OVN","message":"[OVN] Add the \u0027uplink-status-propagation\u0027 extension to ML2/OVN\n\nConflicts:\n    neutron/common/ovn/extensions.py\n\nCloses-Bug: #2032770\nChange-Id: I5f11b568bab4518617409b7c3a02c7222df80a92\n(cherry picked from commit 0545f40a0401629e68561a54e4a576a9fb15f7fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62ed750a1596c875bb38e0094a553f11a33d29ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/62ed750a1596c875bb38e0094a553f11a33d29ec"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
