)]}'
{"id":"openstack%2Fneutron~893082","triplet_id":"openstack%2Fneutron~stable%2F2023.1~I31b1a9c2f986f59fee0da265acebbd88d2f8e4f8","project":"openstack/neutron","branch":"stable/2023.1","topic":"feature/reduce-lock-contention-on-subnets","attention_set":{},"removed_from_attention_set":{"26832":{"account":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"last_update":"2023-09-28 03:24:05.000000000","reason":"Change was submitted"},"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-27 09:28:10.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"hashtags":[],"change_id":"I31b1a9c2f986f59fee0da265acebbd88d2f8e4f8","subject":"Reduce lock contention on subnets","status":"MERGED","created":"2023-08-29 15:38:09.000000000","updated":"2023-09-28 03:25:28.000000000","submitted":"2023-09-28 03:24:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893082-feature/reduce-lock-contention-on-subnets","meta_rev_id":"7e89c9f28cd7785106869416f89149ba8571a7f4","_number":893082,"virtual_id_number":893082,"owner":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-28 03:24:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-09-27 09:28:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-09-28 01:24:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-09-28 01:24:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-29 15:38:09.000000000","updated_by":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"reviewer":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"state":"CC"},{"updated":"2023-08-29 16:04:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-29 17:45:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-22 07:56:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-09-28 01:24:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"9c568e94f7840f08be5efe4bdc428a90556d4cbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"date":"2023-08-29 15:38:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89dac3d337523b5fdcf73a6f969fda04ed1db6a9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"date":"2023-08-29 15:48:28.000000000","message":"Topic set to feature/reduce-lock-contention-on-subnets","accounts_in_message":[],"_revision_number":1},{"id":"8a6025cb9fe649decca2ed06ab00edf142ccf127","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 16:04:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a79c607f1c194e1d87b5b27d776d371f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1e94f7afe77456283b851bd4b48d42d : SUCCESS in 25m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3c51c738a26a478cbf07f81f2276d803 : SUCCESS in 20m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94ec8d4d6831087410c8345efe1babf40e948c60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 17:45:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64cd277d85a24d989a77bc1435d9676f\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/d68f300508c34e2fbaa69db57cc04742 : SUCCESS in 1h 28m 25s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/b0c8fa1972414626a4fd1df7b6baed22 : SUCCESS in 1h 26m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/64ecc2fe00074bff89187ac8a03c0c83 : SUCCESS in 1h 56m 39s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/364499f3636e445f81e22b5d34789425 : SUCCESS in 1h 30m 02s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/0c7ab1f451fa4687b02727277a261e3a : SUCCESS in 45m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdd0b0cce71b416699f6e290c278a53e : SUCCESS in 38m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d7dac93dd254c49af7bb97055606de7 : SUCCESS in 11m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9d3763463324874891ad5886c284c50 : SUCCESS in 26m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b72b44087c74087be6149339def0b5e : SUCCESS in 23m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d641a468b1c049178642e118eef0d4b1 : SUCCESS in 10m 09s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/0cbe3e27e3c7483e9fa6bb129129072a : SUCCESS in 21m 22s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/79f4139ab82c4df0abfbc91fef442bb7 : SUCCESS in 46m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c33124a2370d4249ba9da162030c5d1f : SUCCESS in 2h 01m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/22aaac3975ec40408af45f58c1f1eee6 : SUCCESS in 1h 08m 34s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/281fafd7cb674d7ea931e7ea97fddd85 : SUCCESS in 52m 47s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b1a121cfdb55448ca25cca091df38cf3 : SUCCESS in 1h 11m 15s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e60ee85ccb624f9cb6543ff4fec0d5d1 : SUCCESS in 1h 07m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/613f0da76ebd47ec8ea299273fdc5ee5 : SUCCESS in 1h 20m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/197b30a7899c4e6a93cd60ee8cf22275 : SUCCESS in 1h 28m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d188cfa29f09486aa09920e642625591 : SUCCESS in 1h 16m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9db8a8143e114c078525f4031a25e266 : SUCCESS in 1h 35m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f86d422da47a4039b7a5afa140b88835 : SUCCESS in 1h 40m 06s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2ea670f09cd34179a358729f43ed2aea : SUCCESS in 1h 20m 02s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/bc440152961f45d4bd41f90a13f43f7c : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3fa6ed039417690f7de5b15e681cbc972b229ca0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-22 07:56:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8948b748bf7fb4a40147ae2dd74b6c5a0a248379","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-22 07:58:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e8dcc834d4ce196d15d17d9dae95492e25a252b3","author":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"date":"2023-09-22 08:47:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ba190184f73091ad847610a707c129b8e650b3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"date":"2023-09-27 09:14:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"589dd9fb4053e7bbece7c5e0537b5c8ce083fa2b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-27 09:28:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"161a7cdad246a6c2cd248f25748df2471a3c1cfe","author":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"date":"2023-09-27 09:41:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"694847ca985132829f9824de128edc0c0e2c9c4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 10:16:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13c71ae65d3847a7bd20952581342aa4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/16d35ad6c31448f18b7098284282dbfa : SUCCESS in 51m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/54f12624b34d4904b1b15fd2e08c8193 : SUCCESS in 54m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b0909618d45d574551ee9d047b80adf33a489d0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 11:23:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/461f0167bb814836886eaa8c506828e6\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/43ce09b3b08f4f088dc03713c1046b92 : SUCCESS in 1h 40m 29s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/e9c804a231b844e3909da78cf27d36ad : SUCCESS in 1h 47m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/f0f6ed2abb7741e786a8a3611b468707 : SUCCESS in 1h 57m 12s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/c6b81b1c68e344e5a89c38190a0f1a0f : SUCCESS in 1h 50m 55s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/4b4b7c7d10ab4aa7985a6a68455beed0 : SUCCESS in 46m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f87f621099e045c0b776ac754f50e3f0 : SUCCESS in 34m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f44e50c8470845a49b56785fc8bd4340 : SUCCESS in 12m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30aee79e3d63415ab4c6222f32b0e93c : SUCCESS in 23m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e0174f60cce421c996740cad4097240 : SUCCESS in 19m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed7c564c02b74b85aa9768c019f31433 : SUCCESS in 10m 51s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/09184fe56f334f8fb3de68fe71d29531 : SUCCESS in 20m 28s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/76fe5b64190a4f02a4fe032efd452357 : SUCCESS in 51m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed62ff1073c543fbb8867bb76e681a50 : SUCCESS in 2h 03m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ee243687c4f4b628218fe28508d3393 : SUCCESS in 1h 02m 59s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/60797e4c6cf5456facb2548650f6aa5a : SUCCESS in 1h 00m 30s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/339bdb2ccb2349da8147f46a312fe605 : SUCCESS in 1h 15m 15s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/995fa97606d44403bdfec7b1f3362c05 : SUCCESS in 1h 03m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9f19954d1c0b4c88b18ed1f2bae36e92 : SUCCESS in 1h 28m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0d0c74a50d324b7d9ab7dfab876c6367 : SUCCESS in 1h 24m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8982248cfccb405c984fbf060ce097da : SUCCESS in 49m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ead69c447fb64ceb8d0736a41b3130c0 : SUCCESS in 1h 16m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/483ed30ca9b84d2f910f9c9a64f5836a : SUCCESS in 1h 42m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/71df180317784b8ea298c20a1c2b0953 : SUCCESS in 1h 02m 05s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/55fc4308fa2848cf8e4ef65a2470f0d6 : SUCCESS in 50m 31s","accounts_in_message":[],"_revision_number":2},{"id":"99618fc82c7400b4ea41523008895b0d6f9646df","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-28 01:24:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8e62ddf58b1046e116dfc90c39696b67ed170d84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 01:25:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"db3a3fb3583a3beba8b392e0f616e13c80391af8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 03:24:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62ecdd41f9384b8eb4f9176de027d8c1\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/77e840e9dc1040fca1851391292a602c : SUCCESS in 1h 45m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c559ff5d4a34e9e84f695b127054893 : SUCCESS in 13m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6015593be35b4458b3fdd2e7cccbfaed : SUCCESS in 21m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9867cd6f6c54749a52844eb51b199ee : SUCCESS in 25m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6710a86360e74c25bd5d782bd2d64e13 : SUCCESS in 12m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06ba081494314560a92d36f64074f7ab : SUCCESS in 37m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b6656269e944590aa9cf855b3483805 : SUCCESS in 1h 58m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/794b788122e14f899abcf02fa63b140a : SUCCESS in 1h 22m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b527682849b74fa6829f94ebc11ec0bc : SUCCESS in 1h 05m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e141c26c07074585b14665b0ac3f7dcc : SUCCESS in 1h 12m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2e95c9d63fd147b8b6e051e89744c8d0 : SUCCESS in 1h 19m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4cb93735165e4376b929b42cdc490c32 : SUCCESS in 1h 22m 15s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/8a9f68bbc7d5496980e6847dea0a4c19 : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":2},{"id":"a9feeaae300374cd55a36e6b6a2d4d627839ab91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 03:24:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7e89c9f28cd7785106869416f89149ba8571a7f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 03:25:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c7bce608870450bbf67987765bb9c51\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730c2d4db64e4fac83f759de3428295a : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c89d028a955721c8132ca23c44e84a59d5fd99ce","revisions":{"b8f74158399de645e050a6249ca854cdb6f8bcd5":{"kind":"REWORK","_number":1,"created":"2023-08-29 15:38:09.000000000","uploader":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"ref":"refs/changes/82/893082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/893082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/893082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/893082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/893082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/893082/1"}}},"commit":{"parents":[{"commit":"6dc93c5072f586ffbf245e4740f314d3de1e95f1","subject":"[OVN][Trunk] Set the subports correct host during live migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6dc93c5072f586ffbf245e4740f314d3de1e95f1"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-03-01 15:14:18.000000000","tz":60},"committer":{"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","date":"2023-08-29 15:34:53.000000000","tz":120},"subject":"Reduce lock contention on subnets","message":"Reduce lock contention on subnets\n\nin [1] a lock was introduced with the goal of preventing subnets from\nbeing deleted while ports are being created in them in parallel.\nThis was acheived by aquiring an exclusive lock on the row of the\nsubnet in the Subnet table when adding/modifying a port or deleting\nthe subnet.\n\nHowever as this was a exclusive lock it also prevented concurrent port\nmodifications on the same subnet from happening. This can cause\nperformance issues on environment with large shared subnets (e.g. a\nlarge external subnet).\n\nTo reduce the lock contention for this case we split the lock in two\nparts:\n\n* For normal port operations we will aquire a shared lock on the\n  row of the subnet. This allows multiple such operations to happen in\n  parallel.\n* For deleting a subnet we will aquire an exclusive lock on the row of\n  the subnet. This lock can not be aquired when there is any shared\n  lock currently on the row.\n\nWith this we maintain the same locking level as before, but reduce the\namount of lock contention happening and thereby improve throughput.\n\nHINT: This isn\u0027t a clean backport, as we keep the subnet in-use field.\nWe can\u0027t backport the db update that would remove the field.\n\nThe performance improvement can be measured using rally test [2].\n(improving from 21 to 18 seconds).\nAlternatively it can be tested using 250 parallel curl calls to create a\nport in the same network. This improves from 113s to 42s.\n\n[1]: https://review.opendev.org/c/openstack/neutron/+/713045\n[2]: https://github.com/openstack/rally-openstack/blob/master/samples/tasks/scenarios/neutron/create-and-delete-ports.json\n\nCloses-Bug: #2009055\nChange-Id: I31b1a9c2f986f59fee0da265acebbd88d2f8e4f8\n(cherry picked from commit c0af5b3b5ea89d3147adf1054625f29d5b01b309)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8f74158399de645e050a6249ca854cdb6f8bcd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8f74158399de645e050a6249ca854cdb6f8bcd5"}]},"branch":"refs/heads/stable/2023.1"},"c89d028a955721c8132ca23c44e84a59d5fd99ce":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2023-09-27 09:14:00.000000000","uploader":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"ref":"refs/changes/82/893082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/82/893082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/893082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/893082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/82/893082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/82/893082/2"}}},"commit":{"parents":[{"commit":"493bfd0554f0357ca5b9d55aff7b193f55dfd3e8","subject":"Merge \"Check the device ID and host ID during virtual port binding\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/493bfd0554f0357ca5b9d55aff7b193f55dfd3e8"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-03-01 15:14:18.000000000","tz":60},"committer":{"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","date":"2023-09-27 09:13:49.000000000","tz":120},"subject":"Reduce lock contention on subnets","message":"Reduce lock contention on subnets\n\nHINT: This isn\u0027t a clean backport, as we keep the subnet in-use field.\nWe can\u0027t backport the db update that would remove the field.\n\nin [1] a lock was introduced with the goal of preventing subnets from\nbeing deleted while ports are being created in them in parallel.\nThis was acheived by aquiring an exclusive lock on the row of the\nsubnet in the Subnet table when adding/modifying a port or deleting\nthe subnet.\n\nHowever as this was a exclusive lock it also prevented concurrent port\nmodifications on the same subnet from happening. This can cause\nperformance issues on environment with large shared subnets (e.g. a\nlarge external subnet).\n\nTo reduce the lock contention for this case we split the lock in two\nparts:\n\n* For normal port operations we will aquire a shared lock on the\n  row of the subnet. This allows multiple such operations to happen in\n  parallel.\n* For deleting a subnet we will aquire an exclusive lock on the row of\n  the subnet. This lock can not be aquired when there is any shared\n  lock currently on the row.\n\nWith this we maintain the same locking level as before, but reduce the\namount of lock contention happening and thereby improve throughput.\n\nThe performance improvement can be measured using rally test [2].\n(improving from 21 to 18 seconds).\nAlternatively it can be tested using 250 parallel curl calls to create a\nport in the same network. This improves from 113s to 42s.\n\n[1]: https://review.opendev.org/c/openstack/neutron/+/713045\n[2]: https://github.com/openstack/rally-openstack/blob/master/samples/tasks/scenarios/neutron/create-and-delete-ports.json\n\nCloses-Bug: #2009055\nChange-Id: I31b1a9c2f986f59fee0da265acebbd88d2f8e4f8\n(cherry picked from commit c0af5b3b5ea89d3147adf1054625f29d5b01b309)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c89d028a955721c8132ca23c44e84a59d5fd99ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c89d028a955721c8132ca23c44e84a59d5fd99ce"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
