)]}'
{"id":"openstack%2Fneutron~893084","triplet_id":"openstack%2Fneutron~stable%2Fzed~I31b1a9c2f986f59fee0da265acebbd88d2f8e4f8","project":"openstack/neutron","branch":"stable/zed","topic":"feature/reduce-lock-contention-on-subnets","attention_set":{},"removed_from_attention_set":{"26832":{"account":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"last_update":"2023-11-08 21:09:04.000000000","reason":"Change was submitted"},"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-11-08 21:09:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I31b1a9c2f986f59fee0da265acebbd88d2f8e4f8","subject":"Reduce lock contention on subnets","status":"MERGED","created":"2023-08-29 15:41:26.000000000","updated":"2023-11-08 21:10:47.000000000","submitted":"2023-11-08 21:09:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893084-feature/reduce-lock-contention-on-subnets","meta_rev_id":"69a5ff0ad61838d4b7dcc6e2c69c76c95f2ee46c","_number":893084,"virtual_id_number":893084,"owner":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-08 21:09:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-09-28 11:12:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-11-07 22:49:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-11-07 22:49:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-29 15:41:26.000000000","updated_by":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"reviewer":{"_account_id":29074,"name":"Felix Huettner","email":"felix.huettner@digits.schwarz","username":"felix.huettner"},"state":"CC"},{"updated":"2023-08-29 16:30:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-29 17:30:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-22 07:56:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-09-28 11:12:07.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-11-07 22:49:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"030e974acbeac7c5c13e9303f07eef42db63df1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"date":"2023-08-29 15:41:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"831c21dff82b58d4054e97d899f6fb1027809ede","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"date":"2023-08-29 15:48:35.000000000","message":"Topic set to feature/reduce-lock-contention-on-subnets","accounts_in_message":[],"_revision_number":1},{"id":"3a475e93bccf591d24d3a435effbf3309f7a01d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 16:30:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/346a3c226902412899e0c24976089fec\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b27f4a3dd0a4ab38dfe951d34d2e50b : SUCCESS in 19m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8770a9a2236c4e66bf0e8d4e1256075e : SUCCESS in 47m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"251c6837e28a30340b77dc33391a0b0ec3fc3149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 17:30:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/379eb96f10864ef2a423549d6cb44b36\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/7b33e6639dad4780a1f264bada997d1b : SUCCESS in 1h 44m 57s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/1ef949a691f642afb85d543e80380271 : SUCCESS in 1h 34m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/95428a51555d4e8b982af2337b27eb6a : SUCCESS in 1h 29m 55s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/16622daccccc48dabdebcd02e8392d69 : FAILURE in 1h 46m 46s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/de3e36d2bc9040388a6d8e38a83cc679 : SUCCESS in 47m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e9d36f7bdce4252873b3758f47d84a2 : SUCCESS in 31m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/073685093e62477f80d521b0fc09e712 : SUCCESS in 13m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52b34172f65648c1987e4ca8469fb71c : SUCCESS in 27m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a26197a7b42d4bccad45df4c7720aa60 : SUCCESS in 35m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edba8d9f9820410a82272e49e5898795 : SUCCESS in 35m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0b43a53fa946fa91000d6977bba2cd : SUCCESS in 10m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c32de8e5354d43bba009e45966285a92 : SUCCESS in 45m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/71d0c0e5b13e48c798a23926b3249604 : SUCCESS in 1h 44m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1101f0feeba94269bd0e23c25bcd65dd : SUCCESS in 1h 17m 30s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/86e066926ebc4697b935e8b5318c74b6 : SUCCESS in 1h 04m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2f1197ed89034860bfdc7e7cf5ece8f3 : SUCCESS in 54m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/23e52c1bcf784b4aa4a58127453a3471 : SUCCESS in 1h 10m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/53c236e4af4a468e954a107bbe7f9bdc : SUCCESS in 1h 26m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edfb93b382614399b47e960b2a7376d7 : SUCCESS in 1h 08m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fa1b114202cc4cfaa0fcc7f97675892b : SUCCESS in 1h 23m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79c6455c21b397604f4fa9a074bb17ddd76a4936","author":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"date":"2023-08-30 09:40:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc085cc37ff6c4f07c3dfb738eefa66750c4d391","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 10:00:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee4b83833d934b30bba5cb62a2dc14cc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/306c5c763ce647bb8ac3a9df03f6dbe7 : SUCCESS in 18m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b3050c5eb77460c99b918bc931c3844 : SUCCESS in 18m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83710d53f9c4089ade06b0b99559971b20ed6efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 11:38:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be6dede5f8a04d80ac46203d4f818bfa\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/64131698cab242c1b86d942b2e846f90 : SUCCESS in 1h 34m 22s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/ad516eb3bf4b402e9f9ff23b107263c5 : SUCCESS in 1h 50m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/7847f36fb5364f3481a2950026a22eae : SUCCESS in 1h 40m 13s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/eff0670e849a492197b39df9bda1ab79 : SUCCESS in 1h 48m 42s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/6a3a8ad4dfe04aa5a071e25accb94640 : SUCCESS in 29m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e32940b32e9f4e52a7475de985129c06 : SUCCESS in 33m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9768b8d0d3d4054bd3d5aacfdfddc4f : SUCCESS in 13m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/533a0b0643a146a5befa2dc1d68287d1 : SUCCESS in 25m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9cb48a0d8614d2ab878d52c035186b7 : SUCCESS in 25m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1d7b878ae204587b8ff47e34b3161ef : SUCCESS in 23m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0063d83e81d4b5f829aed6f17470160 : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd6182721cca41db8bd0415e6d5e64b9 : SUCCESS in 55m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee9e7f8d88434f1e82849ae54e91f78a : SUCCESS in 1h 53m 37s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9c5cf6c655134c0cb9d4a2524cbb05fa : SUCCESS in 1h 23m 12s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/34ce4ae77a1145ee8fc66e6a9397698e : SUCCESS in 1h 03m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/664e0f74497448e2ad540f02cb48662a : SUCCESS in 1h 01m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/06252667d1e44825a3e383e577e8d6ad : SUCCESS in 1h 04m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4209cf6b539f441e9abd737e9bc96f0d : SUCCESS in 1h 21m 53s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d6a1684087b428388daa390065c78ac : SUCCESS in 58m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/62776bf353cb40bb9fb42bb50484bc50 : SUCCESS in 1h 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16ff8c51fd6172951703f86297771bce408e6c38","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-22 07:56:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5d8c7615133041737335b7e468355f8730a9d059","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-22 07:58:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42fc08e29f8922dce441c93d3b739eeafb48994b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"date":"2023-09-27 09:15:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"deb087b8eb381208854b830b7b9b06e2c4b55b7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 09:59:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8787ab3d9d324a719cd8373b56354f96\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4e80100fcba454abc403bed7c50c4e9 : SUCCESS in 38m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/424573d5ff9148be82e42fbbc777836f : SUCCESS in 35m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05aa5ba1262ce40d01c06e73d63a7c56c8bda212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 11:28:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56e2e70f9def4be394de36be8289d4cc\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/a7f3fb790014456daebe9091d60b11b4 : SUCCESS in 1h 27m 09s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/d21e99ec544c4266b75e91f54cf22d14 : SUCCESS in 1h 24m 24s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/151674d7730e4bc79c0d1187b72a8535 : SUCCESS in 1h 51m 06s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/2481d2e4431349d28c9bed3bb7c6045a : SUCCESS in 1h 46m 44s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/c26ec42cfd8647e4b35e4b7c5da3af6c : SUCCESS in 29m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1d6c678e4c54ce69d97c9fd3b22890d : SUCCESS in 30m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83f230bd097746538ba7f7fe0a1ffd44 : SUCCESS in 12m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/619f823aafc94992b4faca24956746ef : SUCCESS in 26m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/658feae2954949ac858ce07149b4d134 : SUCCESS in 21m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d06ecb65f7f4cae83ee72f162a9065e : SUCCESS in 21m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9277f90bc61f4538bd3aecaefaaedc6d : SUCCESS in 11m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4528b27d1efd4237b068b384ff1faf0b : SUCCESS in 45m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/65e58990e3a244a4ad52df664145fd41 : SUCCESS in 1h 36m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ffa1b9e0f7f0451b89a92d01a5d59e66 : SUCCESS in 57m 59s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/72590119ae484ecabc57630562083c5b : SUCCESS in 57m 06s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9324d004efc848d7a3a7a9c2662759dd : SUCCESS in 58m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/61af086f75de4290aaf8bc1ad2f29974 : SUCCESS in 1h 02m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f06e53ed7cba42c2a7f2aa087a95ac71 : SUCCESS in 1h 05m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08dca58588c04d40b0e81109036a5176 : SUCCESS in 58m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/06dfecb84ada4552b99135c63f03926f : SUCCESS in 1h 08m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b890add8f7afd57d7d7926156d2c4d01fac7de54","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-09-28 11:12:07.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b8d79e907292f4f96bbfab09bb5ff6254408664","author":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"date":"2023-10-23 11:48:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8a443c48dfe065b0068ec8cba70194a032e141b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 12:58:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed002c61e3eb4e36a0a2b6aabb880fbe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/447f844d2061483b9a10e364cba4f7e0 : SUCCESS in 1h 06m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/175d7c0af7744b87a1f21e2ca674a5a8 : SUCCESS in 1h 05m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8832665bef1411daee7e6920449e26e8ca391976","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 13:41:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/989dabf18e914adf85f64b92764f9623\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/14b4fe95b065425b9cd851e1ea40bcac : SUCCESS in 1h 28m 42s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/c2bf603bb313444eacb5cbeeb14326ef : SUCCESS in 1h 41m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/8221ad6292354c9eb1d1acab27c681ae : SUCCESS in 1h 35m 34s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/73122bb6e41145cdb53d4b173c985e1e : SUCCESS in 1h 42m 43s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/efaf6522fc764c3cb83fe1ecde440c6a : SUCCESS in 34m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/189af3bd716f4b11b17f5e3d0265a4e1 : SUCCESS in 29m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/719c5a3fba0a4d5c887a2f170f45793f : SUCCESS in 12m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dad8c8a792341fb9a270d8b47600b4a : SUCCESS in 22m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8040cf186ff24a718c6e461a707c13fb : SUCCESS in 25m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85f8f6e80c5f47d5b4cbedd5de809ba7 : SUCCESS in 18m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c1bd05611a4747ba5fa141212a61d5 : SUCCESS in 10m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ef4793edd48245969bdeb78e667cd6f4 : SUCCESS in 43m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b9aa904c0534467b523c02d158b6785 : SUCCESS in 1h 45m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/111c3e3fbbf5458cb6535cd797d6c314 : SUCCESS in 48m 02s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f20365dee954256b5a4eb2ce86b5ce7 : SUCCESS in 1h 02m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a01e02b66e074b37b9550398f9e379d6 : SUCCESS in 54m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/542f4c6b213f4065963cb127c6d049ec : SUCCESS in 55m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8cd17fb1c6fe4569be8075cb2328f4da : SUCCESS in 58m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74ccbbd7a2034496a7727da0f6a012d7 : SUCCESS in 59m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c5214f7bba1e414b98c3b31540dcd841 : SUCCESS in 1h 08m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb3e9ce7326a23a2237f51cb892217e50b96687f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-07 22:49:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7c8e7349b051a6301e5f1a7acb5470616d93261b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 22:49:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"20e25c87f31a61398c688aece1152c91a4436a02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 00:56:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/402513f2ed704040b2b23c32beb4312a\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/6799d4f7fc794578af59e6f050ea4bd2 : SUCCESS in 1h 43m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/710363ffeea94638b41aeface891a764 : SUCCESS in 14m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/959d58969e66458da5741320932e5498 : SUCCESS in 20m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/673716fde85846edbe41d0d0be24e72b : SUCCESS in 23m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e9d55d697fd4b3f8b375fab69897b00 : SUCCESS in 10m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/243f71ffaf5c45bda04165a75879d412 : SUCCESS in 53m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/034e882943484354941a3d71dcc8c9aa : SUCCESS in 1h 56m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ee213a6cf1c649f7b79a8d9acb53d0d6 : FAILURE in 59m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a12b4ac176b49f6b1fd0b483c30ab42 : SUCCESS in 1h 04m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/71c3d4a49b8847f68075e977235abaee : SUCCESS in 46m 37s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/380ccb173278481f855e3658b2187cd5 : SUCCESS in 43m 27s","accounts_in_message":[],"_revision_number":2},{"id":"944019f8b6928bb665b882d606dc3a8489b5af35","author":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"date":"2023-11-08 14:05:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b3b059930d1169a53c86c8a038f8f76d2086f10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 14:24:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/501050718ea144a1a2d02a58fc262b0a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d7a9ae0701849409c4b4057776633dc : SUCCESS in 17m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c419f694bc64145ae93eae54fb6e1dc : SUCCESS in 17m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"981f697b9dd295f16b2082b18e6e8eba917141da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:29:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7eb507ebba254f87a2d4ccf8c86b0309\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/b6ff210c6d2c4e19b4a05655d49ed1cd : SUCCESS in 2h 15m 43s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/c0e2674692de4a20b4c7db6ac0244906 : FAILURE in 1h 30m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/2b59ee60e1da418086543be895638d2b : SUCCESS in 1h 44m 02s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/2991bdcb3ef54a23af996c15eac9c427 : SUCCESS in 1h 30m 26s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/563f1b2d1a78448eb11f71989f734b53 : SUCCESS in 45m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f81ce1a45af44e1e8a2ae42c98ee022d : SUCCESS in 31m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/802798afd2c44ad686e9267e9972e4e6 : SUCCESS in 13m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcfab5e07d934d0e9f10aace2c04d51d : SUCCESS in 22m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d5f9b124092439ebcddf92b889ec8ac : SUCCESS in 23m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa4e55ff0e584d49a58d6ea614a9494f : SUCCESS in 27m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f77fceed3c5e485991497a2e195c9836 : SUCCESS in 10m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/56a4c40d9d814932ae2a16576942f68b : SUCCESS in 54m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9d9530c7179d4c6caa5605b1e6a7baed : SUCCESS in 2h 12m 29s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b1960303491649e0a7a2becc31674c27 : SUCCESS in 1h 08m 49s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3957aa50fe3b4b3bafb27820e056a319 : SUCCESS in 1h 03m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2c6938c07d904d9b8108bacef5f95193 : SUCCESS in 1h 04m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d5dd447954444e929e0da96605cecb24 : SUCCESS in 1h 09m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/98096702cd574b8d968499092043a709 : SUCCESS in 1h 08m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e94bb35d2cc4ffe8586621e96cd7d21 : SUCCESS in 52m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cca31df999fe42a780b36ce77025ba4f : SUCCESS in 1h 02m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa809a35f94b0f6dad4420fadedb87eb8367c1c8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-08 16:33:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e0ebb7e153ddb43446d55dfd93d4bfa8df1a3728","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 17:12:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5283fe0ae93f441490905548ab40b6c4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89734f456f6340cdaa223e0beb908049 : SUCCESS in 38m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9ec5fb362cb48b388966b7642b028bd : SUCCESS in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c50c52e577ac8a256d5e5e714ff9c2c318663936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 18:57:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdc2a8d065a84d018695e695c809625e\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/06ef83cbd48b4c56ae71a435b7b13d63 : SUCCESS in 2h 21m 26s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/fa9a246aa03a4b98b7a3062ab9d2b649 : SUCCESS in 1h 15m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/4fba2a8722064969b37f412d7182e186 : SUCCESS in 1h 31m 34s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/6e6a2785cce14c4ea2b53a0d7d0338c1 : SUCCESS in 1h 27m 43s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/594f858386154b46b1c78b52baeb00d6 : SUCCESS in 32m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e98abb82b1b4a65a9955980dd389cce : SUCCESS in 33m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a853cc5d61b4ba9a2b1c99abb22e4df : SUCCESS in 14m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6a1c60b4ea54aebb76d3c3e198917a3 : SUCCESS in 23m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89e0beb69bd342f4919b3c315848e935 : SUCCESS in 23m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e779f47e33b4e7881ce496650159f48 : SUCCESS in 23m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/795637d801d4449fb33b8e6fa696b8ba : SUCCESS in 11m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7dc25188baa740eb9cfe8d12a632c08f : SUCCESS in 46m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8561726639254d8d83dff82ccc7e0a3f : SUCCESS in 1h 43m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fb44b801befe4bf38ab5c2428a95e58e : SUCCESS in 1h 03m 07s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/add99749a53e469d8c2156f9771c05ce : SUCCESS in 59m 34s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/60f198a0026c4665b110de671a4941af : SUCCESS in 47m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/75262b110e1f4fe5ae25fe8882816178 : SUCCESS in 1h 17m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/008b1e2ffb074f5ba514887a9bab0a07 : SUCCESS in 1h 27m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e823562b2224ac89e5f235503cee474 : SUCCESS in 53m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/27db817cf57d41128d023532b308fec1 : SUCCESS in 51m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86953e260cde9cb6b7b593c25f49ed8c79aec7c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 18:57:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a87f7f3788de0c13d947c660ffd9a090d240b997","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 21:09:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8bf7feea9834235ad309eaf77a5534b\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/031f0c990f5344d7b5d75e0cc3408fa7 : SUCCESS in 1h 23m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d34e84c7da549f7b4369acd095ca816 : SUCCESS in 12m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55a2bdc62c974e10b3cad1a1195d71c4 : SUCCESS in 21m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76569780b6094a4bb728bd68ee7b71d4 : SUCCESS in 24m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13070d73c17e4745b1471daaef073ed8 : SUCCESS in 9m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d08a1234360342e782df1303889f3148 : SUCCESS in 53m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/933a61aac89a43d299474856694c5ab6 : SUCCESS in 1h 39m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bd54dbf08ace4f89bfbad3bcf793e51b : SUCCESS in 1h 02m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a26742574804dfc8e5ececf6b119363 : SUCCESS in 1h 05m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7eac2b9fadd6402eb977302dbc8e99cd : SUCCESS in 49m 31s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab3bdf1517884cf2a7a71a4d2acaad3f : SUCCESS in 39m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9f3c41fe165089424630d320474ce415c9a2dce2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 21:09:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"69a5ff0ad61838d4b7dcc6e2c69c76c95f2ee46c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 21:10:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/109c54d2deff44a49786798b6e9f7190\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de82908fb644e51823f59eb73f42115 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d25c129ec2d541986b5dcd12bc2bd48d51b87111","revisions":{"950e64be08dd9baf36152712167d49ecf3cf60f4":{"kind":"REWORK","_number":1,"created":"2023-08-29 15:41:26.000000000","uploader":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"ref":"refs/changes/84/893084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/893084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/893084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/893084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/893084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/893084/1"}}},"commit":{"parents":[{"commit":"6a9990dba1777dc5510d4fc2b249d8adc98e3e12","subject":"[OVN] Skip the port status UP update during a live migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6a9990dba1777dc5510d4fc2b249d8adc98e3e12"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-03-01 15:14:18.000000000","tz":60},"committer":{"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","date":"2023-08-29 15:40:22.000000000","tz":120},"subject":"Reduce lock contention on subnets","message":"Reduce lock contention on subnets\n\nin [1] a lock was introduced with the goal of preventing subnets from\nbeing deleted while ports are being created in them in parallel.\nThis was acheived by aquiring an exclusive lock on the row of the\nsubnet in the Subnet table when adding/modifying a port or deleting\nthe subnet.\n\nHowever as this was a exclusive lock it also prevented concurrent port\nmodifications on the same subnet from happening. This can cause\nperformance issues on environment with large shared subnets (e.g. a\nlarge external subnet).\n\nTo reduce the lock contention for this case we split the lock in two\nparts:\n\n* For normal port operations we will aquire a shared lock on the\n  row of the subnet. This allows multiple such operations to happen in\n  parallel.\n* For deleting a subnet we will aquire an exclusive lock on the row of\n  the subnet. This lock can not be aquired when there is any shared\n  lock currently on the row.\n\nWith this we maintain the same locking level as before, but reduce the\namount of lock contention happening and thereby improve throughput.\n\nHINT: This isn\u0027t a clean backport, as we keep the subnet in-use field.\nWe can\u0027t backport the db update that would remove the field.\n\nThe performance improvement can be measured using rally test [2].\n(improving from 21 to 18 seconds).\nAlternatively it can be tested using 250 parallel curl calls to create a\nport in the same network. This improves from 113s to 42s.\n\n[1]: https://review.opendev.org/c/openstack/neutron/+/713045\n[2]: https://github.com/openstack/rally-openstack/blob/master/samples/tasks/scenarios/neutron/create-and-delete-ports.json\n\nCloses-Bug: #2009055\nChange-Id: I31b1a9c2f986f59fee0da265acebbd88d2f8e4f8\n(cherry picked from commit c0af5b3b5ea89d3147adf1054625f29d5b01b309)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/950e64be08dd9baf36152712167d49ecf3cf60f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/950e64be08dd9baf36152712167d49ecf3cf60f4"}]},"branch":"refs/heads/stable/zed"},"d25c129ec2d541986b5dcd12bc2bd48d51b87111":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2023-09-27 09:15:04.000000000","uploader":{"_account_id":26832,"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","username":"shoffmann"},"ref":"refs/changes/84/893084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/893084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/893084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/893084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/893084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/893084/2"}}},"commit":{"parents":[{"commit":"86bc3761d1ba60a17fdc1bc1f717a872be3817f6","subject":"[PostgreSQL] Subnet entity with ServiceType grouped by both tables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/86bc3761d1ba60a17fdc1bc1f717a872be3817f6"}]}],"author":{"name":"Felix Huettner","email":"felix.huettner@mail.schwarz","date":"2023-03-01 15:14:18.000000000","tz":60},"committer":{"name":"Stefan Hoffmann","email":"stefan.hoffmann@cloudandheat.com","date":"2023-09-27 09:14:54.000000000","tz":120},"subject":"Reduce lock contention on subnets","message":"Reduce lock contention on subnets\n\nHINT: This isn\u0027t a clean backport, as we keep the subnet in-use field.\nWe can\u0027t backport the db update that would remove the field.\n\nin [1] a lock was introduced with the goal of preventing subnets from\nbeing deleted while ports are being created in them in parallel.\nThis was acheived by aquiring an exclusive lock on the row of the\nsubnet in the Subnet table when adding/modifying a port or deleting\nthe subnet.\n\nHowever as this was a exclusive lock it also prevented concurrent port\nmodifications on the same subnet from happening. This can cause\nperformance issues on environment with large shared subnets (e.g. a\nlarge external subnet).\n\nTo reduce the lock contention for this case we split the lock in two\nparts:\n\n* For normal port operations we will aquire a shared lock on the\n  row of the subnet. This allows multiple such operations to happen in\n  parallel.\n* For deleting a subnet we will aquire an exclusive lock on the row of\n  the subnet. This lock can not be aquired when there is any shared\n  lock currently on the row.\n\nWith this we maintain the same locking level as before, but reduce the\namount of lock contention happening and thereby improve throughput.\n\nThe performance improvement can be measured using rally test [2].\n(improving from 21 to 18 seconds).\nAlternatively it can be tested using 250 parallel curl calls to create a\nport in the same network. This improves from 113s to 42s.\n\n[1]: https://review.opendev.org/c/openstack/neutron/+/713045\n[2]: https://github.com/openstack/rally-openstack/blob/master/samples/tasks/scenarios/neutron/create-and-delete-ports.json\n\nCloses-Bug: #2009055\nChange-Id: I31b1a9c2f986f59fee0da265acebbd88d2f8e4f8\n(cherry picked from commit c0af5b3b5ea89d3147adf1054625f29d5b01b309)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d25c129ec2d541986b5dcd12bc2bd48d51b87111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d25c129ec2d541986b5dcd12bc2bd48d51b87111"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
