)]}'
{"id":"openstack%2Fneutron~893140","triplet_id":"openstack%2Fneutron~stable%2F2023.1~Ibed9e0d77500570c3d0f9f39bfe40cb9239d0d7a","project":"openstack/neutron","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-08-30 22:04:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibed9e0d77500570c3d0f9f39bfe40cb9239d0d7a","subject":"[OVN] Cleanup old Hash Ring node entries","status":"MERGED","created":"2023-08-30 10:33:27.000000000","updated":"2023-08-30 22:06:10.000000000","submitted":"2023-08-30 22:04:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893140","meta_rev_id":"6ce9279533aa97f17ea7497df3ad32305bb85b95","_number":893140,"virtual_id_number":893140,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-30 22:04:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-08-30 12:43:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-08-30 20:06:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-08-30 20:06:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 10:55:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-30 12:43:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-08-30 13:00:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 20:06:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"60837a52921fcf239391fd3de4f45c45136d871b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-30 10:33:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3608780722da6ee904f125c33a531c2237de5c7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 10:55:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9174ed3797334da3986491520dd50245\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6a11a0dca174557939e0d5b42ce92b2 : SUCCESS in 18m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8caa453ef5254388a4ebf2b330c6bc1b : SUCCESS in 19m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72733b3cfc4959cb1864419e9ed2db0630f3a195","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-30 12:43:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c447a69a7e22549dbf6e9b4ef647889f48879c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 13:00:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/176f8641a1ed4c5fb097f1a66d7a49eb\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/ca36499b96934b44a402716780e20439 : SUCCESS in 1h 59m 06s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/08037a38b72d44dba12733eb4901a0de : SUCCESS in 1h 45m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/215d47bc95864ebab8c635f992321418 : SUCCESS in 1h 32m 37s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/29ba05137b634515a90c1aad0b09fabb : SUCCESS in 2h 16m 42s\n- neutron-tempest-plugin-designate-scenario-2023-1 https://zuul.opendev.org/t/openstack/build/165aadd0dea2407a8d7895336a1e3598 : SUCCESS in 44m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/872ea886ebaa428b90ee2c3869491e98 : SUCCESS in 32m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a911ed3a1e34f559156a599cb3f6ff7 : SUCCESS in 12m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d56c0a457119421aae3458aeda0e5166 : SUCCESS in 25m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c03331b0c5fa4982b2ac8c2ea39f5b15 : SUCCESS in 26m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ea0f174abf64679b7dc19906f982bea : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d70dea8023eb4d21b319c93b0954f093 : SUCCESS in 7m 44s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/67dd1ac966a84fe4b25f81f750c76293 : SUCCESS in 27m 00s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0d32a6651ce04006b999641d3e8eb570 : SUCCESS in 43m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6fb0d901e26474585e0f163e22717ac : SUCCESS in 2h 07m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/41499390d64448e3887a877a405fb145 : SUCCESS in 1h 02m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1bc950a23b294c5d85dcbfae9fac5ed6 : SUCCESS in 1h 15m 20s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9c2fdf09ef0049a9a0c495fee38c16c2 : SUCCESS in 1h 26m 00s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/78be8a98186040d389fe413d57fbac3b : SUCCESS in 1h 00m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/903e509625c64fd287c3ba8fe8d6d91a : SUCCESS in 1h 41m 31s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b719f6456e61459abed9c5230551c682 : SUCCESS in 1h 20m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fdf1a5f32730458dbab0ef480c40e364 : SUCCESS in 1h 09m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/99fc551c328841f6a739080b55f73b3e : SUCCESS in 2h 19m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2a2f63c2b46e4def8b88d06dd2a86af0 : SUCCESS in 2h 08m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/526800cbf4fd48298a48c3698c3f1c0e : SUCCESS in 1h 14m 10s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/3a829c0f54c44fcb9a7f7ea93cc7ef03 : SUCCESS in 1h 20m 51s","accounts_in_message":[],"_revision_number":1},{"id":"15d3444be2cd983dcf17d1b19b96cf6c5545d93a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-30 20:06:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0c71f3c20c806df287ef2ba65b82ad0262a9ba71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 20:06:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"164859597bc797d367b9e482bb00be630c48cf63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 22:04:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25d001e01abe4a28ab18d80c8f154733\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/9235af7100e8454882f67726c82dc0ff : SUCCESS in 1h 43m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df2c2b0f6905493fa2063e91a538a441 : SUCCESS in 14m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c471cc742e8d4098ac796af9881e8341 : SUCCESS in 26m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0e7b53e2c1642ceaadd30601ec55f0b : SUCCESS in 25m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e094c0c5c1b4fe783031d0220fad8ec : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e07ee818bb6c47468a5511598bb395f4 : SUCCESS in 7m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/416b3ceced81458f865a1f33ceefb194 : SUCCESS in 53m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6113e3b70dda4b6bb768346d442c6677 : SUCCESS in 1h 57m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5e9d138ca9604ee9832d9313fdb2532b : SUCCESS in 1h 22m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/275db859693a411fa6057693b8050116 : SUCCESS in 59m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a311c7f864364e0aa2b1e0e8ba3beff5 : SUCCESS in 1h 04m 41s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8a1b73be94b84460bac38aa603c409d1 : SUCCESS in 55m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/911b368ed32e46fea2e743a3a362a3ea : SUCCESS in 1h 08m 43s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b8ba147b2e3c4d24badf664dd526ab02 : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8110b94832babec6e78e84b8721b5ff343920e46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 22:04:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6ce9279533aa97f17ea7497df3ad32305bb85b95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 22:06:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb4e39f1e64d4b35b0245850bcea1505\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6e6a02e604e4f40b49b479479566df2 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5ae4ae6358f4d3c909f61d21e7b4106 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"adfe0659c619ad82761c9622bdcfdffed09248da","revisions":{"adfe0659c619ad82761c9622bdcfdffed09248da":{"kind":"REWORK","_number":1,"created":"2023-08-30 10:33:27.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/40/893140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/893140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/893140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/893140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/893140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/893140/1"}}},"commit":{"parents":[{"commit":"8ddf72df9c8144fc509a06bd301eca47716868f0","subject":"Merge \"[OVN] Disable the mcast_flood_reports option for LSPs\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ddf72df9c8144fc509a06bd301eca47716868f0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-29 10:37:52.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-30 10:32:57.000000000","tz":60},"subject":"[OVN] Cleanup old Hash Ring node entries","message":"[OVN] Cleanup old Hash Ring node entries\n\nThis patch introduces a maintenance task that runs once a day and is\nresponsible for cleaning up Hash Ring nodes that haven\u0027t been updated in\n5 days or more.\n\nChange-Id: Ibed9e0d77500570c3d0f9f39bfe40cb9239d0d7a\nCloses-Bug: #2033281\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 7f777c223e557f7bb0e0290090019eaa60095b50)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/adfe0659c619ad82761c9622bdcfdffed09248da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/adfe0659c619ad82761c9622bdcfdffed09248da"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
