)]}'
{"id":"openstack%2Fneutron~893141","triplet_id":"openstack%2Fneutron~stable%2Fzed~Ibed9e0d77500570c3d0f9f39bfe40cb9239d0d7a","project":"openstack/neutron","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-08-30 22:05:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibed9e0d77500570c3d0f9f39bfe40cb9239d0d7a","subject":"[OVN] Cleanup old Hash Ring node entries","status":"MERGED","created":"2023-08-30 10:33:54.000000000","updated":"2023-08-30 22:06:17.000000000","submitted":"2023-08-30 22:05:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893141","meta_rev_id":"a3df5bf9d1afdc1db1cfbb238047f893fd1cd11b","_number":893141,"virtual_id_number":893141,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-30 22:05:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-08-30 12:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-08-30 20:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-08-30 20:07:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 11:22:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-30 12:38:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 12:43:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-08-30 20:07:12.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"104e6f56c40905814164de89215286800d226c62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-08-30 10:33:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"062270858d19b487482bd9327ee4023ca4e72073","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 11:22:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d850a04588c4a619d251b86bb8b7220\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91df4dda28ea4853b73173a747fc311d : SUCCESS in 45m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3b9e8b9631fe4162b45f9ca8e380d648 : SUCCESS in 20m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"226dba504c2fbaba20f45cfe7f4ce944c303f2c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 12:38:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd8cfd2af52c466389a12f569e2c255b\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/2cc5e2327bd844ad8178892d4d5a8f1d : SUCCESS in 1h 49m 04s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/bdde2b8ad1a64dcf832e7623e4caf050 : SUCCESS in 1h 20m 21s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/bfe30014f02a4d5fbc3f21bb0bb144ff : SUCCESS in 1h 51m 14s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/7a59cb688ce943f4959b7066e984cdf2 : SUCCESS in 2h 01m 41s\n- neutron-tempest-plugin-designate-scenario-zed https://zuul.opendev.org/t/openstack/build/30aeb863775643239d84278eaf26990d : SUCCESS in 29m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c281fb897ef48d3af35f83056e02b3e : SUCCESS in 37m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a89c362a34fd45289af1fb4eed222b69 : SUCCESS in 13m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa8295f64d41458e967feaf87c86290a : SUCCESS in 25m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2264cabb5b74e18b0d4bf9c0374aef5 : SUCCESS in 24m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab09f4f6af5a468db9faf676d6a020b6 : SUCCESS in 27m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a5556f9df844b8cb6453042c8195ea5 : SUCCESS in 12m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25c12479163b4175987342ad2e140739 : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/800eeade834643d081c8e44b18e8bfe1 : SUCCESS in 40m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecbbe88d33b741b49c3d55d8f7f6b9d1 : SUCCESS in 1h 56m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dbb0a3fce3814e3682aeb8b439cd6505 : SUCCESS in 1h 14m 07s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7de40bbb06634694a0a4bad5d8f131f6 : SUCCESS in 56m 54s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d595f62ea6d24912b8fffc7fd6e48eb7 : SUCCESS in 1h 05m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bd4b34b33e1a4676891a336d9bdcc373 : SUCCESS in 1h 17m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/443f738d4d4b4be680af775bdd111ed4 : SUCCESS in 1h 35m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/032243a851194216988e8f3f11fefcfb : SUCCESS in 1h 08m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/67d45d1e3d234191bef4e99b88d7ab7a : SUCCESS in 1h 21m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93912203294aa268f29c90bf16af30798bb7e3c7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-08-30 12:43:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f20eed533eba8554a771db2dbf1c987cdbbf972f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-08-30 20:07:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb4f7510ba25526fb5b1d39a66d822cdf67e7575","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 20:07:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89ff7e68ac98a24bc3a269b499d7121b19a9300b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 22:05:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf1a3ac9c5f9496f91cfdc4ba321395e\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/c5420eb06e3942a19425e994317df1b8 : SUCCESS in 1h 45m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0565680bc56642baafe71d90ffa79182 : SUCCESS in 13m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5eeb7f0bbde04b7ca231ca679404c909 : SUCCESS in 24m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8056a59075f0465ea79e2ca56fd9f407 : SUCCESS in 22m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b3aad3e154c4219a46cf7cdbac6efc6 : SUCCESS in 13m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/145bef68217c49ab99b490f8d8a53ae9 : SUCCESS in 9m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/73247401e1ec4ad49d5244430314c85a : SUCCESS in 42m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/35846d6aeac84a42bde984ce80309be3 : SUCCESS in 1h 52m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3af00e5824d540a082ee239e494a29bf : SUCCESS in 1h 20m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7ef62c43b6e4094a5e241134b2a28c6 : SUCCESS in 1h 04m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a2ab0325be5a4225bc1aa52818bb4e45 : SUCCESS in 1h 00m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afcb3c27c4e64d549fca31f6c5a2ff7b : SUCCESS in 1h 07m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7a4cc9531e58af3a2643f116d2a5e8ee4d1837ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 22:05:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a3df5bf9d1afdc1db1cfbb238047f893fd1cd11b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 22:06:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/459cfc713c7a4005851d5a91d8f351fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce2ded6d6204076999495adae0b5844 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acfc2e52a3c9406ba33a218e89ee32d1 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"850da2af847742e483c70c1cb36b11154e5e141a","revisions":{"850da2af847742e483c70c1cb36b11154e5e141a":{"kind":"REWORK","_number":1,"created":"2023-08-30 10:33:54.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/41/893141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/893141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/893141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/893141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/893141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/893141/1"}}},"commit":{"parents":[{"commit":"cfd139ef198f29bc8d39a96141da4831a22ae2af","subject":"Merge \"[OVN][Trunk] Set the subports correct host during live migration\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfd139ef198f29bc8d39a96141da4831a22ae2af"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-29 10:37:52.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-08-30 10:31:55.000000000","tz":60},"subject":"[OVN] Cleanup old Hash Ring node entries","message":"[OVN] Cleanup old Hash Ring node entries\n\nThis patch introduces a maintenance task that runs once a day and is\nresponsible for cleaning up Hash Ring nodes that haven\u0027t been updated in\n5 days or more.\n\nChange-Id: Ibed9e0d77500570c3d0f9f39bfe40cb9239d0d7a\nCloses-Bug: #2033281\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n(cherry picked from commit 7f777c223e557f7bb0e0290090019eaa60095b50)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/850da2af847742e483c70c1cb36b11154e5e141a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/850da2af847742e483c70c1cb36b11154e5e141a"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
