)]}'
{"id":"openstack%2Fneutron~893185","triplet_id":"openstack%2Fneutron~master~Id5b0b230e342074d8ca27e4fc4ce1c1f0c1b0d34","project":"openstack/neutron","branch":"master","topic":"2023-aa-l3-gw-multihoming","attention_set":{},"removed_from_attention_set":{"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2024-03-05 17:26:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5b0b230e342074d8ca27e4fc4ce1c1f0c1b0d34","subject":"Add release notes for l3-ext-gw-multihoming and adjacent features","status":"MERGED","created":"2023-08-30 13:35:47.000000000","updated":"2024-03-05 17:27:46.000000000","submitted":"2024-03-05 17:26:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893185-2023-aa-l3-gw-multihoming","meta_rev_id":"06849de2ec23dfe72beb209587ca27a685f80860","_number":893185,"virtual_id_number":893185,"owner":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-05 17:26:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2023-12-04 06:18:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-03-01 08:36:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":2,"date":"2024-03-05 14:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"date":"2024-03-01 08:36:09.000000000","_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":1,"date":"2024-03-05 14:48:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-30 13:53:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-04 06:18:21.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-03-05 14:48:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"158cfef0a0313fc49ed28777b0788ac06f3f2c92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-08-30 13:35:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46d54cd44edef823fbef130491e6c9473a9ba456","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2023-08-30 13:37:17.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a378b60b1511d1e712b0482573feb2efb739c9fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 13:53:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27b8f57036c440bdb33c0ffb7163fd8f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab32f22160f49c59b761d5fd88128e3 : SUCCESS in 11m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f3b718e3444e3b843e4e2a0f54d73f : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab3fcde7636b4c89bb89d109aaa2515e : SUCCESS in 9m 59s","accounts_in_message":[],"_revision_number":1},{"id":"39becf5a5e14c07b0c173dba1a4c450bc4564827","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-12-04 06:18:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b93be9c0a22b9cc0e36c24d551ded9c052f1ff4f","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2024-03-01 08:36:09.000000000","message":"Patch Set 1: -Workflow Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f91b3fa949e5cabbd486af53d2d58405a3eb6aa3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-03-05 14:48:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c367ae43dd3e32993be0da8079a3fd846ee0d783","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 14:48:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ddcae4ec19dc444eb3d8da0a811c4ed279a4542a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 17:26:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62b2b6fbacc64daf8a70f440ff4d8634\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3f097ece69e4692965862252ef35a19 : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34d56ec6c1e54d678dea0f51b8c93939 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d049011cc7c4537b5f8d9295a586f00 : SUCCESS in 13m 54s","accounts_in_message":[],"_revision_number":1},{"id":"dbd2c3bed00040ae21cc7c6b6e14f71fb7cadb85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 17:26:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"06849de2ec23dfe72beb209587ca27a685f80860","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 17:27:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0957f396acb043eebfa36ad5c0adf3e9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d34b0ae49684ac3818519c628987a7e : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fecd8f14caed42cf9c0d495cf3f9c03d : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f16310515860beac14ec04dddb97573a69181ecd","revisions":{"f16310515860beac14ec04dddb97573a69181ecd":{"kind":"REWORK","_number":1,"created":"2023-08-30 13:35:47.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/85/893185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/85/893185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/893185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/893185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/85/893185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/85/893185/1"}}},"commit":{"parents":[{"commit":"c6b6ecc75196bd47ebcea0284e7d714c861fa630","subject":"Drop release notes for l3-ext-gw-multihoming and adjacent features","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c6b6ecc75196bd47ebcea0284e7d714c861fa630"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-08-30 13:29:46.000000000","tz":120},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2023-08-30 13:35:23.000000000","tz":120},"subject":"Add release notes for l3-ext-gw-multihoming and adjacent features","message":"Add release notes for l3-ext-gw-multihoming and adjacent features\n\nThese features was added to Caracal.\n\nPartial-Bug: #2002687\nChange-Id: Id5b0b230e342074d8ca27e4fc4ce1c1f0c1b0d34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f16310515860beac14ec04dddb97573a69181ecd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f16310515860beac14ec04dddb97573a69181ecd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
