)]}'
{"id":"openstack%2Fneutron~893333","triplet_id":"openstack%2Fneutron~master~Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b","project":"openstack/neutron","branch":"master","topic":"fdb_aging","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-10-24 16:00:17.000000000","reason":"Change was submitted"},"34451":{"account":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"last_update":"2023-10-24 16:00:17.000000000","reason":"Change was submitted"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-10-24 09:42:33.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-10-24 16:00:17.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2023-10-24 13:02:14.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-10-24 16:00:17.000000000","reason":"Change was submitted"},"23804":{"account":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"last_update":"2023-10-24 16:00:17.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-10-24 13:57:02.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2023-10-24 16:00:17.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-10-24 16:00:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b","subject":"Add support for FDB aging","status":"MERGED","created":"2023-09-08 09:11:17.000000000","updated":"2023-10-24 16:01:36.000000000","submitted":"2023-10-24 16:00:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":75,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"893333-fdb_aging","meta_rev_id":"42e6fe75fadc7b51b4263d022014d2585eadbf5a","_number":893333,"virtual_id_number":893333,"owner":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":1,"date":"2023-10-24 09:20:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-24 16:00:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-24 09:42:33.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-10-24 13:02:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":2,"date":"2023-10-24 13:57:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":1,"date":"2023-10-24 13:57:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"CC":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-08 09:42:30.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-08 09:46:50.000000000","updated_by":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"reviewer":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"state":"REVIEWER"},{"updated":"2023-09-08 10:42:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-08 11:04:28.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2023-09-08 12:25:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-08 13:45:40.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2023-09-08 16:02:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-10-10 12:36:47.000000000","updated_by":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-10-12 21:24:50.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2023-10-16 08:57:37.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2023-10-16 09:49:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-10-16 13:49:04.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"}],"messages":[{"id":"c6407aef69069b110ea21d6045f7abc5b2da8cfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-08 09:11:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cf13252cbb82485027831fa20918f0887746ecc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-08 09:21:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ff64b1f054c83c3d220f9743c20286065f4cd3f2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-08 09:42:30.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/33/893333/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"6727185c50214a002028e209e6ce9b763d77c014","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2023-09-08 09:46:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4521cb1943fe67c9caf2696b2df16e92d9da382d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-08 09:54:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"bcf0cfc2edbafb52ff6a9c798500261f118ce55f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-08 10:14:32.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/33/893333/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"dd3ebc055f0b80f14f425c09b17302ac8f735573","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 10:42:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e72e1cea2d904a6a9c70132adaab5d88\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f673f67e45284350a3ae042ee3f88036 : FAILURE in 23m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60a922d5f89643028199b5500e5510b9 : FAILURE in 23m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b7241d8b2cd54ed889f6b37504b2d9c6 : FAILURE in 45m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d3a76c3fdc1a43588c9ed7069b65fa81 : FAILURE in 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8267067f83d2eea4f513b40d9c03d9df35544fd7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-08 10:46:41.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/33/893333/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"7a47b1bcd05f3217b52cfbc825555aab23972e9e","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2023-09-08 10:47:06.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c3f38f87120c577a8ba032e658bd6b679a9ebcce","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-08 11:04:28.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d2d71199c1c9066d72bfb9575c8e89dd084ee6f1","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-08 11:05:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e0514263597f72e2c1476d695597ac24b401f466","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-08 11:07:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fab89f916a9dd248134a53aacc4f99a48ac12727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 12:25:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae99665c5c244a57b92a72409a4f5f56\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/285deadc82a144efafcaab4b00f6e8fc : SUCCESS in 1h 34m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c5dd458628d44ec2aedf94d23ff4f755 : SUCCESS in 1h 52m 58s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7ab85d60aa414c27a155bd2fd58d6a35 : SUCCESS in 1h 36m 12s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/671207a8327f4f2a9dca8e8c7ca0112a : FAILURE in 21m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6dfc3b3b670d4182a78784e445ddd336 : FAILURE in 21m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f4750a3039b4a9aa9ed98d070c1c40f : FAILURE in 47m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61b61f86252246fd9d384eb7cd8b81ff : SUCCESS in 10m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9527fa4df0044a55853c5c279f884297 : FAILURE in 30m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56c6ab59766f4dfd893af71ec5ff63ff : FAILURE in 24m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ee2654a3c7b46e6a70ec0e893273925 : FAILURE in 25m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7fc6ecaa32d4c65a2fc2605b3317fab : FAILURE in 18m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da34ee44e364d6b96ba6263eaeed160 : SUCCESS in 13m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f39b70a262a44defa87ec38404a9ff79 : SUCCESS in 10m 24s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6c7597a2a7f24b9bbfc3b64b3eb165c8 : FAILURE in 27m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5411589753cd443e8c78a2e2b01678af : FAILURE in 42m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b4de413e2e3b462693ea797d0cef6d9e : SUCCESS in 2h 24m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c959e40f6ef45c4b6fd7cc5b74a2154 : SUCCESS in 58m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bd8c854beb8846a69746dc5b28bee024 : SUCCESS in 1h 02m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f0d9d968569a4b80a78853224d9884df : SUCCESS in 1h 23m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/322dbf86039a4e368db3aec2f6234d81 : SUCCESS in 1h 17m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9d0c3d0383b2464c98cc87215ae55594 : FAILURE in 21m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ed739479b49a4d75aef042f4a44c3d05 : SUCCESS in 1h 26m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f9f4c165949d45459449f4ef2db8abb8 : FAILURE in 25m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3db625c19bd340059f2106acc509c15b : FAILURE in 24m 50s","accounts_in_message":[],"_revision_number":3},{"id":"9a0f123a9d0b6c120302a67d3adfdfcef2ad31e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-08 13:00:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"94eab9b1564463e805f2f8eb605f4e66e3445fa9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-08 13:34:46.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/33/893333/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"1bd18b2969c33253315fc56de23d49239eb72576","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 13:45:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57303b60e0c84dba8cc3b1916fc800b5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a164b070a9814e3e97c95e7e41aaece7 : FAILURE in 27m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75b80d60812840e2beab7cd57514ffa5 : FAILURE in 27m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a193c818430f4c4bab17cf2f4ff9965b : FAILURE in 41m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b5d5798a38a74cb192d8b907f66b007d : FAILURE in 31m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7cd6900a0ba2e7310fbb2357c8e283d9d395f5ad","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-08 13:45:40.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2f548a964001433a6755d817f4f353fdfd2e2f1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-08 14:51:01.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f8df78f4aa220387256a998dd869ce852a423cec","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-08 14:51:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4c01e7c178a9527cf9b0a411e427d82d813d0f0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 15:29:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc3d598112914965810a2c1da3b66eef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f9dc57528e7e42a4a74640d850cffcf3 : FAILURE in 26m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7b3298911a045a385afe720674f28fe : FAILURE in 35m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eee246cbe2b940109774d6c28fe7e9f7 : FAILURE in 19m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b9e05b60b2fb42d1bc027d3db9180dfa : FAILURE in 21m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"85a03139d42fb124d06628b4607ccdc3f5598f1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 17:10:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73d6ec20c5cf4d0b9a7a233e987cbcd8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6b4bae30a35f4904aa5bdadf3bbab0a1 : SUCCESS in 1h 32m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6bb817bcfa3d4df0bf13ffa7e78b7ea4 : SUCCESS in 1h 55m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9326aa9d448647c099c391f900176337 : SUCCESS in 1h 37m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2045177643b04055b913aaa0f5ac56be : FAILURE in 21m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7dd26f5ecae74646a656457de4951ab8 : FAILURE in 21m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35aff96d11cb44f4aa7ea6bac2dcc7e5 : FAILURE in 37m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e555fb2185143d7bb23c07ec5fdb570 : SUCCESS in 10m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/174636e604924be69b47a362026978c5 : FAILURE in 23m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e57b9a10dfbe405781fcaa6fb8995004 : FAILURE in 28m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64e4d92769854931a21c1f5d1f1f8fa0 : FAILURE in 34m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2ad54e5391b40e28dd0cdee865dd891 : FAILURE in 26m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb53364e4c494fed92b9151e1911f107 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd512b6fe23f4f4b87e9aa3c150afa69 : SUCCESS in 14m 01s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/944c3854e913497a8b68da66cc30fe34 : FAILURE in 28m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/334a3d62a2ab416c961ca35dc1759e09 : FAILURE in 55m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c27d7b4e6dce445aa8365c9f062a33f5 : SUCCESS in 2h 11m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5062a1aa224477ea252c12bb6872bc4 : SUCCESS in 1h 02m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2186dfcfbe8642bfb685a3d17bc05fb1 : SUCCESS in 55m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d3cbbcf05cc747699c95852d35fb2b50 : SUCCESS in 1h 36m 43s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3047c9d012c341b0bb861d7b8bdf559d : SUCCESS in 1h 33m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/790e2e5bd02c4cb1aad7561047afc34d : FAILURE in 31m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/40cd70a6bf804d9d874d936220ff5f58 : SUCCESS in 1h 45m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7603286f58324088b8cb9bf617bff8c3 : FAILURE in 34m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5ea33ca4365547d6bc5920b7db6b6e43 : FAILURE in 17m 14s","accounts_in_message":[],"_revision_number":5},{"id":"7b21bb52dda8ecdfc34d425611fe990217bddeef","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-08 18:19:50.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/33/893333/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"35deea766dd817d5dca890e1c06bfae9e67b401e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-12 16:45:07.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f54465c668bc677a0904746542703bc2a3d3295a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-12 16:59:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bf7063edcfd48eb7a90a8941d0f7b39583cc4f35","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-13 07:44:37.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ab6ac4aa176e1738cc7b6c057a2e2fecf20ab420","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-13 08:04:48.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0c3d946e5cbb08b6bb675b786cb24f1f3f9da147","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-13 08:27:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"af3a3ee1912f0475f9efd4f99a0e753d76cea7a3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-13 08:36:46.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/33/893333/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"dbfb5f24f5ea74c4b0d5f1712a05dfe233e636bc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-13 09:09:33.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/33/893333/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"dc0b22bc16d75c57510104df00280f7e677a5b93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 09:11:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bffa91ca4084b9888d76317a57d8942\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53f20caa561648beb129312b1274cfb4 : SUCCESS in 38m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5208a979bd284fc6af91c453d63714fc : SUCCESS in 41m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b983c6fd2de54e94b76dc8ca1b2e8d07 : SUCCESS in 40m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3d296702bcd946a6b02f655a52960290 : SUCCESS in 35m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"68aea33c7c1b20232c283929a91fa9cb146e1940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 10:50:58.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/372a57507c0240d58bc7963fb019e975\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/db69c23de6114710b69cced562e323ad : FAILURE in 1h 46m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8c7f094a53cd4e6288e89800d4cee218 : SUCCESS in 1h 49m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/41639d5066f1405c82d482e72f316d92 : SUCCESS in 1h 39m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4f5be556132748b48f9db99358972c02 : SUCCESS in 1h 42m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fe807f4d9194491084e8250bd732f0d4 : SUCCESS in 31m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b040966e48de4eeb8972f37f2dff64b3 : SUCCESS in 36m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b8f7332513a4c3491d9961cc709ec16 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90485b2b6a234f939d81cb8612c554c2 : SUCCESS in 22m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adf18a5c8c7c42d9863cacda47c54172 : SUCCESS in 27m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d978681f33b04315a477d8613e6f3ec6 : SUCCESS in 23m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/887efc56d4ad47c1862cc7a65411cd2f : SUCCESS in 18m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e63a19dde9479aaa52ec60b829a454 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a6e18a5199545978b298b05005bdab6 : SUCCESS in 10m 30s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/829a10d0cab54a03864835de4e99ee17 : SUCCESS in 31m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f01e8943302438c960846f76ce61216 : SUCCESS in 1h 03m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9e5597ca869b47cf9748834bc69a58ae : SUCCESS in 2h 17m 59s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/31658905b57244da81623a036a4d7b03 : SUCCESS in 59m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2e9a92fb68b642be8270ab38cbdae23a : SUCCESS in 1h 03m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2414df3cb954429abd49557886245801 : SUCCESS in 1h 36m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b10d90a80d514fd592d0a1777a0353d0 : SUCCESS in 1h 04m 44s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3c06c40c0b6c428a8d33f4818fd44179 : SUCCESS in 49m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3b21bbcf094f41edb80a542f0eedf121 : SUCCESS in 1h 56m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f65c587149ac4efabdf2786495de59ff : SUCCESS in 2h 03m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/25cf4c12038d4bfd9ef3d997020e52f3 : SUCCESS in 1h 13m 33s","accounts_in_message":[],"_revision_number":7},{"id":"2e405cd662f80025c766d00be0426de6ee2d70b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-05 14:44:01.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"7505c1dfdd78be9a94e6c7558003985fb3b775bc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-05 15:34:19.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/33/893333/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"eab6de29db8233a9db9b5cb57d46acf5cb280c11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 17:05:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5e21205047d445bb420f591da1fa29f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b4a28808cb6e4545930ba29ca3b9ab5d : SUCCESS in 1h 41m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3a49883d9a524edab7a6acc9b323e608 : SUCCESS in 1h 40m 26s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/adab7122af45473e817596a6d87d2434 : SUCCESS in 1h 38m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/614df9620dfc408dadb0ab18bee95050 : SUCCESS in 1h 26m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5a266ed543040de9d1664a46e315fd4 : SUCCESS in 41m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dd582d7123241dbbe1e01835589897d : SUCCESS in 38m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7ba93d3a7c4ab49da91b12ce696924 : SUCCESS in 11m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41614f1e3d7c4c6b8e66c56f200a9ae8 : SUCCESS in 25m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a213688231464b5199a806344859369b : SUCCESS in 26m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce6e4d6ac8764748a4e8f5e544e854e8 : SUCCESS in 34m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b044fdc5a6814373bb5efcfbab30ab99 : SUCCESS in 30m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf69fa302ae4f69941760cc5f6d247f : SUCCESS in 15m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa0f10c8a730475b98d8c77610851c6c : SUCCESS in 19m 53s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b632ebd5f43349519ff9931f28b85247 : SUCCESS in 30m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/defc7dcd6403451098b9e0d9035d94b3 : FAILURE in 57m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/152c7e14a64c4b0fa4e07a2abc649b39 : FAILURE in 2h 13m 04s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/df81f8ea83fd42b8b32ea7a08650954f : SUCCESS in 41m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/78ec869c7b4b461284232292337e22ef : SUCCESS in 54m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3b87151113c54c8d9afcd9e9e874c792 : SUCCESS in 1h 12m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bb46487caa234678a036e52da9364e6b : SUCCESS in 1h 24m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bef990139b8e4108aa5b8eb5f2fb26a2 : SUCCESS in 1h 05m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3c6efab9658e4398a1d8be9a85032234 : SUCCESS in 2h 05m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8501b23ff52c4d6eb9203708190a41fc : SUCCESS in 1h 19m 34s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9f4d9db0639749079522f11adbeaffd8 : SUCCESS in 1h 18m 16s","accounts_in_message":[],"_revision_number":8},{"id":"f10dfcfa4ce61cfaf8e6477b9e7622fb0aaabd5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 18:47:47.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46e6e87b2d224ef3abfbd3117cc8d229\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48bd632abcba4f04a16f1e1187025dea : SUCCESS in 53m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c53c41309a24b91b6b11d848ff1cbb1 : SUCCESS in 43m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d4cb88620eda43b38e7988bbbb545e6a : SUCCESS in 40m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/831a6447368a4a6fa0983b4768f5ecc0 : SUCCESS in 48m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fdcb8823a57c02207733567e88c8138f12ada3ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-06 05:39:54.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"cae36ec7248bad99949956ac15aab9d32ef7ae69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 06:51:09.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9d47eac62e94e7ca5508ef217f335b4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e18e6815ea45480e870cf2a21c22b8c2 : SUCCESS in 1h 06m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9add6ea09922423eb5cdfb70b8fa0c0b : SUCCESS in 1h 06m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e3850f8fec20448093d8a4c3a7285102 : SUCCESS in 1h 08m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/002eac560a5f436abeca47c2f2645f71 : SUCCESS in 57m 06s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"744cfaa706b0b128d5c3dfe9f5e1dddf10b559c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 08:32:16.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10284d209d8148fa9e1da375da6ccb93\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/196c2af4176a48a39597102a76c3a4c8 : SUCCESS in 1h 26m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8b7319d9f03b43a4b7d99aad1ca66665 : SUCCESS in 1h 48m 15s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ede54a6ac3b945cca7daa3d18be36971 : TIMED_OUT in 2h 50m 27s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/526d2936a1a2416194f5fd4963caa782 : FAILURE in 22m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9f60a656a1474c26b5f80df43213a26d : SUCCESS in 37m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fdba72fc4d3458a8641166d0dae090f : SUCCESS in 38m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/691dbbe2b4bd40e59a8e56d037baa31b : SUCCESS in 10m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c58647c316f46a5bbd3e6ec17d116d5 : SUCCESS in 24m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/113a12a184cb4bc98c43be4d3e39858e : SUCCESS in 28m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5637cf3c0404471a221fd723b5c26a3 : SUCCESS in 24m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f44793718cee43b288ad31fdaa1eb155 : SUCCESS in 24m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0b84b94ee145d79359c5fd5147ad32 : SUCCESS in 11m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/126a9b907ad64b75bd75d56c842d8003 : SUCCESS in 17m 09s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/83a5753cac1f48778e625d69713a54c4 : SUCCESS in 27m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd7f79ff69e445abb4034246de27fb88 : FAILURE in 38m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a755850c9ba47e6b05946ee00d93932 : SUCCESS in 2h 22m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f464b6204aec45d292989a52c818ce36 : SUCCESS in 54m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/958359705fea45c780c0dc4c14b3adf1 : SUCCESS in 1h 08m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f9ae84bed311438e9cb7f023fab4a5c6 : SUCCESS in 1h 08m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2fe36ac19ae24090a23f8cc52c2b27af : SUCCESS in 1h 24m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5988e335aa804cbdb111a32db8f1b119 : SUCCESS in 41m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/12999e1add1f4f669fafa2c6bef16b7e : SUCCESS in 1h 42m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5091c2a53ba74088a9184b3c1f85f32f : SUCCESS in 1h 19m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/52c88f2ede3343c49bbf1098a2e3154f : SUCCESS in 1h 15m 33s","accounts_in_message":[],"_revision_number":9},{"id":"432d7184e6cdab5fd6d8e6f1fa4683a4533c24a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-09 05:41:47.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"22ffa437728d9f98b882a5a2931399ad61d6fd89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 06:32:05.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7944dc3d93a43478184ee24a1476d03\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ac5ddb498d14247ae40c0373e10faf8 : SUCCESS in 44m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1af6b9a8c68b40019caf643721e940cf : SUCCESS in 48m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5fe0788b5ac344e5a5886fe5f49badcd : SUCCESS in 37m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/abe8b7fe3f2b42cd819f4e04c979e4d9 : SUCCESS in 41m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0caaaba2bdeb9013f1218303cb815b120b7ddd92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-09 06:42:28.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"f4e8d1be9c8b5c5f23b934ad86745fc04d47bea0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 07:47:46.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c77aa15aad8482eb1428180cc481ba2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c88eceedeec5401ca221d009465db9ae : SUCCESS in 59m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d87ce4fb8c484e31a58471da74fbf113 : SUCCESS in 1h 01m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2d3d3129fc144d1aa74be157df96b575 : SUCCESS in 1h 03m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/077009d4b2264790be58ce4430a07c93 : SUCCESS in 52m 52s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2e3ce5cf5e6369eaee61bd9cd790b6e91d0d7071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 09:46:12.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05f9faff7fb14ca788b6a30d3a5902a4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/680730d7544a4e78aced2beca4af9b29 : SUCCESS in 1h 08m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d4352d0296294585b5e6ab693e4e09f8 : SUCCESS in 1h 32m 41s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/73f61a48a08343259a8fb909a087a469 : SUCCESS in 1h 11m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eec6cb52e05f41d1aaa8313eb09b898a : SUCCESS in 1h 33m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d00c4a18c2a14c8796edbdae6041c228 : SUCCESS in 41m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3241fa75f060472ab4656537ec79eb56 : SUCCESS in 38m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d364ec355f244c0eb7ebfdf4c0d52420 : SUCCESS in 11m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a69e1c867c0a40e594f6e40961eeb575 : SUCCESS in 32m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf1d25e7572c4c3b9eefccab9a228d0d : SUCCESS in 29m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1419b9ed205468da83cbab076244ca5 : SUCCESS in 30m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e951369eb0f4118970263427cb46940 : SUCCESS in 23m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3733f1156964dc19a05330ef4f4457a : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a11e7078ccf4408abd211ad0d675e00 : SUCCESS in 18m 24s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b809fe4302db459db50566753dcbe631 : SUCCESS in 22m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/175dc78a16c34a67b7f66a85607b9e2e : FAILURE in 1h 04m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c8041cc1df44d17a1e416d3b954963c : TIMED_OUT in 3h 02m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d1594c8ebe9348b6bc00fbb925d10736 : SUCCESS in 1h 05m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aebe7d80dc1140d0b3d239ae54907a3c : SUCCESS in 59m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f78533b3ff974622a1a9aaed0a25e827 : SUCCESS in 1h 10m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5fb0bc07ebc6480ea526771c5a5226b7 : SUCCESS in 1h 24m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ced579992b3f4c349a29537aeb9ba33c : SUCCESS in 49m 06s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/383d6ef2bf054806bce51ae78a12dcc1 : SUCCESS in 2h 02m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/61c991755ea34fa8b52a15238fddd1d7 : SUCCESS in 1h 56m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6879ac65fc6a4a5fa8db84f029baeda9 : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":11},{"id":"66211619bcd3b7954e1f9716e14c6b3057145e34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-10 09:18:37.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"da33482f093597551c86d1e0718972d53066b6d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 10:42:07.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4811e0a1eeb48b1960be849e21ac7ad\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95db054808284c6fa0d77bbffb6f8825 : TIMED_OUT in 1h 21m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebdce33f78a34b3da804238485996d0b : TIMED_OUT in 1h 21m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/63885e505cc345b393bec2db770555c4 : TIMED_OUT in 1h 21m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fb947b51adc143ddbeec808dac1c4bab : SUCCESS in 1h 17m 11s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"11664ab992a38376301ed1ff6159df55205bc33c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 12:05:58.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fa3cc08ca3840fcb8dc5b93d95d6cf7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/542ffe3347ad474f92d8789e8c2f88f9 : SUCCESS in 1h 38m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1d56984bfb7048b5a36606cce69d5728 : SUCCESS in 1h 24m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ab7413a566144c03b18d8d3b714f8425 : SUCCESS in 1h 26m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a90a6b4aa79640ecbc7b5860c2f4bf6e : SUCCESS in 1h 47m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5eb7e7b3b2a04e24a73c79742c55ecce : SUCCESS in 31m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25ecb468a2c940609052e4a6dac949c9 : SUCCESS in 37m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/644a9d95fff341f9a45de0e6d8170fc7 : SUCCESS in 11m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0a81d6abf2943fe9fafe980b4287909 : SUCCESS in 26m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d55c134d41d47aaa272e82642b65b87 : SUCCESS in 22m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27497949c43040f981328a0b866d5b0d : SUCCESS in 22m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab46b741fefc4e768ff322131039add6 : SUCCESS in 25m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d13933774714994aa6bb2fe44d89ab1 : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b918f81fa9f648aa913b78527fab900f : SUCCESS in 19m 03s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f356a40348424656ba7ec7f2a51270d1 : SUCCESS in 29m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0605da57d3d64562ab96916f5764f324 : SUCCESS in 50m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/921be94c2a8e460f9e7017ff5f1ce4c7 : SUCCESS in 2h 39m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e82ba72e4484dc7a402d9ef22c75ef5 : SUCCESS in 1h 04m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3322feaa4f0a4004bd26ca58281697aa : SUCCESS in 53m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1e05bce36f304c428d8b3da33db8d756 : SUCCESS in 1h 30m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3514069d4c9e4181bac99e0ddd0e0235 : SUCCESS in 1h 26m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/196d9f292b2f48fdb40a1f7ea73e612c : SUCCESS in 1h 01m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/dc25de7d264b4aae883fbef6e0e39b09 : SUCCESS in 2h 01m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d809d87d823d4b24970f53a0c2213e4e : SUCCESS in 1h 59m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f4fbaedc2b274253a4bbb37beef34fd1 : SUCCESS in 1h 12m 11s","accounts_in_message":[],"_revision_number":12},{"id":"549757363c4b69d93eeea22ebbf257b51732534c","author":{"_account_id":34451,"name":"Fernando Royo","email":"froyo@redhat.com","username":"froyo"},"date":"2023-10-10 12:52:38.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"439b78fc772b939316719d1293d6f52038efec8e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-16 08:57:37.000000000","message":"Patch Set 12: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"283d96abeffebeb77bab56c84d4b928c8f9b97fc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-16 09:49:29.000000000","message":"Patch Set 12: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"e517f7a911c57778caabf726c855f27afe58fb82","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-16 10:08:03.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"44816f72ab42a20b7a4bcab5d41c98b37ecd83f0","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-16 13:49:04.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4c146ef2c22c35490cac83ccfec4427ff78fb82d","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-17 06:51:43.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"472adc7d08deef72b3f5c0164f5a1fae6801cb37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-17 06:51:50.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"d085d124dd45d066a9a900cc2a5e6b378221e9f1","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-17 06:53:19.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"90242ee7e2d8301b404f201145e490a8bcb2b1a5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-17 07:26:20.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/33/893333/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"383bcf2f6c04b0dbd50f10d1f7e16253e6cbc5f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 07:49:42.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/959e76eb62974b97bad7e709fedfc4d9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/828d5be29dd645a9b5df60bb1c42dda2 : SUCCESS in 55m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1f0afcf2ef7472cacc5b9b314908c26 : SUCCESS in 54m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ae294ffed634424a29a54f568973fa5 : SUCCESS in 53m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f182c4d9ad3e4fdd81b631a2e6e43f0f : SUCCESS in 50m 32s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"004d195c407869674569df2c3a22e36cab94bf2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 09:23:35.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/428596d7507d46838628fa6bcd5e8345\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4bffdc00d87d43058d5cd5f727f53df3 : SUCCESS in 1h 36m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d9e6170d3b7f4565b8757dff74c12729 : SUCCESS in 1h 34m 22s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c41175b5f8d44f19a98b96ae5a3f963a : SUCCESS in 1h 34m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/433fbc24c3ce4215a638fa05bfe8b62f : SUCCESS in 1h 41m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d70663fdafcc4e8f920399292f34eec5 : SUCCESS in 36m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fdbbebf68c834be3944d3b0c92395d2c : SUCCESS in 30m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc4b61a25d24a4f9d6dc3e2728df84d : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52c8e6825abb41e2a1d1c9afe8dfae5e : SUCCESS in 21m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3064a53005ea4821bddeb42f1c951030 : SUCCESS in 26m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb3e071338a242e2976d9cf59121d375 : SUCCESS in 24m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97b24441856c4de98971d3503148f867 : SUCCESS in 22m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f5459779448422896149f5c276e54a1 : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d32b382f6504139b8edaaa5add102dd : SUCCESS in 10m 56s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/13c7af6a4a0443f9b44e2f861bef3ddf : SUCCESS in 24m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ce62df626d143c89ee1ca075bf2e1b6 : SUCCESS in 58m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c61b6e3e4a7e4aa98f8a9faba07bc2dc : SUCCESS in 2h 24m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d7bccdbf30549f18b5f5280ce995b4e : SUCCESS in 41m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/60de4a078c804354976263d0f1145c40 : SUCCESS in 59m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3f59f0b2fe904645be80cf92240f456c : SUCCESS in 1h 11m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/34b8bcd903414c9aa0f0a21e610ee124 : SUCCESS in 1h 26m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/66f3a2f6c16e45c0821c1f6ff83571d2 : SUCCESS in 44m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9c308818ead3451c8e353a30780d36de : SUCCESS in 1h 52m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0cab5f56fb2b46cebbc88353cb45093a : SUCCESS in 1h 31m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5ba8130842194e3e99430f1ebe39bece : SUCCESS in 1h 20m 19s","accounts_in_message":[],"_revision_number":13},{"id":"f2d467251cee0deb29fe51ec9a96d7e2687512b0","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-18 14:29:08.000000000","message":"Patch Set 13:\n\n(6 comments)","accounts_in_message":[],"_revision_number":13},{"id":"4dac688ccfcef0d0889ea3ee71bf42ee5afa0233","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-18 14:48:25.000000000","message":"Patch Set 13:\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"21e28676de6dbc9e0f9b3fef89d179462162d35f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-18 15:24:00.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"b251a7a495b702046093db601f10b6ad1108fcb6","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-18 15:25:17.000000000","message":"Patch Set 14:\n\n(7 comments)","accounts_in_message":[],"_revision_number":14},{"id":"abc258b3bf7d4883af4ddb0902f944f09ec92573","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-18 15:46:47.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"1ba9580c7a643c31f37265792855efc689aa3b7a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-18 16:12:05.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"24ae12c0df08df812416ce8becdfb5d8d74d1512","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-18 16:47:41.000000000","message":"Patch Set 14:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/33/893333/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"3f16cb789b25bfd76416936d2cfd16822d5a9864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 18:02:33.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfe6f681acbc4a1196d565b0ac074c74\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/331cef6173594ac69932a9c048f06e2b : SUCCESS in 1h 53m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4b280ad65ad04cf0b1490e42ba7bc77a : SUCCESS in 1h 48m 10s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c2bd212802e5404f8ece6da5e7550cdb : SUCCESS in 1h 55m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f4188a651c0a422886c89d013796b5c1 : SUCCESS in 1h 39m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/14998aee25ce4d55bda031702ba61dd9 : SUCCESS in 49m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af073737e39248f0bd0f851c7e65877e : SUCCESS in 44m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66794c7ace7346acb255a083391abe25 : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/440632d921164ea7946e9914e6fc7782 : SUCCESS in 24m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/190d1dac71a74ac494991308dd2140af : SUCCESS in 26m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18477ef8bd274b579f31b26ad0c65ddf : SUCCESS in 28m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a088beb6d7a94ec681c88e88718e8d77 : SUCCESS in 23m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c029a23d1b4dc29ea98510303c2923 : SUCCESS in 12m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1598fde086542a4b57865231b0c9500 : SUCCESS in 17m 16s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a6d0dc3e963d41679845ccd7d0cf81ee : SUCCESS in 36m 17s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/99710a753f1a4cc98cedf011cefc337f : SUCCESS in 1h 04m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2daad76e1f224072a605adea56b6c04d : SUCCESS in 1h 06m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d0ca48f8cd644a9a5e4278e66a37839 : SUCCESS in 59m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3c62ed6bf0a4ef59656f2dca431299e : SUCCESS in 2h 32m 20s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ac1f36cff3c240d59e48c3b5a1dec071 : SUCCESS in 44m 54s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/94c7db3f6a55444888a7077aaffba1ff : SUCCESS in 1h 04m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/52abbadaa9184e1788d0bd2630c68876 : SUCCESS in 1h 26m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d34c907da7904362ac107e1c3b954406 : SUCCESS in 1h 10m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/bfb52157a24e42418d14facdf7d4e419 : SUCCESS in 1h 05m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7d786fd951b54a2c9c9ca7e8d7481cb8 : SUCCESS in 2h 13m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d6dc851aea6a4a72aa6799ce756c4ccb : SUCCESS in 1h 49m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/efcb0718ce8847f98adde582c9d504b7 : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":14},{"id":"aae64f5bd6d301c4c2f0f55b0826b68ad29bdeb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 19:44:20.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5d33c9d31b140309176fca84acc0301\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c38803ea096e410dad441e1683b95002 : SUCCESS in 42m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c8e875a4e0143cf8d21413e0e6f0c17 : SUCCESS in 50m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f8427ddf84e64f10a98f65ab055e3f53 : TIMED_OUT in 1h 21m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b13a9c7b398440a7ad459f3ac2accdcc : SUCCESS in 45m 54s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"31444188fe12513d59ab10023127ce0c7060f7ae","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-18 20:39:17.000000000","message":"Patch Set 14:\n\n(7 comments)","accounts_in_message":[],"_revision_number":14},{"id":"df819c7ed4ca6ef9dbbd4a88f81a2950e690eda4","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-19 06:42:11.000000000","message":"Patch Set 14:\n\n(10 comments)","accounts_in_message":[],"_revision_number":14},{"id":"276f0219b4853c3a977d6e7376c35c82e84416c2","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-19 06:43:03.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"5ca23aec70539151c7c7695ebccaf3317ea6bfde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-19 06:43:47.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"9f27f6ea3f08e486c48aa34ffcc0a5b65be1041c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-19 07:15:57.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/33/893333/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"4294da4fd459e2bdd86892cac03eba3dee1a86c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 07:56:40.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffa62a7570704317be2df5a561371a24\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f28b2d3c17bd4f3c904b4433a4028784 : SUCCESS in 1h 08m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9da53d2a02143608c1de8888e952c24 : SUCCESS in 1h 10m 45s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/57ad10a959334e71beb342b3826d2090 : SUCCESS in 1h 10m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/182cd9663c2b4a2aabb0020f5e0f2063 : SUCCESS in 57m 01s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"98762783e8ac2f76b3d6df807f12e0f8261b0d03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 09:10:46.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ba5c900b47b43909c59a4038d0c36ef\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fcaa2f9badc94bf98911aacb31260453 : SUCCESS in 1h 47m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/bd88fd51c25d4353ac11ddcec650c4d9 : SUCCESS in 1h 52m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bf7a08a03a8f4c669c68ec5890a230a1 : SUCCESS in 2h 12m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0b69a4869f8746d8bbefe12d1253cecf : SUCCESS in 2h 07m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5235daca96154f5d989e32dad07fa307 : SUCCESS in 40m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/191407e30f6c4ed6a5d4c72953bebfb9 : SUCCESS in 41m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d095b5e894e432aac58187877a6f21e : SUCCESS in 10m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/375e971132954872b7c78780f451423f : SUCCESS in 26m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7fd03b433514001abef43a4c1d6e12f : SUCCESS in 20m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07aab6f4df7f447aba3a0c7aeb048755 : SUCCESS in 28m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b260039710a54d6694acb861142a2b8a : SUCCESS in 23m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/722cc3e90c7f4da2b72771bf3da00542 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4806f5606aa4179b8206cac4a764cd0 : SUCCESS in 13m 00s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/abd73b305d45462c9c638c8854510f7f : SUCCESS in 22m 28s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2630deb3b5294d7f9e3a33137873eb15 : SUCCESS in 55m 27s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b35a6bdb311a4f4c98d946b493565892 : SUCCESS in 1h 08m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ba211df11d54b13a4fb0918ead64e05 : SUCCESS in 55m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f84b8c969306477cad48794cea5e3ca3 : SUCCESS in 2h 21m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3812836b4e224d1c969b087bc648000f : SUCCESS in 48m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dce569143a854a579316814073341780 : SUCCESS in 42m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d17d034b33bd4c54bf94fd484dfed759 : SUCCESS in 1h 21m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b8899cf9cfe04ab38f1e4a3bd8f1b580 : SUCCESS in 1h 26m 13s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c7516bad092241ae82fd622d6fe77c64 : SUCCESS in 1h 01m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0fcc0cab7d7a494387da1a50fe29b641 : SUCCESS in 2h 00m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4e051c1a0fc24f7e998dd1ddb7c78272 : SUCCESS in 1h 17m 40s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7c67f6857dab44c492ada6a8f96ba69d : SUCCESS in 1h 13m 07s","accounts_in_message":[],"_revision_number":15},{"id":"60a40ee7303aa385a7480018746de89a650c8975","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-19 09:59:29.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"caffe76116bd5ba034cc0d3b8229cb57a35be803","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-19 12:51:47.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"8a8f5c9f2ca55b1af4d440c8a582e086c2b6e1bf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-24 07:41:49.000000000","message":"Patch Set 15: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"4b58d234bb984b67c823bc4caa24f8927b0c3daf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-24 08:25:03.000000000","message":"Patch Set 16: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"5030975bc8a99899fcd6161633a86468a148f8f0","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-10-24 08:25:25.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"9edaaeeeea6d8bf43f050ad694fb8c712a766eee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 09:18:45.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd33a6d1634346b487aadca86bcb605d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/18132e0069d34771b06a5345305da9db : SUCCESS in 19m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08cda9e57c50483e85b1536a11dfc03c : SUCCESS in 43m 57s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8c8d57f773f0499b87a782cba3667cab : SUCCESS in 28m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1273e262604f447fbb3b22778eb1bd65 : SUCCESS in 50m 57s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f2e34e59e494f05677ee315cd7b3fd62f0610040","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-24 09:20:46.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/33/893333/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"3dcbf08aa57f83fa3e3c3aa355ab5754900ec957","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-24 09:42:33.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"28fc096f9c915427db7ab5bd9fbebe813f8564a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 10:46:05.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b6953e40e5a4e109b38d9ca3d536636\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fab4ba31bbe94c538476e65ae7c258dc : SUCCESS in 1h 52m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7e68052ba2b44c24848e7255208be5a2 : SUCCESS in 1h 57m 17s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7c8fdb593e1f47ac81276eff69a014bc : SUCCESS in 1h 48m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9e5b4574be074e279397d6d0d2fac3b4 : SUCCESS in 2h 19m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea43a3103eeb47f19f41dd9ed8938548 : SUCCESS in 48m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27e7cd7c30874b2cbffb6cb432bef5a1 : SUCCESS in 30m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c43157089bcf4f91b49ac6b8c11b822c : SUCCESS in 9m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e07882d76dc746679f93b70d5ff2376a : SUCCESS in 20m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5f7d93e5ec1483897feacdbbbcb7ba4 : SUCCESS in 21m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40571b6f03fa4e8a8a5f859f62ebbe31 : SUCCESS in 21m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d718bf1b4a8c4e20a3e78c1524a77ac9 : SUCCESS in 19m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b212d8c3ad4463893b158d182f76bf9 : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43cc5b6f03a24122844274c284d534e3 : SUCCESS in 13m 00s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a0551c7428014552ab25cf094ffbde86 : SUCCESS in 30m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d032e8d05f8942b5a3d11f0b5be2dae2 : SUCCESS in 43m 00s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e19d0098fb864d059b12a3888c120f3d : SUCCESS in 1h 07m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e02d7c8ff94348f58e1a6fc70d60d79a : SUCCESS in 1h 02m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d85dabba774b49b0a015e0962cb0ea6f : SUCCESS in 2h 14m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c387d9fc47041fdb2561380c9716be2 : SUCCESS in 46m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1cbc616e732046b3afbc5b68330b8b26 : SUCCESS in 1h 07m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f4a5345d09c346f6a1a7ce1984024acf : SUCCESS in 1h 39m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d6f4894ddbf24bc6b0b83541da989129 : SUCCESS in 1h 19m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/749ca81f391249edbc031a859010404d : SUCCESS in 55m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f8a6b92b792046e1a68fa7e69ae5348d : SUCCESS in 2h 08m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fea2f382a9c44fa9a384c106b5e400be : SUCCESS in 2h 01m 02s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/60d788ff60174006aa49dd7b7cd122b1 : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":16},{"id":"08d689be7c91bf48363473d22a1381503c2cb8a3","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-24 13:02:14.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"b6fcbfd3aee70fb27c368b2cce86acf5c597ef0e","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-24 13:57:02.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"19897c65f2847c6509c38c0ebc0035a2c13c77de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 13:57:34.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"e5a495447b83f70aaa6194689179a6a715e33412","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 16:00:17.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1d8bd70f74d40209b1d26623d9c40e6\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/11f32decaa5a4f6296b3fd4a8c9ad0eb : SUCCESS in 1h 54m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ea2862f801174c958c944b889c648dec : SUCCESS in 1h 47m 40s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1ea5a740875f42ddaf513d2d482dc69a : SUCCESS in 1h 59m 56s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9e35ecdc5e5843c29e860f8a4cf1fd77 : SUCCESS in 1h 41m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f132a3ab18d1429692f49cf97c2943af : SUCCESS in 11m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/921692e912b94125a716a811979c834b : SUCCESS in 26m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68702e6b620d476c83173b98f2882d53 : SUCCESS in 28m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fabbe4749364fcdbd76c4091907c360 : SUCCESS in 31m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f02966338e5d427092ed208e637a3a5a : SUCCESS in 23m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c1d8a5b6574ea0b508b87feacc3c4d : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05dc0326011343a6a8a6bed3e6ea5872 : SUCCESS in 11m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab96f468e42e46c5bb8e89bcf74a4ae3 : SUCCESS in 1h 09m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6742775109bd400c8d4b1ec81ddc50ab : SUCCESS in 2h 01m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5882c27d1a3440539f78753721766525 : SUCCESS in 1h 10m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e0a093e7af3435aac8fb7ab978879d6 : SUCCESS in 1h 17m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c152b78cfca5420bb9d6314f1374d632 : SUCCESS in 1h 06m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/99fb42cd65c44fdf9b5beef5818249ff : SUCCESS in 1h 03m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8678c5d74d3042e5b6cbf226ab28e125 : SUCCESS in 1h 21m 36s","accounts_in_message":[],"_revision_number":16},{"id":"50867215b44ef6dd52bfa0ebc0f823e1e5a4b299","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 16:00:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"42e6fe75fadc7b51b4263d022014d2585eadbf5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 16:01:36.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08c822aee96a4e96b7c281ee84d35aa7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e2829063cd14acf89785612f24cfb91 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/893cc5c1dd094747b453e63dcff4fe7c : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"1e9f50c73638171403b71d742321464dcd5ef7ed","revisions":{"ac687a04c0bcf2aba615d2d240c0032c5c983949":{"kind":"REWORK","_number":1,"created":"2023-09-08 09:11:17.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/1"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 09:07:37.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, this new parameter should be\nappropriately configured too: `fdb_age_threshold`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac687a04c0bcf2aba615d2d240c0032c5c983949"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac687a04c0bcf2aba615d2d240c0032c5c983949"}]},"branch":"refs/heads/master"},"187f06b866aec02281f1ee5aa080279da7081155":{"kind":"REWORK","_number":2,"created":"2023-09-08 09:21:39.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/2"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 09:21:28.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, this new parameter should be\nappropriately configured too: `fdb_age_threshold`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/187f06b866aec02281f1ee5aa080279da7081155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/187f06b866aec02281f1ee5aa080279da7081155"}]},"branch":"refs/heads/master"},"fcafa2c6621349528357ad0a59612522d6a35ba4":{"kind":"REWORK","_number":3,"created":"2023-09-08 09:54:09.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/3"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 09:53:59.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, this new parameter should be\nappropriately configured too: `fdb_age_threshold`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fcafa2c6621349528357ad0a59612522d6a35ba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fcafa2c6621349528357ad0a59612522d6a35ba4"}]},"branch":"refs/heads/master"},"2e212764d37e20034f59fa6014758351bf1fe955":{"kind":"REWORK","_number":4,"created":"2023-09-08 13:00:51.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/4"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 13:00:26.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, this new parameter should be\nappropriately configured too: `fdb_age_threshold`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e212764d37e20034f59fa6014758351bf1fe955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2e212764d37e20034f59fa6014758351bf1fe955"}]},"branch":"refs/heads/master"},"1c795bd7789609ecb1c1bd098e8b731af8624c29":{"kind":"REWORK","_number":5,"created":"2023-09-08 14:51:01.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/5"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 14:50:12.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, this new parameter should be\nappropriately configured too: `fdb_age_threshold`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c795bd7789609ecb1c1bd098e8b731af8624c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c795bd7789609ecb1c1bd098e8b731af8624c29"}]},"branch":"refs/heads/master"},"a7f0005a927c3a7922c1843e4fe51a5ea6d372d5":{"kind":"REWORK","_number":6,"created":"2023-09-13 08:04:48.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/6"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-13 07:59:38.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, the new configuration parameters\nshould be appropriately configured too: `fdb_age_threshold` and\n`fdb_removal_limit`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nCloses-Bug: 2035325\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7f0005a927c3a7922c1843e4fe51a5ea6d372d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7f0005a927c3a7922c1843e4fe51a5ea6d372d5"}]},"branch":"refs/heads/master"},"536f9eefbb6ccc830378e639779ed4a62abbb04f":{"kind":"REWORK","_number":7,"created":"2023-09-13 08:27:41.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/7"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-13 08:27:23.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, the new configuration parameters\nshould be appropriately configured too: `fdb_age_threshold` and\n`fdb_removal_limit`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nCloses-Bug: 2035325\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/536f9eefbb6ccc830378e639779ed4a62abbb04f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/536f9eefbb6ccc830378e639779ed4a62abbb04f"}]},"branch":"refs/heads/master"},"b01577618858ebf4f55f39949a311b0a4596e122":{"kind":"REWORK","_number":8,"created":"2023-10-05 14:44:01.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/8"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-10-05 14:43:40.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, the new configuration parameters\nshould be appropriately configured too: `fdb_age_threshold` and\n`fdb_removal_limit`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nCloses-Bug: 2035325\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b01577618858ebf4f55f39949a311b0a4596e122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b01577618858ebf4f55f39949a311b0a4596e122"}]},"branch":"refs/heads/master"},"bd352f5926bb82783a9de43235d0e30e79c65bf8":{"kind":"REWORK","_number":9,"created":"2023-10-06 05:39:54.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/9"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-10-06 05:39:40.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, the new configuration parameters\nshould be appropriately configured too: `fdb_age_threshold` and\n`fdb_removal_limit`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nCloses-Bug: 2035325\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd352f5926bb82783a9de43235d0e30e79c65bf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bd352f5926bb82783a9de43235d0e30e79c65bf8"}]},"branch":"refs/heads/master"},"294c6877daba108ffaa62632eb983489d9c29dd5":{"kind":"REWORK","_number":10,"created":"2023-10-09 05:41:47.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/10"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-10-09 05:41:32.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, the new configuration parameters\nshould be appropriately configured too: `fdb_age_threshold` and\n`fdb_removal_limit`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nCloses-Bug: 2035325\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/294c6877daba108ffaa62632eb983489d9c29dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/294c6877daba108ffaa62632eb983489d9c29dd5"}]},"branch":"refs/heads/master"},"4c2ca4c874119a8d1a4e34b4fd446cb44fdc4c29":{"kind":"REWORK","_number":11,"created":"2023-10-09 06:42:28.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/11"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-10-09 06:42:19.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, the new configuration parameters\nshould be appropriately configured too: `fdb_age_threshold` and\n`fdb_removal_limit`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nCloses-Bug: 2035325\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c2ca4c874119a8d1a4e34b4fd446cb44fdc4c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c2ca4c874119a8d1a4e34b4fd446cb44fdc4c29"}]},"branch":"refs/heads/master"},"dbc3ca925aafff2b435e5b1ed842b7692d861c00":{"kind":"REWORK","_number":12,"created":"2023-10-10 09:18:37.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/12"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-10-10 09:18:19.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, the new configuration parameters\nshould be appropriately configured too: `fdb_age_threshold` and\n`fdb_removal_limit`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nCloses-Bug: 2035325\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbc3ca925aafff2b435e5b1ed842b7692d861c00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbc3ca925aafff2b435e5b1ed842b7692d861c00"}]},"branch":"refs/heads/master"},"e18aeda68c367676be4851bc0fc29f7a2eae9c80":{"kind":"REWORK","_number":13,"created":"2023-10-17 06:51:50.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/13"}}},"commit":{"parents":[{"commit":"2bc9c3833627da0dfdf901e15f78b9be397014e0","subject":"Merge \"Introduce ovn_nb_global config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bc9c3833627da0dfdf901e15f78b9be397014e0"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-10-17 06:50:17.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, the new configuration parameters\nshould be appropriately configured too: `fdb_age_threshold` and\n`fdb_removal_limit`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nCloses-Bug: 2035325\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e18aeda68c367676be4851bc0fc29f7a2eae9c80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e18aeda68c367676be4851bc0fc29f7a2eae9c80"}]},"branch":"refs/heads/master"},"7698d0e46dcabda27d2e1df979d4a2501ed45f39":{"kind":"REWORK","_number":14,"created":"2023-10-18 15:24:00.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/14"}}},"commit":{"parents":[{"commit":"2bc9c3833627da0dfdf901e15f78b9be397014e0","subject":"Merge \"Introduce ovn_nb_global config section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2bc9c3833627da0dfdf901e15f78b9be397014e0"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-10-18 15:23:17.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, the new configuration parameters\nshould be appropriately configured too: `fdb_age_threshold` and\n`fdb_removal_limit`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nCloses-Bug: 2035325\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7698d0e46dcabda27d2e1df979d4a2501ed45f39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7698d0e46dcabda27d2e1df979d4a2501ed45f39"}]},"branch":"refs/heads/master"},"155fd68ce49d43ef6e86056b0e2680985988d4f6":{"kind":"REWORK","_number":15,"created":"2023-10-19 06:43:47.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/15"}}},"commit":{"parents":[{"commit":"55c20cdf1ad6e3a8e8e55e3d58dab8ad0354fa17","subject":"Merge \"Add policy enforcer for \"tags\" service plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55c20cdf1ad6e3a8e8e55e3d58dab8ad0354fa17"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-10-19 06:43:28.000000000","tz":120},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.06 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, the new configuration parameters\nshould be appropriately configured too: `fdb_age_threshold` and\n`fdb_removal_limit`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nCloses-Bug: 2035325\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/155fd68ce49d43ef6e86056b0e2680985988d4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/155fd68ce49d43ef6e86056b0e2680985988d4f6"}]},"branch":"refs/heads/master"},"1e9f50c73638171403b71d742321464dcd5ef7ed":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2023-10-24 08:25:03.000000000","uploader":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"ref":"refs/changes/33/893333/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/33/893333/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/33/893333/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/33/893333/16"}}},"commit":{"parents":[{"commit":"55c20cdf1ad6e3a8e8e55e3d58dab8ad0354fa17","subject":"Merge \"Add policy enforcer for \"tags\" service plugin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55c20cdf1ad6e3a8e8e55e3d58dab8ad0354fa17"}]}],"author":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-09-08 08:40:32.000000000","tz":120},"committer":{"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","date":"2023-10-24 08:25:03.000000000","tz":0},"subject":"Add support for FDB aging","message":"Add support for FDB aging\n\nIn [1] we added support for FDB learning. In order to avoid issues\ndue to that table increasing without limits, which will impact OVN\nperformance, this patch is adding support for its aging mechanisms\nwhich was added in OVN 23.09 in [2]. By default is disabled, so if\n`localnet_learn_fdb` is enabled, the new configuration parameters\nshould be appropriately configured too: `fdb_age_threshold` and\n`fdb_removal_limit`\n\n[1] https://review.opendev.org/c/openstack/neutron/+/877675\n[2] https://github.com/ovn-org/ovn/commit/ae9a5488824c49e25215b02e7e81a62eb4d0bd53\n\nCloses-Bug: 2035325\n\nChange-Id: Ifdfaec35cc6b52040487a2b5ee08aba9282fc68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e9f50c73638171403b71d742321464dcd5ef7ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e9f50c73638171403b71d742321464dcd5ef7ed"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
