)]}'
{"id":"openstack%2Fneutron~893552","triplet_id":"openstack%2Fneutron~master~I642c3eb1771463db73248a8c50c0db6f7467a6d5","project":"openstack/neutron","branch":"master","topic":"bug/2027605","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-05 16:50:29.000000000","reason":"Change was submitted"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-09-05 13:09:18.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"34118":{"account":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"last_update":"2023-09-05 16:50:29.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2023-09-05 16:50:29.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-09-05 16:50:29.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-09-05 16:50:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I642c3eb1771463db73248a8c50c0db6f7467a6d5","subject":"Revert \"[OVN][Trunk] Set the subports correct host during live migration\"","status":"MERGED","created":"2023-09-05 10:50:35.000000000","updated":"2023-09-13 18:16:48.000000000","submitted":"2023-09-05 16:50:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":888776,"submission_id":"893552-bug/2027605","meta_rev_id":"01f7c5728dfd954995032ee8b3a35fb7173ec9c6","_number":893552,"virtual_id_number":893552,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2023-09-07 13:29:47.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-05 16:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":2,"date":"2023-09-05 13:09:18.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-05 14:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-05 14:12:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"CC":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-05 10:50:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"state":"REVIEWER"},{"updated":"2023-09-05 10:50:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2023-09-05 10:50:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-05 10:50:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-09-05 10:50:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2023-09-05 10:50:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2023-09-05 10:50:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-05 14:12:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-09-13 18:16:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"08218fd6f3370de30e2de43974470af4bc260fe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-05 10:50:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d06934d476a3d0741cbdbe7b79b729f0ee0d88eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 11:39:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d881be45b89049ddbe3ec3efbd475430\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/73f1b8557c0a4054823e2f2e3ce371fe : SUCCESS in 26m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d3938f5efdd421c815fe7604c75755c : SUCCESS in 46m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1fcd385841dd442e815f59db9b03bb38 : SUCCESS in 24m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/00442456133b467880c4bf6fac93d401 : SUCCESS in 39m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b524148609ca440718d8d9c10929f885c83243d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 13:03:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ffe714a7a5c402ab973113959c9c1ab\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ef50cb5c8a3941f5bceb3986973ed3a7 : SUCCESS in 2h 10m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8b723057564459b9bfc4850cae21669 : SUCCESS in 35m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84c6d4bd7cf84a1f9359ce906cf7bf34 : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba10b52db2094f2b8cbafb80bdae27a1 : SUCCESS in 26m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e9a357622a14b33965f77396d228bb8 : SUCCESS in 25m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d4532d9086c47808391f176d962dfa5 : SUCCESS in 24m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5241b56e96584a40aa02461eb42fc4f4 : SUCCESS in 22m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21c4b7e283ec49cab8f104f7eaa80fad : SUCCESS in 11m 21s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/da9a15ebbf8a49f79e215978878514d4 : SUCCESS in 29m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/685762c6b5a745e5a7189dc285449a04 : SUCCESS in 59m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6104a765409c4349b781d6b3a2924235 : SUCCESS in 2h 07m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/6494c67a9310450ba74bb0cbc111b2a2 : SUCCESS in 1h 07m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fc75b21225574a4a8c9c81b012023f55 : SUCCESS in 2h 05m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7ba3b06bd8c54bef8a3cefef631571ff : SUCCESS in 1h 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"058d4b1e93aae3d03ec77d1754c40e4bb2a26181","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-05 13:09:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"83bc64ba35ce8f681526b41b772116a60c1a2c14","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-05 14:12:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"399bbb9c04354791ec7a8f1b522bd2982fdf5b40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 14:12:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"169008ed9114e92c0e78e5caf84b8ee08dbc4b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 16:50:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/134242b0226c475cbfd0e6bec5dfad5d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c0f26649613740a795fd09bebc61c4e4 : SUCCESS in 1h 55m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/409cf68f595c498c8c8bf3a7bf227a24 : SUCCESS in 10m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c37d82539ce42de97e7867ddd89f30f : SUCCESS in 23m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bc16564025646b588610dcdad8a21ba : SUCCESS in 25m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9451418b5ac42f89f12207892eac045 : SUCCESS in 24m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41d1a32987a14d0e9b74780b7b156729 : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c1b0fb791dd7403d840370724d672a80 : SUCCESS in 47m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1226d356c2984e07b79f6f083f8615fc : SUCCESS in 2h 37m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d91190bb7cb5422cafcefc0b9b6c3f9c : SUCCESS in 54m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8ca956e6b8c74169873c4c6305d92a46 : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a4dc13a7f8c8e9fe76b25ce164607e957b6b3551","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 16:50:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6d0bb8b36434dff0f386a098e73b7346ac15dad4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 16:52:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8722b60b6dd40d0b344bfb07e4e6bf5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cee4fc3131d24bf7b7c6e779cb83a554 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1},{"id":"c383a12a78b3df66c29a6c509cbd6bdd4094cfec","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-07 13:29:47.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/52/893552/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"01f7c5728dfd954995032ee8b3a35fb7173ec9c6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-13 18:16:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34e441e06cd57c95ce4c5ef155103f356324c56c","revisions":{"34e441e06cd57c95ce4c5ef155103f356324c56c":{"kind":"REWORK","_number":1,"created":"2023-09-05 10:50:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/893552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/893552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/893552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/893552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/893552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/893552/1"}}},"commit":{"parents":[{"commit":"7ed79c1f7890456488a4b44769ed84fea23c9a39","subject":"[OVN][Trunk] Set the subports correct host during live migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ed79c1f7890456488a4b44769ed84fea23c9a39"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-05 10:50:35.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-05 10:50:35.000000000","tz":0},"subject":"Revert \"[OVN][Trunk] Set the subports correct host during live migration\"","message":"Revert \"[OVN][Trunk] Set the subports correct host during live migration\"\n\nThis reverts commit 7ed79c1f7890456488a4b44769ed84fea23c9a39.\n\nReason for revert: the port binding handling done in this patch is\nincorrect and leads to issues during the cold migration process with\ntrunk ports in ML2/OVN.\n\nChange-Id: I642c3eb1771463db73248a8c50c0db6f7467a6d5\nPartial-Bug: #2033887\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34e441e06cd57c95ce4c5ef155103f356324c56c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/34e441e06cd57c95ce4c5ef155103f356324c56c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
