)]}'
{"id":"openstack%2Fneutron~893554","triplet_id":"openstack%2Fneutron~stable%2Fzed~I779b13f10f20f95c4963e0e1667566e4589beb76","project":"openstack/neutron","branch":"stable/zed","topic":"bug/2027605","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-08 13:52:18.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-09-06 08:43:30.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-09-05 19:25:53.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I779b13f10f20f95c4963e0e1667566e4589beb76","subject":"Revert \"[OVN][Trunk] Set the subports correct host during live migration\"","status":"MERGED","created":"2023-09-05 10:52:26.000000000","updated":"2023-09-08 13:53:37.000000000","submitted":"2023-09-08 13:52:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"revert_of":892884,"submission_id":"893554-bug/2027605","meta_rev_id":"1891e18af818198ed51a34c5f5e386d2b5b2f30c","_number":893554,"virtual_id_number":893554,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 13:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2023-09-07 07:22:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-09-05 19:25:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2023-09-08 11:05:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-05 10:52:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-09-05 10:52:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-09-05 10:52:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9334c787ab0ba2f36dbfce548a1f3497e3dd5463","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-05 10:52:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f0f86e85993d6fbf599a202d05963e9cda8be0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 11:33:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a423e1abff9429bab8fe7624cde19a3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da365e7b8c2045718d2b9125dc24b469 : SUCCESS in 26m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd57f3042c0641399829bb209c9ae7ed : SUCCESS in 39m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35fc977a7effffbb240a06f6c022797f29f72042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 12:45:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a7522cbcccf3479f89915582f85c8470\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/a85aeaf9a0b444fea81fa724bfe86387 : SUCCESS in 1h 21m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5fce910fe4140ab9d120920de0fb6d8 : SUCCESS in 34m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/544f4f616cab4d8ca303c6a555918f7b : SUCCESS in 13m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/480c773d0e234f03982ed66c0f7cec1d : SUCCESS in 24m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2327d9f666b47568ad0942527791ce8 : SUCCESS in 26m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c6d360e09d7485fbed8d0671d3c6138 : SUCCESS in 26m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/450756bf1ff74587b427a2f66146752a : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b81462e4c69445ea815376f152023252 : FAILURE in 54m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee1ff924255f4197994acc517901962d : SUCCESS in 1h 47m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a85cd43f23bb431d9cd349245831abc1 : SUCCESS in 1h 07m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5a7b5796429e4dedb9c2b8e1762392bf : SUCCESS in 1h 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9053377ea8fe11c724ca840169d58c6fd5325777","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-05 19:25:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5c980b638267683648e99d29a31535df0f1603a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 19:48:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/423672023ae24b8986e42ed0755e273e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3af1c6f38e74971b5da939916ee8b07 : SUCCESS in 17m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/256a9b84dd6942b5a42732c06b5636fc : SUCCESS in 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48df6fde6c2d4e526b9665a55baf13bee257cd3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:19:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e59f13a9297476b81af8fa7996ca30f\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/1758e4062c934640aa2247468517458c : SUCCESS in 1h 43m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/897ad7d6a646457880d6d364519a357c : SUCCESS in 33m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd1e788c5a6e493ebf2d48d61ced0239 : SUCCESS in 13m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c91f430673464a519e3367a8545cf7cb : SUCCESS in 25m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6660d418ae574930ac00ae013e9c6925 : SUCCESS in 26m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3cbdf5ae01e34ef297b28dfe597d1614 : SUCCESS in 26m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f1c3fce90c413c98aaa6743be6a8c9 : SUCCESS in 10m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc80755a4c2940518da29645c4c48fff : FAILURE in 35m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e7620fcc5256403982506c4c3811613a : SUCCESS in 1h 48m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43a2fa482c0d4065ad7063059e525d52 : SUCCESS in 1h 00m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4a0f1cf3efad48e8be2515319c19c934 : SUCCESS in 1h 04m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3ccb844a9d942e4259c06f3f5c2edf022e8c61f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-06 08:43:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c868aa621958464e60dfb8c7b97af42990f9f3e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 09:03:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc6dd5681df9463799e33eef51c36fda\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1159115e761e45f680b58e458185ca18 : SUCCESS in 19m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0bb6ee5b0f2a4018b982a8a7aaf82b34 : SUCCESS in 18m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7efaf98d88841af37a7f1257707e120a1f614da4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 11:00:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddd20313b56b4b1ea3f13bee366967ae\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/bf84036bbb634dc9b2f665995923d4b3 : SUCCESS in 2h 15m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/586cd3c455804506aab35da703fe61cf : SUCCESS in 36m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b0e44513dfd4c369a010b077406c486 : SUCCESS in 14m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33edec124db34e1a91f72f27f28f474e : SUCCESS in 28m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/987cfcc06de7488683bf8ddd215207fe : SUCCESS in 28m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fc471b9791f4c408578ea0846f5757e : SUCCESS in 24m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd4947549e64628b11767db69585ea9 : SUCCESS in 9m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/127fabd1fbe14aeeacd07daec710959c : FAILURE in 48m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0831763454f741dc98e051f206b36d26 : SUCCESS in 1h 38m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5718d4cd98864acc939bc4cf4e703a1d : SUCCESS in 1h 03m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/472f03acb7c940c6a916044a971e9303 : SUCCESS in 1h 16m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07c103ec26ab7341cde9629d3f7a7c389edf06a0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-07 07:22:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab0e6ecd3bab90a73edf2838435d917ebc57ab93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 08:08:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caeb6a5b7a1d4af5a52a190386f43091\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/565659f5dd5f415b8201d0f11be46e65 : SUCCESS in 41m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/340696b0f8b943bda296aea8988c3470 : SUCCESS in 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b39fd6ce35f5ed99dbd408f2c08b49cdb7e43f4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 09:14:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d96c0a6deee94fa7bb7f22237ba33a40\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/5157f6ec82f14d20ae856da0ad9e5de1 : SUCCESS in 1h 41m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63d18a96ad4d4455ab202df60b3179b3 : SUCCESS in 34m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d60b43c9fe484d95b97f358704e7708c : SUCCESS in 12m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27abccf3360c44d48530e4234f268ceb : SUCCESS in 25m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ae722d2c6414b2ebe9d53d6a0a7dcb6 : SUCCESS in 20m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc3e71c40116406f9ffcea7bb189241d : SUCCESS in 21m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a95e31f0bb84cbba5608d7aaaec68c2 : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0fae2fb94a2f459197f7303d004de382 : FAILURE in 40m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb627ad9086b4b93923fc7d5f06b7437 : SUCCESS in 1h 45m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0c0cb7dd499e412eb5b2b96d22f284a3 : SUCCESS in 1h 08m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a26bbdf5e58244a4ad8e11d89e439182 : SUCCESS in 54m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62b268fade485151931d5608ee854a99909aa0c7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-08 09:55:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b63ff8dfca648ffd92c7302e795b76671689fb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 10:38:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6d3ca90b2c84a09ba5d70d4982716d4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/05cff40751d542b7be1648c5e9e253e6 : SUCCESS in 40m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb78723757a44c25b2af3aec74638cd4 : SUCCESS in 41m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c4de68d77d94601dc86135581e29e4d5464082f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-08 11:05:21.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3c7f965fa74ea890797fd1a578c0b32193aae4ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:57:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca3e1a19b47549ce805fa6d841fd4c60\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/91ee4fcd51664420ac2ac477c20875d4 : SUCCESS in 1h 24m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4d51fbd62a74bce83de8ae07bf0076a : SUCCESS in 40m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe4106ceb1b74a6eac0f17337391346d : SUCCESS in 14m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05670e2ad7b04e19975a6d9abce9bab4 : SUCCESS in 21m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a2ab9f94445422783acb3e9bc06e7fe : SUCCESS in 28m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1cc03c2c3b8f4d1c9b24098a957ce575 : SUCCESS in 24m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea1cea502a984fd38cf76c16bf0ee502 : SUCCESS in 9m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a5999963f8647c28d7d76d9e4a45776 : SUCCESS in 50m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6776d9cdbd89460fbe313ce43a438fd2 : SUCCESS in 1h 54m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04556127d70948fdb3b6e183a720f4ac : SUCCESS in 1h 04m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8972a683fe50448199e297de49f4f14e : SUCCESS in 1h 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"68e408127658623cc3b8d0994265bb65576d40f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:58:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3418ea257654b0bed9780f837772dc4c40551acc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 13:52:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0eda1dce8d54de59afef72aa7831d9b\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/cf887b8cf61146069b3df099c97cd8c6 : SUCCESS in 1h 22m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f33ce0af0f742e398eaeb5f7f068159 : SUCCESS in 13m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6522be909b264d74b1ab813939b78a37 : SUCCESS in 25m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb1602c9013f4ef2a7fd0f6762274f8a : SUCCESS in 24m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec64032c1694438949623a532e6a895 : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6c093448b421453b8e977984f27db15e : SUCCESS in 1h 00m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/60433602f7374a0db9f966077b409aab : SUCCESS in 1h 43m 04s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f507c9b2d7634a72af1aaa2c04d755af : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":1},{"id":"b42ab853c8372eac38e521a4959f409a3080c27c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 13:52:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1891e18af818198ed51a34c5f5e386d2b5b2f30c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 13:53:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ee4c9fdefbc4f8a8a0329e28a95515e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af359835df5345bb9f09310f4046203f : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f3409f6b6ccc624cea284bf6f76594afd1bee688","revisions":{"f3409f6b6ccc624cea284bf6f76594afd1bee688":{"kind":"REWORK","_number":1,"created":"2023-09-05 10:52:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/54/893554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/54/893554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/893554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/893554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/54/893554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/54/893554/1"}}},"commit":{"parents":[{"commit":"1ce5ef7f83776512487980e3d1b0fd1364ffb0f1","subject":"[OVN][Trunk] Set the subports correct host during live migration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ce5ef7f83776512487980e3d1b0fd1364ffb0f1"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-05 10:52:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-05 10:52:26.000000000","tz":0},"subject":"Revert \"[OVN][Trunk] Set the subports correct host during live migration\"","message":"Revert \"[OVN][Trunk] Set the subports correct host during live migration\"\n\nThis reverts commit 1ce5ef7f83776512487980e3d1b0fd1364ffb0f1.\n\nReason for revert: the port binding handling done in this patch is\nincorrect and leads to issues during the cold migration process with\ntrunk ports in ML2/OVN.\n\nChange-Id: I779b13f10f20f95c4963e0e1667566e4589beb76\nPartial-Bug: #2033887\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3409f6b6ccc624cea284bf6f76594afd1bee688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f3409f6b6ccc624cea284bf6f76594afd1bee688"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
