)]}'
{"id":"openstack%2Fneutron~893575","triplet_id":"openstack%2Fneutron~master~I91070ad6addb30ffdedba5d561984d2f6626e2b7","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-12-02 01:24:41.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2023-11-30 18:14:36.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-11-21 21:54:50.000000000","reason":"removed on reply"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-12-02 01:24:41.000000000","reason":"Change was submitted"},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2023-12-02 01:24:41.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-12-02 01:24:41.000000000","reason":"Change was submitted"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2023-12-02 01:24:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I91070ad6addb30ffdedba5d561984d2f6626e2b7","subject":"Add support for OVN MAC_Binding aging","status":"MERGED","created":"2023-09-01 22:34:23.000000000","updated":"2023-12-02 01:25:52.000000000","submitted":"2023-12-02 01:24:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":78,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"893575","meta_rev_id":"729a1e424e0430ef39530964fc15e5e79bbe8163","_number":893575,"virtual_id_number":893575,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-11-21 18:59:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-02 01:24:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-01 23:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2023-11-30 18:14:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":1,"date":"2023-11-29 06:09:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-11-21 21:54:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-01 23:14:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-01 22:47:38.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-09-01 22:47:38.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2023-09-01 22:59:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-02 01:37:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-05 09:05:11.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-09-05 14:47:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-09-07 01:12:04.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-16 09:58:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-10-18 13:53:15.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2023-10-19 02:47:09.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-11-29 06:09:51.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-12-01 23:14:10.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"301de8bc66bd407c6e44a727cf4c34c6e4a1c9a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-01 22:34:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b36549d968156ef1ad9bf86fa31ab15f8f42fda","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-01 22:47:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d08633fa60b200de95d32cd4001687945671af74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 22:59:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d72de8b4ad3455b92fccd819cadf014\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f5eea20765d4bc294244f308e8ed106 : SUCCESS in 18m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e728850df7a498689ba00e51332c561 : SUCCESS in 18m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/383edbbd925248628dde8ccec56cca25 : SUCCESS in 17m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d8050af71a8f43fd91e6b9b74bf87e17 : SUCCESS in 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b72a770ef8367617ab0915a6c48f0bda8b79ec50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 01:37:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/46b0e9928f9d41c9ae19d1510320f3d3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ba1980b973764bbba141d5c5475e73f0 : SUCCESS in 1h 38m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b3748a5d1b764eab9fca517779552b3a : SUCCESS in 1h 38m 02s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b4f899a87d6f4d83881234bddb74f8de : SUCCESS in 1h 16m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/440dbd030ef74796bcba8c90e1a7804f : SUCCESS in 1h 49m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/086a1e25c20547c2ab4351133bca3f85 : SUCCESS in 37m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/850637303c8b4eee87e46785c6ca53d0 : SUCCESS in 29m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2c4880bd2d946099f739654f39556c5 : SUCCESS in 10m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b1dd4d263be4b7a9ae89060cef1eab1 : SUCCESS in 20m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa1499192d694504ba6686bb38cef292 : SUCCESS in 27m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e84cc355b174f7bbbbb92aaaebef697 : SUCCESS in 24m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/848b3fe403aa4aefa7be666f0fea4039 : SUCCESS in 23m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac4f25be85f4490cbc6d43c5888d931b : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12d04ded1905477f8039e9dc8b1bf06c : SUCCESS in 7m 15s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7080966c74fa42a8bfecd2347efb5394 : FAILURE in 31m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6fe4b50431bb4b7aab6808e94f13e2dd : SUCCESS in 1h 03m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ba35a649b804e43b0d1ff1242ce9b1e : TIMED_OUT in 3h 02m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9ac21fe3c3f4e77a3bff66208804b45 : SUCCESS in 46m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aa41028fa7a04e83aafbb4cabe9a2c23 : SUCCESS in 41m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/110f700a2a1a40e89ca486e7c97c1e97 : SUCCESS in 1h 06m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/13ce90d0c34f48ecad3cc112e339e2f4 : SUCCESS in 1h 00m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/66027dc00a7c45d5a792677c06337979 : SUCCESS in 1h 04m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2f4e393c383643098ac199fdd0166744 : SUCCESS in 2h 05m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/284dc8c165844267958989558527909d : SUCCESS in 1h 31m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c03afcb1492c45bf9638eea4086e1a24 : SUCCESS in 1h 30m 00s","accounts_in_message":[],"_revision_number":1},{"id":"35c4293f49aaf32834339d37d77ee30fe372f2c5","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-04 06:44:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ceba525e7a2627a51f548eb0a68bf1b330f32cc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-05 14:47:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b5d64399f428a469c2cec0ede2d3407337421327","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-06 14:57:23.000000000","message":"Patch Set 1:\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"63b11147e3aa7ed635ab0b9c2774a1c95d1d4a82","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-06 16:24:37.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fb431e6adaf052e4e34a22c599c40ce42f8572b8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-07 01:12:04.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/75/893575/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"5137c311224dac8be2ebcef85f82ef75e612c054","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-08 06:44:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d35943d35c0128e8017f1bc0dab18c95f7f344b3","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-08 18:27:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"22bd987f6af10d4fff9ba34e238c338032808a4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-09 00:10:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f7d300d0119611629a76a558365cb956bfe093e4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-09 00:42:28.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/75/893575/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"b3772fb613ac3511e72dd7a896b87619e75f9413","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-09 00:54:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ebde3772fb94c2db622f90d34a3365d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/868a688e43c24de4b16b8e654b865047 : FAILURE in 21m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d88bb4834f4448d0a7d08c61038cf337 : FAILURE in 22m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9961516171d6463da82e1570124afbb1 : FAILURE in 40m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c80fc423e33c4e258bc5ee43a349706d : FAILURE in 30m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e04aae4cf53588034534e9d4f69f59540568d864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-09 02:04:42.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bddf38f15d0a4d2d9e07bd54b8a976aa\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e7ae04fec55f429a9d400545fc61a3c0 : SUCCESS in 1h 23m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/556d208341214507b0df594fcfbaf83e : SUCCESS in 1h 36m 04s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/da55e5c9654f408a96e5961f91ef8d57 : SUCCESS in 1h 17m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/25f24735b5114de78c819f10f696239e : FAILURE in 19m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/78dbc323f1a8440f912d714cfd8f9955 : FAILURE in 20m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/267cf2c442744f51a287bb332268d78e : FAILURE in 26m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9d044d6e0b4953a0f5242c3f47d922 : FAILURE in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f773f38836ab45dd889f0a8178913b1f : FAILURE in 19m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f29f664cb4e14fddb2d27f9c32ab97b0 : FAILURE in 27m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60398db8cfed460f97b0ae4bc78dd8ad : FAILURE in 17m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbe16270c5714b0b9882123b8017c7e5 : FAILURE in 24m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba84e697f05473e999c73c67edb923a : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/652bdeda4b614a6f9d14d231491f80ca : SUCCESS in 10m 26s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/74f87592e8b14cfdb55cdda0aa3efacc : FAILURE in 31m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9064228fa14b4d31926e84fc70f49a1f : FAILURE in 38m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2ac3879fe8f145e0a720a6fcabce48fa : SUCCESS in 1h 46m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3bfceac56c0e417998552ac3de1d7a37 : SUCCESS in 56m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f0f15f8ae8e94a5b9bbb3e759feeef56 : SUCCESS in 1h 01m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/141e90b28bdc4a31bdaf757e1c7a01e2 : SUCCESS in 1h 25m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/65abc832c24949f682e876a529b1ed7e : SUCCESS in 1h 27m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d0d935cba5e74547b4941b631bd7d394 : FAILURE in 26m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9dd04902d90f4000a035ba20cc6ca4b0 : SUCCESS in 1h 21m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d7509b6ddc91412b9133e91e244aa5d2 : FAILURE in 18m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/724b1fe94e7a40eca148c33fa2483a8d : FAILURE in 28m 46s","accounts_in_message":[],"_revision_number":2},{"id":"784fc30600a4acffb344c4463d9f66aaa4a24344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-11 20:24:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b404d55466f3213e62d1519a86b107a58da687c1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-11 20:54:41.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/75/893575/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"41def185c9ab7470dee807395d3c79ebc3a0da44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 21:35:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15e4453f13ae43938bf6afc5877d797d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f02975a9c204d0880ad17115c69f53c : FAILURE in 40m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02d1621503934094a7b551ceb6304de4 : FAILURE in 50m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/87a088028e2446508f23867e6fc8a98d : FAILURE in 52m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7636a67deccf4e428ed12ddbd9732f18 : FAILURE in 48m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ab77bf99699921ad16343bf38a4c9c776918de8","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-11 22:16:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c416631d99cf5992121671e3da5f56ed16991ae5","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-11 22:19:00.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7747e7fa5836a12d8065ca9e316abf7e0120ab45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 22:48:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63cf2e6d4aee452fa557203cb0f4dbf7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/585c5fe14e924ff8a70c146e4e64a852 : FAILURE in 1h 35m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/601d1f425a21491099b4aeb75656923e : SUCCESS in 1h 44m 35s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7ab4dea758b0446eb04512e0182502a5 : FAILURE in 1h 24m 26s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e68a2196a5df4a78ae9a7f91984b0fd3 : SUCCESS in 1h 24m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0af568c225884f519c2edf80d4ea9fca : SUCCESS in 31m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf615b178acf43eea0d6249f5a7ac1ba : FAILURE in 30m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0649a7ec9652488c86bd4419d5977f9d : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbc5687d272b422a9c40a4730fc04578 : FAILURE in 23m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75fc610883024b80abd1090b13e329da : FAILURE in 25m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f55b74b689e4429965543b0d79c7322 : FAILURE in 25m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d98e322342ae4b6d9759976be6f767bb : FAILURE in 24m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2232dce72082439985d80db66b569bb2 : SUCCESS in 11m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d507e19ffeb4dbf88b8946e091006cb : SUCCESS in 8m 01s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/33fd0394a3294a5480e42edbaa19bde7 : FAILURE in 32m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c9695b9a7324cd6a68012f2997bc8cd : SUCCESS in 1h 05m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9fa5428f6a834f5f881aea895bff3397 : SUCCESS in 2h 23m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6d3fb5f59674ac2befa2d64565f626b : SUCCESS in 56m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/45aa5a332b4a40dd9f28583e7da6fe66 : SUCCESS in 1h 05m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f0c249747ff84ea2be27bb48e2ca53f4 : SUCCESS in 1h 29m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8e131fd4095141669265af796334898d : SUCCESS in 1h 20m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1224484298f94361ad2bd65e31bf359e : SUCCESS in 45m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/50d1ec7bf26b4658b4fd79d825602caf : SUCCESS in 1h 15m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d5a11f29896d4e959d0b456ca400836f : SUCCESS in 1h 52m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/54d8e4c960f3470090d1191c62ac2748 : SUCCESS in 50m 31s","accounts_in_message":[],"_revision_number":3},{"id":"255f42b616baf41e9299e3e1070f028a00db7123","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-12 14:10:21.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9440c33d019ad924a6d43129ba61f969aaf4e0b7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-12 14:56:24.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/75/893575/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"5ed2eeae37051ab7f350463bee8b68c05c7b69c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 16:35:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c6ba2cf29c94062a737ff1f59809af7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bebcc71a6c7414dbb3254fc7c49b563 : SUCCESS in 45m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a821ad983b6040138ec4b2a879decc69 : SUCCESS in 40m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c0ac3d1ba5fe41099c3283c4e93ae076 : SUCCESS in 48m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/31e88566b74440938bae974f3e60ec03 : SUCCESS in 37m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"58ca0e797429ef176e70e0c4965eaca4c068439c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 17:00:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ae8a1263513401697aeb2475b5dd728\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6bd71b443e654d1ab68ce613342fe8ff : SUCCESS in 1h 41m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8b9c82bb80644af9c311b6fdfe126fe : SUCCESS in 1h 37m 41s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d675b8355cdf42be9b63264f6f7f95c3 : SUCCESS in 1h 12m 23s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c3547e87ba6f4ed5a0c6cd49e17f893e : SUCCESS in 1h 50m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90d99401e8af4c3dbe56b5c60ff1273a : SUCCESS in 39m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c413d94cc2c4d29a3a215bce91bedef : SUCCESS in 35m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4430244741214c179999f62881671b9d : SUCCESS in 11m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fc1967cb3b04a33b45acfbc8d5b5dda : SUCCESS in 25m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ff3bb691f7a4fa3939562997ac37942 : SUCCESS in 22m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6a0979d766740328985ef25bbf00f0f : SUCCESS in 21m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2759666458c44e19a693566d27e534e2 : SUCCESS in 25m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c0d1a3694ad4a23a6d0a787fb3649b0 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00c2b90a89e84a42b3e56cfe8d10a718 : SUCCESS in 9m 02s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/838904c9ab53458f97031ea5bb144296 : SUCCESS in 33m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4e41d59b56742caa06c5f0c07de0458 : SUCCESS in 1h 02m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/76ad2734a4934f3ab0808ba279561f11 : SUCCESS in 2h 42m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fffa903735b64ceb8a8da5db929549dd : SUCCESS in 59m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/97ab3defc5614f7f90c82185d8237124 : SUCCESS in 1h 01m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3d9128e9371546a6924e59091dd4ad1f : SUCCESS in 1h 30m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5a59847b3955418394e3f581bda7d92c : SUCCESS in 1h 11m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/162fcaa2938f45688be5d30bc43112d9 : SUCCESS in 1h 10m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e3ab2f5f549f44179f0ec78485a6ba71 : SUCCESS in 1h 47m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1805342154874cbf930ad622c97060ed : SUCCESS in 1h 26m 38s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/77ba1a2526b842d1b3f1444082fa856d : SUCCESS in 1h 22m 35s","accounts_in_message":[],"_revision_number":4},{"id":"c0f5cd11269d727f19b8354abb21a254573c4f02","author":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"date":"2023-09-13 07:41:28.000000000","message":"Patch Set 4: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"542563057efbd661c580e127715e7dae5b38b8ed","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-13 14:46:30.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b783bfbf9e1daa370166181f987b6dcfb618baca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-13 14:46:40.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ccfb15f15322864776df16a119b221ff76c79f71","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-13 15:21:57.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/75/893575/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"210226c3399d1ce02bb347313c5a8697836ed010","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 16:09:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/527216ae57de4f28b1d0ad09f71b9466\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01763954e007460c885d8631f05e1f99 : SUCCESS in 42m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1cb4ae062a854f2e806d021f11ec0697 : SUCCESS in 43m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9da626cb2913492f9fccdb8ee6999057 : SUCCESS in 51m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad0f175132584b8db02631940efa2776 : SUCCESS in 36m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"588c677d135ba54003551761eddf85da7105fcb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 17:22:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45c59dd9989843dda123b11e47b3818a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/89c8bef065ea4cedb855157a40b3b6cb : SUCCESS in 1h 38m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0268a03835834aef863ab5d5ead7688f : SUCCESS in 2h 14m 43s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f7e63d59e48d4f31914a3b9d7e0c55f7 : SUCCESS in 1h 37m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3dd8308783194772b6a3f0945e31017d : SUCCESS in 1h 53m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/44e49e2679f54c33bc6d089e36b880c8 : SUCCESS in 43m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/701b084ebb2644d99d8b9a01e72caa77 : SUCCESS in 40m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216b9c14dd944590887f78b50b1dffef : SUCCESS in 11m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/918e214f99fd40a7af0cba030c80b907 : SUCCESS in 24m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/267477b1d299446ebbbbfd0a9169113b : SUCCESS in 26m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66e57e7fd5bf4567840609299edd72a6 : SUCCESS in 23m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6dff7d93df1479eaf82b8409bf321b7 : SUCCESS in 22m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f72203a313143c5b2fe5fd60821bb27 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdeec2445b9c46698506730aa740440c : SUCCESS in 11m 57s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6f20baa616494f07a96bd9711e6f4bd7 : SUCCESS in 28m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bc96da75128c4dbe90ccd76c96041d48 : SUCCESS in 1h 04m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf670e9ffa4c4e0d95e6db93ae724f35 : SUCCESS in 2h 27m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c3a08b4b979440493b4f5c5f92208e3 : SUCCESS in 1h 03m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/79588e6df7774e94bf1d07f46c822abf : SUCCESS in 1h 09m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/33728fc8e67442af8434b2a357873277 : SUCCESS in 1h 27m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2e0071af6da647b79890bff836237275 : SUCCESS in 1h 23m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d8926094d0e7484387497a1ac114f236 : SUCCESS in 58m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/17234798de0a43909d307f5401d7a5c2 : SUCCESS in 1h 37m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ad619ff22c7d4852a49eb43875277c53 : SUCCESS in 1h 47m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b95a87b77f1f48acb102861d36e1f1c0 : SUCCESS in 1h 15m 38s","accounts_in_message":[],"_revision_number":5},{"id":"76b75933fca5c65e42648e72bc9a5b5acb2d01a3","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-13 21:30:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ded5be296f85448b5821931abd86a547eb0f3e3e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-27 16:30:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4061b1b07c4f86d45c677dadeb785ff4bf5cc219","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-29 18:27:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3a3c88dea924fe969c52907afa8ae060ccf528a1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-16 09:58:45.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"761e18cc01e3e090b87f7309d5e3ef8fcc2a4790","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-17 20:05:15.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"97b0ac2e6e21ee7deaa86123960b9748cc4cb00c","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-17 20:10:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a2e25876c3a492491dfc2fc165d523bb4e30f110","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-17 20:32:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6e433d7fb757b900aa257758cb258b8a78ecb9ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-17 23:02:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"6b367c27af2239da2877ac30057f4dcf5e26e4bb","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-17 23:04:35.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f45558e162879cc384726dc150661bc25d70bad2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-17 23:36:20.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/75/893575/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"474b21d143e23cb60a2cfead2fcaa8e0c6a0612b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 00:36:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aaaf229f4a7a4220b51b547a6031f55f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c9907c694d84013ab1be45aafe5c9fa : SUCCESS in 43m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/923c458fb288452a89457a0adc4a4034 : SUCCESS in 41m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5bf3924cc3b24280a671bbc75d710ae0 : SUCCESS in 56m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d48e35ff0444565951ceba7a37ed7aa : SUCCESS in 49m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1cc987f5f8c00b3be6ce4b7278cbc647d43f75a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 02:06:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/afdf1fddd5944fd48b7a07e0d4a0edaa\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7287ca3667b94307b53ef49e38547bb2 : SUCCESS in 1h 23m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ed3dbedbb5aa4fce8e62423f2d1a3d40 : SUCCESS in 1h 38m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/97335c9973a841809a4dbc8b5515e7b0 : SUCCESS in 1h 43m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/aca5709091594e0a83198dac2b42edb9 : SUCCESS in 1h 37m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e94cb3eacc7545e48b65f4e6aaa1fb99 : SUCCESS in 34m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4094c49207ce4bff88e433ee5ae11549 : SUCCESS in 40m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7886f1dc2d415297bae2d5496ce92d : SUCCESS in 9m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddb16604f90b44a286f3ce970d8b1acd : SUCCESS in 28m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/100ae164f2e54394bf49401f70d3fd99 : SUCCESS in 25m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca1fba5c5eb14bc198ec75ece9dc0b02 : SUCCESS in 30m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a149317a9f34bb7a3793db676fa91ad : SUCCESS in 22m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d48bce31cce84592b386857af7aa0096 : SUCCESS in 12m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b73c76311d484468ac67ae1a025c0b00 : SUCCESS in 13m 21s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1d58243281034f6bb8c5235d4495737d : SUCCESS in 24m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/047df205c382458ba90a361dcbe7f382 : SUCCESS in 45m 08s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b1de9764b8104d4a961f91aa98ae0552 : SUCCESS in 57m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/88ac895f53d54859aab24dd132b3eaba : SUCCESS in 48m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b1ab0f328944c06917a033f73b10a65 : TIMED_OUT in 3h 02m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e834ff926ffc4f8db42059d1db468f37 : SUCCESS in 50m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c6c4126b54154d6a8f397b2ae6823e1c : SUCCESS in 1h 04m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/659728d0ecbf495f9c27ddeda78fd72c : SUCCESS in 1h 24m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f9af8e812ed04aaf8d06905f5407e010 : SUCCESS in 1h 26m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d284e62435724a6da4a35d33ce6b96af : SUCCESS in 1h 02m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/de075261b67643389eaa4d0380dc1699 : SUCCESS in 1h 12m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cc74d39888184378bf78dcb723bbdb94 : SUCCESS in 1h 58m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/92944c4f29064ee5851cb1a34dbee4af : SUCCESS in 1h 09m 54s","accounts_in_message":[],"_revision_number":6},{"id":"b4535de061e0c4666ef3822b5d3345b7716f265b","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-18 02:08:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4ca0bf1a4df55a81e877648c3a7019c0f2babf62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 04:06:40.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8557c09508cd49bbb70ce01bc6281e10\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab0093fcfef641979605b0cd21b30881 : SUCCESS in 1h 11m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e532ff8cffb94b229e44228226d58de2 : SUCCESS in 1h 11m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7fa9f6f9578948d488d30cf7f3651053 : SUCCESS in 47m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/59910a89fd95472d97c964ec8e9559e6 : SUCCESS in 1h 04m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cbcae95c9ee6cc8b884536904f817ff00c007754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 04:52:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9843c01ce0aa4e9a9de225a45405c09f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/025e97386b264c548fadc480895996a7 : SUCCESS in 1h 42m 55s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9193abacc6214181929968f729d911cd : SUCCESS in 1h 32m 05s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2c52f7c3d66e4d10a03cb179c8ff78eb : SUCCESS in 1h 24m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b2f66750e677412db32d5f5653e43482 : SUCCESS in 1h 33m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f6109f012e814de39b93900290fd32ed : SUCCESS in 32m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68933a90c8794d05b01874ea04869420 : SUCCESS in 45m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb13488bc8134849b6ad348cddfa712c : SUCCESS in 11m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6334eb8bcdbb46c7b5272c3912c5e0a7 : SUCCESS in 26m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1e5a6da207d404f8876fb55189f85a7 : SUCCESS in 26m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/062802ad97b4420d925645b3140fa629 : SUCCESS in 27m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6f0630562ae494392497172c6d988bf : SUCCESS in 22m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e7c639f1fb944b7a4f9aeff0cde468b : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef03c5bf447b4c6a9141a29538383390 : SUCCESS in 16m 16s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/79ba1922ad2c47e0aef449c1cc31bc8c : SUCCESS in 26m 48s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/579d1e54e5354b41b67d72aedb9e2185 : SUCCESS in 1h 05m 31s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fe81ae131bfa4377b6d4fe285d69746c : SUCCESS in 1h 10m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/71e878a7331d4dc8aaa0db50939fabbb : SUCCESS in 59m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa1dcf4673bf44b780cb3dea42e0a77c : SUCCESS in 1h 56m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0998c4be14134b81afc71bc9ef94b511 : SUCCESS in 51m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5f15f26e695b4034842178c3c4e1d44c : SUCCESS in 1h 00m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/29984ff228fd454d9d716296e5520e27 : SUCCESS in 1h 21m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/581964afc01b48cba01082d056148c7b : SUCCESS in 1h 12m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a3a7ff5a9f744c34930358a97b09d2f2 : SUCCESS in 54m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1d3eb36914b148cc8e17962252c197f9 : SUCCESS in 1h 56m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1560262946eb4da6952abad63f167820 : SUCCESS in 2h 10m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4439f7d3519e4436b2faa8fe3c5e5319 : SUCCESS in 1h 11m 12s","accounts_in_message":[],"_revision_number":6},{"id":"223513411bdd48a30ad33f29ac6dc6f95afb9ba1","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-18 14:08:34.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"aca04be95662b45247d85d4cf61439c06241eac9","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-18 14:17:35.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a811d3c615a5ed7321acfc655ad97b7178c1b661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-18 14:18:26.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"07a9fe9ea2ef52df485555564ca6e681474fc12b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-18 14:45:56.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"ca0da6556189d9d669fec54ede515e8c5acc74b9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-18 14:50:20.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/75/893575/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"4c4164617fc404b99ab0b645333f45a69c7dc2be","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-18 14:51:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fb3c15b08425c55ff19cd9c558c6658ad0b6aae1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-18 14:51:46.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"706b3f24f1cfbf5832972405c94640305a733403","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-18 14:53:59.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"1f51583009b74b414bcd1d4d9ac7b9426f11ddd9","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-18 15:38:25.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"17f167bd4795aa3b2ff85dbada4dd132ae71e2fd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-18 15:55:06.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/75/893575/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"f36692c50fa7223a66113daf70a5fde73b1635d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 16:46:08.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de5bd4bf7475466d847e2cb61aa9c223\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bdd6c0bd87f4a29bd60638931c00fe8 : SUCCESS in 47m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d412a6f9aa04d79916630a1c87a324d : SUCCESS in 1h 19m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c51115338b97471cb21147dc550b1013 : SUCCESS in 45m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/34b84819884a412781a0f414818dc795 : SUCCESS in 1h 09m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ce7a0df007ea1f5ea841783e0d2f04a226296896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 17:55:29.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86f7a0b868fc41939a27bb2d8cc5ea44\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4d4008c078d545199091605d124673b8 : SUCCESS in 1h 39m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/82a9b470a6ba4f82a6dd291256176c2f : SUCCESS in 2h 04m 48s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7344a7d5bef143d7941083ddd949ac4a : SUCCESS in 1h 22m 35s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5a567a266e2f420793c706352aef008a : SUCCESS in 2h 10m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d4cb7c26c3b74372819a9af9865e677a : SUCCESS in 49m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e058677fffa4bd68dcfafdbbde26492 : SUCCESS in 39m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d749c812db0494caf124bc8b19db421 : SUCCESS in 11m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75d35917f40647658efa1e67130bb966 : SUCCESS in 26m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20b2a83f2d944461a9d8cf772e32801a : SUCCESS in 27m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c56d3d52a99147bdbcfe3b595347efd2 : SUCCESS in 22m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5a2142f765443d99aaf6b1696eccaa5 : SUCCESS in 24m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f15d51e11e614c56abd936da3c0fbabc : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0aa6308630a546979ff9fbff5469065d : SUCCESS in 14m 52s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0a03d2deade0438881ced1209c15ca54 : SUCCESS in 31m 27s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/10c6f4fe1e4847ad9681f2ad0a359ef6 : SUCCESS in 1h 04m 55s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c878947872a04bd092d70783468563a8 : SUCCESS in 1h 06m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ef3b87038d54ee6aa7b182bf2a1f86d : SUCCESS in 1h 05m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/690cc7d2cddd4953912fdff75026bcfb : TIMED_OUT in 3h 02m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/83ac6d9cbc0a4bc08470d9cd45f00379 : SUCCESS in 58m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1977d0b227a348cfae45829f16c27377 : SUCCESS in 1h 07m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8464be0ebeba4fb6a08dc8d9f1cbc01c : SUCCESS in 1h 34m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6dbe6e2914ad4747b92f27c96ce51697 : SUCCESS in 1h 15m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/415c4975d32f4c79b994bd837c9e7f4f : SUCCESS in 1h 02m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/29fc8e6332e844e3b8ea2c561652ab48 : SUCCESS in 1h 43m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/00358801284d4907b71bbd93b15f9b7c : SUCCESS in 1h 47m 56s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/46711af64e024f4ab8e961078ecbf9f1 : SUCCESS in 1h 21m 14s","accounts_in_message":[],"_revision_number":8},{"id":"2a5e938e1f66112313e9c01e0a58d24d59d18fbf","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-18 20:21:31.000000000","message":"Patch Set 8:\n\n(8 comments)","accounts_in_message":[],"_revision_number":8},{"id":"03c00daa10217589391857725f9b8b388d42595a","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-10-18 20:46:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f198c941a26ab780d3153e7fa9d255097b08b650","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-19 01:41:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"90254e1df2890f595e24a728bc860f7afab2e92d","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-10-19 02:47:09.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"29b9c665531f4e4f6e6cbc0f6099c7b8da9a0f22","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-19 07:28:47.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3cd2caa0977d7188241aec51e2030486dd4ab342","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 08:53:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b81f53b61a34cfbab93d2c029f19603\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b465f7013c4340e28d10cca8c13f021b : TIMED_OUT in 1h 21m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3f893dfa9354c62b495b952aca885db : SUCCESS in 1h 22m 04s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/19f40bdaaddb4331a32efb36a68f705d : TIMED_OUT in 1h 21m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/97f565b4e7b9463487f889a5a7e8e651 : SUCCESS in 1h 02m 13s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d933bd09c8458c3c988f1a237153e0f3b9939eb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 10:32:17.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f66c8284f67446b9af7091fdd32b10d8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8f00a9b55e8b4b04b49ac38084355ead : SUCCESS in 1h 31m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c6792907f1df4ffb9331ec1904da5403 : SUCCESS in 2h 34m 58s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/988e6bff24fd409b9d76b77e36641536 : SUCCESS in 1h 43m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b256919d0c6d4cf7826fe8387413a7b2 : SUCCESS in 2h 24m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/24a44d576cb147169be6f407dd0f3705 : SUCCESS in 46m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a2b5768423f482e923971d06974b095 : SUCCESS in 40m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a62477a74f4198b5737640d8248a71 : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b677959f0ba4f95b8557a3e8ec7c0a5 : SUCCESS in 30m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fe470877fbf41fdb8112c508b8afa5e : SUCCESS in 26m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f9bf1f57a114d0683b3f1acb39175cb : SUCCESS in 31m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a33ed322a384332b2b134581acc191c : SUCCESS in 24m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6eb057598fd4bb9af75f09e973ed8de : SUCCESS in 11m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85cc3e7b0c0941fcafa1bd018e28f957 : SUCCESS in 17m 52s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8e9cf44085a6488784ed0cf7a3910783 : SUCCESS in 29m 27s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/02db63dd9b0a4f179c742dc7939a3d46 : SUCCESS in 1h 04m 55s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c994467b3c0a49f2a0eb1ba3bf9a7074 : SUCCESS in 1h 17m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/641ab4acb6cd4b30a7cc8ff260ea8bea : SUCCESS in 1h 10m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/433474502b12461bb8745aff6a60375c : TIMED_OUT in 3h 02m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e6b467be41f045a1ae5360e8d80ce31e : SUCCESS in 1h 03m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cc67e3150c564d0f86dabf5164b0134d : SUCCESS in 58m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/499efff9491a42e48353ae89fc60023c : SUCCESS in 1h 51m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/21595e486c8c428c8ccf506d65802fe1 : SUCCESS in 1h 18m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/24e2aef6941f43ab96388f2461c95592 : SUCCESS in 1h 08m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/45d2ca50d20a4c50893524237e57833d : SUCCESS in 1h 47m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bffaaab020414e909c5518486fb4a5ad : SUCCESS in 1h 51m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/90da48c5cc13411aa41b4e474cd8fd13 : SUCCESS in 1h 18m 35s","accounts_in_message":[],"_revision_number":8},{"id":"1a7fbc09abbe28e1d6227bc50e59e7c80788ac42","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-11-13 17:24:25.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9dc9a9c2339120b11e8c84114a48ddd565186a58","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-11-17 15:15:40.000000000","message":"Patch Set 8:\n\n(8 comments)","accounts_in_message":[],"_revision_number":8},{"id":"aa48f25afd65e47a978ad98e41d59c7dc697787f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-11-17 15:15:47.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":9},{"id":"9f69d80a236e73f631721d7fc76b83e7f8e6cb62","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-17 15:48:24.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/75/893575/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"9884ee157704b636836ba57b733cfc92ac670bdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 17:56:57.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2de6736343d4b659ea395281123c6f1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8e4095c3fec47ca809d2cfc0219a20c : SUCCESS in 50m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e6f3e1b14af417badcc68511abba2d7 : SUCCESS in 23m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/593ffe9dbf9a43dfb96d3dd2e270b3fa : SUCCESS in 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e3a7a89753570ae08a3ca95c23c459e7e705aeaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 19:48:15.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7162b15e9004fa6befc818bc035c23e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8885781ccec242a68d660690eac45a35 : SUCCESS in 1h 38m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/820f2b83c7014e71aa713786312919bb : SUCCESS in 1h 37m 57s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/c043534a1722437dbb8f89545e9e1914 : SUCCESS in 1h 38m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/28392234fdb94fd68a71e7c3ecac71d2 : SUCCESS in 1h 30m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/29b31b0662f240bb9750dce59de1ce36 : SUCCESS in 34m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9dffec64f007410a816355677364f810 : SUCCESS in 40m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6780ae30a5e54267ab5fcb3698ee2d5d : FAILURE in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a422ce8afc7648acb2bef529e82d90eb : SUCCESS in 30m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/75709e53740e4b35a21335176b7eb45b : SUCCESS in 26m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7f0e4350a6d40fcb10a09a7c031e550 : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56cb949bfa3f489083e00f1223ee846e : SUCCESS in 17m 55s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/213d11b95fee4ba78a500631572182b1 : FAILURE in 30m 58s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/9b3f88a7a29a4443a203b811b52ea6ab : SUCCESS in 1h 05m 22s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/021f74a274e341fb87964a017bf5b59b : SUCCESS in 1h 15m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/89501dafa59347a886af5d752239921d : SUCCESS in 54m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd7b97bc1fe34389b2c0e64535b2f1fd : SUCCESS in 2h 15m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8cd0c5e13fdb46d89fa85f060da00d50 : SUCCESS in 1h 05m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b46a628a688a4522abcf29b2305e441a : SUCCESS in 1h 16m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/accc90232f644385b107e2d6ef3b5349 : SUCCESS in 1h 29m 27s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d73bd2481cb6488296765eded01bfab1 : SUCCESS in 1h 25m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3f4b23e76f434525835570e0ad59a7f8 : SUCCESS in 1h 24m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4459b9b75621466984cdbf7b59c0fab3 : SUCCESS in 2h 17m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/19dbb62e91034c6687c8be358a898b94 : TIMED_OUT in 2h 33m 16s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/01faeac65a4d439a86016f4ddae85267 : SUCCESS in 1h 29m 53s","accounts_in_message":[],"_revision_number":9},{"id":"daed4702658c8a02e74804054b1b08c8e1023ea4","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-11-21 17:49:16.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b7c12e055599c76bc6ba3ea1305cbae0c4f5f7f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-11-21 17:55:08.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"b20517f5b31681f16b2920f707cb3b0b11d3b3df","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-11-21 17:55:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"805fb5d835160a30626be47e2e042bef0202781c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-11-21 18:06:26.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"c68c13467b31c102a3234070ce2a6b23c8322c8c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-21 18:27:36.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/75/893575/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"c437ee1c2c3d1c45d3c3215002d5036db12d6102","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 18:37:48.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ab1ac4d0ebe42418ffed70c716d8965\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d81c4f8c55ff4e79ba547f2ef6982913 : SUCCESS in 28m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af05bd82a16a4f37b6a9b249ea571e33 : SUCCESS in 17m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/431dd6ec10a44fc697578640291edad6 : SUCCESS in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c5dbc9a63752fc217c635d2b0f4aaa4a4c86b1f6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-21 18:59:54.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/75/893575/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"2b0840dc34c691540ca5fdc5f28f0c2a8f78f22e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 20:09:22.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5cd68bbc1f549ceb604cf07bc471eec\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1b433d4e94b7429dab50106dfbcad208 : SUCCESS in 1h 38m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/00c3f43895024718b314ba72b2b63387 : SUCCESS in 1h 28m 54s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/edb35bed417e47d5a22f6b2b63a2b36f : SUCCESS in 1h 49m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/673c1131f20942c58d800c72afb7dab6 : SUCCESS in 1h 55m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/192e8a0137814fb597ddb99756f9cfa6 : SUCCESS in 41m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd3913e54fe34c6399e274e0b9a4e152 : SUCCESS in 49m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b867d786c64c748d32f4af6deac8ff : SUCCESS in 8m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b53f3d38fa354c34926f07e890e9e22f : SUCCESS in 37m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2733beeda1e64699b3cb27173e1f1746 : SUCCESS in 17m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89a5f7ccc4e64df0b165f9f987ea6284 : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fb7cc7cd95f4dfb897c8ea6303d3d00 : SUCCESS in 18m 07s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f3f7d13c5af54c338f88900ff8996185 : SUCCESS in 27m 21s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/71fc50e66ab94e49b39c95677ec6137a : SUCCESS in 1h 04m 08s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6fcb29c7ffc24538bf34b88a9412ea50 : SUCCESS in 1h 07m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a9a229de1a646bc86ab2a452c6caf8f : SUCCESS in 1h 04m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8755885053484bf59ed42d04e43688e6 : SUCCESS in 1h 56m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a7a09d9985ce45b0af029e6f0d32024c : SUCCESS in 1h 00m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2f8a8edfccfa47e197f155f830479fe5 : SUCCESS in 1h 05m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/47175d05f51f4e2c9cacce896d29e7f8 : SUCCESS in 1h 09m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ffca892ed8964ce7bc69a1b17458042b : SUCCESS in 1h 23m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/22242b9ac22a4f839c33701e34846495 : SUCCESS in 56m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/131e2417a219459bbd27033b59b6e6b3 : SUCCESS in 1h 31m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c02fe55f0046417295f0c94b60637579 : SUCCESS in 1h 37m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6004b7e0344545e59a7af2dc78de4436 : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":11},{"id":"9a153d61216c7af31d6330f2f6d250d516ef41fb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-21 21:54:50.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"6262b6e632c578b3a534fdbaa92e6b752d2cfed0","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-11-29 06:09:51.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"7eb2b033987b67322820c6d25ce21383721ea68b","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-11-30 18:14:36.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"4f943d6cda46db8909b6cb052b4fefd26b2b52b8","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-12-01 23:14:10.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"9f757c374a5827bd7b59c9bc5045f98783da45b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 23:14:55.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"8b2a09ddda032d7d76942301de6f0f5a7b1f9d55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 01:24:41.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/735fc3ad71fe46d4a86dfcdaa1da0048\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/66752ec81eb640eda5386850d5ffede1 : SUCCESS in 1h 39m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ba58026a723a470a91d2d13a6251d737 : SUCCESS in 1h 43m 17s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b5c98c278c3647b5b465541badd537c7 : SUCCESS in 1h 33m 28s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/e97d5c1407d2478884a83772f6a29ad1 : SUCCESS in 1h 40m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5cf23a97e244e799246189c2a327c3d : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f824f6454b5f40ad9180a6c795f1172e : SUCCESS in 31m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/645728e20d314fbd97c39ff6bfbd837c : SUCCESS in 23m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f4f077906948478bd25eb1104000d9 : SUCCESS in 11m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a30a83d5cfe141bea8aa7f792ab9210a : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ef1a8f59a1c4e47bedbca2a4a3e4eb8 : SUCCESS in 1h 03m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/30fdabc8d327496f9d8d2ae93e28a43f : SUCCESS in 2h 01m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0d1ae44c9ecd4d7da815ca4701422d6d : SUCCESS in 1h 09m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/71f2ad75b57d47d0842e045f9fe6dec2 : SUCCESS in 1h 04m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fc6dd14e88144776ba1903324bf97d63 : SUCCESS in 44m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1a3199133317429d9fec1bdf615bd406 : SUCCESS in 1h 04m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/12be50ccb4e24348939fb9074002cea6 : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":11},{"id":"033128c5fb06a3208f578851b296efae9f3eefdd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 01:24:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"729a1e424e0430ef39530964fc15e5e79bbe8163","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 01:25:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e049ef98d994e9198ec1977cbee9944\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbfbf439d1714c79a8ff20dc42cd0019 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8730b98b64524ca59f4a28248288999e : SUCCESS in 42s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"0a554b4f29a800694539b285c75eb7f8a2a69193","revisions":{"f40bd262349f3d750b93cc52dd98f2e86c53b2db":{"kind":"REWORK","_number":1,"created":"2023-09-01 22:34:23.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/75/893575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893575/1"}}},"commit":{"parents":[{"commit":"9716e0356d863a6fc7ea152f4e0da0e080545384","subject":"Merge \"[OVN] Fix rate and burst for stateless security groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9716e0356d863a6fc7ea152f4e0da0e080545384"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-01 18:05:25.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-01 22:34:21.000000000","tz":0},"subject":"Add support for OVN MAC_Binding aging","message":"Add support for OVN MAC_Binding aging\n\nOVN added support for aging out MAC_Binding entries [1][2].\nWithout this feature, the MAC_Bindings table can grow indefinitely.\n\n[1] https://github.com/ovn-org/ovn/commit/1a947dd3073628d2f2655f46ee7d3db62ed15b55\n[2] https://github.com/ovn-org/ovn/commit/cecac71c0e49f9bfb6595bc03a13f3f7644dd268\n\nCloses-Bug: 2033932\nChange-Id: I91070ad6addb30ffdedba5d561984d2f6626e2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f40bd262349f3d750b93cc52dd98f2e86c53b2db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f40bd262349f3d750b93cc52dd98f2e86c53b2db"}]},"branch":"refs/heads/master"},"079ea0654cf03321bb2ccb76f3bc100d045c8a39":{"kind":"REWORK","_number":2,"created":"2023-09-09 00:10:59.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/75/893575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893575/2"}}},"commit":{"parents":[{"commit":"9716e0356d863a6fc7ea152f4e0da0e080545384","subject":"Merge \"[OVN] Fix rate and burst for stateless security groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9716e0356d863a6fc7ea152f4e0da0e080545384"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-01 18:05:25.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-09 00:10:46.000000000","tz":0},"subject":"Add support for OVN MAC_Binding aging","message":"Add support for OVN MAC_Binding aging\n\nOVN added support for aging out MAC_Binding entries [1][2].\nWithout this feature, the MAC_Bindings table can grow indefinitely.\n\n[1] https://github.com/ovn-org/ovn/commit/1a947dd3073628d2f2655f46ee7d3db62ed15b55\n[2] https://github.com/ovn-org/ovn/commit/cecac71c0e49f9bfb6595bc03a13f3f7644dd268\n\nCloses-Bug: 2033932\nChange-Id: I91070ad6addb30ffdedba5d561984d2f6626e2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/079ea0654cf03321bb2ccb76f3bc100d045c8a39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/079ea0654cf03321bb2ccb76f3bc100d045c8a39"}]},"branch":"refs/heads/master"},"6b99b1ff78ae655acba3b2bedd15d2c4f16b7ee3":{"kind":"REWORK","_number":3,"created":"2023-09-11 20:24:06.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/75/893575/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893575/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893575/3"}}},"commit":{"parents":[{"commit":"9716e0356d863a6fc7ea152f4e0da0e080545384","subject":"Merge \"[OVN] Fix rate and burst for stateless security groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9716e0356d863a6fc7ea152f4e0da0e080545384"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-01 18:05:25.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-11 20:23:57.000000000","tz":0},"subject":"Add support for OVN MAC_Binding aging","message":"Add support for OVN MAC_Binding aging\n\nOVN added support for aging out MAC_Binding entries [1][2].\nWithout this feature, the MAC_Bindings table can grow indefinitely.\n\n[1] https://github.com/ovn-org/ovn/commit/1a947dd3073628d2f2655f46ee7d3db62ed15b55\n[2] https://github.com/ovn-org/ovn/commit/cecac71c0e49f9bfb6595bc03a13f3f7644dd268\n\nCloses-Bug: 2033932\nChange-Id: I91070ad6addb30ffdedba5d561984d2f6626e2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b99b1ff78ae655acba3b2bedd15d2c4f16b7ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b99b1ff78ae655acba3b2bedd15d2c4f16b7ee3"}]},"branch":"refs/heads/master"},"05851e2c3ff5524185384a77fcd868da4a3b4963":{"kind":"REWORK","_number":4,"created":"2023-09-12 14:10:21.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/75/893575/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893575/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893575/4"}}},"commit":{"parents":[{"commit":"9716e0356d863a6fc7ea152f4e0da0e080545384","subject":"Merge \"[OVN] Fix rate and burst for stateless security groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9716e0356d863a6fc7ea152f4e0da0e080545384"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-01 18:05:25.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-12 14:10:10.000000000","tz":0},"subject":"Add support for OVN MAC_Binding aging","message":"Add support for OVN MAC_Binding aging\n\nOVN added support for aging out MAC_Binding entries [1][2].\nWithout this feature, the MAC_Bindings table can grow indefinitely.\n\n[1] https://github.com/ovn-org/ovn/commit/1a947dd3073628d2f2655f46ee7d3db62ed15b55\n[2] https://github.com/ovn-org/ovn/commit/cecac71c0e49f9bfb6595bc03a13f3f7644dd268\n\nCloses-Bug: 2033932\nChange-Id: I91070ad6addb30ffdedba5d561984d2f6626e2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05851e2c3ff5524185384a77fcd868da4a3b4963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05851e2c3ff5524185384a77fcd868da4a3b4963"}]},"branch":"refs/heads/master"},"c155c755c88b13f6ea6c83ea9b9bce7f4a937b71":{"kind":"REWORK","_number":5,"created":"2023-09-13 14:46:40.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/75/893575/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893575/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893575/5"}}},"commit":{"parents":[{"commit":"9716e0356d863a6fc7ea152f4e0da0e080545384","subject":"Merge \"[OVN] Fix rate and burst for stateless security groups\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9716e0356d863a6fc7ea152f4e0da0e080545384"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-01 18:05:25.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-13 13:45:56.000000000","tz":0},"subject":"Add support for OVN MAC_Binding aging","message":"Add support for OVN MAC_Binding aging\n\nOVN added support for aging out MAC_Binding entries [1][2].\nWithout this feature, the MAC_Bindings table can grow indefinitely.\n\n[1] https://github.com/ovn-org/ovn/commit/1a947dd3073628d2f2655f46ee7d3db62ed15b55\n[2] https://github.com/ovn-org/ovn/commit/cecac71c0e49f9bfb6595bc03a13f3f7644dd268\n\nCloses-Bug: 2033932\nChange-Id: I91070ad6addb30ffdedba5d561984d2f6626e2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c155c755c88b13f6ea6c83ea9b9bce7f4a937b71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c155c755c88b13f6ea6c83ea9b9bce7f4a937b71"}]},"branch":"refs/heads/master"},"cb2f690964d53fbd4fb35d617a103c9704037f1f":{"kind":"REWORK","_number":6,"created":"2023-10-17 23:02:41.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/75/893575/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893575/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893575/6"}}},"commit":{"parents":[{"commit":"849930ba5bdf4bd1675577f48f3fc3ec1d65cef5","subject":"Merge \"Bump skip-level lower version to stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/849930ba5bdf4bd1675577f48f3fc3ec1d65cef5"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-01 18:05:25.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-10-17 23:02:31.000000000","tz":-300},"subject":"Add support for OVN MAC_Binding aging","message":"Add support for OVN MAC_Binding aging\n\nOVN added support for aging out MAC_Binding entries [1][2].\nWithout this feature, the MAC_Bindings table can grow indefinitely.\n\n[1] https://github.com/ovn-org/ovn/commit/1a947dd3073628d2f2655f46ee7d3db62ed15b55\n[2] https://github.com/ovn-org/ovn/commit/cecac71c0e49f9bfb6595bc03a13f3f7644dd268\n\nCloses-Bug: 2033932\nChange-Id: I91070ad6addb30ffdedba5d561984d2f6626e2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb2f690964d53fbd4fb35d617a103c9704037f1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb2f690964d53fbd4fb35d617a103c9704037f1f"}]},"branch":"refs/heads/master"},"308a5a94da419aeed04f57145c740511d885f1c2":{"kind":"REWORK","_number":7,"created":"2023-10-18 14:18:26.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/75/893575/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893575/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893575/7"}}},"commit":{"parents":[{"commit":"849930ba5bdf4bd1675577f48f3fc3ec1d65cef5","subject":"Merge \"Bump skip-level lower version to stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/849930ba5bdf4bd1675577f48f3fc3ec1d65cef5"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-01 18:05:25.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-10-18 14:17:19.000000000","tz":-300},"subject":"Add support for OVN MAC_Binding aging","message":"Add support for OVN MAC_Binding aging\n\nOVN added support for aging out MAC_Binding entries [1][2].\nWithout this feature, the MAC_Bindings table can grow indefinitely.\n\n[1] https://github.com/ovn-org/ovn/commit/1a947dd3073628d2f2655f46ee7d3db62ed15b55\n[2] https://github.com/ovn-org/ovn/commit/cecac71c0e49f9bfb6595bc03a13f3f7644dd268\n\nCloses-Bug: 2033932\nChange-Id: I91070ad6addb30ffdedba5d561984d2f6626e2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/308a5a94da419aeed04f57145c740511d885f1c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/308a5a94da419aeed04f57145c740511d885f1c2"}]},"branch":"refs/heads/master"},"e486f1532c690b9121b7011ed70fd26967413881":{"kind":"REWORK","_number":8,"created":"2023-10-18 14:51:46.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/75/893575/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893575/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893575/8"}}},"commit":{"parents":[{"commit":"849930ba5bdf4bd1675577f48f3fc3ec1d65cef5","subject":"Merge \"Bump skip-level lower version to stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/849930ba5bdf4bd1675577f48f3fc3ec1d65cef5"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-01 18:05:25.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-10-18 14:50:48.000000000","tz":-300},"subject":"Add support for OVN MAC_Binding aging","message":"Add support for OVN MAC_Binding aging\n\nOVN added support for aging out MAC_Binding entries [1][2].\nWithout this feature, the MAC_Bindings table can grow indefinitely.\n\n[1] https://github.com/ovn-org/ovn/commit/1a947dd3073628d2f2655f46ee7d3db62ed15b55\n[2] https://github.com/ovn-org/ovn/commit/cecac71c0e49f9bfb6595bc03a13f3f7644dd268\n\nCloses-Bug: 2033932\nChange-Id: I91070ad6addb30ffdedba5d561984d2f6626e2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e486f1532c690b9121b7011ed70fd26967413881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e486f1532c690b9121b7011ed70fd26967413881"}]},"branch":"refs/heads/master"},"37c9ebd7c37b3a0a1a18c059b23e8d5dfb4bdce5":{"kind":"REWORK","_number":9,"created":"2023-11-17 15:15:47.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/75/893575/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893575/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893575/9"}}},"commit":{"parents":[{"commit":"6514e37e47bc572fa10dcb925c0d81918d393a8c","subject":"Merge \"[CI] Remove py310-with-sqlalchemy-master job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6514e37e47bc572fa10dcb925c0d81918d393a8c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-01 18:05:25.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-11-17 15:14:44.000000000","tz":-360},"subject":"Add support for OVN MAC_Binding aging","message":"Add support for OVN MAC_Binding aging\n\nOVN added support for aging out MAC_Binding entries [1][2].\nWithout this feature, the MAC_Bindings table can grow indefinitely.\n\n[1] https://github.com/ovn-org/ovn/commit/1a947dd3073628d2f2655f46ee7d3db62ed15b55\n[2] https://github.com/ovn-org/ovn/commit/cecac71c0e49f9bfb6595bc03a13f3f7644dd268\n\nCloses-Bug: 2033932\nChange-Id: I91070ad6addb30ffdedba5d561984d2f6626e2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37c9ebd7c37b3a0a1a18c059b23e8d5dfb4bdce5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37c9ebd7c37b3a0a1a18c059b23e8d5dfb4bdce5"}]},"branch":"refs/heads/master"},"9ba5c6db9d758cb24cff89c95a8b9613a286db48":{"kind":"REWORK","_number":10,"created":"2023-11-21 17:55:08.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/75/893575/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893575/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893575/10"}}},"commit":{"parents":[{"commit":"6514e37e47bc572fa10dcb925c0d81918d393a8c","subject":"Merge \"[CI] Remove py310-with-sqlalchemy-master job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6514e37e47bc572fa10dcb925c0d81918d393a8c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-01 18:05:25.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-11-21 17:55:00.000000000","tz":-360},"subject":"Add support for OVN MAC_Binding aging","message":"Add support for OVN MAC_Binding aging\n\nOVN added support for aging out MAC_Binding entries [1][2].\nWithout this feature, the MAC_Bindings table can grow indefinitely.\n\n[1] https://github.com/ovn-org/ovn/commit/1a947dd3073628d2f2655f46ee7d3db62ed15b55\n[2] https://github.com/ovn-org/ovn/commit/cecac71c0e49f9bfb6595bc03a13f3f7644dd268\n\nCloses-Bug: 2033932\nChange-Id: I91070ad6addb30ffdedba5d561984d2f6626e2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ba5c6db9d758cb24cff89c95a8b9613a286db48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ba5c6db9d758cb24cff89c95a8b9613a286db48"}]},"branch":"refs/heads/master"},"0a554b4f29a800694539b285c75eb7f8a2a69193":{"kind":"REWORK","_number":11,"created":"2023-11-21 18:06:26.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/75/893575/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893575/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893575/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893575/11"}}},"commit":{"parents":[{"commit":"6514e37e47bc572fa10dcb925c0d81918d393a8c","subject":"Merge \"[CI] Remove py310-with-sqlalchemy-master job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6514e37e47bc572fa10dcb925c0d81918d393a8c"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-01 18:05:25.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-11-21 18:03:15.000000000","tz":-360},"subject":"Add support for OVN MAC_Binding aging","message":"Add support for OVN MAC_Binding aging\n\nOVN added support for aging out MAC_Binding entries [1][2].\nWithout this feature, the MAC_Bindings table can grow indefinitely.\n\n[1] https://github.com/ovn-org/ovn/commit/1a947dd3073628d2f2655f46ee7d3db62ed15b55\n[2] https://github.com/ovn-org/ovn/commit/cecac71c0e49f9bfb6595bc03a13f3f7644dd268\n\nCloses-Bug: 2033932\nChange-Id: I91070ad6addb30ffdedba5d561984d2f6626e2b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a554b4f29a800694539b285c75eb7f8a2a69193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a554b4f29a800694539b285c75eb7f8a2a69193"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
