)]}'
{"id":"openstack%2Fneutron~893775","triplet_id":"openstack%2Fneutron~stable%2F2023.1~I6da1b135c83ecfc41ec91e907ebf8500325a7a80","project":"openstack/neutron","branch":"stable/2023.1","topic":"check_haproxy_running-stable/2023.1","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-01-04 08:35:08.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2024-01-04 08:52:16.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-01-04 10:44:18.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2024-01-04 10:44:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6da1b135c83ecfc41ec91e907ebf8500325a7a80","subject":"Metadata: handle process exceptions","status":"MERGED","created":"2023-09-06 14:35:59.000000000","updated":"2024-01-04 10:45:36.000000000","submitted":"2024-01-04 10:44:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893775-check_haproxy_running-stable/2023.1","cherry_pick_of_change":890986,"cherry_pick_of_patch_set":23,"meta_rev_id":"878321ceee85d0a6c7d70b8ab162b751d47fcc63","_number":893775,"virtual_id_number":893775,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-04 10:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-01-04 08:35:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-01-04 08:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-01-04 08:52:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 14:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-06 16:24:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-06 16:55:11.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-12-04 11:05:32.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2023-12-19 11:50:56.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-01-02 19:37:43.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"404deecadc28de74da12dd7792ca343f378399b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-06 14:35:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"08818fe1beb2a9ae88b9dd33b2a7e297e26f7916","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 14:59:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2b2a1b52bed46f38550221fdb1c4f87\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50995062dfad433c8496a5794aaa9ceb : SUCCESS in 22m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cc1d44081eb44e328df3087621cd258c : SUCCESS in 21m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4cb5079034eac18c2151dea2295135db3b25bcc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 16:24:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf302db82497497fa10cf7b8b9d39842\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/10f95c4e5d9544bd90b3ae002b285d26 : SUCCESS in 1h 29m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbbb033e74564ed4b3cd0ca3e273b552 : SUCCESS in 36m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de54f3b2310a44458bb1b7f178ef144f : SUCCESS in 14m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/423e125590eb446ea88f119f677fe872 : SUCCESS in 27m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/606307fbe5234283a82a822d0de1f823 : SUCCESS in 27m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d422caad094856956b7d1e508d4352 : SUCCESS in 10m 48s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/08995207480b4969b461511fb284f437 : SUCCESS in 26m 52s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/228adf6a66424a14b9ce55e256da3fd4 : FAILURE in 50m 11s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/26e0acc7c6e342e099c2de179b322d9b : SUCCESS in 56m 54s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c492be3a41ba4fb98a6e309b541fc3af : SUCCESS in 1h 06m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b8200805f884dbdab48d1623d967cf4 : SUCCESS in 1h 46m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ba31037889f844b89b4a5b589b1dd473 : SUCCESS in 1h 07m 51s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/638afbd4499f4709b0347ab290d16afd : SUCCESS in 55m 50s","accounts_in_message":[],"_revision_number":1},{"id":"4adde0e6a3cea0ac2639cd75df54b4eb8db868cc","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-06 16:55:11.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5282ee3ca288e4e2a7f938d9cc89c80b2c05b09","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-06 18:29:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a978c7af0a05909966a1a2ac613e141384cbeac","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-12-04 11:05:32.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"fc6b6ccce0889d58a1662fb9d8db1799a39638cb","tag":"autogenerated:gerrit:restore","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-06 14:54:27.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"382dbebdce84b4f1d4031ef0825a360ac0b78cd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-06 14:54:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0222c77665cc1c055172b0b8e9bcd49dad0438ae","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-06 14:59:59.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"032df7b10961538c61325a99fd150a2b56d8ccf7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 15:09:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a9769a0fcbf4507ac78dfe340e4faf0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51970d9c6f03458e81a1d762d0fd176b : FAILURE in 11m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/867611fcd0d049f88ba0acda9967be93 : FAILURE in 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"269b353f7df92416241f70fcc1542406ae44d4bd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-06 15:19:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"98be8ee0274d65ebec22430b9c1d5314b435aa3e","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-06 15:26:20.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"09497f370d9175f09e44c05e5b79b5af2d848332","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 15:36:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5566777a0ea4b768f18986bb7f67bbd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7eca0a2139c1450a9ae9303de57fa3d7 : FAILURE in 16m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1663818e908469bb2e47105586ea99d : FAILURE in 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9d17d519c81c24f815d4e17ea9f0a815e1c3edec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 17:19:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1b17afe7eef4b6480e92b6068d2c21b\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/5d8827bcbb344fb7979dc3ac51d6efa5 : SUCCESS in 1h 31m 49s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/5107fd30b7354e84a41b12ed8694f661 : SUCCESS in 1h 11m 16s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/7658e2965c4f4d519d6817994830850d : SUCCESS in 1h 17m 56s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/9b03161ca19b43eeaeac9998f3d31ed2 : SUCCESS in 1h 27m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d36bf307b9e44a5ca05504a2aee91886 : FAILURE in 34m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00da2d8228804c5b8d008a20a060b3b0 : SUCCESS in 11m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67f3a92df9704c6bb49ff97cdff6f2f4 : FAILURE in 26m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d13eec70a8f47b0a7a68e1f053466e8 : FAILURE in 23m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b4744aa4b00444f97abe8a44ff08fa4 : SUCCESS in 12m 25s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/6546b33fff9c40c7b4a6ccc574bed7ae : FAILURE in 23m 20s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5999fcf4b424b98922617099724fb91 : SUCCESS in 56m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/17f3128a252343faa21480d3b1941262 : SUCCESS in 1h 58m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f60e64bd31314d9384e923a131262881 : SUCCESS in 51m 49s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2aa32c4fef0f429fa367772a0fde420d : SUCCESS in 52m 37s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b84572a0393a4d838652566b18c8fa6b : SUCCESS in 1h 00m 21s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5d7d687c689f45b6bcd0bf0cf6808db8 : SUCCESS in 46m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c08cbfa75aba45db9ac1b31a976a9eac : SUCCESS in 1h 24m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6d1ad34f426f4b7abdbe6ce20016de24 : SUCCESS in 1h 13m 56s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8b7711abcb724a408c4074ba9ce5b593 : SUCCESS in 45m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b79979803a5b462788f3dadbfef68089 : SUCCESS in 1h 24m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1d493634259c43259e293c210915d99b : SUCCESS in 1h 30m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a33c602b3f144a538f3c7b9b0db88f0e : SUCCESS in 1h 00m 22s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b82719df7dbb43f99e764e86668d4e78 : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":3},{"id":"b95770366bdedf2fbc7be02ae68d792f24ec43c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-08 20:14:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"64449071c400e26c2997d1c3490692dc8cdb0363","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 20:33:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ea7e3515924803981c4f12ba9142d7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4793d6b9c322409da1350e395b348335 : SUCCESS in 17m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad6b6a38c4be4bffb1255df4de4419ee : SUCCESS in 17m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5f1206a3eb27494e16f369cb3d61795a960db898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 22:16:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06ca00dea74e442aacba18448e71135a\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/2e1b16da3e9a4146a53daeb651665f65 : SUCCESS in 1h 59m 26s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/7a7ca303a4834f669a42bbee7d009f25 : SUCCESS in 1h 55m 12s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/55c047513960439b8b455e9a45f49d6a : SUCCESS in 1h 27m 14s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/e9339bbf1d6e4d3faa501adc5c28af15 : SUCCESS in 1h 44m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73c2fa37b5ec4dc68c2f852a28c0de9f : SUCCESS in 34m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15b32dfbc15645b5b631b13e2a8e4d2f : FAILURE in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc9aa598db694e97b0178aef74dd91ad : SUCCESS in 23m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bac8f5134f141fc926da554bf759f4b : SUCCESS in 26m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4b486522c4742778c69e4ec24d97b51 : SUCCESS in 10m 29s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/7e4d74b85b004be197387f7db64e5941 : SUCCESS in 20m 39s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aa056fab061846f18205ff3c62b1f7af : SUCCESS in 46m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9d798958987c4bc48b11233669f738f2 : SUCCESS in 1h 42m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/36ed6c181a454353b703e8135fc95948 : SUCCESS in 55m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/852173bd948e41ce8e144a638d7d368f : SUCCESS in 50m 00s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/14cf4b07e49848f58f14ba7233f3ff20 : SUCCESS in 1h 17m 51s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/807a7903a8ab47419aa9aca09f08b6f6 : SUCCESS in 56m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7c7db08ef81a408c9e9a8f7c64f218e3 : SUCCESS in 1h 09m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/35c4af5da5684897b3d09f6754113955 : SUCCESS in 58m 11s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2d1e343f53604460a947e42cec2a4027 : SUCCESS in 1h 02m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b59a313cf12842f8bb699a0297077aa8 : SUCCESS in 1h 19m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/409a10110a3f478b94e802d95a34cd6a : SUCCESS in 1h 43m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/46012a6fdbb042baae6cd372118eaab5 : SUCCESS in 1h 08m 13s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/3c150702d44a45e8bf36e5995fcecc8c : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":4},{"id":"112d17113fa1b03ab65dec36cc7710c08fa6a202","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-18 20:27:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7e599d8f5f1efc30337accbd4246fc468422563c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-12-19 11:50:56.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"106d09a75b7fac30eb0de8d167c3fc843cd05473","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-20 15:41:59.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c1d74b9b5ec62a4cf54cf3c1b67d50b58cdf28e1","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-20 15:43:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"eadfced871395efaf402bed5ecdc7f56791102c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-02 17:35:26.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"794acea4bcdc3fc3c1834f0bd1421a25ef3ee259","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 18:13:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d37dccba0e204bea908a0abe5adb7948\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f3f25f625ac4ff6ab391c5230b4e6ed : SUCCESS in 36m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/31f182cda98a4290bc1823460a07db0f : SUCCESS in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"920266b0837bc913a630d2b5c34956172192db94","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-02 19:37:43.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3236187685c924ed0fb8c5bae700e61b4e78ece7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 19:57:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b04df7423ad74f9f8ee9874a1a547af0\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/c9daaa3c62e2489d9eb3b7c135e8e7aa : SUCCESS in 2h 19m 43s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/748244d3f15841ed98b8a39b9cf2834c : SUCCESS in 1h 30m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/4aa398cd8ad342b096e658b00fd21ee8 : SUCCESS in 1h 38m 33s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/8ee270f7d92b4206878ca4924af8d3cf : SUCCESS in 1h 57m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4f62182bf69415ca947da87bf15eae8 : SUCCESS in 41m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d1b1a744cd42a792433bf99b634b29 : SUCCESS in 14m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/323268bd78744e7d8d65f5973f7c826e : SUCCESS in 20m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/603b828e662944d5b8156161c7409b28 : SUCCESS in 21m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87bb06401ab24ff69e44748e46021703 : SUCCESS in 9m 41s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f5f35b1173ca4a2496ec03922b6369c6 : SUCCESS in 22m 08s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b62c96fbad5847b1b25e164b5bec71fe : SUCCESS in 35m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c03d9c4e6e94d63b6bdcf547c04b575 : SUCCESS in 1h 49m 26s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb32336365e44b2f844c03a5286450e8 : SUCCESS in 1h 05m 42s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a67ce2c71e0e4599953a8a5ac4f92761 : SUCCESS in 1h 06m 20s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/02ecc3d416c349508b555d84b50dd282 : SUCCESS in 1h 06m 30s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/13c22d90546e4e7a9d8464570302ce88 : SUCCESS in 1h 06m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cbef451d72334f85aedc558f4f395731 : SUCCESS in 1h 20m 41s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1bc25c8de37a4e9d948ec40a16f54aa7 : SUCCESS in 1h 19m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a0e18fb6c8224ce6a7618c4a778db3b6 : SUCCESS in 57m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ea634ed0fd8e42e4a0aac28f7f9ac066 : SUCCESS in 1h 25m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b67831d09edd4e6eb2636bcbb5acb88e : SUCCESS in 1h 42m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e5165fcfe926425ca5bfb7c6f9532ffc : SUCCESS in 1h 29m 09s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/230e421ebb034e3b912085a0822ae779 : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":6},{"id":"479bc83035fe271180f247d84121f81e747f831f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-01-03 09:32:06.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a8fe333b7c4de0cc24223ba6dba28ca5e9727caf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 09:33:01.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"137539f87d3eaab0a4548b7c214c3a6be1514958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 11:34:59.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fdb382bf2b8b423aba3125371d737f89\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/40f28099d0e74b709a3227d78fde64b4 : SUCCESS in 2h 00m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/157841dc233c43588bca5362333b0b5f : SUCCESS in 12m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc0073c0e0e84946b76a6a02a9a90fa3 : SUCCESS in 24m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49b4efa687004a20bfaca8dd5524dbc7 : SUCCESS in 24m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35f4507471c845f38b54fecd1db22fa0 : SUCCESS in 10m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e46bc7c14ba64662b4eda8c2e6c3b06b : SUCCESS in 49m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ebce355e78a445eeb6241dee6b4b0ddc : SUCCESS in 1h 32m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4e87df5685c64d6ca922a48fd60b569d : FAILURE in 1h 06m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f482380376b49e889bce1980e7b1e71 : SUCCESS in 1h 06m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b2fcb4303dd94e6794f30b345f6988ac : SUCCESS in 1h 00m 02s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/711b7591e03f452993f958fdab3aa4b9 : SUCCESS in 43m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/26a6141b7c894d008b524eac2aa4bacf : SUCCESS in 1h 13m 19s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/26ce042aff874627bc209f5bff8f131f : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":6},{"id":"63cfdd5149a7dd5ceda47999a705cf7630fc139d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-01-03 13:44:37.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fc858e4cc0b361b7d0e1670772fcecb0bd775427","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 14:22:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d40b1acab3904559b6102fd52b099c7b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d5f8d08372144e08fda0f800460240d : SUCCESS in 17m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cea80898a7e9472ebfd85bb46ca3b373 : SUCCESS in 35m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"84bfba4f0a91b8e5ad1f8d14f86cec3072b5216d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2024-01-03 14:52:02.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":7},{"id":"3464dd2dc9867c14dfc04ea11b5165ab1d65e8ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 15:28:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0399624341a4b4ea12d78f968567514\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e224b6d5a7734fc6a172002df5e7de76 : SUCCESS in 17m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/07300ee5e47e483bbc8fc456f5a32ac8 : SUCCESS in 34m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"27105f326269b56321aad8de27dc8f3d8346491a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 17:15:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09acfd1027884b608cb1b8460812b041\n\n- neutron-tempest-plugin-linuxbridge-2023-1 https://zuul.opendev.org/t/openstack/build/f9f2aae037bb45749d2911b013b43776 : SUCCESS in 2h 21m 07s\n- neutron-tempest-plugin-openvswitch-2023-1 https://zuul.opendev.org/t/openstack/build/ca2880d34123493699f506893e8e84ab : SUCCESS in 1h 23m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-2023-1 https://zuul.opendev.org/t/openstack/build/5c3eb4bd46d54e0aa1e8c5c465f0f479 : SUCCESS in 1h 38m 00s\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/3122d7aee1384c0089f1081421176f40 : SUCCESS in 1h 56m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce186ba88ea0431f82ed4475e97f09eb : SUCCESS in 24m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e024d596404b64a2f7a7552fbb0740 : SUCCESS in 13m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2d2993455db4effb5c916014b5fab84 : SUCCESS in 17m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c4e55a1fed7405184bd66f67b171bae : SUCCESS in 20m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c565179cc3dc4f45a5c540bd513c6c76 : SUCCESS in 10m 37s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/db10c1f3e85a4b5e9b12b9f85c443460 : SUCCESS in 25m 09s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a7a7c775b184da8a3b8565e1bd449be : SUCCESS in 40m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0bad45c92aea4c4a92c695d5d3c963be : SUCCESS in 1h 32m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e4be1ed232c4e96ab59b0a9be354f59 : SUCCESS in 49m 11s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d71864f4c731448b88b775d89c8f464a : SUCCESS in 58m 19s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1b247b4d541446f787e03177a0afacee : SUCCESS in 1h 17m 06s (non-voting)\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1df08871e79d4eec92a621fe1611cd9f : SUCCESS in 43m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/18651cba6b1f4bdc8ea7977562e62ed3 : SUCCESS in 1h 04m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2f74cbd3df7b4cf89c99ae214236933d : SUCCESS in 1h 22m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ff9574e8fc3240a3b64cd96b5d821454 : SUCCESS in 1h 04m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/180f0998001f488f971e61f335a92d2f : SUCCESS in 1h 11m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/875c41419378495bb7a53f96e6695838 : SUCCESS in 1h 22m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3dfef6cf0b6d4682924c2d7fa378cd62 : SUCCESS in 1h 19m 51s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/02457fd6cdd646cd9477db007ce46e23 : SUCCESS in 46m 29s","accounts_in_message":[],"_revision_number":7},{"id":"b9c9b90099f53ad4d1594820e4cce401529860f0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-04 08:35:08.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"16e32feb1a46ba2e808ea776292133a29556a19c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-01-04 08:52:16.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1fa4e29cc15c76644fe13570dd78b4cbeeaae3dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 08:52:43.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f152e7d3dab7f9f11c3a71f543627cc3c1e331a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 10:44:18.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef98dd2862d44c54a4817050c596576a\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/05653de581414e4287a148f12fb452c7 : SUCCESS in 1h 50m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00ca331eba6c413e8e87cf99caf057e8 : SUCCESS in 12m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea3141dfde2e41eda8195633bbfc9fd1 : SUCCESS in 24m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f6092e39e2e47949298ea01598dca81 : SUCCESS in 24m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f01d8af0e354993a5bb699a81574c3f : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/016f992a276d4d71a62261a5923f3a42 : SUCCESS in 52m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/543c9a0623eb482195b9b319dda01f5c : SUCCESS in 1h 36m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2f1c42d0d4d84e6184f24d3a3daa62e9 : SUCCESS in 1h 25m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fa5d2c78d464909a6dfbcb740b13fa9 : SUCCESS in 1h 08m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/12ea4efaf587426299385c10a2470793 : SUCCESS in 1h 06m 06s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/2df4e3f165294bb698e124d0d217a22a : SUCCESS in 1h 06m 46s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cd8fd86844cd4955832b007f1a907306 : SUCCESS in 1h 11m 37s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/a5abdeb0b40a4e3fb89f711654550fb3 : SUCCESS in 45m 53s","accounts_in_message":[],"_revision_number":7},{"id":"dda56f472cf83a2aefce86fd2fec2fd12007e531","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 10:44:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"878321ceee85d0a6c7d70b8ab162b751d47fcc63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-04 10:45:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea577d409998466597d68eac90ba2389\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d71d71bd4d34264a9b1f7548c2dd71d : SUCCESS in 56s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"63c16672a5ff3a82607bf1912bc51b5db21a1571","revisions":{"8aceb4afdc3d185cb74b5eaac20afc3e2621785f":{"kind":"REWORK","_number":1,"created":"2023-09-06 14:35:59.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/75/893775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893775/1"}}},"commit":{"parents":[{"commit":"fbda749de7069e9fc0376139ca93e9eae000c5da","subject":"Improve the ``PortBindingUpdateVirtualPortsEvent`` match filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbda749de7069e9fc0376139ca93e9eae000c5da"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-09-06 14:35:59.000000000","tz":0},"subject":"OVN Metadata handle process exceptions","message":"OVN Metadata handle process exceptions\n\nOVN metadata agent should handle process exceptions incase\nan exception happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8aceb4afdc3d185cb74b5eaac20afc3e2621785f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8aceb4afdc3d185cb74b5eaac20afc3e2621785f"}]},"branch":"refs/heads/stable/2023.1"},"828390fcb102348501ad6c592e5f48e02f207caa":{"kind":"REWORK","_number":2,"created":"2023-12-06 14:54:49.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/75/893775/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893775/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893775/2"}}},"commit":{"parents":[{"commit":"5fd71ac96e3b100e5c58785df112c5da6e6fb697","subject":"Merge \"Add constant to identify OVN LB HM ports\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fd71ac96e3b100e5c58785df112c5da6e6fb697"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-12-06 14:54:49.000000000","tz":0},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n(cherry picked from commit ed0515737be75745933f4994b5e6b652f9e0a2be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/828390fcb102348501ad6c592e5f48e02f207caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/828390fcb102348501ad6c592e5f48e02f207caa"}]},"branch":"refs/heads/stable/2023.1"},"f99a91b48fa58ae5575ca58c955c509f47ef4f7a":{"kind":"REWORK","_number":3,"created":"2023-12-06 15:19:44.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/75/893775/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893775/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893775/3"}}},"commit":{"parents":[{"commit":"5fd71ac96e3b100e5c58785df112c5da6e6fb697","subject":"Merge \"Add constant to identify OVN LB HM ports\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5fd71ac96e3b100e5c58785df112c5da6e6fb697"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-12-06 15:19:29.000000000","tz":-300},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n(cherry picked from commit ed0515737be75745933f4994b5e6b652f9e0a2be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f99a91b48fa58ae5575ca58c955c509f47ef4f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f99a91b48fa58ae5575ca58c955c509f47ef4f7a"}]},"branch":"refs/heads/stable/2023.1"},"c1d2d04af41a96377e6c20f932b86403e3598246":{"kind":"REWORK","_number":4,"created":"2023-12-08 20:14:10.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/75/893775/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893775/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893775/4"}}},"commit":{"parents":[{"commit":"ac997a629bbe07bf627972657e48c8bd1d9b4356","subject":"Remove obsolete PID files before start","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac997a629bbe07bf627972657e48c8bd1d9b4356"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-12-08 20:12:15.000000000","tz":-300},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/903149\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n(cherry picked from commit ed0515737be75745933f4994b5e6b652f9e0a2be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1d2d04af41a96377e6c20f932b86403e3598246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c1d2d04af41a96377e6c20f932b86403e3598246"}]},"branch":"refs/heads/stable/2023.1"},"d60667897660749f4d56579e10ef4bd2794908b3":{"kind":"REWORK","_number":5,"created":"2023-12-20 15:41:59.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/75/893775/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893775/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893775/5"}}},"commit":{"parents":[{"commit":"ac997a629bbe07bf627972657e48c8bd1d9b4356","subject":"Remove obsolete PID files before start","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac997a629bbe07bf627972657e48c8bd1d9b4356"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-12-20 15:41:45.000000000","tz":-300},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/903149\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n(cherry picked from commit ed0515737be75745933f4994b5e6b652f9e0a2be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d60667897660749f4d56579e10ef4bd2794908b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d60667897660749f4d56579e10ef4bd2794908b3"}]},"branch":"refs/heads/stable/2023.1"},"27498f957a95e692edc95ade3c819b3cdf279e4b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-01-02 17:35:26.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/75/893775/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893775/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893775/6"}}},"commit":{"parents":[{"commit":"f0c722290651126f4305ebdfd8b027d0764448e7","subject":"Merge \"Remove obsolete PID files before start\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f0c722290651126f4305ebdfd8b027d0764448e7"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-02 17:35:26.000000000","tz":0},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/903149\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n(cherry picked from commit ed0515737be75745933f4994b5e6b652f9e0a2be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27498f957a95e692edc95ade3c819b3cdf279e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/27498f957a95e692edc95ade3c819b3cdf279e4b"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"},"63c16672a5ff3a82607bf1912bc51b5db21a1571":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-01-03 14:52:02.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/75/893775/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/75/893775/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/75/893775/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/75/893775/7"}}},"commit":{"parents":[{"commit":"f0c722290651126f4305ebdfd8b027d0764448e7","subject":"Merge \"Remove obsolete PID files before start\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f0c722290651126f4305ebdfd8b027d0764448e7"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2024-01-03 14:52:02.000000000","tz":0},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n(cherry picked from commit ed0515737be75745933f4994b5e6b652f9e0a2be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63c16672a5ff3a82607bf1912bc51b5db21a1571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/63c16672a5ff3a82607bf1912bc51b5db21a1571"}]},"branch":"refs/heads/stable/2023.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
