)]}'
{"id":"openstack%2Fneutron~893776","triplet_id":"openstack%2Fneutron~stable%2Fzed~I6da1b135c83ecfc41ec91e907ebf8500325a7a80","project":"openstack/neutron","branch":"stable/zed","topic":"check_haproxy_running-stable/zed","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2024-01-03 12:40:38.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2024-01-03 12:40:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6da1b135c83ecfc41ec91e907ebf8500325a7a80","subject":"Metadata: handle process exceptions","status":"MERGED","created":"2023-09-06 14:36:19.000000000","updated":"2024-01-03 12:42:13.000000000","submitted":"2024-01-03 12:40:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893776-check_haproxy_running-stable/zed","cherry_pick_of_change":890986,"cherry_pick_of_patch_set":23,"meta_rev_id":"68f6ea898dcb652a3d79faf7a01c4afa8895f4fe","_number":893776,"virtual_id_number":893776,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-03 12:40:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-01-02 19:38:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-01-03 09:33:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2024-01-03 01:58:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-01-03 09:33:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-06 15:00:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-06 16:08:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-06 15:34:37.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2024-01-02 19:38:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-01-03 01:58:22.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2024-01-03 09:33:14.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"30d091b37628926dbf33c433e5a207f73ab61690","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-06 14:36:19.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"6fa6062a33eb68623179a1a5d2e26f41541e976c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 15:00:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87dd2e5bb44547b49b14de0ebef1ab62\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1885ed623ff849698282a0f69b4e0924 : SUCCESS in 22m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9db98e80e2b4f63bbb9d22927fed621 : SUCCESS in 22m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4dcfe8c3d9f7f6d3bae716b32a00382a59e71253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 16:08:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1de604f82ef49d585a7fc1b1184b960\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/c11b8924f7914b25a3808f81cfc14728 : SUCCESS in 1h 30m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ce12f7960114195bd424cd3d3d2d067 : SUCCESS in 36m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3689366ab4814a899a5b569cd1a91278 : SUCCESS in 13m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d53726a5bb344c31ad003b358db85705 : SUCCESS in 23m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47e11630f3074081b9645be0528171b3 : SUCCESS in 26m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7c7981ea92344118b0df8cecbe5c3f9 : SUCCESS in 27m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50374623273e4f3e8c8eda54d741f155 : SUCCESS in 11m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2ff315cedb04828a7d4e425f04f5e39 : FAILURE in 48m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a05f6ac072f84a9fb8e258f805f98654 : SUCCESS in 1h 07m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a665ca120a654746a7220ca842715947 : SUCCESS in 1h 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e055fa1b811c7aa33e7821e3febf710f446ecec","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-11-14 19:40:27.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ef5fc6ec635b1e792ae39612d438bd0aa17d06fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-06 15:27:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"11630d357cd9c8538972e00fe51ca2fa344b51f9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-06 15:29:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ff163128dd6285f9129ddd8eb10ba8c835a66ad0","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-06 15:34:37.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"18d6640f4989279389ef846d7727921fb344c180","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 16:14:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5108f905154b48ca9ba286d7accb1d07\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b102640f7924dcea6d95f01966f357f : FAILURE in 41m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5cb968a45433490dad056c0070a65501 : FAILURE in 42m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"55ad3da520cd7ada57202d827f6b473e9d14c75d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 17:22:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c29df57b7e6045e1a100d32621998492\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/4b9a9159e35f4f8493c398d7a7daf649 : SUCCESS in 1h 39m 03s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/246ad4fe329d429dacacdbea71e2ede6 : SUCCESS in 1h 28m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/70ffe8044a054c9fa511ba9e545d5d2e : SUCCESS in 1h 40m 24s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/c3fae20a6dc44624b5cd3f43220658cf : SUCCESS in 1h 44m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/998ff56798f74e618c6500c135f4aabb : FAILURE in 31m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dc6459c10fb44d4815a516764f94b18 : SUCCESS in 13m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5c143bcd8a24c29b80dabeff1d3da2d : FAILURE in 25m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91a2ff4f1afd4fc6a32fd0ef51690698 : FAILURE in 19m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10500b3bc48c4b06b59360e97025f881 : FAILURE in 30m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c9df966e52e46b8b0e580fd5a9179e3 : SUCCESS in 10m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b9cc784f4aa4a209e726183b01dc83c : SUCCESS in 50m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/33fca2f72a11427aafe765c985245daf : SUCCESS in 1h 38m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/43f89a8e373b47b48b5b49ed78847f1d : SUCCESS in 1h 06m 39s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8420ebfd0f154e7487845802ac1a0d77 : SUCCESS in 1h 08m 35s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bf3c99f28a294538acf2c3f1f929caec : SUCCESS in 1h 07m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/571f9fa69b0941a8a54cbd4c6f211b08 : SUCCESS in 1h 02m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/647b6915282e4d90be13205e2a6f1923 : SUCCESS in 1h 11m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bbb26bf2f594ebb968b1e675556071d : SUCCESS in 1h 09m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/73163bd0118249be9e49bfb9ad0b1873 : SUCCESS in 54m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3c8854b66fd92f01550e19ec27f1b9b50b812f1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-12-08 20:37:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d26831fbb17361d6cf1949f565673ac6f4f16e6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 21:17:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2335d850cb524a09a912037fe2b87a5f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7459c4d326f9477dbaf97eabd1f84ab7 : SUCCESS in 17m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6892ebf92b7040558b6ad70deb6b3e19 : SUCCESS in 37m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ed4d178107d5061e72e78cc5af38d7590ad8abde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 22:39:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbf336629aaa47128c802ce762a4ed49\n\n- neutron-tempest-plugin-linuxbridge-zed https://zuul.opendev.org/t/openstack/build/7df96e8af54c45ab9874eaf5f95af691 : SUCCESS in 1h 39m 22s\n- neutron-tempest-plugin-openvswitch-zed https://zuul.opendev.org/t/openstack/build/6556ef6f322a410e9736559cd2eb0ff3 : SUCCESS in 1h 48m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid-zed https://zuul.opendev.org/t/openstack/build/5850eb376f5d4f7fb50de72ef652ab24 : SUCCESS in 1h 39m 52s\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/d84739a5a6e245d68db3f7214294831d : SUCCESS in 1h 59m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20c1e581820048b3b45084ea6caa4481 : SUCCESS in 30m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49df4b062f354cb5a3e1efb8b793c771 : SUCCESS in 13m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e13d3aee4c44853a7a48fa2ecbc12c3 : SUCCESS in 24m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16d5e75e2be5489697a3720ec17ece03 : SUCCESS in 28m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2260e2ad40674b8984f9c11c40f42046 : SUCCESS in 17m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afcf921401aa418f8e2d9c8b85a8b1b3 : SUCCESS in 11m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/91435c11f6474632bf74d8f3fad8e892 : SUCCESS in 50m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a8129be7ac83427ab44329761fe600d0 : SUCCESS in 1h 38m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/07b532b3f5b240649ad3f8f407823796 : SUCCESS in 53m 40s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a2d16df26de4798b4ed7d4c643ac49c : SUCCESS in 1h 07m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2111c6c6c110436586c5c0b456f8af08 : SUCCESS in 49m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3818d3bb11014dcb8adfb4aea3a45c44 : SUCCESS in 1h 00m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/715697fb3a58469bbd046f7d4189f8bd : SUCCESS in 1h 08m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ee4cc5062014d1d93170cd7d49def89 : SUCCESS in 47m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/df7d356caf5a4033973e3a6c354d50ac : SUCCESS in 1h 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"70d61e9ffdc89c557b08ac4ae97653c2248b0a3c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-01-02 19:38:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"54d4069236b0e65a1c9af831bc0667929fb1904f","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-01-03 01:58:22.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ee963886dd2b53acd92bb72ca522863ab4f13ff0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-01-03 09:33:14.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"46e9c21b0cb1cdf9dda8166cff9154c7b9c3cc9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 09:34:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f3cbc73ec70e4784c467d542cd23d712bf55dae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 12:40:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7395bef021a43f0a4514536f53c4e0b\n\n- neutron-tempest-plugin-ovn-zed https://zuul.opendev.org/t/openstack/build/249f0882ab85491cb8588b6ac44bf428 : SUCCESS in 1h 53m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b39797f227074184b109f5773f8bfc17 : SUCCESS in 11m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79b63226f2484d4fb4882433dc2f10ea : SUCCESS in 23m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9282862eba44c5eb39349ec3dddb766 : SUCCESS in 21m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f8c7577aee14d5195af0ae5094345f5 : SUCCESS in 10m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b3429f5a0af4fb79c851b7e7138332d : SUCCESS in 1h 01m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a7efa5e86d04ce3a6bb6adcccc4adbb : SUCCESS in 1h 33m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5741b7dd40fa4d5cb2e539c6c9d18c4f : SUCCESS in 1h 18m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3611a90d626b49e5bf2e5678a3c11e63 : SUCCESS in 1h 00m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f263f368857b4a3caf3ed7de6a3464f5 : SUCCESS in 1h 05m 19s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/600673f889a340f6bd375e548010613e : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":4},{"id":"e7192cf494ae62bd65b880a67fbf9b0d57d5ffba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 12:40:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"68f6ea898dcb652a3d79faf7a01c4afa8895f4fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 12:42:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28bc9e8918764277be09114df401d73f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f88dfae520114beb9213a018e25dd4e9 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4698dd899fa149088160372b6f2efcd3ce973cbb","revisions":{"72cc9e800c18540adb946d063d374912ef1997b3":{"kind":"REWORK","_number":1,"created":"2023-09-06 14:36:19.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/76/893776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/893776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/893776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/893776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/893776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/893776/1"}}},"commit":{"parents":[{"commit":"810c813adf0ab175a847805adb6a5f20ec586577","subject":"Improve the ``PortBindingUpdateVirtualPortsEvent`` match filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/810c813adf0ab175a847805adb6a5f20ec586577"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-09-06 14:36:19.000000000","tz":0},"subject":"OVN Metadata handle process exceptions","message":"OVN Metadata handle process exceptions\n\nOVN metadata agent should handle process exceptions incase\nan exception happens when spawning haproxy processes\nsuch that the agent can continue its operation for possibly\nother haproxy processes\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72cc9e800c18540adb946d063d374912ef1997b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72cc9e800c18540adb946d063d374912ef1997b3"}]},"branch":"refs/heads/stable/zed"},"d885eef562dea730fb51cd2233370a3f7c4f8a08":{"kind":"REWORK","_number":2,"created":"2023-12-06 15:27:34.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/76/893776/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/893776/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/893776/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/893776/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/893776/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/893776/2"}}},"commit":{"parents":[{"commit":"6e0805b17a7de3115ea21d2e37d79f238ea6ac48","subject":"Merge \"[DHCP agent] Fix route to OVN metadata port for non-isolated networks\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e0805b17a7de3115ea21d2e37d79f238ea6ac48"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-12-06 15:27:34.000000000","tz":0},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n(cherry picked from commit ed0515737be75745933f4994b5e6b652f9e0a2be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d885eef562dea730fb51cd2233370a3f7c4f8a08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d885eef562dea730fb51cd2233370a3f7c4f8a08"}]},"branch":"refs/heads/stable/zed"},"f9dfdb7fd6b69edb410cdc28419953c7a051bd90":{"kind":"REWORK","_number":3,"created":"2023-12-06 15:29:53.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/76/893776/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/893776/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/893776/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/893776/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/893776/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/893776/3"}}},"commit":{"parents":[{"commit":"6e0805b17a7de3115ea21d2e37d79f238ea6ac48","subject":"Merge \"[DHCP agent] Fix route to OVN metadata port for non-isolated networks\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e0805b17a7de3115ea21d2e37d79f238ea6ac48"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-12-06 15:29:11.000000000","tz":-300},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n(cherry picked from commit ed0515737be75745933f4994b5e6b652f9e0a2be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9dfdb7fd6b69edb410cdc28419953c7a051bd90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f9dfdb7fd6b69edb410cdc28419953c7a051bd90"}]},"branch":"refs/heads/stable/zed"},"4698dd899fa149088160372b6f2efcd3ce973cbb":{"kind":"REWORK","_number":4,"created":"2023-12-08 20:37:07.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/76/893776/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/76/893776/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/893776/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/893776/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/76/893776/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/76/893776/4"}}},"commit":{"parents":[{"commit":"6e0805b17a7de3115ea21d2e37d79f238ea6ac48","subject":"Merge \"[DHCP agent] Fix route to OVN metadata port for non-isolated networks\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6e0805b17a7de3115ea21d2e37d79f238ea6ac48"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-08-09 21:34:34.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-12-08 20:31:49.000000000","tz":-300},"subject":"Metadata: handle process exceptions","message":"Metadata: handle process exceptions\n\nBoth metadata agents (OVN and non-OVN) should handle\nprocess exceptions when spawning haproxy processes\nsuch that the agent can continue its operation for\nother haproxy processes.\n\nNote:\nneutron/tests/unit/agent/dhcp/test_agent.py#L836\nAdded `mock.call.get_pid_file_name()` to the expected\nlist of process calls. This call was removed in stable/2031.1\nand later.\nhttps://review.opendev.org/c/openstack/neutron/+/903149\n\nCloses-Bug: #2033305\nChange-Id: I6da1b135c83ecfc41ec91e907ebf8500325a7a80\n(cherry picked from commit ed0515737be75745933f4994b5e6b652f9e0a2be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4698dd899fa149088160372b6f2efcd3ce973cbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4698dd899fa149088160372b6f2efcd3ce973cbb"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
