)]}'
{"id":"openstack%2Fneutron~894026","triplet_id":"openstack%2Fneutron~master~Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f","project":"openstack/neutron","branch":"master","attention_set":{"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2024-09-03 22:04:18.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_4694\u003e using the hovercard menu","reason_account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}}},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-12-15 11:32:42.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2023-12-15 13:10:12.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2023-12-15 13:10:12.000000000","reason":"Change was submitted"},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2023-12-15 13:10:12.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-12-15 13:10:12.000000000","reason":"Change was submitted"},"23804":{"account":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"last_update":"2023-12-15 13:10:12.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-12-15 13:10:12.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-11-02 20:56:19.000000000","reason":"\u003cGERRIT_ACCOUNT_8655\u003e replied on the change","reason_account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}}},"hashtags":[],"change_id":"Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f","subject":"[ovn] Add support for IPv6 metadata","status":"MERGED","created":"2023-09-07 09:08:56.000000000","updated":"2024-09-03 22:04:18.000000000","submitted":"2023-12-15 13:10:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":89,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"894026","meta_rev_id":"49950d12771ac14069224a090fb1dcd0688875d0","_number":894026,"virtual_id_number":894026,"owner":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":1,"date":"2023-12-08 22:50:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-15 13:10:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-12-15 11:32:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-12-15 11:43:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2023-12-15 11:43:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-07 09:11:36.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-09-07 09:47:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-07 12:13:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 14:11:41.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-09-08 07:10:18.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-08 15:49:25.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2023-09-08 18:40:38.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"CC"},{"updated":"2023-09-15 03:38:51.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2023-09-20 19:34:16.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2023-09-25 06:48:16.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-10-23 13:41:09.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2023-10-30 12:49:30.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-12-01 18:55:58.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-12-15 11:43:17.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"4fbb98c1cc1e3edba1803fdd1263e5173f1849fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-07 09:08:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95239d69bc5708351a438fe43a9abc7ec5ee8dee","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-07 09:10:08.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a795f55029aad0553f4e843e2b3fd44aa9940035","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 09:47:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05037c1bab9344e08c390ec0435599c8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba959429dc734911b6c409b50e0a96f4 : FAILURE in 19m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db07322d18ac49b3a794a56417cf63c0 : FAILURE in 19m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3affca54292d4a2ebf8b22c5d388575d : FAILURE in 19m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f9f0226385064afdb0c613c4b2e715a3 : FAILURE in 36m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e26a85ba1124850d87b616a0e75ed2f97d9ca7b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 12:13:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4ec861f7d5644a3a06d4f12ff2a0c16\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b8395308fa2a4e6995a0c0d2734484df : TIMED_OUT in 3h 03m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/352cc881cb2a49a89c4cabeccd59475e : FAILURE in 36m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03f0a6c9a3034000a3995b4bba7ade62 : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11ce74f7e6e04e189ae7210307a2d7aa : FAILURE in 28m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f64d10fbfb144dbda35d5142a4e5ecce : FAILURE in 29m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b04d89a0828e4355b936500d4820958a : FAILURE in 23m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0cb14cc63af74ff6ba622dc21de70133 : FAILURE in 24m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1bc6ccc2f2f4a48a2684f8ea185cfeb : SUCCESS in 12m 27s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a6b2b6ad876f4d258f9000d6bfd6aff9 : FAILURE in 28m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5411b43b38524860b7495cfa22b87d40 : FAILURE in 53m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/88a5a6917cf2481384b09aef00ea8b74 : FAILURE in 47m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eb394fc6bcde4eb4b8aae25bab1ece96 : SUCCESS in 1h 26m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9b4bac9456544ee88fc4759dc31314a0 : FAILURE in 1h 21m 35s","accounts_in_message":[],"_revision_number":1},{"id":"877af31c697a74da46a14c37986d064d6c95b066","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-07 13:42:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ede7a2a3e055c3c0830311e3cf94ebe5c2084cb4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-07 14:11:41.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5bbe7b69b7a9c646e3510d9edeb4d1715429b363","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-07 14:12:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"679d965231d8549b744a85c04744b92462a05a00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 14:25:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d41f56d47c3241dbbd006a1594536587\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b551474d48de4177acfe5a15ec86d484 : SUCCESS in 18m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e30bb5eb62004888b4f2c353c708c0ad : SUCCESS in 19m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f9d6c24b6cd348a5b5d06cba9e30d550 : SUCCESS in 41m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1351c3a9a6be4d29ac116725837697f0 : SUCCESS in 21m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5005e7296cac8bf6f6032f1f2b1cfa3904caa12f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-07 15:20:10.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"941cad2c571d7fbd8babfdc656e79b59068b4f17","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-07 15:21:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5fdb29b7db2f8187960459e2a2ad62db60729c4d","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-07 15:26:50.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"36ed3205a0d299eb0c88cce8f52be0e3f4fa3df7","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-07 15:29:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a7b2ef0a1da88cd29338f8caabd2853d9c9ecb92","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-07 15:33:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"acf01506d159538622220e17eaf5a4be106b382b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-07 16:42:40.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b0eafa9ea78f339ce5594edcd1610c3821013e36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 16:48:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/944ca7c7a8944cd3a2f3baa4fcf39bdf\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2d622aa37587407da1d92d9477477fcd : TIMED_OUT in 3h 04m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dfa8c2a543d41569c76803ad56af061 : SUCCESS in 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22587dcb553146079aa8e4a8f450be7d : SUCCESS in 10m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bbc5b732651430fa3f284d0caf569c3 : SUCCESS in 27m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b60e345ec6d74cd7b136572a8edc84e1 : SUCCESS in 24m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c03ffeebf3cd40b2b0e43320e65eb483 : SUCCESS in 19m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/605e1878d80a45e5a9705268f3cb2de2 : SUCCESS in 21m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a89f0cad92a4a2e97892ed36450377c : SUCCESS in 9m 51s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d81bbfbcfa854c31b777ba73838cd7a3 : SUCCESS in 21m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2b8554160c74c4da6dc05615eb02647 : SUCCESS in 1h 00m 09s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9121914900a848e8aee8f7ca19a69d72 : FAILURE in 49m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b2ea03553a0841888fd83789df4ffc2d : SUCCESS in 2h 01m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2ce9a486fa0546cf8e19021530fbbb1c : FAILURE in 1h 04m 22s","accounts_in_message":[],"_revision_number":2},{"id":"3775c96f9bbad28902809b6c83d748d81f0b9b20","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-07 20:37:07.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c62b128ab8a459d4a44f945d9d5179a679ac3de0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-07 20:38:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e81eaaff0d6945bc0ff91da64f0a07c166c5bdf2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 21:28:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89c3090e5ab047f88cde40c1bda9ac8e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f283c9072633456f833e147cca3f2470 : SUCCESS in 44m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fc3ec6fe5f14fb7a6843ddc6173f9cf : SUCCESS in 18m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/85ce05498c6c4c44b59679626d5f1712 : SUCCESS in 43m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5c540041780943eebd192e22f49479df : SUCCESS in 25m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e0d39d58f6f9c1c41d90e14900f5dddcd679b1df","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-07 22:14:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9ffdd8b98632b265710bd743b183d86b36d7c708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 22:32:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/603f827444cd46088e616f7140e445e3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/698b075bd59448ee824ca530edc22fe7 : FAILURE in 1h 52m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99b051f748a249d1b582e0e3dbf3c73e : SUCCESS in 39m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c4de68e92ac4f39b93ea26afe89caaa : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/750f811084814ccaae6d4de32b356ca5 : SUCCESS in 28m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49ce0eccce794057a1cd9d476a82cd74 : SUCCESS in 28m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e9a94a775ec4ad1b991caff8ee54f81 : SUCCESS in 28m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f94e8bcbb1c04090a620c945c476b9bb : SUCCESS in 23m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f9724dc1d01423d887059ba18de8767 : SUCCESS in 12m 09s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7cb40e25fd884773a053061b8dc46695 : SUCCESS in 20m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/42b233041cc644fc86e7a03f70cfc222 : SUCCESS in 56m 23s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a54676313052488fbeec4a442a208dc4 : SUCCESS in 56m 30s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b06ab4c456a146979300d97cf4aed36f : SUCCESS in 1h 15m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b7c218b43f304def8fdd5606d8aea093 : SUCCESS in 1h 27m 13s","accounts_in_message":[],"_revision_number":3},{"id":"704e9ade26a2bd9d9ecbc09a984d6e255b921b47","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-08 07:10:18.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/26/894026/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ddb4cb5a75d79c5eef9ef4345eee8e0afb03a41e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-08 07:41:07.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/26/894026/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"d322a7c3087cf15fd8b5d2722309e120d801a5ac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-08 08:11:22.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/26/894026/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"1b6a116bda6b2768969cdbf1d2921a234549d3ab","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-08 08:14:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f338f93645db2f64c8cd4666f0d4e354f18c40c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-08 14:01:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5662b36184d859b1936e43b6e10972b64f200a2f","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-08 18:40:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ea53a579aa840287353352124e8e04ec0a268e90","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-08 18:58:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"24316f84351cdc3fc044c279c670325ddedc1860","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-08 21:30:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d351d8eb4e5dfa6617128378b5ba4d60b072a6e5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-11 13:06:25.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6915e1cdfac56afaafa9a1388e2d7beac3df71a9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-13 09:46:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"275edc7c715e559d94231a64708488a1a5a30ed6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-13 15:37:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"50c477d21047645fcc0607f96af1d12f30a535c8","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-13 15:39:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8e6935407233897e995b22685a5e26c11766374f","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-13 15:41:59.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"17b471dc3244f10359dffde51bbe185952e943bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-13 15:58:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8d125e7d83598660bcac8938800d7849f7ab2ba2","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-13 15:59:46.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0a79d4508e605fec9fc28e4787ad2a0d71dc9801","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-13 16:00:04.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"68018eb8b1c5098d0cd75398e84e08a7cf2faaff","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-13 16:07:05.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/26/894026/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"70cb8fbfd42671dc9ace858828c89f56e24d87f0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-13 16:37:35.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/26/894026/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"89baf0dd4decf51c92f75f3e93a29b45c3059140","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 17:31:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d99acf7cdffe4730971d28745ac565c1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c813dcbd0bad40bebfe5abe0a3909537 : SUCCESS in 1h 06m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d675dca567584611af9fbc061e393871 : SUCCESS in 1h 03m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4cf969d4a1b4ca6898e5e0ba84f5362 : SUCCESS in 1h 04m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1c99100d19f24c8191454cd29f749850 : SUCCESS in 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d84ad1e5fb65d6f5fc4511bde1300851f2ab8bd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 18:33:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24b383e931ca4338b77ce17d3ef6bea8\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3ad0020796bd469a9d86c0adfda5741b : FAILURE in 2h 32m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0396bba961b40eab629f85878dd0831 : SUCCESS in 38m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d8dce9e6031419fb321ddc34f6bbc63 : SUCCESS in 11m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58a77e2e4fdf4dae9e408f219e4f1483 : SUCCESS in 24m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/714f3396ec9c476a9e6b3c9467a93e47 : SUCCESS in 24m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47a2c44631cf43d08f2f34e158d14ade : SUCCESS in 25m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4439dfab95bb48fb9a011f05c5e98f6f : SUCCESS in 20m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d149381bf2e843cc8b7fc04583e94b91 : SUCCESS in 12m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b903d30ee32454aa89ce9bc9f034080 : SUCCESS in 12m 03s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b56cb720a6ad448281eb10e2c9407b0d : SUCCESS in 25m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b97f16020f3b4f4da364e2f5817d912f : SUCCESS in 41m 47s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4114b32be0d340ccb9be11b57d8cd8dd : SUCCESS in 46m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9d2e300ca78240d58b617b3e706d8315 : SUCCESS in 1h 46m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a0d6f2f55dda4977a98aa6862e255215 : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":5},{"id":"a1a0e421359646e785a07d7a6c4bd41a37859793","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-14 05:29:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b43bed36d1c05b6158f6c6fb323ea7bf199d3524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-14 05:37:06.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"4469ec81226afc4a73e2056069ea0875803fb5c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 06:29:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a025e04699456aa286bb68e5fe29d6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/40034a9c902e4e7b904fdf2cd6b4a949 : SUCCESS in 49m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e14cd1b026a14ae3bef2a856d9290a94 : SUCCESS in 50m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/884ca4cc124d4cd7bf5a75834e93753b : SUCCESS in 49m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/57397d7c8f6447288f86860cc952165b : SUCCESS in 44m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"45e632959072326c85ae87e3d822036369bc654b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-14 06:34:19.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/26/894026/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"08968a118270accd6a9bb9147c8672d8f0704445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 07:48:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a384851bb949ff9d91915bdb70d73c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/52f5c3d4589b4a36a8b3baab9987f3c7 : SUCCESS in 1h 38m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d57f722c59a449c58f1df8e24ad9e779 : SUCCESS in 40m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b764b7cc8d4bceb72c11c5bb6c02d4 : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ba0a69320664182a70cfbe324d225e1 : SUCCESS in 23m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cc65c5bdc3b4ec48bb0ed8674c898c5 : SUCCESS in 28m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1b5bba5b9544c47be170e9a028cd6cf : SUCCESS in 25m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1e2144e4c804319bbe3418cfbe2a6c8 : SUCCESS in 19m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d29c0c698f37400b98f265ffea01ffc2 : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15f09dbe87e54173898d88acb89a3da4 : SUCCESS in 8m 56s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/07cbfbecc24a489e892e177e033ade76 : SUCCESS in 30m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c54df55fb6a340698e8499dc7c43aacd : SUCCESS in 1h 01m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/81e7ef63628a4b1989c2fb10758211c3 : SUCCESS in 57m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ed1da9661d994ce0adfa6855b9d19de3 : SUCCESS in 2h 03m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/309d20cee4ce46a0818e670a8191b9d2 : SUCCESS in 1h 18m 17s","accounts_in_message":[],"_revision_number":6},{"id":"f98df91a128b8494b7f7a8eb4314c6b4d2b0c561","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-14 16:33:27.000000000","message":"Patch Set 6: Code-Review+2 Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"f94367d5151258204b15a91697b5ed5cd90f167f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-14 22:21:59.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"5213e046045be433fa140fcb74a2b0ceefb70c9f","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-15 03:38:51.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e1caf38d08aefdc6ece10a05686a25e040a801fb","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-15 15:34:01.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"38806c6ea0ff2842a7d6845bf4e3d3ee8552703c","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2023-09-15 15:39:16.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"047660e2a23ed19eb1a42f04a7735caec6255fa9","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-15 20:06:21.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6cd3a354d206f3b634fe135f9ad70f7f67785c01","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-20 19:34:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"848743702519bad7885f99204ac26c381c1a9220","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-20 21:05:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5bd7b570ea8a3d5870fb6f5c035515483ea0263f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-21 16:08:36.000000000","message":"Patch Set 6: -Workflow","accounts_in_message":[],"_revision_number":6},{"id":"c3f19d287d1c3ba6864c2f752bb77ff227211437","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-09-25 06:48:16.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"11fd7f21e06f566198cb5c178be753d92cb19ed0","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-10-04 12:03:32.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c2f29a88cbd282a509d3f40ca0f6e50036466563","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-10-25 10:48:19.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3e7bb1bcad6bf6dff620a66b35359e8e43bdefb2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-25 23:26:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"66c657f3adbe101fd39f4cc89798a21602f9362b","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-10-26 07:59:34.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"704d0d018a406aff5bde5611b7b96366c1cb57e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 19:00:18.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"9e8090f731eaf76aeaf2c2d020c223e50e44cc0f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 19:00:36.000000000","message":"Patch Set 6: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"659495703772574c77ddba836ed23d806c838d83","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-27 19:31:53.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/26/894026/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"9d21a08556b4fe76b14dab0cb8dbe85d0399eea5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 20:01:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1073e88244048f898184f15385aaeb3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c187a52ac134e53ae07922a0c1bce63 : SUCCESS in 58m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ba493c70f8c488aa1d317e5793d62f4 : SUCCESS in 59m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/188b74abe72641edbb7f6e9dce7da57f : SUCCESS in 50m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f722c4265a88e3df88b62d90d4ecc4dee3d7f334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 20:43:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/077f89d4ccea4bb680c70ff35f6aa3a4\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f2e7b5e71c5e4a90ba895eb1626c29fe : SUCCESS in 1h 29m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a9744291710498d8936327b7916c7ea : SUCCESS in 29m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63c70f46b10f4e6fa41db91ef780ca7e : SUCCESS in 9m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/866c103d8f914181b0d804b89ffd9291 : SUCCESS in 29m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/442f6db69da947599e7a7e4966cc9e1d : SUCCESS in 19m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5bb048af7a345d495a38332b9e360d4 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91823105d2dd4fc98da06940b853652e : SUCCESS in 12m 01s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/4f3be3381fb94674978477db874123eb : SUCCESS in 33m 58s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/6d66f2539fda4cc3ba4dfb46d502ace3 : SUCCESS in 53m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a88bc925f3e745268fd420eb59c89caa : FAILURE in 46m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c181d528497040c1a4f4e85262990232 : SUCCESS in 1h 03m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7a704fb08daa4753b9c4d3601fd6d613 : SUCCESS in 1h 41m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/86ad51588abe447ab6d7eb5284652377 : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":7},{"id":"7934e406f8f2b9e30a368752dde57467c91023c1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-27 21:23:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c45183493139aafc59233714b9e2ae5bd3d40307","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 22:06:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2253813dfdc4a0c830099b7ac350c6f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00d49a8267694c339a0d95281b8f7926 : SUCCESS in 37m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fcc1c4937b81492cac856ed6abca52b6 : SUCCESS in 40m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fbc02c4e54454b03af9f823108afbfbc : SUCCESS in 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1c25dd32af7f4e6422c63c01072e3e861bfd74fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 23:11:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb97379d6d174153b49d4e9f8555d29e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bbedd303647d4012ad9f78d866d8e924 : SUCCESS in 1h 46m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd05f474f1e04462be6d252702375262 : SUCCESS in 32m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/366d1a1bdc3a4fe4b7bacb78d7b9629f : SUCCESS in 8m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f16d0a658f9f4bd9a3cf76b18e71df6a : SUCCESS in 20m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/553eaa72e8d648b3aedaacfeedc0563e : SUCCESS in 17m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f9d4c9394964b4a9821cfc4e8bc2f97 : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0417892425e46ed97efeb7db9eee655 : SUCCESS in 12m 15s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/315142522bbe4855ba87494d9f2d2ab7 : SUCCESS in 23m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/f88eaffa5835482bbc937163c9867598 : SUCCESS in 47m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d67245a71bd84b4ca13f6900c57b25a7 : SUCCESS in 54m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/08d401c5e0854b0b84ef985c8f7feab9 : SUCCESS in 45m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0b596cd1afcc4f9488b3a613966fe804 : SUCCESS in 1h 24m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/617f02bd8c2b4ac393284ebf3c8cbf30 : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":7},{"id":"0726e95db276ec2cd1323717e9b1376227003cb8","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-30 12:49:30.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"feaa0a4eda85fa0057cd77b67e2aae57e59182d1","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-30 12:50:17.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5e7bfd271d8f03f520369de36caa1c5b36d82ddd","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-30 12:56:51.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bf934b6d2cc6124f36f50533615e96de9fe4c97b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-30 17:23:07.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"5f9ee258b7cf71e7013246c32b8ad5a39520d748","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-10-30 20:13:57.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"d9cac8fbc16883381d09d16c69a43485a86a7e6d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-30 20:44:18.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e3d83df65599284ae79bb41190872e2f6c139f0c","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-31 21:56:12.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c2ba8d0c452193134c5df5aada89740f726ab9bd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-01 01:47:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e6f4d85808e0306dfbdbe00e84d4696ac7082e74","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-11-02 20:56:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fdbff5f4f30d825f6f0b1584a255197089fc8f9f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-05 23:46:41.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"074b405b00133c3244eec74196af182291a95a0d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-06 21:49:00.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"25631b524cbc8b9286dd3d516969846f576741b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-06 21:53:14.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"82bc326b4a6783f52770db20695629cddd98eee1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-06 21:54:09.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c367c0c1e28a91e33e462ce06752bba8bc3d7d48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 22:20:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/264a9c5e083e4492b2e9396e7e27e12c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/efdfd0532eff4dc8830de2d427981e36 : SUCCESS in 18m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74439708171b4aedba01971c45b8d9c8 : SUCCESS in 26m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cb29e8d5ad694448b992b1fb9bbef4f2 : SUCCESS in 18m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3e2bef3b60e671cc9295b47a60a99a83c9ca0420","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-06 22:26:34.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/26/894026/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"baf483e249335f24662e158f3acd78332f42569a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 23:59:28.000000000","message":"Patch Set 8: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/492ecfe1b5c146bdaf929f2afeb37cfd\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0d7c83f1c6be44fc9022c1578de026a2 : SUCCESS in 1h 28m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/656b5bf912224a2cb422dd8f5caaf104 : SUCCESS in 40m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa476986c0734072979ebca0093321ae : FAILURE in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54f4171457164cda944e1ddd79313929 : SUCCESS in 27m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79c9752600304e639bec073f05d1658d : SUCCESS in 23m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5048c12e400408f944c8ae6b1403cf3 : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2582ac9a7a0448ee8348ec8a29ad851c : SUCCESS in 16m 05s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/5a4183c8465a4d39937566a4583cc888 : SUCCESS in 28m 46s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/d743d0e0ec4f4be4a34674d3af8b00fa : SUCCESS in 1h 11m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/af91d7d84e304c449fc9f00b9d9fa3d8 : SUCCESS in 1h 02m 49s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/06781a153f064ad89fd555c729a852ff : SUCCESS in 1h 04m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d32e577ef962489f9020669e79af6eb0 : SUCCESS in 1h 59m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3671e15fc47d457393890f13144178e4 : SUCCESS in 1h 00m 17s","accounts_in_message":[],"_revision_number":8},{"id":"10ce7cc1ab68547a4742f70973273e75cd410a1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-07 18:43:26.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"c763c9875a612da4695f123fb0e71fad91e6bb2e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-07 19:21:21.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/26/894026/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"e0e3f68d86e0dd2ed12ac25371ec7552baa7d99d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 19:23:59.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8408cd35c0c546bf84f63a6d4658e7f5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5cde59847f6493a91e38c96e405402f : SUCCESS in 19m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/636a0f98cb28464194f836feb18f3536 : SUCCESS in 18m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e9cdb3b395404cf38388213dd0ae063e : SUCCESS in 38m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2475162ac74c7b127cf134c2e5e7e58a416545c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 20:56:42.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc64fa35b1044f73bd7476b4263c5a91\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f96fc248fa274cb790fa33e16d988138 : SUCCESS in 1h 46m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaf4e77f869640b4b3360393e8d6afa7 : SUCCESS in 38m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c902c618ce244f50b4aad023709dc8e9 : SUCCESS in 11m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be902f0ee4fc4988894d5f9a51022351 : SUCCESS in 26m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bcf8a2eec666403185a0d3e5220c646d : SUCCESS in 19m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224311c59e38472a9e457e63bb29c1ff : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2029bc5c4bfa441fb064a1c82cadbd4b : SUCCESS in 12m 39s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/24a90fb0179540fc801fbb7ea4c15a5f : FAILURE in 7m 06s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/1d5df0680be14cdd89b2057647fc3a7f : SUCCESS in 54m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d83435774c3c4694a7b6688cac396971 : SUCCESS in 1h 01m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/21fc2ae104ca4f2a89ad91641de41dea : SUCCESS in 57m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f23b3569de1d4698a290813f9137d958 : SUCCESS in 2h 07m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bb2e4f415260449285d05b98b19bd450 : SUCCESS in 1h 14m 56s","accounts_in_message":[],"_revision_number":9},{"id":"1c09383eee13cc25e5a428261fb35c519da3b6dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-15 00:11:13.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"2578f3fe928e1501d122294bfda48c64234197ff","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-15 00:43:01.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/26/894026/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"abb42bbaced1ea273388aff682b5bfb528bcba54","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-15 00:46:20.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e1ad1703130e6fc6458850b0693d9736e161e86f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 00:56:40.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cc9d7bdd0754673b93fb6898356d084\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4676a85ffcd5443eb47243136c527ccb : SUCCESS in 20m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c03d9f72ad5542c1ae07b3d7e24f80f1 : SUCCESS in 43m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cfee493bcceb43a68d0f6d30d9bf5d99 : SUCCESS in 17m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"27864e6e2df21a4708e1887c11cda84c55038c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 02:18:06.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21213912615e4113b20a074f185fdce4\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/47a114330693408da32f2ecd09db0479 : SUCCESS in 1h 21m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e74bfe240d04074bac3b4fe6f3425ac : SUCCESS in 35m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d375ef30a811402cb5a1a7f4c521f194 : SUCCESS in 11m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24c46a01304644e7819e97eca63bc7c6 : SUCCESS in 24m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b20e1b7a6187414385ca13bed9589373 : SUCCESS in 30m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816b79d94c1e4a7596e3cb091b492f0a : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d1ab0ee3ffc4f19b6a5d7274f098d00 : SUCCESS in 19m 04s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/74b6fb157c6f4ff58d0d8bf529684eb2 : SUCCESS in 30m 21s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/fa68b4de4fd14bc8ad8679f490747acc : SUCCESS in 1h 10m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b1b5206d8e341b286bd55a0ffe780f9 : SUCCESS in 1h 02m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e9329cff55f343e4823999ab5189a142 : SUCCESS in 1h 04m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7f9465a87ebe4520bd0393f72b04dac3 : SUCCESS in 1h 58m 23s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7f272a221fb149cbb5182b8e9208c0cc : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":10},{"id":"8985617968efec7b2bbce036e330e162c4cc9789","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-11-21 09:44:17.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4302d00f82b8fb3e91f88a6573cb2c136ad475e7","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-21 15:32:45.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b5f76eb5290ce92b3bbd634051612b3df17705e4","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-30 19:20:26.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"019c3678007da7480d22e7a3138548235a54877d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-30 19:20:35.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"d769d5a08722aab500ddf2ce54de82e0360cdde0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-30 19:52:27.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/26/894026/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"7eddfe873a202ee7b24b528c5fdec416d36ecfc9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 19:52:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14be7bfdf56246a6b29b0546bdc56e6f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/135ab0731b564554a9c0f21aa8562be3 : SUCCESS in 18m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2aa25f01468540639b902c8f64ef5b88 : SUCCESS in 17m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/973b7eb628a54deda42776e2aae1428d : SUCCESS in 30m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b8ef8e53b56d51b7f1807b102f21b8b053970135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 21:17:58.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/801a9b3b49be49a8892b806d23c8d74c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/430e4f84a0bb48669a9c7c922df7c535 : SUCCESS in 1h 55m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/134b7e795b7a4d1cafbc25a370e139c0 : SUCCESS in 37m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d5e0f65cdc48718029edf43f0aafe0 : SUCCESS in 8m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7626c885871b4cccbcfc1a25ab41a1bc : SUCCESS in 23m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12ee718b5e94496cb83e59b205cc2195 : SUCCESS in 25m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d68206e21f443c4ba91e9b2ff855d11 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/282db86d031a4d2c9abc83f261fa9152 : SUCCESS in 14m 27s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/bf4c32a71ab9437d8c66fcca63b9aab0 : SUCCESS in 20m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/0dccf9d57498460e9f25ef59ca6498f9 : SUCCESS in 1h 09m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8e89b24138a4664b515c2256254ff21 : SUCCESS in 57m 18s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f128ff4eae5d487c8a3430aaaf123d27 : SUCCESS in 1h 06m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/56fa4766a5484c1699a2065ac58961ef : SUCCESS in 1h 26m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/507a8ed5b77947d9a726a41aea352578 : SUCCESS in 1h 09m 13s","accounts_in_message":[],"_revision_number":11},{"id":"51be68c60684d39626e21951ce1a01867e0849c2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-01 15:35:26.000000000","message":"Patch Set 11: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"38a63ea671dc0605d557521b5e0b083bbe51479c","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-12-01 18:55:58.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"0e445995748301802539d5bbcffa2b9da590b935","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2023-12-04 15:06:13.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"bf500511cd0e288c0da96a609af9f95ceb34bda9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-05 14:09:35.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"4fc8ee7da72155ce2236c230e5dc6d3825abf160","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-08 22:19:11.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"a39d63355d50052beec2183a6f7dbeb1d4a71f04","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-08 22:19:40.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"382868a82ab84ec0f76801449b39028aa87df1d7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-12-08 22:50:18.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2023/Arista_Tempest_Testrefs/changes/26/894026/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"1cd1c4baeb9604e99a2a491cc7681633f780b64c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 23:05:41.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/209b4ffdf8194dd29b3ba49922ee63c3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d74f4d2ef8d04b65b7e5acab3bfc837d : SUCCESS in 18m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3b9a0407607842408e2dc9a5115411bc : SUCCESS in 43m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/adb4dcc9940c4970921f8097e52b6da7 : SUCCESS in 15m 31s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f91cddb88ccab2b474d87cdf7c34a7d6cbf8465e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 00:15:48.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e26d4da3f74409dbf57b31ef33f15ca\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/40b06e5ef08644d1a45121820296c698 : SUCCESS in 1h 50m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2d66b7645604149a6e4bd3959725d3e : SUCCESS in 41m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9b380bee6b84a4fa51b3e96f26c7040 : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ba371bb7a5c40938c13594165802744 : SUCCESS in 26m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4220dc6d94d4686b1f2441d01f65f3c : SUCCESS in 27m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1da4847ff034c4ba4beec2d7a892b5e : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1679fdeb540f4d31a1721f0dc6051eb9 : SUCCESS in 17m 04s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/bf7ec4d0f5bf436983f7f54e501e49d5 : SUCCESS in 27m 39s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/96286f8a26594820bd2728ef65f0e385 : SUCCESS in 1h 08m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a88528970a0443efa08bc62295ac43b3 : SUCCESS in 40m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/0e6d431406524220847923fda1f6b7a7 : SUCCESS in 1h 11m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6d5b801a99164ddd96644f5fbbd782fa : SUCCESS in 1h 18m 14s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/72f29fe4549f4a7697b364c1f2c5b27e : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":12},{"id":"964c6f9fae8be81e6b78c85285dea90c301b69d5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-15 11:32:42.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"6c28d9a4ad0450569554df12386f6fab75f66ee0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-12-15 11:43:17.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"100aeb85135e8c0eb25454e93f3ee7c28d3c74b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 11:43:50.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"e3ec719c2de0b27656ff4e967917417131679751","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 13:10:12.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51a8b934ef974c019ad031c5d55861e7\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/52803a8cd3e448c8aa4d251f7c7f8668 : SUCCESS in 1h 25m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7274e7698574da4925561011dca0328 : SUCCESS in 11m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5924072df60644358294a5eebb7a93bd : SUCCESS in 19m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b198612729414461b5dbc9b0da282ae5 : SUCCESS in 23m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f484a2d06306468a859aa885384f21a1 : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a78a81d2f5a4d149ac76b16e38b5d4c : SUCCESS in 19m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d1d014285f249a7983ffbfdae901392 : SUCCESS in 59m 51s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7e189366e289494f990c06e69a5215b9 : SUCCESS in 43m 41s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4938f88d3ba9434bbb656fdbcbe44aa6 : SUCCESS in 54m 21s","accounts_in_message":[],"_revision_number":12},{"id":"2c71a237db5a50890b19ee9cc4db47fe2d73df25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 13:10:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"d8c860ebd941df5cb1d83e1d77c2749e90324b6c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 13:11:45.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ed15c8119064c85811362b1205da48b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/835fb22ebaaf42bdb4582317757334d3 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74de27914d8841839b5afa1add08204a : SUCCESS in 50s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"d9c8731af36d4eb53d9266733fec24659f2dc5a8","revisions":{"6c507655131e6b0cf79ee81b2c2f26193ac12d37":{"kind":"REWORK","_number":1,"created":"2023-09-07 09:08:56.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/26/894026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/894026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/894026/1"}}},"commit":{"parents":[{"commit":"f7489abaf822574f6d8c01b72495788e07d756d3","subject":"[UT] Adjust autogen_process_directives for alembic-1.12.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7489abaf822574f6d8c01b72495788e07d756d3"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"subject":"[WIP][ovn] Add support for IPv6 metadata","message":"[WIP][ovn] Add support for IPv6 metadata\n\nTODO: add tests\n\nChange-Id: Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c507655131e6b0cf79ee81b2c2f26193ac12d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c507655131e6b0cf79ee81b2c2f26193ac12d37"}]},"branch":"refs/heads/master"},"fb154c6663b6b60fd417511c0100847d0202d44b":{"kind":"REWORK","_number":2,"created":"2023-09-07 13:42:29.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/26/894026/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/894026/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/894026/2"}}},"commit":{"parents":[{"commit":"f7489abaf822574f6d8c01b72495788e07d756d3","subject":"[UT] Adjust autogen_process_directives for alembic-1.12.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7489abaf822574f6d8c01b72495788e07d756d3"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 13:33:23.000000000","tz":0},"subject":"[WIP][ovn] Add support for IPv6 metadata","message":"[WIP][ovn] Add support for IPv6 metadata\n\nThis patch adds support for IPv6 metadata service in ML2/OVN.\nThe changes include:\n\n- Add the \u0027fe80::a9fe:a9fe/64\u0027 address to the interface of the\n  ovnmeta- namespace so that it\u0027s reachable from the guests\n\n- Identify the port of the VM by looking up the source MAC address\n  of the metadata request\n\n- Restarts the haproxy instances to honor the configuration changes\n  upon start of the metadata agent. In particular, haproxy now also\n  binds on the \u0027fe80::a9fe:a9fe\u0027 address\n\nWhen the VM requests metadata from its LLA, the traffic will reach\nthe ovnmeta namespace associated to its network. This patch also\nuses a \u0027/64\u0027 (vs /128) CIDR for the metadata IPv6 address so that\nit can communicate to the VIF without adding any static routes.\n\nChange-Id: Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb154c6663b6b60fd417511c0100847d0202d44b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb154c6663b6b60fd417511c0100847d0202d44b"}]},"branch":"refs/heads/master"},"5b2f37a1b851f635eab020e18981370ba130db0b":{"kind":"REWORK","_number":3,"created":"2023-09-07 20:38:14.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/26/894026/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/894026/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/894026/3"}}},"commit":{"parents":[{"commit":"f7489abaf822574f6d8c01b72495788e07d756d3","subject":"[UT] Adjust autogen_process_directives for alembic-1.12.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7489abaf822574f6d8c01b72495788e07d756d3"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 20:37:56.000000000","tz":0},"subject":"[ovn] Add support for IPv6 metadata","message":"[ovn] Add support for IPv6 metadata\n\nThis patch adds support for IPv6 metadata service in ML2/OVN.\nThe changes include:\n\n- Add the \u0027fe80::a9fe:a9fe/64\u0027 address to the interface of the\n  ovnmeta- namespace so that it\u0027s reachable from the guests\n\n- Identify the port of the VM by looking up the source MAC address\n  of the metadata request\n\n- Restarts the haproxy instances to honor the configuration changes\n  upon start of the metadata agent. In particular, haproxy now also\n  binds on the \u0027fe80::a9fe:a9fe\u0027 address\n\nWhen the VM requests metadata from its LLA, the traffic will reach\nthe ovnmeta namespace associated to its network. This patch also\nuses a \u0027/64\u0027 (vs /128) CIDR for the metadata IPv6 address so that\nit can communicate to the VIF without adding any static routes.\n\nChange-Id: Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b2f37a1b851f635eab020e18981370ba130db0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b2f37a1b851f635eab020e18981370ba130db0b"}]},"branch":"refs/heads/master"},"55557dbff2ef61169a6fdd52cb42cc003d729b6c":{"kind":"REWORK","_number":4,"created":"2023-09-13 15:37:05.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/26/894026/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/894026/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/894026/4"}}},"commit":{"parents":[{"commit":"f7489abaf822574f6d8c01b72495788e07d756d3","subject":"[UT] Adjust autogen_process_directives for alembic-1.12.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7489abaf822574f6d8c01b72495788e07d756d3"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-13 15:28:47.000000000","tz":0},"subject":"[ovn] Add support for IPv6 metadata","message":"[ovn] Add support for IPv6 metadata\n\nThis patch adds support for IPv6 metadata service in ML2/OVN.\nThe changes include:\n\n- Add the \u0027fe80::a9fe:a9fe/128\u0027 address to the interface of the\n  ovnmeta- namespace so that it\u0027s reachable from the guests\n\n- Identify the port of the VM by looking up the source MAC address\n  of the metadata request\n\n- Restarts the haproxy instances to honor the configuration changes\n  upon start of the metadata agent. In particular, haproxy now also\n  binds on the \u0027fe80::a9fe:a9fe\u0027 address\n\nWhen the VM requests metadata from its LLA, the traffic will reach\nthe ovnmeta namespace associated to its network.\n\nThe IPv6 metadata tests are passing and enabled in Tempest by\nthis patch:\nhttps://review.opendev.org/c/openstack/neutron-tempest-plugin/+/894027\n\nBesides, this patch ensures that the link-local address of the\nmetadata interface is present so that the metadata IPv6 endpoint\nis reachable. It also fixes a bug that was causing the wrong LLA\nto be present as the interface was set `up` first prior to changing\nthe MAC address. Now this order is inverted so that the proper LLA\nis configured.\n\nChange-Id: Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55557dbff2ef61169a6fdd52cb42cc003d729b6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/55557dbff2ef61169a6fdd52cb42cc003d729b6c"}]},"branch":"refs/heads/master"},"10d49d65cc87990697d42c6885287fdfa1f99e12":{"kind":"REWORK","_number":5,"created":"2023-09-13 15:58:38.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/26/894026/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/894026/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/894026/5"}}},"commit":{"parents":[{"commit":"f7489abaf822574f6d8c01b72495788e07d756d3","subject":"[UT] Adjust autogen_process_directives for alembic-1.12.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7489abaf822574f6d8c01b72495788e07d756d3"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"committer":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-13 15:57:49.000000000","tz":0},"subject":"[ovn] Add support for IPv6 metadata","message":"[ovn] Add support for IPv6 metadata\n\nThis patch adds support for IPv6 metadata service in ML2/OVN.\nThe changes include:\n\n- Add the \u0027fe80::a9fe:a9fe/128\u0027 address to the interface of the\n  ovnmeta- namespace so that it\u0027s reachable from the guests\n\n- Identify the port of the VM by looking up the source MAC address\n  of the metadata request\n\n- Restarts the haproxy instances to honor the configuration changes\n  upon start of the metadata agent. In particular, haproxy now also\n  binds on the \u0027fe80::a9fe:a9fe\u0027 address\n\nWhen the VM requests metadata from its LLA, the traffic will reach\nthe ovnmeta namespace associated to its network.\n\nThe IPv6 metadata tests are passing and enabled in Tempest by\nthis patch:\nhttps://review.opendev.org/c/openstack/neutron-tempest-plugin/+/894027\n\nBesides, this patch ensures that the link-local address of the\nmetadata interface is present so that the metadata IPv6 endpoint\nis reachable. It also fixes a bug that was causing the wrong LLA\nto be present as the interface was set `up` first prior to changing\nthe MAC address. Now this order is inverted so that the proper LLA\nis configured.\n\nChange-Id: Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10d49d65cc87990697d42c6885287fdfa1f99e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/10d49d65cc87990697d42c6885287fdfa1f99e12"}]},"branch":"refs/heads/master"},"1bd41edf5b617e61a47984d17feacf755caa7fbf":{"kind":"REWORK","_number":6,"created":"2023-09-14 05:37:06.000000000","uploader":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"ref":"refs/changes/26/894026/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/894026/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/894026/6"}}},"commit":{"parents":[{"commit":"f7489abaf822574f6d8c01b72495788e07d756d3","subject":"[UT] Adjust autogen_process_directives for alembic-1.12.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7489abaf822574f6d8c01b72495788e07d756d3"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"committer":{"name":"Daniel Alvarez","email":"dalvarez@redhat.com","date":"2023-09-14 05:37:06.000000000","tz":0},"subject":"[ovn] Add support for IPv6 metadata","message":"[ovn] Add support for IPv6 metadata\n\nThis patch adds support for IPv6 metadata service in ML2/OVN.\nThe changes include:\n\n- Add the \u0027fe80::a9fe:a9fe/128\u0027 address to the interface of the\n  ovnmeta- namespace so that it\u0027s reachable from the guests\n\n- Identify the port of the VM by looking up the source MAC address\n  of the metadata request\n\n- Restarts the haproxy instances to honor the configuration changes\n  upon start of the metadata agent. In particular, haproxy now also\n  binds on the \u0027fe80::a9fe:a9fe\u0027 address\n\nWhen the VM requests metadata from its LLA, the traffic will reach\nthe ovnmeta namespace associated to its network.\n\nThe IPv6 metadata tests are passing and enabled in Tempest by\nthis patch:\nhttps://review.opendev.org/c/openstack/neutron-tempest-plugin/+/894027\n\nBesides, this patch ensures that the link-local address of the\nmetadata interface is present so that the metadata IPv6 endpoint\nis reachable. It also fixes a bug that was causing the wrong LLA\nto be present as the interface was set `up` first prior to changing\nthe MAC address. Now this order is inverted so that the proper LLA\nis configured.\n\nChange-Id: Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bd41edf5b617e61a47984d17feacf755caa7fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1bd41edf5b617e61a47984d17feacf755caa7fbf"}]},"branch":"refs/heads/master"},"bf80fced1b42940e594e9117db9b72bc34271306":{"kind":"REWORK","_number":7,"created":"2023-10-27 19:00:18.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/894026/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/894026/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/894026/7"}}},"commit":{"parents":[{"commit":"f7489abaf822574f6d8c01b72495788e07d756d3","subject":"[UT] Adjust autogen_process_directives for alembic-1.12.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7489abaf822574f6d8c01b72495788e07d756d3"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-10-27 18:59:31.000000000","tz":-240},"subject":"[ovn] Add support for IPv6 metadata","message":"[ovn] Add support for IPv6 metadata\n\nThis patch adds support for IPv6 metadata service in ML2/OVN.\nThe changes include:\n\n- Add the \u0027fe80::a9fe:a9fe/128\u0027 address to the interface of the\n  ovnmeta- namespace so that it\u0027s reachable from the guests\n\n- Identify the port of the VM by looking up the source MAC address\n  of the metadata request\n\n- Restarts the haproxy instances to honor the configuration changes\n  upon start of the metadata agent. In particular, haproxy now also\n  binds on the \u0027fe80::a9fe:a9fe\u0027 address\n\nWhen the VM requests metadata from its LLA, the traffic will reach\nthe ovnmeta namespace associated to its network.\n\nThe IPv6 metadata tests are passing and enabled in Tempest by\nthis patch:\nhttps://review.opendev.org/c/openstack/neutron-tempest-plugin/+/894027\n\nBesides, this patch ensures that the link-local address of the\nmetadata interface is present so that the metadata IPv6 endpoint\nis reachable. It also fixes a bug that was causing the wrong LLA\nto be present as the interface was set `up` first prior to changing\nthe MAC address. Now this order is inverted so that the proper LLA\nis configured.\n\nChange-Id: Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf80fced1b42940e594e9117db9b72bc34271306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf80fced1b42940e594e9117db9b72bc34271306"}]},"branch":"refs/heads/master"},"94d3a9718466788a82d44f6ffc451ae7cefd9713":{"kind":"REWORK","_number":8,"created":"2023-11-06 21:53:14.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/894026/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/894026/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/894026/8"}}},"commit":{"parents":[{"commit":"f99da4d15c8efd74e7466e6585c17bbd441afc7a","subject":"Merge \"get_hosts_mapped_with_segments add filter agt_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f99da4d15c8efd74e7466e6585c17bbd441afc7a"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-06 21:52:58.000000000","tz":-300},"subject":"[ovn] Add support for IPv6 metadata","message":"[ovn] Add support for IPv6 metadata\n\nThis patch adds support for IPv6 metadata service in ML2/OVN.\nThe changes include:\n\n- Add the \u0027fe80::a9fe:a9fe/128\u0027 address to the interface of the\n  ovnmeta- namespace so that it\u0027s reachable from the guests\n\n- Identify the port of the VM by looking up the source MAC address\n  of the metadata request\n\n- Restarts the haproxy instances to honor the configuration changes\n  upon start of the metadata agent. In particular, haproxy now also\n  binds on the \u0027fe80::a9fe:a9fe\u0027 address\n\nWhen the VM requests metadata from its LLA, the traffic will reach\nthe ovnmeta namespace associated to its network.\n\nThe IPv6 metadata tests are passing and enabled in Tempest by\nthis patch:\nhttps://review.opendev.org/c/openstack/neutron-tempest-plugin/+/894027\n\nBesides, this patch ensures that the link-local address of the\nmetadata interface is present so that the metadata IPv6 endpoint\nis reachable. It also fixes a bug that was causing the wrong LLA\nto be present as the interface was set `up` first prior to changing\nthe MAC address. Now this order is inverted so that the proper LLA\nis configured.\n\nChange-Id: Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94d3a9718466788a82d44f6ffc451ae7cefd9713"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/94d3a9718466788a82d44f6ffc451ae7cefd9713"}]},"branch":"refs/heads/master"},"543835fe1b1730b59ef44096c55573075b558aa3":{"kind":"REWORK","_number":9,"created":"2023-11-07 18:43:26.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/894026/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/894026/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/894026/9"}}},"commit":{"parents":[{"commit":"f99da4d15c8efd74e7466e6585c17bbd441afc7a","subject":"Merge \"get_hosts_mapped_with_segments add filter agt_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f99da4d15c8efd74e7466e6585c17bbd441afc7a"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-07 18:43:16.000000000","tz":-300},"subject":"[ovn] Add support for IPv6 metadata","message":"[ovn] Add support for IPv6 metadata\n\nThis patch adds support for IPv6 metadata service in ML2/OVN.\nThe changes include:\n\n- Add the \u0027fe80::a9fe:a9fe/128\u0027 address to the interface of the\n  ovnmeta- namespace so that it\u0027s reachable from the guests\n\n- Identify the port of the VM by looking up the source MAC address\n  of the metadata request\n\n- Restarts the haproxy instances to honor the configuration changes\n  upon start of the metadata agent. In particular, haproxy now also\n  binds on the \u0027fe80::a9fe:a9fe\u0027 address\n\nWhen the VM requests metadata from its LLA, the traffic will reach\nthe ovnmeta namespace associated to its network.\n\nThe IPv6 metadata tests are passing and enabled in Tempest by\nthis patch:\nhttps://review.opendev.org/c/openstack/neutron-tempest-plugin/+/894027\n\nBesides, this patch ensures that the link-local address of the\nmetadata interface is present so that the metadata IPv6 endpoint\nis reachable. It also fixes a bug that was causing the wrong LLA\nto be present as the interface was set `up` first prior to changing\nthe MAC address. Now this order is inverted so that the proper LLA\nis configured.\n\nChange-Id: Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/543835fe1b1730b59ef44096c55573075b558aa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/543835fe1b1730b59ef44096c55573075b558aa3"}]},"branch":"refs/heads/master"},"aee2930583790ae87636230d81022d8ee626a1c9":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-11-15 00:11:13.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/894026/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/894026/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/894026/10"}}},"commit":{"parents":[{"commit":"0027928c348ecc437a460b059d1ade3a2d536d18","subject":"Merge \"Remove obsolete PID files before start\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0027928c348ecc437a460b059d1ade3a2d536d18"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-15 00:11:13.000000000","tz":0},"subject":"[ovn] Add support for IPv6 metadata","message":"[ovn] Add support for IPv6 metadata\n\nThis patch adds support for IPv6 metadata service in ML2/OVN.\nThe changes include:\n\n- Add the \u0027fe80::a9fe:a9fe/128\u0027 address to the interface of the\n  ovnmeta- namespace so that it\u0027s reachable from the guests\n\n- Identify the port of the VM by looking up the source MAC address\n  of the metadata request\n\n- Restarts the haproxy instances to honor the configuration changes\n  upon start of the metadata agent. In particular, haproxy now also\n  binds on the \u0027fe80::a9fe:a9fe\u0027 address\n\nWhen the VM requests metadata from its LLA, the traffic will reach\nthe ovnmeta namespace associated to its network.\n\nThe IPv6 metadata tests are passing and enabled in Tempest by\nthis patch:\nhttps://review.opendev.org/c/openstack/neutron-tempest-plugin/+/894027\n\nBesides, this patch ensures that the link-local address of the\nmetadata interface is present so that the metadata IPv6 endpoint\nis reachable. It also fixes a bug that was causing the wrong LLA\nto be present as the interface was set `up` first prior to changing\nthe MAC address. Now this order is inverted so that the proper LLA\nis configured.\n\nChange-Id: Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aee2930583790ae87636230d81022d8ee626a1c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aee2930583790ae87636230d81022d8ee626a1c9"}]},"branch":"refs/heads/master","description":"Rebase"},"5f626bdaf6ae504e7ea450b30f0997f6ff6c3897":{"kind":"REWORK","_number":11,"created":"2023-11-30 19:20:35.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/894026/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/894026/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/894026/11"}}},"commit":{"parents":[{"commit":"f99da4d15c8efd74e7466e6585c17bbd441afc7a","subject":"Merge \"get_hosts_mapped_with_segments add filter agt_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f99da4d15c8efd74e7466e6585c17bbd441afc7a"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-11-30 19:19:36.000000000","tz":-300},"subject":"[ovn] Add support for IPv6 metadata","message":"[ovn] Add support for IPv6 metadata\n\nThis patch adds support for IPv6 metadata service in ML2/OVN.\nThe changes include:\n\n- Add the \u0027fe80::a9fe:a9fe/128\u0027 address to the interface of the\n  ovnmeta- namespace so that it\u0027s reachable from the guests\n\n- Identify the port of the VM by looking up the source MAC address\n  of the metadata request\n\n- Restarts the haproxy instances to honor the configuration changes\n  upon start of the metadata agent. In particular, haproxy now also\n  binds on the \u0027fe80::a9fe:a9fe\u0027 address\n\nWhen the VM requests metadata from its LLA, the traffic will reach\nthe ovnmeta namespace associated to its network.\n\nThe IPv6 metadata tests are passing and enabled in Tempest by\nthis patch:\nhttps://review.opendev.org/c/openstack/neutron-tempest-plugin/+/894027\n\nBesides, this patch ensures that the link-local address of the\nmetadata interface is present so that the metadata IPv6 endpoint\nis reachable. It also fixes a bug that was causing the wrong LLA\nto be present as the interface was set `up` first prior to changing\nthe MAC address. Now this order is inverted so that the proper LLA\nis configured.\n\nChange-Id: Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f626bdaf6ae504e7ea450b30f0997f6ff6c3897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f626bdaf6ae504e7ea450b30f0997f6ff6c3897"}]},"branch":"refs/heads/master"},"d9c8731af36d4eb53d9266733fec24659f2dc5a8":{"kind":"REWORK","_number":12,"created":"2023-12-08 22:19:11.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/26/894026/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/26/894026/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/26/894026/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/26/894026/12"}}},"commit":{"parents":[{"commit":"bfe05ec58f83b25965e9a803b8389e36a9d79f1b","subject":"Merge \"Remove check_router_mac_binding_options maintenance task\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bfe05ec58f83b25965e9a803b8389e36a9d79f1b"}]}],"author":{"name":"Daniel Alvarez Sanchez","email":"dalvarez@redhat.com","date":"2023-09-07 09:01:42.000000000","tz":0},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-12-08 22:15:42.000000000","tz":-300},"subject":"[ovn] Add support for IPv6 metadata","message":"[ovn] Add support for IPv6 metadata\n\nThis patch adds support for IPv6 metadata service in ML2/OVN.\nThe changes include:\n\n- Add the \u0027fe80::a9fe:a9fe/128\u0027 address to the interface of the\n  ovnmeta- namespace so that it\u0027s reachable from the guests\n\n- Identify the port of the VM by looking up the source MAC address\n  of the metadata request\n\n- Restarts the haproxy instances to honor the configuration changes\n  upon start of the metadata agent. In particular, haproxy now also\n  binds on the \u0027fe80::a9fe:a9fe\u0027 address\n\nWhen the VM requests metadata from its LLA, the traffic will reach\nthe ovnmeta namespace associated to its network.\n\nThe IPv6 metadata tests are passing and enabled in Tempest by\nthis patch:\nhttps://review.opendev.org/c/openstack/neutron-tempest-plugin/+/894027\n\nBesides, this patch ensures that the link-local address of the\nmetadata interface is present so that the metadata IPv6 endpoint\nis reachable. It also fixes a bug that was causing the wrong LLA\nto be present as the interface was set `up` first prior to changing\nthe MAC address. Now this order is inverted so that the proper LLA\nis configured.\n\nChange-Id: Idcef6de33ed2a73cb3c426db1c55fa9cd06de63f\nSigned-off-by: Daniel Alvarez Sanchez \u003cdalvarez@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9c8731af36d4eb53d9266733fec24659f2dc5a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9c8731af36d4eb53d9266733fec24659f2dc5a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
