)]}'
{"id":"openstack%2Fneutron~894265","triplet_id":"openstack%2Fneutron~stable%2Fzed~I0ac92e3692fb1ba9667fee2f6fbf531a57b7b5b8","project":"openstack/neutron","branch":"stable/zed","topic":"bug/2023634-stable/2023.1-stable/zed","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-09-08 21:29:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0ac92e3692fb1ba9667fee2f6fbf531a57b7b5b8","subject":"Functional: assert multiple calls for update_virtual_port_host","status":"MERGED","created":"2023-09-08 06:30:07.000000000","updated":"2023-09-08 21:30:41.000000000","submitted":"2023-09-08 21:29:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894265-bug/2023634-stable/2023.1-stable/zed","cherry_pick_of_change":894264,"cherry_pick_of_patch_set":1,"meta_rev_id":"397fd93a10a679a7c98c3c972c6187d566b88e01","_number":894265,"virtual_id_number":894265,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 21:29:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-09-08 09:59:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-09-08 20:29:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-09-08 20:29:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-09-08 09:59:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-08 06:30:07.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-09-08 07:29:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-08 09:59:28.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-09-08 20:29:14.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"bc29035806bac44f1c858df38b9198194dfdd49b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-08 06:30:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"1c8c69884d45da5728ed37c9e3980e4659bd9b66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 07:29:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/528ca916a08c40248c5f557585dd0267\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/698193796f0c48e580f100fa2067dda6 : SUCCESS in 13m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2472f10a27ca4a6385d366e53e626d11 : SUCCESS in 8m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a775fe7581da4f33b6bfa8995fb85cb4 : FAILURE in 52m 24s","accounts_in_message":[],"_revision_number":1},{"id":"17383f12679d165b36b2893a13a1c0c1e8225ef6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-08 09:19:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"787b95722bfc3c09a67bd019eb34f8fd21488469","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-08 09:59:28.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"b82c750a6a8bb8ee88fefe451af6933b34adcb24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 10:13:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41ebe03106bf41e485d6a57570c773cb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85dfab29da3b4304aaff456e406dfa7c : SUCCESS in 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b73b280b1c24ab0ba375b2240ed4aaa : SUCCESS in 9m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e1c8076f3c214e8ba97100b2153a2c47 : SUCCESS in 53m 43s","accounts_in_message":[],"_revision_number":2},{"id":"67fa5fbaad6a0d143975913b253dc35353f3716d","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-08 20:29:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"df2aa48833ce09b8fd30d8ff8053a78d1672d1e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 20:29:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3526eb9ad98a33666f6cc3ac8ded92385d229e3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 21:29:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21febc1372a44a91824d023db24ec60f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/221bbdf6cb144d048987e1c5339b1309 : SUCCESS in 10m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f217e03a1f614d8a9bee4ae496953560 : SUCCESS in 7m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d2cc4742a694f1ab3c307f0fab09a03 : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":2},{"id":"9d2e6adb746abf8a97b7e861139b32ed9090b50e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 21:29:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"397fd93a10a679a7c98c3c972c6187d566b88e01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 21:30:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd67d95525794c7196930f2880f18b6f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2194040357ea43b0895551d859217e2b : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"40c2afc2dc89fe17fc150d11b25338445f64af20","revisions":{"6d5b67f8731989df85954bad16425a55c288d713":{"kind":"REWORK","_number":1,"created":"2023-09-08 06:30:07.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/65/894265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/894265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/894265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/894265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/894265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/894265/1"}}},"commit":{"parents":[{"commit":"810c813adf0ab175a847805adb6a5f20ec586577","subject":"Improve the ``PortBindingUpdateVirtualPortsEvent`` match filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/810c813adf0ab175a847805adb6a5f20ec586577"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-06-14 13:15:06.000000000","tz":120},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-09-08 06:30:07.000000000","tz":0},"subject":"Functional: assert multiple calls for update_virtual_port_host","message":"Functional: assert multiple calls for update_virtual_port_host\n\ntest_virtual_port_host_update assumed update_virtual_port_host\nto be called once, but the test started to fail when [1] was\nmerged and OVN version was bumped to v22.03.3 (not yet released).\nPortBindingUpdateVirtualPortsEvent with event ROW_DELETE\nreceived twice in the failing case.\n\nBackport Note: It\u0027s required since [2] merged, ovn version deployed is\nnot related to the failure.\n\n[1]: https://review.opendev.org/c/openstack/neutron/+/880890\n[2]: https://review.opendev.org/q/I34caf7d0212ccb4bd7259c4414e7c3994bd8da4d\n\n\n\nCloses-bug: #2023634\nChange-Id: I0ac92e3692fb1ba9667fee2f6fbf531a57b7b5b8\n(cherry picked from commit 593278550a8185dfc52fc6c327ca1198df11e422)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d5b67f8731989df85954bad16425a55c288d713"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6d5b67f8731989df85954bad16425a55c288d713"}]},"branch":"refs/heads/stable/zed"},"40c2afc2dc89fe17fc150d11b25338445f64af20":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-08 09:19:26.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/65/894265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/65/894265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/894265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/894265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/65/894265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/65/894265/2"}}},"commit":{"parents":[{"commit":"37d09f95cd58143d0b2de9586158d3a6dfda800b","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/37d09f95cd58143d0b2de9586158d3a6dfda800b"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-06-14 13:15:06.000000000","tz":120},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-09-08 09:19:26.000000000","tz":0},"subject":"Functional: assert multiple calls for update_virtual_port_host","message":"Functional: assert multiple calls for update_virtual_port_host\n\ntest_virtual_port_host_update assumed update_virtual_port_host\nto be called once, but the test started to fail when [1] was\nmerged and OVN version was bumped to v22.03.3 (not yet released).\nPortBindingUpdateVirtualPortsEvent with event ROW_DELETE\nreceived twice in the failing case.\n\nBackport Note: It\u0027s required since [2] merged, ovn version deployed is\nnot related to the failure.\n\n[1]: https://review.opendev.org/c/openstack/neutron/+/880890\n[2]: https://review.opendev.org/q/I34caf7d0212ccb4bd7259c4414e7c3994bd8da4d\n\n\n\nCloses-bug: #2023634\nChange-Id: I0ac92e3692fb1ba9667fee2f6fbf531a57b7b5b8\n(cherry picked from commit 593278550a8185dfc52fc6c327ca1198df11e422)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40c2afc2dc89fe17fc150d11b25338445f64af20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/40c2afc2dc89fe17fc150d11b25338445f64af20"}]},"branch":"refs/heads/stable/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
