)]}'
{"id":"openstack%2Fneutron~894766","triplet_id":"openstack%2Fneutron~master~I48b5358f4c3ac492d68b00a4c0153370332e9249","project":"openstack/neutron","branch":"master","topic":"bug/2028846","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-12 16:06:57.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-09-12 19:16:40.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-09-12 16:31:52.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}},"hashtags":[],"change_id":"I48b5358f4c3ac492d68b00a4c0153370332e9249","subject":"Add release note with known issue with FIP PFs and vlan tenant networks","status":"MERGED","created":"2023-09-12 14:04:05.000000000","updated":"2023-09-12 19:17:53.000000000","submitted":"2023-09-12 19:16:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894766-bug/2028846","meta_rev_id":"812133c6bb19ca2671c8fde39018bf4088dd36f9","_number":894766,"virtual_id_number":894766,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-12 19:16:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-09-12 16:06:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-09-12 16:31:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-09-12 16:31:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-12 14:10:48.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-09-12 14:48:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-12 14:54:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-09-12 16:31:52.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"0c01793845367ce0b5bcac500ace38b9f300fb44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-09-12 14:04:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f300b45259e9131df49313116469ddde4917e89e","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-12 14:10:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5a240fc94870c1f026dce4f4b7271c8d199f3d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-09-12 14:13:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1c71175812b873b44819d26a24ed37229dd1366c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-09-12 14:13:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4b8d4e6a87c85c69c843bb614655345576b18c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 14:48:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/718f3b0c7d43498385b1932db99f9753\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b70a1c648f974423b6d232be9ec14b58 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b493027d61854cb19e2a51187b613d3f : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/716c432e70534d48908c874ff087366e : SUCCESS in 10m 59s","accounts_in_message":[],"_revision_number":2},{"id":"bc667c0b2e7d1d77d92afd6641567feedb104d9d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-12 14:54:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"603d7449654d1e6043e5499b8d2c0f4b17faa151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-12 16:03:51.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"632fc7ea5d40631022c331dfdef5a713ec402553","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-12 16:04:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d9b62338d751cd55cdd1681884ef10ac9fdb0393","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-12 16:06:41.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5660c93f01cc01f026d9cc8ecfc90b246d3214b5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-12 16:06:57.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5e7ef0dea8b95160c3082518a54af14dd36d2a54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 16:19:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47f779c8eca4417ab26ede595425c86d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54e0909cdd4141e58651b5e051962f79 : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a97be008d2d4e08920aaad67b3e9aa8 : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31557901b0f54d159653e2facc2f75ab : SUCCESS in 10m 48s","accounts_in_message":[],"_revision_number":4},{"id":"ae27dcab250790eee660fa64c64778f86996a122","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-12 16:31:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"59831443885d1cc9fd87f418a3c72773e78e2047","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 16:32:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"aaf87bd0f50fc4885b880a3ac969a45eaf8cd40e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 19:16:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d9fafd2a7ab4fe785ba35a4ce5e8d66\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b922ea90a6b54159a308086f59352adf : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb17e22633a4557a86f5d8087d2f2d5 : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe7daaa765814c87b7b47d071a81c6da : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":4},{"id":"2860d2c35c1fece729099d8740bcad1b19bc197f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 19:16:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"812133c6bb19ca2671c8fde39018bf4088dd36f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 19:17:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbdd1e8085a343c795b49e607008823f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd22be851b6245f083e2ea036e0ac43f : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/710ab108c209481e97eafe492aa374d0 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b2ceb8b854648b1072a017c022e3d7c83bb89122","revisions":{"c2bf6f1e47f6a218418e81ecfcf5086960f2d2ae":{"kind":"REWORK","_number":1,"created":"2023-09-12 14:04:05.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/894766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/894766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/894766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/894766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/894766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/894766/1"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-09-12 14:02:02.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-09-12 14:02:02.000000000","tz":120},"subject":"Add release note with known issue with FIP PFs and vlan tenant networks","message":"Add release note with known issue with FIP PFs and vlan tenant networks\n\nThis patch adds just release note with the known issue to describe known\nwrong configuration which cause FIP Port Forwarding to not work properly\nin ML2/OVN backend and distributed FIPs enabled.\nFor more details, please check related bug.\n\nRelated-bug: #2028846\nChange-Id: I48b5358f4c3ac492d68b00a4c0153370332e9249\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2bf6f1e47f6a218418e81ecfcf5086960f2d2ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2bf6f1e47f6a218418e81ecfcf5086960f2d2ae"}]},"branch":"refs/heads/master"},"fa0baf755374b348460cda6ff3cc9fc04eb783e3":{"kind":"REWORK","_number":2,"created":"2023-09-12 14:13:24.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/894766/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/894766/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/894766/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/894766/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/894766/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/894766/2"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-09-12 14:02:02.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-09-12 14:13:24.000000000","tz":0},"subject":"Add release note with known issue with FIP PFs and vlan tenant networks","message":"Add release note with known issue with FIP PFs and vlan tenant networks\n\nThis patch adds just release note with the known issue to describe known\nwrong configuration which cause FIP Port Forwarding to not work properly\nin ML2/OVN backend and distributed FIPs enabled.\nFor more details, please check related bug.\n\nRelated-bug: #2028846\nChange-Id: I48b5358f4c3ac492d68b00a4c0153370332e9249\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa0baf755374b348460cda6ff3cc9fc04eb783e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fa0baf755374b348460cda6ff3cc9fc04eb783e3"}]},"branch":"refs/heads/master"},"0f33c3971947f5fc497e335456334dc4b8905a7a":{"kind":"REWORK","_number":3,"created":"2023-09-12 16:03:51.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/66/894766/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/894766/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/894766/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/894766/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/894766/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/894766/3"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-09-12 14:02:02.000000000","tz":120},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-09-12 16:03:45.000000000","tz":0},"subject":"Add release note with known issue with FIP PFs and vlan tenant networks","message":"Add release note with known issue with FIP PFs and vlan tenant networks\n\nThis patch adds just release note with the known issue to describe known\nwrong configuration which cause FIP Port Forwarding to not work properly\nin ML2/OVN backend and distributed FIPs enabled.\nFor more details, please check related bug.\n\nRelated-bug: #2028846\nChange-Id: I48b5358f4c3ac492d68b00a4c0153370332e9249\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f33c3971947f5fc497e335456334dc4b8905a7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f33c3971947f5fc497e335456334dc4b8905a7a"}]},"branch":"refs/heads/master"},"b2ceb8b854648b1072a017c022e3d7c83bb89122":{"kind":"REWORK","_number":4,"created":"2023-09-12 16:06:41.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/66/894766/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/894766/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/894766/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/894766/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/894766/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/894766/4"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2023-09-12 14:02:02.000000000","tz":120},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-09-12 16:06:41.000000000","tz":0},"subject":"Add release note with known issue with FIP PFs and vlan tenant networks","message":"Add release note with known issue with FIP PFs and vlan tenant networks\n\nThis patch adds just release note with the known issue to describe known\nwrong configuration which cause FIP Port Forwarding to not work properly\nin ML2/OVN backend and distributed FIPs enabled.\nFor more details, please check related bug.\n\nRelated-bug: #2028846\nChange-Id: I48b5358f4c3ac492d68b00a4c0153370332e9249\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2ceb8b854648b1072a017c022e3d7c83bb89122"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b2ceb8b854648b1072a017c022e3d7c83bb89122"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
