)]}'
{"id":"openstack%2Fneutron~894767","triplet_id":"openstack%2Fneutron~master~Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e","project":"openstack/neutron","branch":"master","topic":"bug/2037294-external-port-scheduling","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-11-15 10:34:18.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-11-15 10:34:18.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-11-15 10:34:18.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-11-15 08:26:11.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2023-11-14 17:10:02.000000000","reason":"\u003cGERRIT_ACCOUNT_9656\u003e replied on the change","reason_account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-11-15 10:34:18.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-11-14 13:33:32.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e","subject":"[OVN] Enhanced external port scheduling","status":"MERGED","created":"2023-09-12 14:18:34.000000000","updated":"2023-11-15 10:36:00.000000000","submitted":"2023-11-15 10:34:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":148,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"894767-bug/2037294-external-port-scheduling","meta_rev_id":"3b83ceb05f13809713c92a43b676f92e0d3075a7","_number":894767,"virtual_id_number":894767,"owner":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-11-14 09:25:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-15 10:34:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2023-11-15 08:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2023-11-14 17:10:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2023-11-14 13:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2023-11-15 08:26:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-12 15:30:00.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-12 16:58:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 20:00:47.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"CC"},{"updated":"2023-10-12 17:10:08.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2023-11-03 15:56:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-11-08 10:22:39.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2023-11-09 20:38:56.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-11-09 20:38:56.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-11-13 11:28:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-11-13 11:28:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-11-13 11:28:55.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"3b4a93fe33b7e76d1c129329b00e1d6af4d9d061","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-12 14:18:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59684da7b26bf620edfeca0c22fb8f35c6d5a8df","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-12 15:30:00.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/67/894767/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"4d6ea6a3c1f8b01c7d9bf4c1722d968167e8f9f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 16:58:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c04dde286ea24d59a8b781a2f0d819cd\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e540d911958b4309831d42a250057abb : SUCCESS in 2h 06m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/38a9091e8003473bbf84e7483e34b6dc : SUCCESS in 38m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/540759a7f8aa46a482f79a60371c2871 : SUCCESS in 48m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78d9460baade439e9b759f5b6c8331fd : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1390d9b8394e4742833595ab05cf03b0 : SUCCESS in 26m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/119fd5491afe4520a85583fc41ba9790 : SUCCESS in 33m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44a496e353d34cd6be18af6b59e813d3 : SUCCESS in 26m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a131ed1c13b44c42a0ff6eb132d98059 : SUCCESS in 27m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56522c5b3eee432fbdcae2e1f05dce50 : SUCCESS in 9m 05s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/13cca7fa4cc8432f87f7788205ce9e7f : SUCCESS in 30m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/663e25e6f7fa4980a99c4a77da40ea37 : SUCCESS in 1h 01m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0acc02e5b73142de9255b940ffed6b5f : SUCCESS in 2h 32m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d345fca407404dc18d527f0b9bbc1903 : SUCCESS in 41m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fb439512f319413bb964646a6cddfe33 : SUCCESS in 1h 51m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8b32fd5a494748a6a9a4b04b19516fd1 : SUCCESS in 1h 33m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0a4b122132f7efc6ac71ce125517c87a6474a95c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 17:08:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e59316bb30b74bf39f1c85750babc3ef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/33fb0038cf5844a783ba25be60de059a : SUCCESS in 45m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1746a7cab0e4d9c92ceebbe092fe23c : SUCCESS in 56m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4234bb2ee352461396e577671745353d : SUCCESS in 43m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d26aee310bdf48f7b6043a97f057b1d4 : SUCCESS in 50m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e47dbad89ed64ee27ec44518ee8ceaa6e903371c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-20 09:20:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"82da1225e74b818bb0cd0037081e9701be5cb5f2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-20 10:02:17.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/67/894767/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"63d9f3d300fe0e4840113f8341d53833c9846967","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 11:37:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45e1db69a9bd44d2a40337d52c5e0faa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00a2d9364765428aae9fdd8a827028c4 : TIMED_OUT in 1h 21m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1840555cbe324286981fac09865e1fb7 : SUCCESS in 44m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/14a41c4585b847f88b93a4cc47131191 : TIMED_OUT in 1h 21m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/04ab2add77b7434891ba27c2674bcd32 : SUCCESS in 36m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"024e4bb57dd7ecf8decc29ae9f66b247e2ead597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 12:00:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36fad43f501744a8a39f4afa3c0c9a4e\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e2c0d8811dbe4cf28a26fac4c0d3d079 : SUCCESS in 1h 27m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c0fda4d11bd646c8a9da2b87bbe85dbe : SUCCESS in 32m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7aa16fc3753e44fdadab793e7d7aaa6b : SUCCESS in 34m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f07d3180999e480cb1d361ea93001f8b : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7be189578163420388873a2bf833b199 : SUCCESS in 29m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f56e94cdb1fe42d88bc5d2c142a346b6 : SUCCESS in 28m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9662ca303e445b4ba587a4aa015ab0d : SUCCESS in 24m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72b32e705a844830ab637766fdde1a8c : SUCCESS in 19m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ae1aa6f1ad649e2825ec59a11903427 : SUCCESS in 11m 34s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ad0c735c53c74a6397a8fbc668baa413 : SUCCESS in 25m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2417f20bcf7b4e70b9139998f182d111 : SUCCESS in 1h 15m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5593ae02042647a7a84d1701fd89e517 : SUCCESS in 2h 34m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/5c63afeae2d0469d898d7b39e2576f3b : SUCCESS in 1h 11m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ba8f3777871b4816837001903d04645a : SUCCESS in 2h 20m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/de04ae7f1382484ea0dad3aa7ec70efd : SUCCESS in 1h 11m 09s","accounts_in_message":[],"_revision_number":2},{"id":"88a9761f6ccd6437f4f16fd6b893ddb07b5455b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-21 09:02:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d8ce6783dca222cc1ea6b63e3c6b10530025f81f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-21 09:36:30.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/67/894767/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"ac638cc292fd276a4627bd3ae070738f5a76f74c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 10:39:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dcdada291274690b17d20b3290bb54f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72d51e34b3584938a46e4a1f2a5794d9 : SUCCESS in 57m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a423d2eb9f540be82f3f4a1a5483757 : SUCCESS in 55m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a3c5e1f359a64c7ba5d964aedd2a7eee : SUCCESS in 44m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aff06dbaf7284cfdb24a767e8f66fa41 : SUCCESS in 37m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"36d8566c489cbec64bd836c4b34b69de85bc852a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 11:35:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/618c95472d4f4d69a504f18d4b5be396\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8dcec9b13caa43f08497362fc9043ae1 : SUCCESS in 2h 03m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57148527070444f191b528a8921eee1f : SUCCESS in 34m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c671676f5b414e7a92eb6c2777d78da7 : SUCCESS in 35m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d055c56cbe44323953c09dabd142534 : SUCCESS in 11m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ec14d97313d498da016d641228e473e : SUCCESS in 24m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccd551a809874ce9b386218d3e61b6ef : SUCCESS in 25m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21d859902a5e4811bab27e1e42e7d6ce : SUCCESS in 22m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/873fa89bff2a43beb4f56e440170e4f0 : SUCCESS in 19m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ab681b60ac3406eb32f53d97f4c70b2 : SUCCESS in 12m 01s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0f48d8b457a548e69a2293fef9a76eb3 : SUCCESS in 25m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5945b4723c474489b6ca683c92154a28 : SUCCESS in 50m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d8592725bfc4efc8f0390c2381907de : SUCCESS in 2h 25m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/dbf81528deac4cf799b67d86f6a0ac56 : SUCCESS in 1h 01m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/69bb01bd4df04cfa82921aea5043b2c4 : SUCCESS in 1h 24m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4435f334546b43d2b1961f4167e4e5e6 : SUCCESS in 1h 07m 56s","accounts_in_message":[],"_revision_number":3},{"id":"c591f83b1ab660102540ddcc71423fe03185b4ff","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-09-21 20:00:47.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"404e2df7e3ffc64e446b8209248b9477732e2562","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-22 09:38:30.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"34c96dd100ad8c5c2be341152d716a76177e3451","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-22 09:40:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c970af12a3b32585a8807a62b876cce466f336fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-22 13:31:10.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"cff39ea4493063d4ac24f34c68cdc03f840f2e38","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-22 13:33:38.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"771f656465d79a6137310b333ff6c1280d293a98","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-22 13:33:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2c0f7231716d2d6b8f8685547f0aa32e0b2e2802","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-22 14:02:57.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/67/894767/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"8ebace4e36ed2eef58125953cb4443b0c9efe1f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 15:35:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e9815c2e14e404a9c5f00fae23810cd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e437549f5e240aeadf27a62d07b3ebf : SUCCESS in 1h 08m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/810dfc2f0af44566a9375984973b07d2 : SUCCESS in 42m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/75c5dc70b7e740c1b20928be4653eac2 : SUCCESS in 1h 07m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9670efa4373b4f3db0307b5733b48b75 : SUCCESS in 1h 02m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"adb52084de900afb5441ebe94c20ab9681ba7266","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 15:54:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9477bfa394947aca997467f24930ce3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7ddf45c4109f47839cdab2ae0cb9151b : SUCCESS in 2h 03m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/35cc59fec6724cbbb23d493723fa1028 : SUCCESS in 35m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cd674ee28a945ccb5925fdec78774a0 : SUCCESS in 39m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e02a6d1786d74f9282147b449483a32a : SUCCESS in 11m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd65aeaf6ccf4b8ca227518547d5bc1b : SUCCESS in 26m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da3cd33ec3764a4fa9b7bca8450add30 : SUCCESS in 25m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adf70182c5ed4befbf514592de09cdc4 : SUCCESS in 26m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/821914edc170415d84936c77dbbe813c : SUCCESS in 22m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e85091a1f184f25b3fccf5bd5001c06 : SUCCESS in 12m 40s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/868676f7c7dc481eb3af8f0097e783b7 : SUCCESS in 31m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/caf025d0ed9d40578f8e0dc9911f835c : SUCCESS in 58m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6017af2df67847009cfdb54ef81ddeac : SUCCESS in 2h 17m 28s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7236679bacda4afc9e7075b54415182d : SUCCESS in 1h 03m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/edb44d5549c34bd4872a6e4ed3c9cbdb : SUCCESS in 2h 04m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c8f29ad23fbb4f29a9c48612e2737bf8 : SUCCESS in 1h 13m 47s","accounts_in_message":[],"_revision_number":4},{"id":"b34d16f534b02743b7f81cac0f7fbea1b7a7e50e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-28 09:36:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"80f6be672cb37b09db92efed407972118d65b261","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-28 10:12:16.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/67/894767/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"d3df2bdf960ba135e753bb312be181f02f1cf787","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 10:51:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a515fa11d0364acc9ada582188c1be6f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c326bbbad51413b828b1dd1a1750e79 : SUCCESS in 42m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/adf7745d950c4c6c8c81a364bafec7b8 : SUCCESS in 52m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f41748d7fbe741bd9bc8fd7c004afe4a : SUCCESS in 53m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0fad04b82e204021aebe68417524ca80 : SUCCESS in 38m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"995c79d958171187700b38c47c0a0f4438e60c5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-28 10:54:28.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"2402bd22956644d33b7bb0839ecff03ec85a7af3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-28 11:26:33.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/67/894767/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"1e5877f92d476c62c7b33e5d21227dfba50736f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 12:19:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfcc561ea63148aca928cede0b540aa4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/673030f3c77f489da35d5518da45dfc2 : SUCCESS in 41m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f7097458f0d4096b09dd1a8af03d5c2 : SUCCESS in 57m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8a44050e209545549db51a91c907b3d5 : SUCCESS in 59m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7502844309354bd297b4af687281147d : SUCCESS in 37m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"44d5b9ac95677cc96a84d97e0e6e87d1994f1c1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 13:58:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c5105a317414331b077ab4a2b615dff\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/43a5e14f4ab042a7856d8b6aded07735 : SUCCESS in 1h 42m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e2c3c61b2d24e1f83526cc706a37856 : SUCCESS in 37m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76358d77656c4ed4a58d2064b217f9cc : SUCCESS in 31m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/271055fe9b174f79821292f0a70bd5ba : SUCCESS in 11m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa82b212c45340d592e958d0ca83266f : SUCCESS in 27m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f1796c104ac45e28dfb523cc822d308 : SUCCESS in 21m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36830c20a4e64e159a3999629f17fa48 : SUCCESS in 27m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/75905b22c3504dbeaef65ddd683d1aab : SUCCESS in 23m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/941712dbf8ca4b51a636818ca144f060 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d516b4f472174a17b83ccef9fdf47afc : SUCCESS in 15m 19s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/62d0dadd69e741ff849c607df4c17c3e : SUCCESS in 28m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f8ab91c49fd49da9072eeeb2a81b0db : SUCCESS in 52m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/10277253bccd46a48a70bb593d7e601f : TIMED_OUT in 3h 02m 22s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/381739640a1b4d42bfd1e5966aa31094 : SUCCESS in 43m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ab63959775db4f20aac3b122175f2baf : SUCCESS in 1h 49m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d290dba8cc7d41a09fb609cf413a4dad : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":6},{"id":"05ceff6e44d41d63354a3abaabc47f27bd6b4683","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-29 09:18:56.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"8a0ff79377b285434dd10b310d7076d750c29c46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-29 09:22:29.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fe53168097beba851ac5c3d484c34ebdf44ab779","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-29 09:50:48.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/67/894767/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"958c1a8ae8191db268afd75b46d066896ffcb755","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-29 10:22:01.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/67/894767/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"13291201052f41f4adf23472232feff1f14793e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:43:26.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/600908e19e254e018576c66f26334dce\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b22232c30ec44e5394f3050dbaa8e65e : SUCCESS in 1h 16m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fee5129ed650421eaf5def9f9b87cc30 : SUCCESS in 1h 17m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9656ef2fa88443a28655864480deceee : SUCCESS in 1h 17m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0a2d501edc1b443fa97f475e54ca97df : SUCCESS in 35m 23s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9ac916e915b6441d6c0d60bde4230fe0040286a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:56:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a10f1b1240e0469d98ad658df74303b3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0d5a4fef36034448b56d98ba365f8217 : SUCCESS in 1h 44m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b828ace574ad4c108c538b49aad65fab : SUCCESS in 31m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0767db7b95d42fe9d47a31697ac11d2 : SUCCESS in 39m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30a22da29b2d485e9fc9352233929fe2 : SUCCESS in 11m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08672a69da5941c7b1aa8d0b105197e4 : SUCCESS in 24m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d70f0a478f6c43e4856a70471b88e262 : SUCCESS in 26m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a85d114533d8417286cd8b6239e24022 : SUCCESS in 21m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/688743e4c4d546f48248b6a16656ee50 : SUCCESS in 24m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9865b420479f412cb7ff37f5c729ca58 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28f519e45c224fdab8364af47834d370 : SUCCESS in 9m 39s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/497cd55d3e0f4128b1630b0216ee8301 : SUCCESS in 28m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/84ea8091ad1044fbbf0560c179a1d4ba : FAILURE in 58m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d75e20e2b0d4229a966d69cd8e9d933 : SUCCESS in 2h 28m 08s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/93a6827fbce84290a04d103be4e7e603 : SUCCESS in 1h 04m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/67b7e713cc614eeeacb61f9f753341fd : SUCCESS in 2h 01m 13s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8f20e24d53d446e9b758cf83b61e786a : SUCCESS in 1h 14m 34s","accounts_in_message":[],"_revision_number":8},{"id":"68b14873fbc5293fdb6ad852d833fda41e37373b","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-29 12:29:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1fb471f26077786952afece655d00f44a2be9c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 14:48:33.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67ed6da41be7475ba3aafe26276abe77\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/11ab9358ec2642b5aa09079e3da212f6 : SUCCESS in 2h 17m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8c3993cbba70441dbc759233f028133f : SUCCESS in 43m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d70e0a858bab4c1891ad2ed047567456 : SUCCESS in 35m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08418bd51d23409e98ba9cef2a6dd069 : SUCCESS in 8m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4af2af5ccb14179bcefc9f3bb05b46c : SUCCESS in 25m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44d5b52813644fd0ad560d5b14b1c409 : SUCCESS in 25m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09034a9f11e24712a6b2ac4d7b96f9ee : SUCCESS in 25m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1391c2a7a51d4fdc95dead63115ea168 : SUCCESS in 27m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1197ccb8f8504cf0b7260987900e36c9 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bcd3efa2dd449e0977998ba612c7ba6 : SUCCESS in 17m 13s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/03d64392abd642f9ad4e41534bebcdbf : SUCCESS in 29m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6fb01b2d520c4604a7815360004ab943 : SUCCESS in 1h 09m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/22bb9b48345b49dd8c84bc1f00cb40a4 : FAILURE in 2h 00m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3ecd37e9e3484dde8652d05cc859eb03 : SUCCESS in 52m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7cd05701f8224979b174a44259d41c31 : SUCCESS in 1h 44m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9311fa300bb44c25b6c19e84f11f1bc0 : SUCCESS in 1h 18m 18s","accounts_in_message":[],"_revision_number":8},{"id":"43b09634223ddad44b18183915f6d83bfa95a831","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-09-29 14:50:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0bcabd3fad4a32a1e657c15d71a42f4e9b117047","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 14:54:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/907461d3ab03488e8d70d0ad2da2e2a6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a5ceb3101274ed6909eb81706e05e7c : SUCCESS in 42m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f266ea8dbaa4de899f01c5591e5f1d7 : SUCCESS in 47m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8a5e544b62de4e4c9860604ece5e3eef : SUCCESS in 46m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2c98c829102b455f95523bd4d0177cbf : SUCCESS in 41m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"bc96c62802e38b920562cfe46e2c55a4f8fd0a50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 16:58:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3592b5eebb414d62b3a39cb63b5ef045\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/60c27f3047454f32816ff8d9e8687a9f : SUCCESS in 2h 05m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dd23426c3040458c805e37b1db1dd7b2 : SUCCESS in 35m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcd487c3a4444f4687f71048a349a321 : SUCCESS in 37m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f9418afab104c95b5fb98839af02440 : SUCCESS in 12m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7092a19beb9c4607a1ebf00030d3b3b3 : SUCCESS in 29m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af9a1371de704ecdbafe70843fe4bfa5 : SUCCESS in 35m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eadbc4e46440481e87f64c6c1cd17e69 : SUCCESS in 26m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc9f3bc42f514ee48cc06d80805b0e5f : SUCCESS in 22m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1409cba3116345b9b3e86f06ee13b698 : SUCCESS in 13m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27052e0031f8400da95054528596c33d : SUCCESS in 15m 11s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/3d63dfd390fc4d42b151835f98e4745b : SUCCESS in 40m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/90087d12e5454d8ea0ad72ff02092a37 : SUCCESS in 1h 03m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/046a2b09c1b44cb5956e11843d263572 : SUCCESS in 1h 52m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/94e970634a5e4e59877884bd4b46f544 : SUCCESS in 45m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6f87807446794a94b39defbdb6c42e01 : SUCCESS in 1h 58m 11s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d9c9b5c4582f4713bc6a42be16cb44a9 : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":8},{"id":"c3e77b01f5a00ad2c243322f8a9e8531c94a85eb","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-03 21:07:21.000000000","message":"Patch Set 8:\n\n(10 comments)","accounts_in_message":[],"_revision_number":8},{"id":"310fd351d3f7eb6a49b5179849247a241fa762b1","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-04 09:36:07.000000000","message":"Patch Set 8:\n\n(11 comments)","accounts_in_message":[],"_revision_number":8},{"id":"66848cf5a6675a198632658aa014d9a81d7edb25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-04 13:09:51.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"7a0aac6f12ce6cb9bddc7631d25b1432b92241c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 15:38:52.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d05dd5544c746ea80b4b79a1c1cf20f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6881431bea4744f3917447a8a97264b6 : SUCCESS in 1h 24m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/396a7c1ac0904bcfa8b587b22b96415e : SUCCESS in 33m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/367e65cd2d83463990ae2f81332273a8 : SUCCESS in 55m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d86d6e056f444029bdaf54e612c79eb7 : SUCCESS in 9m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b47536b633143eab27fd50993bedae9 : SUCCESS in 33m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c77cfe89b5b94aa5af31621fff805d74 : SUCCESS in 22m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3cf6215c86834b4aba8229d4e937cf3d : SUCCESS in 41m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f5381fb6521430faf518d61395bbc03 : SUCCESS in 21m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2206a0cb7d6e4cb5b9953a54225d4d2a : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/256945edbdb645708a4a993c673d9019 : SUCCESS in 19m 58s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b2f8a719e54d4aef9f0a9ffa0ee19b86 : SUCCESS in 33m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f45d6c36a7014cb0b2929cf17126faae : SUCCESS in 1h 04m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0598024beeb40a2af9c45b9c61b7cd9 : SUCCESS in 2h 21m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/650d0c87703f4e3cb1ba7d8efb5cad4c : SUCCESS in 1h 08m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6ae4581fa2804d2f991f41ed37fe2669 : SUCCESS in 1h 45m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a09932f0721a422bbe9dbacc9746413f : SUCCESS in 1h 12m 57s","accounts_in_message":[],"_revision_number":9},{"id":"0f1a683aa50d6576772ce11ce1ebfea57e4ee8f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 16:10:09.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/007e54ca33ff4058a2d018e26fb3ac15\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/929b83bf09484fcb97818ae3dc9e96c3 : SUCCESS in 44m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1453b8ac27c4438fb049a61cf6cb05cd : SUCCESS in 53m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0019978b5e2c4ef7a155651262f688a3 : SUCCESS in 40m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/564f30fecd11454f8241f82e7325f9aa : SUCCESS in 50m 24s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"05d62bc9b15b4c7906ad4c7af81b12912f1cb331","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-04 16:36:10.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/67/894767/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"5e426fb86ed2325cfba3dfc6defda6fe7837f4f0","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-12 17:10:08.000000000","message":"Patch Set 9: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":9},{"id":"e97145c01d2a9f24d687c5eebf247bab435cdf36","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-13 09:38:41.000000000","message":"Patch Set 9:\n\n(16 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4e2057e008c3ee4a67ecbeb846ae5dd8bb866ef0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-13 11:15:46.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"bbeaee9f134b662ce86491375072284db5baa389","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-13 11:46:13.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/67/894767/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"42575cb961b45dac641072e1f97a334ea61ddfca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 12:12:08.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fe3e053aaad4e4e926af5ad3943b664\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f616c9a971794cff89d8d437d3d4f8c9 : SUCCESS in 49m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74eb8e96a7ae45d996ffd403b84096d0 : SUCCESS in 49m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/920c0f9b776f4b6097ad54a81a9449e4 : SUCCESS in 48m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8c3d06b8fe4e458cb27e0fd5dac5cb24 : SUCCESS in 38m 38s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b9b9f5d7ca916566f8cb720e2d7732369c484f5e","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-13 12:31:17.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6a80ce28f3edbd253c6f9674be9a66c34309bfb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 13:17:32.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcd5b557471441729cc29e63472c4059\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1d23725a7f154d0583749eaca2264866 : SUCCESS in 1h 28m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/611dfaee4ac6423788f3a3601618d4a5 : SUCCESS in 33m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93c378ed53e347a799aec1dfc49e4720 : SUCCESS in 33m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed5ace265fa84d67927a3bd543184539 : SUCCESS in 9m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f7be85f949e48ae9c0c1d02a7ab77c2 : SUCCESS in 24m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1abad2eabf9c43deba386ddb5fa46322 : SUCCESS in 19m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/389190da364f4884ba896d0e4407995b : SUCCESS in 21m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2b80f78cc444d04b305af98c4cb0ed0 : SUCCESS in 24m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6afd2e344e194b608595b27e5c128387 : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7aa866d0db1b4e1e8515f2ba2d285553 : SUCCESS in 16m 01s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/fa1a8399729c41508e3d5f4ece14d467 : SUCCESS in 29m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1654e21a1a534d6aa2d5fe13091b6a5e : SUCCESS in 57m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d902b0e610b4331ace7fd03ebb77758 : SUCCESS in 1h 59m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fe1dd47fd9384b71ac5db2df9c73fc3f : SUCCESS in 1h 04m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ada8c78852f1457ab7c0f32579233750 : SUCCESS in 1h 41m 12s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bb991a881c9347e087fa6c333e9d87e1 : SUCCESS in 1h 17m 10s","accounts_in_message":[],"_revision_number":10},{"id":"7cdfa26cfbde378cce96306ddef403e4b3bbabd1","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-20 16:03:05.000000000","message":"Patch Set 10: Code-Review-1\n\n(20 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d0ac3835bf5ed0f7041df767e296b9d9ae000b4b","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-21 00:32:01.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ebbad3d05b118e818d69e47205bbdcd9b1c4e5d4","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-25 10:37:41.000000000","message":"Patch Set 10:\n\n(21 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4cf999e0500a7b00bd3104e29e642eb035bb6ec2","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-25 14:05:46.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6b775ee885ef1f6b850a2086a663e5c0e21da19d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-25 14:06:52.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"e24f433f881bcfb860cb1473bf4ba0bb0f364f22","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-25 14:07:46.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"960b85f734533ae4d8418bb1553002cbaa598208","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-26 13:18:11.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"af0cdb675291dae39fdfdd47ac35d067ee67a281","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-26 13:19:26.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"aabf5894f4c7475daebc1e0df362a7375fe6398c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-26 13:49:40.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/67/894767/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"f4b0b8249202b1f77a8ac0f447caab5f32902364","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 14:15:58.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7530980559614a81b03d8ee9d7b178c0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d803b0e790804ed38ed64d7ea9c5efc2 : SUCCESS in 42m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83e043a984094911999730be34b6afa7 : SUCCESS in 53m 41s\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e41de09f74584b0f96249c961aabf421 : SUCCESS in 41m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6f02f2cf7501419daaf4723a150b267d : SUCCESS in 48m 12s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9d325f9317e04d748f879d9cca8937cef41ca043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 15:55:27.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e31d60b023dd429781b982a956912205\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b8757d96ba244f6090d46d208c752549 : SUCCESS in 2h 14m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c04b4abe60e54e65956f604294fc8f55 : SUCCESS in 42m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37886d169a96422ca8f53d80c4664590 : SUCCESS in 38m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d451a1a57bb941f3a859ae5acaa64722 : SUCCESS in 10m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75d6a5fff6734453a32857a6a7bf60de : SUCCESS in 28m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bb31b1d19a8440f84def0bda1c3a64b : SUCCESS in 28m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69f9bb5dccc14772906dc4dd141cb3cd : SUCCESS in 25m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b4e0124e47e444ab4356803f2116076 : SUCCESS in 22m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b533820ccf6c4a38a3607ce8a0841343 : SUCCESS in 11m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dbb37c3c9f3423f8885e19b044125a9 : SUCCESS in 18m 53s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ff37bb4f9845482b973c86051dd98f21 : SUCCESS in 33m 26s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/112afa7248c34a8eaf11ca9f8798b74e : SUCCESS in 1h 04m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/17f427ba137e4648a48d0147de4d29cc : SUCCESS in 58m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f216f8a9593a416b90982d3be7761bd2 : SUCCESS in 2h 30m 05s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7897b486fc134a6ea7ae3aed55cc7252 : SUCCESS in 1h 08m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c9e30a730eb846d1b6d7646b4f927c32 : SUCCESS in 1h 47m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8a4122febf054c1e99cc7ed8cdc65c14 : SUCCESS in 1h 19m 29s","accounts_in_message":[],"_revision_number":11},{"id":"eb121eb38af3d0d432ccb6d8b41020fe930b9728","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-27 15:08:14.000000000","message":"Patch Set 11: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":11},{"id":"7e970e452c490b54353a52a53c3900458d8c48df","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-27 15:10:58.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"932a2770d84c6cd8a9eb86afa6dd71d6f947ac60","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-31 10:19:27.000000000","message":"Patch Set 11:\n\n(6 comments)","accounts_in_message":[],"_revision_number":11},{"id":"d9ef1e351e353ca38f639e4b06c4c1ff2dcb49b5","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-31 11:24:09.000000000","message":"Patch Set 11: Workflow-1","accounts_in_message":[],"_revision_number":11},{"id":"ebdf6ec6d615ea07dce6b5e0256f4dd3660ab78d","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-31 11:45:03.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e1f52e269b98f1c9b3be71d910f4997da8dcfb96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-31 13:20:37.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":12},{"id":"6bea449f4fa6e0ac444085194912a2eea74a6029","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-31 13:50:06.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"313570ff47de745d5ae08e256f8428ef703f8b25","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-31 13:52:11.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/67/894767/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"925143a74b37245601174ea6f39777e84cb4a4b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 14:32:42.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0035c2e547cc433ba5ec1fb24a853dc0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d0bb42c8abd45e7829b761e6c0f019c : SUCCESS in 1h 09m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7be1e05952554f28898f067c88f23130 : SUCCESS in 23m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/23979ff4d5594c13abc930079d6bfbcf : SUCCESS in 58m 42s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"b74ddf377c11b674ec142bc581fa4a4dc999e24a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 15:31:55.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0612c2bb19e4373956158eb229f591f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9d79904f288744cb971c8b9fc9c3d4d7 : SUCCESS in 1h 41m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/149ec6595828442fb92fef51d87b9824 : SUCCESS in 47m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e70f6c8088144b21968bf6bc538e19e9 : SUCCESS in 35m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a1d9f84bdc6418ebebc43bd08b8287c : SUCCESS in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96b631bc629f4e5791d73820e9c1ab7d : SUCCESS in 26m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/82d24e1af70a4a6bb71a2095d7980f6b : SUCCESS in 24m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/271b7b67139a476bb76c136ac655e288 : SUCCESS in 13m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b46a74be8e2409e9e03352f7722b395 : SUCCESS in 16m 51s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/49ac568e25d04238b78352c8d85d90a3 : SUCCESS in 27m 24s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/aa05bffdebff4c4fbfef7d61bf05557a : SUCCESS in 1h 06m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a0c4b4c929e24a909e5b240a6b150526 : SUCCESS in 51m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2cb84e4ca98645dc9731d28b06bd2e9d : SUCCESS in 2h 07m 20s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4269158e4c3d4cb98067d0cd6896f9ab : SUCCESS in 1h 09m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2731413d1b2940d798886a77df4669fe : SUCCESS in 1h 50m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/963967d4fb0e4a6db3545a0ea895771a : SUCCESS in 1h 15m 06s","accounts_in_message":[],"_revision_number":12},{"id":"10fb8700d93ec0eff20c9fd12a17f10f6932a5bd","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-03 15:56:16.000000000","message":"Patch Set 12:\n\n(6 comments)","accounts_in_message":[],"_revision_number":12},{"id":"aa4cfd4cb18e24babb26664ccee4d3c90ee245bb","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-11-03 16:26:50.000000000","message":"Patch Set 12:\n\n(6 comments)","accounts_in_message":[],"_revision_number":12},{"id":"6b4e7606200660a269beb8bbef10f446f9da4650","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-11-03 16:28:09.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"a574db89541e08f8513311305146b0900a258ceb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:52:42.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44f00539cb9b40a1898c9c8ccb7c0a8c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/769470ae99b04145b7c77d20d03456f0 : SUCCESS in 22m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/26b89d7dbccc42b38d9b946b4050b6a2 : SUCCESS in 23m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1103b43d486e4a948e93895789c3625b : SUCCESS in 20m 59s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fe0aa5722ead3e122c5d9e6262af4d8704e682d9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-03 16:59:51.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/67/894767/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"b7026fd67cbfa25a14e68901c176c841eecda774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 18:37:12.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/459ee447709445979341c808d93e447d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/86c3c1aede3c473d9447b239753078e4 : SUCCESS in 1h 57m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/53ec88e85ce44d09b45be045cf01b66f : SUCCESS in 39m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e7bac2abc9b4d6cb68a40b95f7a5c7c : SUCCESS in 37m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c9a3e5772f34df0aeb38c139201e1c7 : SUCCESS in 11m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67d3db04bd8143f09dea08400878198c : SUCCESS in 34m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02b2538004b841199f9699fb1f4afaf3 : SUCCESS in 23m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43784524703c4cf5afa90c53022ab135 : SUCCESS in 11m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff2523d0d22347e0b8ea4609f0108ce7 : SUCCESS in 12m 48s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ef511f0137f44cd1a9d0e095c0a2db9d : SUCCESS in 28m 52s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/da847ce0d72542d682bcbbec04f94c5f : SUCCESS in 1h 06m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1b9b89657beb4ad0b4a8ca6749dc6e82 : SUCCESS in 1h 00m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/927922294be74a4187137a5c204ba901 : SUCCESS in 2h 08m 10s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ec89cea52bd64ab5b4695e91d9d3b79c : SUCCESS in 59m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a914f358be3a4569a814b75016f72b65 : SUCCESS in 1h 46m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4bbbee68502a4342a15fcbe23c003bd3 : SUCCESS in 1h 17m 29s","accounts_in_message":[],"_revision_number":13},{"id":"7fb6366f49f88489f24ea30db6043716983c5329","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-11-04 01:55:02.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"aed33a28b93a5f677574396e8f6b6660ee93ef9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-11-08 09:32:28.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"920337fef7c6265d6794ac9cf1f6d46b74336e30","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-11-08 09:32:44.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"5a975cac365a866a738fbcace57e0ac64dbc055b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 09:57:46.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/377d0006a4d8488da10ccc6874c9adb1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdab49dc8cc24e198e32f962e4d95a77 : SUCCESS in 21m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bd6cdb9b139f41c29afdaf32da8b8b45 : SUCCESS in 18m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b54b7258c7054a58b800fbcb541201a3 : SUCCESS in 24m 25s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c19a0f0d19a877125470f4f8f74a911906a83dcd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-08 10:05:17.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/67/894767/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"d0863e2dfb16bdd6ce90ebb43fc4cbfd710f2eb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 11:34:56.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/429387d745d44dff9db8468d8fcb4a1c\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b55e53041299411f88b8ddbf381f6207 : SUCCESS in 1h 37m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26c04dd1db8c4b0e8faf32b426860eb6 : SUCCESS in 41m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9803d10af4364931b18446a6dcd0abe9 : SUCCESS in 38m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84e4c74fe9d24486a8b48bb381a0bffc : SUCCESS in 11m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01f7a5a2e5e94ff9a5b73b0ceec25a60 : SUCCESS in 25m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/feda92bfea624e26a9148e7f5e1c13ef : SUCCESS in 24m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c14dfb4c5959434296c4618e35362e82 : SUCCESS in 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0be1cd8cf574463197d004d53826fd96 : SUCCESS in 15m 54s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8291e96a621b4516872bfd0e9bfbc711 : FAILURE in 6m 53s (non-voting)\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/4914ffdd965d4cab8d3bd91e942f5aa7 : SUCCESS in 1h 07m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/019cbd0860094a829bf76dfa5f4d3a7d : SUCCESS in 1h 01m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb1cd6b3bc8044209a0750f6e5c79777 : SUCCESS in 1h 53m 48s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/aa2cf0b540564e0b87f33f9bb5167d95 : SUCCESS in 55m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cfd7e29095b642e793853529ee95122f : SUCCESS in 1h 55m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b79ddb67de744cbda6372e5b341dbe4b : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":14},{"id":"0142bfa148e8edb60cc039faf4eebb3a0d93ab8d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-11-09 20:38:31.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"1c3c2cf8af4c5d9c85ccec5181d69b4c10c6aff9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-11-13 10:58:05.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c34253b2f387476a702fa8d9b812971a293683e2","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-11-13 11:07:04.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"30143c0d56feed7e18832772a67bc22e0c6743c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-11-13 14:13:29.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"6f67021e402624b60a32d356c7e5b3874a85f67a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-11-13 14:13:38.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"b936d6e59465e9b2086717b4d2a4ce102c14c23a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-13 14:49:06.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/67/894767/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"b61613cf96de0f13ce21fe4b960f73d1947e0129","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 15:01:19.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4281717b6ea8477a9d4a234624c618d7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7536ff225e654d9ab0f1ca8c2fcf0bec : SUCCESS in 44m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/80ac501365b44b27b836eec627c9fdd9 : SUCCESS in 45m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f767823172d47be9feb55ba140748cc : SUCCESS in 37m 45s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"4f4aedd9594ece33196351d8be4f195f74d2108b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-11-13 15:09:42.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"cc19d78f09e39c33238bfd6a4e4f75c711aa8a2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 16:20:51.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86831e59385941d69c2f21ca830af6f1\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/72fa66209db14cf186a1b73362b7b814 : SUCCESS in 2h 01m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a7143e2a04394154bc9a6ad44ebd31ff : SUCCESS in 41m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9060369beb9545e9a253c9eee541536d : SUCCESS in 29m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ac589edff6b4dc0bf5f8b873ce3cc0b : SUCCESS in 12m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d27a3e341ea94afc85bc365c0407adf5 : SUCCESS in 25m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08b1d0aad18d4afd979c1a89c0bf82e7 : SUCCESS in 24m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65527e7ae6464d78a12ea695519dc1e2 : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60fc9a705c0e4b0fa06524b6c012058c : SUCCESS in 16m 48s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c2d8b3ab5ef14039953de0dc6d0b07ff : SUCCESS in 29m 04s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/c82eb9b1e870410d9fec5b870925ede1 : SUCCESS in 1h 02m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/db8a903f2b9a4091b6588d5a52c536e0 : SUCCESS in 59m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cbba7ba398404e16ba5560fe0e6bc042 : SUCCESS in 2h 00m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7609c950acf341ecabc6be96e8f6105d : SUCCESS in 48m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/29f7e0a808654bfeb52d9c982f468b35 : SUCCESS in 2h 01m 30s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/943d6a8462324847ae4c10c275edc0f5 : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":15},{"id":"4961abf9376c0976afe520326c1bfb820b9941a5","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-11-13 17:20:48.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"19c7b17370e4ae0a5b3826c1bf62dc7913c7bb0a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-13 17:50:02.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"2931509b47df66afafb8d684b3a83f22175a3da5","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-11-14 08:54:10.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"cbbec08b41d5a48e499868221f17354527d3829a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-11-14 08:55:11.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"bde856f82f5dcc82a24efcf1043dde6541788576","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-14 09:25:56.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/67/894767/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"31ef8df760a93a9d2087e41fc7fb1d2236ed573a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 09:37:38.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d5d5178f1ae42fe86279773ced41d4b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/debab50ef6da4e058808961c2880bbd0 : SUCCESS in 41m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dd05c2027c5c4e6d96be6043d4dfbd7b : SUCCESS in 22m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/76c88a5000174839b25171cdf83d1b45 : SUCCESS in 16m 33s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"e2495ba747cdf8dc3b4cdc5794d25b796df997b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 11:33:57.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71fd37edb9be4c71b6e8ed5001942629\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3d39ce26bca84611ba1d6e57d24cf421 : SUCCESS in 2h 21m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/89a15e08f97b472cb54ccfb90008c4e2 : SUCCESS in 45m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bda19657b4714f8da0301ee6c67ca806 : SUCCESS in 41m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6a4b107d9d14ebdb945aa70d3fe0d29 : SUCCESS in 10m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4ecf77f643845489c07675c77ef61ec : SUCCESS in 32m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8cca61f31f84654bd2f515d03eeefc8 : SUCCESS in 18m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/685560d2fd9541dbace06960b9e1f0f8 : SUCCESS in 13m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b4d4fa44d0d445aabda88606f9e4016 : SUCCESS in 16m 39s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0c624cb0d6b14c23a50c9f06bab58df5 : SUCCESS in 32m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a4ae659643fe4dae8c0959efb637e872 : SUCCESS in 1h 08m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b1e1e88d04e34467bd48f65432bddd3b : SUCCESS in 51m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6bfdbd8b31474a68b2a24f14e0809c04 : SUCCESS in 2h 37m 32s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/68f0689da62c4d86a3e746924de340d9 : SUCCESS in 49m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ad91e0521daa47f6bfdce5f04c358474 : SUCCESS in 1h 32m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6e5850ef201a4720811006a418508a2a : SUCCESS in 1h 19m 55s","accounts_in_message":[],"_revision_number":16},{"id":"26943788ed385197b4114dfc3bab2d66ea527882","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-14 13:33:32.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"1a72663067d4e2e1b40fa7a176e93bac78cceab8","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-11-14 17:10:02.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"de9d82df9df28952023a3da6c4907b9aa70b93bd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-11-15 08:26:11.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"8128ac3619b0556320cc2fbba90b29276b513098","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 08:26:40.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"43d8711348f63e0c72189deb5cf96f7a1d784460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 10:34:18.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67c2bb8c374642758cdd8fac00e5a030\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2ccd647702274a6ebc2b818212b7ace7 : SUCCESS in 2h 07m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/197287388be446bcb42da86c192b4be3 : SUCCESS in 11m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39feef8d7fc94be89fd87124a7f682a9 : SUCCESS in 27m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ce11074de084969ac0c3135aeb24b2c : SUCCESS in 18m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d35e2ca2364c0daa484e68e7d3dddf : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f771d032f5da459aba3dc8d125c69b1d : SUCCESS in 20m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/73d001d6857449ba948e3bfbb745fefa : SUCCESS in 53m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d828bda885734481a4b953febe0bf459 : SUCCESS in 1h 35m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e906099a06594ea9a5ea37479bac0d74 : SUCCESS in 59m 05s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/323d56ac691049688a7a4b5cf2c367d5 : SUCCESS in 1h 21m 14s","accounts_in_message":[],"_revision_number":16},{"id":"292e91546baacb1b296f62c5b90b4def302941c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 10:34:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"3b83ceb05f13809713c92a43b676f92e0d3075a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 10:36:00.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e000aa89710b4b3a9a233b0a81d333f5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbbb4bbd777d4b6d8012e410b779858b : SUCCESS in 1m 19s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6bff3a0eb5640de800210ff5a49263f : SUCCESS in 55s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"770914f98df9abda2b5b8ca2896ff05618ea0181","revisions":{"276e83600131c7a05824a10e297a8fcbdd966bf6":{"kind":"REWORK","_number":1,"created":"2023-09-12 14:18:34.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/1"}}},"commit":{"parents":[{"commit":"8cba9a2ee86cb3b65645674ef315c14cfb261143","subject":"Call the \"tc qdisc\" command for ingress qdisc without parent","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cba9a2ee86cb3b65645674ef315c14cfb261143"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 14:13:12.000000000","tz":60},"subject":"[OVN] External ports scheduling (WIP)","message":"[OVN] External ports scheduling (WIP)\n\nDO NOT REVIEW! This is WIP and is not tested!\n\nThis patch introduces a new flag for OVN to allow external ports to be\nscheduled on different nodes than gateway nodes.\n\nTODO:\n* LP\n* Use the ignore_chassis parameter\n* Tests\n* Documentation\n* Release note\n\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/276e83600131c7a05824a10e297a8fcbdd966bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/276e83600131c7a05824a10e297a8fcbdd966bf6"}]},"branch":"refs/heads/master"},"9cab29a6c2b581d699b72c0c8469a28d7a274618":{"kind":"REWORK","_number":2,"created":"2023-09-20 09:20:11.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/2"}}},"commit":{"parents":[{"commit":"dbe4ba910b3236ff3ac42e33dcb4cc067b1f9177","subject":"Merge \"Open the 2024.1 (Caracal) DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbe4ba910b3236ff3ac42e33dcb4cc067b1f9177"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-20 09:20:00.000000000","tz":60},"subject":"[OVN] External ports scheduling (WIP)","message":"[OVN] External ports scheduling (WIP)\n\nDO NOT REVIEW! This is WIP and is not tested!\n\nThis patch introduces a new flag for OVN to allow external ports to be\nscheduled on different nodes than gateway nodes.\n\nTODO:\n* LP\n* Use the ignore_chassis parameter\n* Tests\n* Documentation\n* Release note\n\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9cab29a6c2b581d699b72c0c8469a28d7a274618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9cab29a6c2b581d699b72c0c8469a28d7a274618"}]},"branch":"refs/heads/master"},"32238978ea8df28510601395bd889aebbfefda1b":{"kind":"REWORK","_number":3,"created":"2023-09-21 09:02:21.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/3"}}},"commit":{"parents":[{"commit":"dbe4ba910b3236ff3ac42e33dcb4cc067b1f9177","subject":"Merge \"Open the 2024.1 (Caracal) DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbe4ba910b3236ff3ac42e33dcb4cc067b1f9177"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-21 09:02:11.000000000","tz":60},"subject":"[OVN] External ports scheduling (WIP)","message":"[OVN] External ports scheduling (WIP)\n\nDO NOT REVIEW! This is WIP and is not tested!\n\nThis patch introduces a new flag for OVN to allow external ports to be\nscheduled on different nodes than gateway nodes.\n\nTODO:\n* LP\n* Tests\n* Documentation\n* Release note\n\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32238978ea8df28510601395bd889aebbfefda1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/32238978ea8df28510601395bd889aebbfefda1b"}]},"branch":"refs/heads/master"},"8829fdda0d55ceb53d900218570301a78f1fa302":{"kind":"REWORK","_number":4,"created":"2023-09-22 13:31:10.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/4"}}},"commit":{"parents":[{"commit":"dbe4ba910b3236ff3ac42e33dcb4cc067b1f9177","subject":"Merge \"Open the 2024.1 (Caracal) DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbe4ba910b3236ff3ac42e33dcb4cc067b1f9177"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-22 13:17:02.000000000","tz":60},"subject":"[OVN] External ports scheduling (WIP)","message":"[OVN] External ports scheduling (WIP)\n\nDO NOT REVIEW! This is WIP and is not tested!\n\nThis patch introduces a new flag for OVN to allow external ports to be\nscheduled on different nodes than gateway nodes.\n\nTODO:\n* LP\n* Tests\n* Documentation\n* Release note\n\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8829fdda0d55ceb53d900218570301a78f1fa302"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8829fdda0d55ceb53d900218570301a78f1fa302"}]},"branch":"refs/heads/master"},"691f8aab9b2560799733c3c563abfe833f895ca0":{"kind":"REWORK","_number":5,"created":"2023-09-28 09:36:17.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/5"}}},"commit":{"parents":[{"commit":"61ac046cf7ed81a7502ddfdfb6fa785f4184554f","subject":"Merge \"contributor docs: Architectural overview for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61ac046cf7ed81a7502ddfdfb6fa785f4184554f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-28 09:15:40.000000000","tz":60},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration is used by ML2/OVN\nto identify nodes that are eligible for scheduling OVN\u0027s external ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the\nOVN CMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to serve\nthose external ports and running them on control plane nodes are not\nideal. This is where this patch comes handy allowing these ports to be\nscheduled on EDPM compute nodes.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/IOVN will continue to schedule\nthe external prots on ndoes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/691f8aab9b2560799733c3c563abfe833f895ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/691f8aab9b2560799733c3c563abfe833f895ca0"}]},"branch":"refs/heads/master"},"618b4ef42093c3b0adfd08a7e4485779d9d7b2cf":{"kind":"REWORK","_number":6,"created":"2023-09-28 10:54:28.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/6"}}},"commit":{"parents":[{"commit":"61ac046cf7ed81a7502ddfdfb6fa785f4184554f","subject":"Merge \"contributor docs: Architectural overview for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61ac046cf7ed81a7502ddfdfb6fa785f4184554f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-28 10:54:19.000000000","tz":60},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration is used by ML2/OVN\nto identify nodes that are eligible for scheduling OVN\u0027s external ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the\nOVN CMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to serve\nthose external ports and running them on control plane nodes are not\nideal. This is where this patch comes handy allowing these ports to be\nscheduled on EDPM compute nodes.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/IOVN will continue to schedule\nthe external prots on ndoes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/618b4ef42093c3b0adfd08a7e4485779d9d7b2cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/618b4ef42093c3b0adfd08a7e4485779d9d7b2cf"}]},"branch":"refs/heads/master"},"ba0c7dab2165f724a6e08dfa65b1057ddfd56ead":{"kind":"REWORK","_number":7,"created":"2023-09-29 09:18:56.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/7"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-29 09:13:33.000000000","tz":60},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration is used by ML2/OVN\nto identify nodes that are eligible for scheduling OVN\u0027s external ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the\nOVN CMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to serve\nthose external ports and running them on control plane nodes are not\nideal. This is where this patch comes handy allowing these ports to be\nscheduled on EDPM compute nodes.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/IOVN will continue to schedule\nthe external prots on ndoes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba0c7dab2165f724a6e08dfa65b1057ddfd56ead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba0c7dab2165f724a6e08dfa65b1057ddfd56ead"}]},"branch":"refs/heads/master"},"dc8a9a2b0f0cdc91e6c741a567aad24419a605b3":{"kind":"REWORK","_number":8,"created":"2023-09-29 09:22:29.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/8"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-29 09:22:00.000000000","tz":60},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration is used by ML2/OVN\nto identify nodes that are eligible for scheduling OVN\u0027s external ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the\nOVN CMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to serve\nthose external ports and running them on control plane nodes are not\nideal. This is where this patch comes handy allowing these ports to be\nscheduled on EDPM compute nodes.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/IOVN will continue to schedule\nthe external prots on ndoes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc8a9a2b0f0cdc91e6c741a567aad24419a605b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc8a9a2b0f0cdc91e6c741a567aad24419a605b3"}]},"branch":"refs/heads/master"},"354a6db7b210125c13dc70bd33bae1eb501c2bbe":{"kind":"REWORK","_number":9,"created":"2023-10-04 13:09:51.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/9"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-10-04 13:09:36.000000000","tz":60},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration is used by ML2/OVN\nto identify nodes that are eligible for scheduling OVN\u0027s external ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/354a6db7b210125c13dc70bd33bae1eb501c2bbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/354a6db7b210125c13dc70bd33bae1eb501c2bbe"}]},"branch":"refs/heads/master"},"886956499b129f3b53394644450f97af5b85d60f":{"kind":"REWORK","_number":10,"created":"2023-10-13 11:15:46.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/10"}}},"commit":{"parents":[{"commit":"ea0bc376a2ec93a78ee5c9d4772a970832dc367b","subject":"Merge \"Fix the common/ovn functional tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea0bc376a2ec93a78ee5c9d4772a970832dc367b"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-10-13 11:02:36.000000000","tz":60},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/886956499b129f3b53394644450f97af5b85d60f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/886956499b129f3b53394644450f97af5b85d60f"}]},"branch":"refs/heads/master"},"90dd48843fde08d3a44e94df9a3adc52167f0925":{"kind":"REWORK","_number":11,"created":"2023-10-26 13:19:26.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/11"}}},"commit":{"parents":[{"commit":"f90d7d2a9e49688f7baee13268c3ba34f59d5a16","subject":"Merge \"[FT] Clear the idl lock in TestMaintenance tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f90d7d2a9e49688f7baee13268c3ba34f59d5a16"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-10-26 13:19:20.000000000","tz":60},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90dd48843fde08d3a44e94df9a3adc52167f0925"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/90dd48843fde08d3a44e94df9a3adc52167f0925"}]},"branch":"refs/heads/master"},"7f021319ac4487dbf7bac1fa74f4e38b5630e3a2":{"kind":"REWORK","_number":12,"created":"2023-10-31 13:20:37.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/12"}}},"commit":{"parents":[{"commit":"fbca7f0104064a3e68a0174f9ce10924b34eaad0","subject":"Merge \"[OVN] Populate the \"router.distributed\" flag in ML2/OVN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fbca7f0104064a3e68a0174f9ce10924b34eaad0"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-10-31 13:14:19.000000000","tz":0},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f021319ac4487dbf7bac1fa74f4e38b5630e3a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f021319ac4487dbf7bac1fa74f4e38b5630e3a2"}]},"branch":"refs/heads/master"},"188bffd74bf5333714034a8433485ec86351ea7a":{"kind":"REWORK","_number":13,"created":"2023-11-03 16:28:09.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/13"}}},"commit":{"parents":[{"commit":"dfa6dc4c0e43684c09ecd1c8066f0d7d0f839232","subject":"Merge \"[DHCP agent] Add route to OVN metadata port if exists\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dfa6dc4c0e43684c09ecd1c8066f0d7d0f839232"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-11-03 16:27:54.000000000","tz":0},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/188bffd74bf5333714034a8433485ec86351ea7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/188bffd74bf5333714034a8433485ec86351ea7a"}]},"branch":"refs/heads/master"},"e8aa91eb47157e75ab7ae9c49ac8dd91b3b970a9":{"kind":"REWORK","_number":14,"created":"2023-11-08 09:32:28.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/14"}}},"commit":{"parents":[{"commit":"26f173423986ca6c110ee517bae8d5db844d6ad6","subject":"Merge \"ovn: Spawn metadata resources on additional chassis event\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/26f173423986ca6c110ee517bae8d5db844d6ad6"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-11-08 09:32:03.000000000","tz":0},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8aa91eb47157e75ab7ae9c49ac8dd91b3b970a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8aa91eb47157e75ab7ae9c49ac8dd91b3b970a9"}]},"branch":"refs/heads/master"},"973c500d4b4ed67e4ab0ece192fa01c14523eda1":{"kind":"REWORK","_number":15,"created":"2023-11-13 14:13:29.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/15"}}},"commit":{"parents":[{"commit":"b1585c52dcb4bf7623bfff2fe34dbb9cff08ab4f","subject":"Merge \"[Fullstack] Drop all linuxbridge scenarios from fullstack tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1585c52dcb4bf7623bfff2fe34dbb9cff08ab4f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-11-13 14:12:48.000000000","tz":0},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/973c500d4b4ed67e4ab0ece192fa01c14523eda1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/973c500d4b4ed67e4ab0ece192fa01c14523eda1"}]},"branch":"refs/heads/master"},"770914f98df9abda2b5b8ca2896ff05618ea0181":{"kind":"REWORK","_number":16,"created":"2023-11-14 08:55:11.000000000","uploader":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"ref":"refs/changes/67/894767/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/67/894767/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/67/894767/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/67/894767/16"}}},"commit":{"parents":[{"commit":"b1585c52dcb4bf7623bfff2fe34dbb9cff08ab4f","subject":"Merge \"[Fullstack] Drop all linuxbridge scenarios from fullstack tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1585c52dcb4bf7623bfff2fe34dbb9cff08ab4f"}]}],"author":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-09-12 10:36:59.000000000","tz":60},"committer":{"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","date":"2023-11-14 08:54:59.000000000","tz":0},"subject":"[OVN] Enhanced external port scheduling","message":"[OVN] Enhanced external port scheduling\n\nThis patch introduces a new configuration for OVN CMS Options called\n\"enable-chassis-as-extport-host\". This configuration can be used\nby ML2/OVN to identify nodes that are eligible for scheduling OVN\u0027s\nexternal ports.\n\nPrior to this patch, external ports were always scheduled on centralized\nnetworked nodes tagged with the \"enable-chassis-as-gw\" flag in the OVN\nCMS Options but, when it comes to deploying OpenStack on OpenShift\nrequiring services such as the OVN Metadata Agent or DHCP Agent to\nserve those external ports and running them on control plane nodes are\nnot ideal. This is where this patch comes handy allowing these ports to\nhave more flexibility in where they are scheduled.\n\nThe patch is also backward compatible and if the new configuration is\nnot present on the OVN CMS Options, ML2/OVN will continue to schedule\nthe external ports on nodes configured with the previous configuration\nlike always.\n\nDocumentation will be updated on a follow up patch.\n\nCloses-Bug: 2037294\nChange-Id: Ic46d847e3aebfe543d5a7ab49d18d1f1abf1342e\nSigned-off-by: Lucas Alvares Gomes \u003clucasagomes@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/770914f98df9abda2b5b8ca2896ff05618ea0181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/770914f98df9abda2b5b8ca2896ff05618ea0181"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
