)]}'
{"id":"openstack%2Fneutron~894784","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Ia7937390867e45af34ebcd65bd76fc89b6adafe9","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/2022914","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-14 12:40:14.000000000","reason":"Change was submitted"},"33237":{"account":{"_account_id":33237,"name":"Maxim Korezkij","email":"maxim.korezkij@mail.schwarz","username":"korezkij"},"last_update":"2023-09-13 06:42:00.000000000","reason":"\u003cGERRIT_ACCOUNT_33237\u003e replied on the change","reason_account":{"_account_id":33237,"name":"Maxim Korezkij","email":"maxim.korezkij@mail.schwarz","username":"korezkij"}}},"hashtags":[],"change_id":"Ia7937390867e45af34ebcd65bd76fc89b6adafe9","subject":"[neutron-api] remove leader_only for sb connection","status":"MERGED","created":"2023-09-12 14:57:07.000000000","updated":"2023-09-14 12:41:42.000000000","submitted":"2023-09-14 12:40:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894784-bug/2022914","meta_rev_id":"caf64d91181b66d6116f114ee3757b8f56b8bc36","_number":894784,"virtual_id_number":894784,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33237,"name":"Maxim Korezkij","email":"maxim.korezkij@mail.schwarz","username":"korezkij"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-14 12:40:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":33237,"name":"Maxim Korezkij","email":"maxim.korezkij@mail.schwarz","username":"korezkij"},"all":[{"value":1,"date":"2023-09-13 06:42:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33237,"name":"Maxim Korezkij","email":"maxim.korezkij@mail.schwarz","username":"korezkij"},{"value":2,"date":"2023-09-13 13:21:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-09-13 13:41:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-09-13 01:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-13 01:58:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":33237,"name":"Maxim Korezkij","email":"maxim.korezkij@mail.schwarz","username":"korezkij"},{"value":1,"date":"2023-09-13 13:21:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33237,"name":"Maxim Korezkij","email":"maxim.korezkij@mail.schwarz","username":"korezkij"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":33237,"name":"Maxim Korezkij","email":"maxim.korezkij@mail.schwarz","username":"korezkij"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-12 14:57:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":33237,"name":"Maxim Korezkij","email":"maxim.korezkij@mail.schwarz","username":"korezkij"},"state":"REVIEWER"},{"updated":"2023-09-12 14:57:45.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-12 17:31:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-13 01:39:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-09-13 01:58:56.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-09-13 13:21:45.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-09-13 13:41:31.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"db3b00b5a6780904f5f84a87b5624cfeea713045","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-12 14:57:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a2451dc72a809b344c1d76d01d1fb46f1649e32","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-09-12 14:57:45.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-neutron-wallaby-direct http://13.74.249.42/refs/changes/84/894784/1/openstack-sriov-neutron-wallaby-direct : FAILURE\n\n* openstack-sriov-neutron-wallaby-macvtap http://13.74.249.42/refs/changes/84/894784/1/openstack-sriov-neutron-wallaby-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"843ad8e2e9af07c9dda9f1881273a31e189513ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 17:31:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/508285d015404a78b92dd1d286c115cc\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/03aa0533060641f9b602a6de07e65279 : SUCCESS in 53m 24s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/356dd353166c46659f11913c40aca5bf : SUCCESS in 2h 27m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23f7ecef30064ab49c19dee52c837a26 : SUCCESS in 51m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4d14d0cc20b4961a534f0c9129a4c5b : SUCCESS in 16m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aac962a38f3e44d38a6fecf26d987ca5 : SUCCESS in 28m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15f52275dbe740598c6d7a782a4faa92 : SUCCESS in 33m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ea63f03839a4a94a7b3dd4fc0a6fef0 : SUCCESS in 27m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fefa02b2fc14ac38c6bd76b8ce05879 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68a0b73e935a4444afc4dd09273b1dc0 : SUCCESS in 8m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/20ac32444a7945a9bffdca58c415ee81 : SUCCESS in 48m 45s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/f31653cd2ea94441b8f95853f8115f34 : SUCCESS in 1h 16m 32s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ea15d66b6d6046ef86bc8cf2227d8b3d : SUCCESS in 38m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3ffbafd70c714267bea1f94a97ff3447 : SUCCESS in 1h 24m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/48aca3e75548495dbb02089795afdc80 : SUCCESS in 1h 33m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c7059fdac27383f1af793b46e8e831f63e3e941e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 18:13:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f63998656c774ca69d4898b1e9b6a931\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/664d102ef1194db2b9126c91d8b1f082 : SUCCESS in 56m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/406ba000dc2e4d3096813bc0849a0599 : SUCCESS in 1h 05m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c90c9cfc1b11771fce29af7649fa1880d5c4ea9c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-13 01:39:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"da5f936e71082f25a2f7942232c4b884e9995e86","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-09-13 01:58:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b82bfb0c07a593a43b44d4e18952810d72982bac","author":{"_account_id":33237,"name":"Maxim Korezkij","email":"maxim.korezkij@mail.schwarz","username":"korezkij"},"date":"2023-09-13 06:42:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3e56f57dda8ac2e88b5ebfe860972a09f78b16a5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-09-13 13:21:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"62c063e2743564cc6a164f38d9341a6548aa404e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 13:22:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe2e446b699af9dc611e13ba2aaed3f7f5fd16db","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-09-13 13:41:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6a20eddee8e8b935d0b2f95debaf32173bdc4b81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 14:47:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba295ceb245444a99b794bc5f7e1a038\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/3796fd21554c46899cc2fa32b6da212a : SUCCESS in 59m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d82e245ebf5417eb39ddb783711f593 : SUCCESS in 11m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/143dc55828834f40854980d4ea24c4b1 : SUCCESS in 33m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/170d13fba14f43d4ac1f0af9ae631f03 : SUCCESS in 27m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/355bb051302f4bfcbd691071d81b8d12 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e66b434f236b4c22b023f4d635e6417b : SUCCESS in 9m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c92042d502a84426b93445762a9ceb98 : FAILURE in 58m 17s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/58d67b7300f143a3be246b807c65a1e9 : SUCCESS in 1h 20m 15s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/3b1ab6691fe64ca0b2524a6ca02d04d1 : SUCCESS in 34m 14s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7ba74a9c69204521a01ba8389f82a959 : SUCCESS in 1h 17m 34s","accounts_in_message":[],"_revision_number":1},{"id":"75f6466ab98af249b840cff4e5cb89e1846c093f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-09-14 08:58:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a0889f7882c343b8983176ac2bb4b1003c9c6ef","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2023-09-14 09:04:14.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-neutron-wallaby-direct http://13.74.249.42/refs/changes/84/894784/1/openstack-sriov-neutron-wallaby-direct : FAILURE\n\n* openstack-sriov-neutron-wallaby-macvtap http://13.74.249.42/refs/changes/84/894784/1/openstack-sriov-neutron-wallaby-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"47a9f81042891df5a3af24923cb4ac0ebf89a8f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 09:56:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a2b8b17dd9417ca11e302b2ce6f66b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3f8b0cf8f6f4ebcaded416108927e06 : SUCCESS in 43m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75519a8ca19742fcb4958f32b4c098be : SUCCESS in 41m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e4f84328e707a9a908be4b1df1193d7c47014e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 11:05:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78f78d2aa9bc45cebb91c5010acedd10\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/6d733c492ce8492eb18e2d5b4950b76a : SUCCESS in 53m 51s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/acc36706672d4b638fe817c12239918a : SUCCESS in 2h 01m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/817a2dd264724bf99bcbea1359daeecd : SUCCESS in 39m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed03d71373894f629e9a23babff3a389 : SUCCESS in 13m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b75ff7d796aa47e08eb6ef3214fc088d : SUCCESS in 28m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/618e920f7390490cbe807007f1e12cfb : SUCCESS in 31m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83fa685da9b34a32b6c027b1f24043e2 : SUCCESS in 24m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa315d53497a4251b38c39ddc6cbf510 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff3af08d0df3471b8fb71cb1fa16ef8b : SUCCESS in 12m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c16efd0fc764c9d95b57e4b2cf73ccc : SUCCESS in 51m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/771528c99ed3436fac914252218dc2c3 : SUCCESS in 1h 01m 26s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/e89f4d48db2442c4a49486d201fcb261 : SUCCESS in 25m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3b32e7d3c58d42b295f0e72c8e57499d : SUCCESS in 1h 31m 22s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/5cebdea5047842c18fa110c85b5f83b0 : SUCCESS in 1h 27m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e8cea02ecd4ef586e32f1d4894663d52d0b55592","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 11:06:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfa034a9e627a6c7a1cbaa9d814a790b5138c97d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 12:40:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff5364a8e47a4abc8632aade2d17daca\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/3b172e85c6e946fe878121a596defb88 : SUCCESS in 46m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7489ce983fa44e81b70efd23467da7d5 : SUCCESS in 12m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1db331a27edf4f4895f773ec24b1e50c : SUCCESS in 25m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17fe27aa654646ac87cf038324783962 : SUCCESS in 23m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1148be2fb65419dbfeae06a63784347 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/831f353867734f8bbd0f5a0c1bafe27c : SUCCESS in 11m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4690613a5f424de1bf335ad22ebfc38e : SUCCESS in 54m 15s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/88411b6785db4399b343717cdd9eca0b : SUCCESS in 59m 19s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/7d426c91ee394da29a9f973617a2d642 : SUCCESS in 43m 25s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/16d52da1337440a693b966716515ec44 : SUCCESS in 1h 27m 20s","accounts_in_message":[],"_revision_number":1},{"id":"92533cebd86505bb8fcb6fba50fe369ebdddb651","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 12:40:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"caf64d91181b66d6116f114ee3757b8f56b8bc36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 12:41:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eba50d119f7f422eafaf879bb952971c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3369c4eaa07740508d0049c4dc821e50 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b99a83760fe4eccbd1b5aeb9a220479 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9869029aef1dafca4637aaf43f587ce4799cade4","revisions":{"9869029aef1dafca4637aaf43f587ce4799cade4":{"kind":"REWORK","_number":1,"created":"2023-09-12 14:57:07.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/894784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/84/894784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/894784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/894784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/84/894784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/84/894784/1"}}},"commit":{"parents":[{"commit":"39593d86d506572ff5ade112e272c47e909f70b6","subject":"Mark fullstack test_l2_agent_restart as unstable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/39593d86d506572ff5ade112e272c47e909f70b6"}]}],"author":{"name":"maximkorezkij","email":"maxim.korezkij@mail.schwarz","date":"2023-06-05 09:20:47.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-09 11:06:35.000000000","tz":0},"subject":"[neutron-api] remove leader_only for sb connection","message":"[neutron-api] remove leader_only for sb connection\n\nThe maintenance worker from the neutron-api uses a southbound\nconnection. Since the southbound does not use any locking and all the\novsdb locking is used for the northbound this changes should not have a\nbig impact.\n\nThis commit removes the leader_only flag for the maintenance worker.\nThis should also enable the neutron api to connect to relays instead of only\nthe sb directly.\n\nCloses-Bug: #2022914\nChange-Id: Ia7937390867e45af34ebcd65bd76fc89b6adafe9\nSigned-off-by: maximkorezkij \u003cmaxim.korezkij@mail.schwarz\u003e\n(cherry picked from commit a9c8bf5c069d3324828294e3e36b5374d5a828fe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9869029aef1dafca4637aaf43f587ce4799cade4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9869029aef1dafca4637aaf43f587ce4799cade4"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
