)]}'
{"id":"openstack%2Fneutron~894791","triplet_id":"openstack%2Fneutron~stable%2F2023.1~Ifc2d37e2042fad43dd838821953defd99a5f8665","project":"openstack/neutron","branch":"stable/2023.1","topic":"bug/2033887","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-28 13:22:22.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-09-28 09:32:54.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-09-28 11:12:05.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"Ifc2d37e2042fad43dd838821953defd99a5f8665","subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","status":"MERGED","created":"2023-09-12 15:30:36.000000000","updated":"2023-09-28 13:23:33.000000000","submitted":"2023-09-28 13:22:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"894791-bug/2033887","meta_rev_id":"dd7f254357f9307fc43fcba36d4a391146102955","_number":894791,"virtual_id_number":894791,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-28 13:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-09-28 09:32:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-09-28 11:12:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2023-09-28 11:12:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-09-28 11:12:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-12 18:10:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-18 11:13:57.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2023-09-18 11:14:02.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-09-19 13:00:48.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"d084e9a6274f00a37798987d7e60af38a4244e53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-12 15:30:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c423e3a5a109848bec96278defbd1088488f3ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 18:10:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1819190cc67f404ea8a98ed83c9d6366\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/ba643bc010fb4ed995c06d3f956da8fe : FAILURE in 1h 26m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8eabff8954594f6895dcf57739780241 : SUCCESS in 33m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f696b3d8aad54de3a6aaf89924708f0f : SUCCESS in 14m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2184b92f83064c17872bc7e9977eedb8 : SUCCESS in 36m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfc2a4e5a4b248f08acff53fd32f5f8c : SUCCESS in 29m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d03c99dc3320441fb1c3087053113d0a : SUCCESS in 10m 08s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/f8fbf0f7621e4b31bf2df842a19e3190 : SUCCESS in 24m 04s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/db5d1cec8a37472a9ce35c86b2fc64dd : SUCCESS in 52m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5663fc78670c4f078c0fcc92f2f60af9 : SUCCESS in 1h 39m 27s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a9fcfd6d21a04dcab85aebb6a1ded3fa : SUCCESS in 1h 02m 13s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/90cef61ce3e7409ea3b4242b62a41e1d : SUCCESS in 1h 17m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/da65dcd92cb846be91556e6068bb4908 : TIMED_OUT in 2h 33m 50s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d817339369924d27835198f84723386b : SUCCESS in 1h 21m 06s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/ba685df10b0645dcaccf656c8fe804a9 : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":1},{"id":"18f4d12deab939449f467be76797ad5720f20ee3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 18:15:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f0b4c0308f54904b111eb20bd80812a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a98d5739664541eaa2ea54bc6048613b : SUCCESS in 43m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad6c62c8066846ec9538ec054232aed6 : SUCCESS in 40m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d0af716ca42dfae417c25e32219b55a4de1f5fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-15 13:00:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b51eb1d506d76f2f7db72d35bf1c1280d1456b50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:59:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e849fb40fde472180cd72fd7dcc07d8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/981626bad94f4b4482553a7eb354b34f : SUCCESS in 44m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d85438ee2e3848f8a612a5aebb23feae : SUCCESS in 1h 09m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c61596994f62e82be32e47e49d9bd8f904c8e6cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 15:40:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5c080bd0c78433aab6859b4ffb1a6d4\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/6f9662ae251c4328a4a6519f0726c443 : SUCCESS in 2h 14m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa1ba9d52a4848f29a58cf52d48eefe2 : SUCCESS in 37m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7261352c44854e3383cff19392f703ee : SUCCESS in 13m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/828dfb3ed44c4ec19e824ffc8ad3e2ad : SUCCESS in 26m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/faac042668b2477c8dceb1e465ea34de : SUCCESS in 25m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0354cfdfce9423ebc41b0d5b4b222ea : SUCCESS in 15m 11s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/a47cef8683c3416686e1194162848936 : SUCCESS in 27m 16s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/af87280b2ccc496993a0b5d71db706ba : FAILURE in 45m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/be3ea1955b62473db98943e5b1a480ed : SUCCESS in 1h 53m 26s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/e8e6220d6c134545aa854be8af6c7fe3 : SUCCESS in 1h 08m 26s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/091d8cad27dd49e1bc578f1aed8fab9a : SUCCESS in 1h 11m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fbbc34ff2a694a45b123ab3c8cf45224 : SUCCESS in 2h 14m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3960125f83be440c8f1661c536cceeb1 : SUCCESS in 1h 10m 43s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/01c9c5bbbbec44399c3b7c8fad8f2a7a : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":2},{"id":"811e2be78d728a917d76d0eaed2751890425a1c0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-18 11:13:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2fb51354d88551e7ce0762e117da9b2be3819284","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-18 11:14:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0a27f9337f73d701dd1e65e545cda7b430ce386f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 13:34:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/545f423628634ef39e544cf19c99cd7d\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/c429f8e8a35e47078aed91ddeeacd5a2 : SUCCESS in 1h 21m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b801d6f583e41d0a99e8f93c79aab69 : SUCCESS in 24m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f416d2d6eeb3488b944e584cc1673be1 : SUCCESS in 10m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fc8a77e233449cc91c20c0761bc6516 : SUCCESS in 20m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82748ce2b32f42739ccd11eb6e667d9d : SUCCESS in 24m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129a7d80ee4b44ac90cf921256953575 : SUCCESS in 11m 47s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/db80babea8244bad8ea8d8d4d331093e : SUCCESS in 21m 05s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/51816377ac91482f9698f3903f5ec682 : SUCCESS in 54m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b058685d3734454b88ce96c565a55dd0 : SUCCESS in 2h 03m 57s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8f7be646f33c4e388b775ee72dc1605d : FAILURE in 1h 01m 39s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b0608f4ee4754e8f897218fe5fb2d43b : SUCCESS in 1h 08m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8302e8ac93b94695b53ff2aa521ad8bc : SUCCESS in 2h 11m 54s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5f82d88cd2d14a0ba63e7af56246d31c : SUCCESS in 54m 37s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/cceb35f45483492a87e04692ccf4b20f : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":2},{"id":"325150685ec15b3214e1b22ee3a9cf9737c77267","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 13:54:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13601e2f550648ffa04e3c8e0ebaf845\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd1a9f28695f4a9a9451c7f3d5d5d799 : SUCCESS in 41m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a36b1567f9cf4842b573f3861ebec524 : SUCCESS in 35m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5b4d4a855c0b27bc3e3460910a1d145448367819","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-19 07:51:27.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"e49c85c82818c03d131448e5c410cca62fa6803a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-09-19 13:00:48.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2fbd82898ce4cfc96d51f188051c6c8759c0c9be","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-19 15:10:16.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"72c54d8d184044fb900f230657d411f71ee977aa","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-20 08:55:16.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"ce1fd74aea7684dc68f46cae3cbabdf4e27a1321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-20 09:06:02.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"96e6819875bf8465e359306e0dfb4051b2a54fe7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 10:52:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba794009f2e84c40864250f869e626c5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be2f9db9f1754559b99fd52fbcd57961 : SUCCESS in 1h 12m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ef530fac558243258b7109224147fe21 : SUCCESS in 42m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5d951ec1f97b1a7e80bd917aee4dffef57ad7cf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 11:08:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7ce4be2a388474e9f688a6180ebece4\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/59513ca9ff574fd6aa33644c6a24aa0b : SUCCESS in 1h 30m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5983dd25ac1648dfb04301c24ef148d9 : SUCCESS in 36m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75b94d96c2314f80935143dbe3774116 : SUCCESS in 14m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a44786afc17f47a4a9777aa4c3aba4ee : SUCCESS in 27m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cf1a117470545338ea488ee3e73ad7f : SUCCESS in 23m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd0b66f8ec0c49bbb937379a528f4685 : SUCCESS in 12m 55s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ced9715d0c854d9390b0eeb27eb2bb82 : SUCCESS in 20m 16s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eb5ab3ceedfa44c78f5016307bb27951 : SUCCESS in 45m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b9962fb065a14a85870f32456d5e64e0 : SUCCESS in 1h 56m 42s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/bf8717dec6bc451ea3291273316ffc47 : SUCCESS in 1h 18m 39s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a327e4ec2bfe4d9eb74c94e32099a9fd : SUCCESS in 1h 07m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/73e85878ef5e4ae4881cf040fd7734f6 : SUCCESS in 1h 44m 21s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7abe1ba2d95b43c3bb0d56aaa9e52ee9 : SUCCESS in 46m 16s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/fb80294913464cf68576d1fd059004e8 : SUCCESS in 1h 07m 55s","accounts_in_message":[],"_revision_number":3},{"id":"27d9549bcb81af970fad3afa4fb8e88771e8d21f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-20 12:47:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"73643c284ed505f48ae001ff199973948eceb3e2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-26 10:52:37.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"f50e48e7db0c89ae4a41013d6a2d8f5ff725866b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-27 14:38:43.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7bf97192e3c87a333d88b4b6141bae4ee07e5878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 16:44:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a59622a3d2f34acd987d5071ae1c36fe\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/2cad2700f0b341db82f6a0c290f1a7f1 : SUCCESS in 1h 44m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e17ff622025f4895a9bb80da34af89cb : SUCCESS in 29m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c45e74e97ef340f6acf9eb8173237e78 : SUCCESS in 13m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/819fb2907c0741e6b3b01d3156301275 : SUCCESS in 23m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c865b3573019414dbf79d7e01cbec51f : SUCCESS in 24m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd5e080808846b0bfc1325477384097 : SUCCESS in 11m 45s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/c2f9a0126f4842fa8276518fcc266914 : SUCCESS in 16m 45s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3c1e552913c14c10a6f79464f71c6b00 : SUCCESS in 49m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/69373d2494df40fb9e5d4bdbeaa2c0ae : SUCCESS in 2h 00m 01s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/b20d18b6cf914ceab38fad0c5f920fc0 : SUCCESS in 1h 18m 30s (non-voting)\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c20bb4b34e4b4d4bbce2d4ee87bf517f : SUCCESS in 1h 05m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6b8eb516c8c047e698cfbf9eb56751e0 : SUCCESS in 1h 47m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/57f0f0bf648f4beea7ac702cdcbe506d : SUCCESS in 1h 02m 03s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/b5dadc7a9178436ea48c786df5ba2541 : SUCCESS in 1h 14m 29s","accounts_in_message":[],"_revision_number":4},{"id":"6ea3164b86724218631403af0beaada7b177842b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 17:30:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfe1ed533e8c437fae9eb978efdaa5c0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/59c163a5826c41e686932405eb0c117e : SUCCESS in 37m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9c207a16cc944c42bb33e2e4bb99031f : SUCCESS in 37m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"029acd6107a7a3baae6117bfab3cc8c8d01f835b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-09-28 09:32:54.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"adcc0121eedc13d68c9b176b095fe771e8d730a3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-28 11:12:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1 Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"ae6815ad287bb9c530bf14fe695c81910d617d22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 11:12:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9196129239c46478de039a55bb58cf2ac8b32a85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 13:22:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e19f5c6b93ce4b888bf5dba83e3db5f5\n\n- neutron-tempest-plugin-ovn-2023-1 https://zuul.opendev.org/t/openstack/build/cdd854234f9043379d12c218344ed5f7 : SUCCESS in 1h 27m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25299b87a90b406fa67a06d1839db27e : SUCCESS in 14m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94d4485aa0f84ad083dabf7d8d74eae7 : SUCCESS in 20m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6d7f6877a5044299691d63832707fd5 : SUCCESS in 23m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8abb040f3d23429ab73c2cbd4d48d19b : SUCCESS in 10m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/20a3753003f34a23a19163c5ca665613 : SUCCESS in 50m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/81abaf7e1ee34523bc8331d51e821dfe : SUCCESS in 2h 08m 37s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/32fb2ea26fbc40049f2c6b0ed3868e99 : SUCCESS in 1h 04m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d77f52c4415a444389e716a54e9900af : SUCCESS in 1h 13m 18s\n- neutron-ovn-tempest-ovs-release-ubuntu-old https://zuul.opendev.org/t/openstack/build/bc3cc05ae95648b99a37ea924823a9e6 : SUCCESS in 1h 06m 56s","accounts_in_message":[],"_revision_number":4},{"id":"80c0f018a9685274ae46d95cba15074333c5e518","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 13:22:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"dd7f254357f9307fc43fcba36d4a391146102955","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 13:23:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8493e8a228dd4d81a0a3c50c7a8d3a4a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/871c669b36bd4b02bdf38d0e2ffb77ed : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"93f223f8c0a8ad0709418820c96e22161a198313","revisions":{"5438714f84b791e93587c6ddc3e77f583232a4ad":{"kind":"REWORK","_number":1,"created":"2023-09-12 15:30:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/894791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/894791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/894791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/894791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/894791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/894791/1"}}},"commit":{"parents":[{"commit":"4a7e7f2d258df8e1f93c3e397d165bdb437354ac","subject":"Functional: assert multiple calls for update_virtual_port_host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4a7e7f2d258df8e1f93c3e397d165bdb437354ac"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-01 11:43:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-09 11:17:58.000000000","tz":0},"subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","message":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\n\nThis reverts commit 955e621167987626806128d259a29d4e62acf102.\n\nReason for revert: the port binding handling done in this patch is\nincorrect and leads to issues during the cold migration process with\ntrunk ports in ML2/OVN.\n\nChange-Id: Ifc2d37e2042fad43dd838821953defd99a5f8665\nCloses-Bug: #2033887\n(cherry picked from commit 6ffad1f31fbb95247d6e5911298439ae61ef4ac0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5438714f84b791e93587c6ddc3e77f583232a4ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5438714f84b791e93587c6ddc3e77f583232a4ad"}]},"branch":"refs/heads/stable/2023.1"},"b6c5bb669594efddfc1feb9b7df7e9b5e3580623":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-15 13:00:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/894791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/894791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/894791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/894791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/894791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/894791/2"}}},"commit":{"parents":[{"commit":"b7a59a51320ace612276d35c0358e7d5ee90de8e","subject":"Switch fullstack/functional fips jobs to 9-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7a59a51320ace612276d35c0358e7d5ee90de8e"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-01 11:43:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-15 13:00:28.000000000","tz":0},"subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","message":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\n\nThis reverts commit 955e621167987626806128d259a29d4e62acf102.\n\nReason for revert: the port binding handling done in this patch is\nincorrect and leads to issues during the cold migration process with\ntrunk ports in ML2/OVN.\n\nChange-Id: Ifc2d37e2042fad43dd838821953defd99a5f8665\nCloses-Bug: #2033887\n(cherry picked from commit 6ffad1f31fbb95247d6e5911298439ae61ef4ac0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6c5bb669594efddfc1feb9b7df7e9b5e3580623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6c5bb669594efddfc1feb9b7df7e9b5e3580623"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"},"cc24e422855b56718615852dd1c141861e2b6df4":{"kind":"REWORK","_number":3,"created":"2023-09-20 09:06:02.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/894791/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/894791/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/894791/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/894791/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/894791/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/894791/3"}}},"commit":{"parents":[{"commit":"b7a59a51320ace612276d35c0358e7d5ee90de8e","subject":"Switch fullstack/functional fips jobs to 9-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b7a59a51320ace612276d35c0358e7d5ee90de8e"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-01 11:43:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-20 09:06:01.000000000","tz":0},"subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","message":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\n\nThis reverts commit 955e621167987626806128d259a29d4e62acf102.\n\nReason for revert: the port binding handling done in this patch is\nincorrect and leads to issues during the cold migration process with\ntrunk ports in ML2/OVN.\n\nChange-Id: Ifc2d37e2042fad43dd838821953defd99a5f8665\nCloses-Bug: #2033887\n(cherry picked from commit 6ffad1f31fbb95247d6e5911298439ae61ef4ac0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc24e422855b56718615852dd1c141861e2b6df4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc24e422855b56718615852dd1c141861e2b6df4"}]},"branch":"refs/heads/stable/2023.1"},"93f223f8c0a8ad0709418820c96e22161a198313":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2023-09-27 14:38:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/91/894791/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/91/894791/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/894791/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/894791/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/91/894791/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/91/894791/4"}}},"commit":{"parents":[{"commit":"493bfd0554f0357ca5b9d55aff7b193f55dfd3e8","subject":"Merge \"Check the device ID and host ID during virtual port binding\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/493bfd0554f0357ca5b9d55aff7b193f55dfd3e8"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-01 11:43:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-27 14:38:35.000000000","tz":0},"subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","message":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\n\nThis reverts commit 955e621167987626806128d259a29d4e62acf102.\n\nReason for revert: the port binding handling done in this patch is\nincorrect and leads to issues during the cold migration process with\ntrunk ports in ML2/OVN.\n\nChange-Id: Ifc2d37e2042fad43dd838821953defd99a5f8665\nCloses-Bug: #2033887\n(cherry picked from commit 1b034f8d62091724a6fb639f34c2f8936329b501)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93f223f8c0a8ad0709418820c96e22161a198313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93f223f8c0a8ad0709418820c96e22161a198313"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
