)]}'
{"id":"openstack%2Fneutron~894793","triplet_id":"openstack%2Fneutron~stable%2Fyoga~Ifc2d37e2042fad43dd838821953defd99a5f8665","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/2033887","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-28 15:20:34.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-09-28 10:17:18.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-09-28 15:20:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifc2d37e2042fad43dd838821953defd99a5f8665","subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","status":"MERGED","created":"2023-09-12 15:30:49.000000000","updated":"2023-09-28 15:22:10.000000000","submitted":"2023-09-28 15:20:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894793-bug/2033887","meta_rev_id":"6bf0d24d46e7607e1483cb9f4ebad1a3fbadbc65","_number":894793,"virtual_id_number":894793,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-28 15:20:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-09-28 10:17:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-09-28 13:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-09-28 13:45:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-09-27 14:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-12 17:25:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-19 08:00:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-09-19 13:06:18.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-09-28 13:45:16.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"682a4f3d3075dbcc9132c078d614654fe3ac7c60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-12 15:30:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ead67d5af8d6ac64b76e513372d2f41bb6f90e0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 17:25:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc0dd8b5b396439eb3f1699dd9c0f9d0\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/a9d2378120684df096996d0e71056440 : SUCCESS in 1h 03m 22s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/fac8b5673a034980bfc203f0a8d96d22 : SUCCESS in 1h 09m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90694df127be46079d0130f4b7802e73 : SUCCESS in 36m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/779084e39c2c455a848e3b86c3a60555 : SUCCESS in 13m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f846cac4fed4742969213d7eaf6f80d : SUCCESS in 28m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7305291002254eb6896b461ac6e5a17a : SUCCESS in 27m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab75d1b75bd4d87ae8f1b9a5a0ac600 : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/80f65fab26f54025bfd9826558a8f6aa : SUCCESS in 1h 05m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/16f99a9586cc456e95cd6718d37600d0 : SUCCESS in 1h 46m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78c796be0d9c400f97518cc039892314 : SUCCESS in 1h 09m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/755df57d5d8341acb160f06b23a1e0ab : SUCCESS in 1h 07m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dee1d878d32496102cf3968d569cc8dc303ab7dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 18:39:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bda542b32f84117bdfeacf97f301928\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d17a205aa37a44b8a51c7e2445d32c34 : SUCCESS in 40m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/094b2ac684e640c38ea8f672bb1c1b9f : SUCCESS in 39m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad329116d73a2038344e81aec51aa3e0d7f346ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-15 13:00:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"905a3c5b8b743ee0f4a683844bf6e5d9404c0851","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:42:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a7e9c3c814342a9bf3c0313d68b8f53\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c2d41d53f4c4024b70abe989d81dd7d : SUCCESS in 44m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7dce966239bf4e679678b792a38bc887 : SUCCESS in 39m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d1f53fc323d045e928b87884533f0f3ac32cab4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 14:52:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c3892bd70b94f7aae7eb969a599f581\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/937f004642cf4bb6987a7db0ddd40226 : SUCCESS in 1h 08m 33s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/a67d40d281ce42c3bd31f1ba3fdaac6f : SUCCESS in 1h 22m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27e54ae7d616456d8a2d11dd78293dea : SUCCESS in 35m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e7a1bb8ef13458cb40a524ea3f43498 : SUCCESS in 15m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/965c28b8856b46638fd2229ea34d19dc : SUCCESS in 27m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54a61296c8b94d8990855cea73a93cdf : SUCCESS in 27m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c1c8408d1a4789bd2a5ae29121f4de : SUCCESS in 12m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0699778f97d7485c8156b1452cdee73f : FAILURE in 42m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a02b1e87f6b6482893d027c1378605d9 : SUCCESS in 1h 46m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/273839a8d78046b3bb00db0eba5e19e9 : SUCCESS in 1h 08m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a00a397f93c94b0b9ae194fc5d808f7b : SUCCESS in 1h 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0349d5f9aaa084a2249fc037b9dbca6e73f18b89","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-19 08:00:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"160f474a8483f6a94b051ea05759d404ff577feb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 08:52:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cc9cc408f2844e18dd5daa94cf21b55\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fab1c80e84e4f20b87504a1a52367da : SUCCESS in 50m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89f514d6d8ac4ea09a0df8bfebb549a8 : SUCCESS in 49m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e39f13193992916bd1e4b3d6574f8953ad16325","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 09:29:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42b139e294ae489a831b3ee6bb0c0ec0\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/c048fe0696ed4df7a22adec0d1c33ad9 : SUCCESS in 36m 24s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/3a3a4b6ee64946a2bb2bb079480072d7 : SUCCESS in 1h 18m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd2b06967819465eba82497a4b95d368 : SUCCESS in 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9529f41badfb448da25deb966254b2a3 : SUCCESS in 12m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7417a87630c94c0eb1fdd5435eee050c : SUCCESS in 19m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37bbb58cd1e4478798864d3e6286c26b : SUCCESS in 22m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c970d9748d94cd7ab6a57d79ab73b93 : SUCCESS in 8m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/060205c8e9264a2cad7ae79e4a25185e : SUCCESS in 43m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca877e02ef3641c1baff57810939637e : SUCCESS in 1h 27m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30cf7af3d0d04ca69c152f67accc2479 : SUCCESS in 1h 06m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6df1ad6366d84aee890be2c572d85a5f : SUCCESS in 1h 13m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5086306c68d5d145ad9b2bfdcea8c11d23600305","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-19 11:23:09.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"d4fd5dd49bfbdff0087712e9db57bd6ef8999b7e","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-09-19 13:06:18.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6ee7cb3d3aacd25d1d9ec94de3407f75d56291c3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-20 08:55:27.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"65a5812db49379b7974a511b1ca35bfa9a86ac9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-20 09:06:40.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5719dfe8b16cd0989d814e5760d145bf60f84b90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 11:00:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1633a73f94024881b85fc9ffa57a4c84\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7edc67020e2540a3a0c92aab06c1b166 : FAILURE in 1h 15m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b304659704b4533a9bb074f8cad9e5c : SUCCESS in 39m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"18e0eb389bdfabffad9428713841ed929906e043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 11:02:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1976bde4dcd4c78acf0a6c4958788a4\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/4b0c8b4dee9d4cc494e797a3e5c9ffcd : SUCCESS in 52m 36s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/ca849abdde884f5bb58fc9fec548ea44 : SUCCESS in 1h 33m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa659099bbbf40e08aa28be5a8b02210 : SUCCESS in 35m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5af3fc4acbea47d0abbee6169d0a676b : SUCCESS in 10m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f6098c1662046beb679aab27de64ac5 : POST_FAILURE in 25m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9024cd4b8a6c41c79698879319a8c4d0 : SUCCESS in 23m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20009340e0d84b51ba9c88cae302e3c7 : SUCCESS in 11m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/31607548b7514d45b6cf65e6e064db94 : SUCCESS in 51m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/888d0ae1af1646e892e92a38adfa2914 : SUCCESS in 1h 48m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65b5733dd5fa4b1e8f33b030c656b5ca : SUCCESS in 50m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8bf77b6708864ff99d64d3b6a42888d3 : SUCCESS in 50m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d85b47440dc95e6e23e161fef899f0b5ac34ba7c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-26 10:53:32.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"42ed77ab0b856154da597354dc203b632dcaf35c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 11:38:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e511213229a491cb752ab3bb5dc09b9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b2a0fedc3c84054828b56d1d001ef58 : SUCCESS in 42m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed802d8a01364a7999cad451e5ed2dfa : SUCCESS in 41m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0120aa4c261e52f342673ba52b7f4034de8597e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 12:49:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/601298d577ed410ab9f34f17a2ad483c\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/2f519549b14b4a8da8a2ed3bd526f296 : SUCCESS in 58m 16s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/e10076d5dd9640ec84e82b821f8b6c51 : SUCCESS in 1h 50m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b48a7226085b4c6e8593b86276297095 : SUCCESS in 31m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9362897a564e4ea93520ab4d0391d9 : SUCCESS in 14m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0e45cf599014414bf792bfe53d5ff24 : SUCCESS in 24m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55c9c7a278864353950f5e9d7cb2af2f : SUCCESS in 26m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8957cdd4a17d4566a2e9b924d80c1dc1 : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/52e2e97df7dc4173912609a12f4709f7 : SUCCESS in 53m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0896192dc32e4ff4a0b982c8dfa8ea4c : SUCCESS in 1h 40m 11s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3aaba2fdc4134f5ab29cc378369848f3 : SUCCESS in 58m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/401a9229b547419e9a88b52e4c40c61e : SUCCESS in 1h 05m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b490a882c5d2b9aaf25f54fc0e9d7f2ba043b484","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-27 14:39:17.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"52b268ca58fad8241d25eeedb64f53145635c6b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 16:23:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b06dea3d21a4a45b2902c61217557bd\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/8370ebbabd004070ac3e1e4c9679153c : SUCCESS in 1h 00m 21s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/0d9d37e0a60c492d9144fa5960b0c045 : SUCCESS in 1h 13m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ce733da4dc2409199aca4c2f66e3220 : SUCCESS in 32m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a34a770aba1d4a3fb6a07d6a1a8c8e8c : SUCCESS in 14m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42b769642b9044a083f93e9970926b4b : SUCCESS in 26m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8e56456f0fe4a219883eb43df2df6ba : SUCCESS in 22m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de4835f4561449508cd7fd622536b587 : SUCCESS in 10m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/21923ada9c9148b29733052158493cf7 : SUCCESS in 36m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a317225d00c046eb8482317be2855c3b : SUCCESS in 1h 38m 58s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2921b924f3646ea9ea5664775b55e77 : SUCCESS in 42m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/943f69f8f6f74d42a49707498be86ffa : SUCCESS in 1h 03m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4e75bbecda73660b5c87c8131ebe6d527913217f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 18:14:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dfc0af92ecb45fca4e0e412628e9c04\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a347490f44c64f1986a7d80398491226 : SUCCESS in 56m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e5178316cee4e339c4e1a71b08c0b24 : SUCCESS in 58m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0c300e7d6cb961be41d0f39e8b4a1ccc372efab7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-09-28 10:17:18.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"556e706d0daef73101be5ecc9a2d8d3c9abfd084","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-28 13:45:16.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b33a660118fb5fc5c470d9e15a6cb73b6bb5c822","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 13:45:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"eb9ec7ad520d99620628352b360cffac5bb58d04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:20:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7dceb268f3e424c96eb2edda20da8fa\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/551ed887e39b4619a040f132d6a0d71f : SUCCESS in 1h 00m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74e79b753aa9454f89ec4902555e03ef : SUCCESS in 15m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3deba5392f764a62bbcb1a1681cd1fb4 : SUCCESS in 24m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/418e737174534e84b0554cf0ee8836f0 : SUCCESS in 20m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83716c7c676a4a5d84587874bc2c97b0 : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/55fbcbd8876c4e349e6a51e3549a51d1 : SUCCESS in 51m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1259b3e2e46540deae1dfe588b5544e5 : SUCCESS in 1h 33m 15s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e7c6d3a81ff434ba7b09f2ab0bb2717 : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":4},{"id":"bddd08b993a851a5158ed0561ea5802f6efcf0cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:20:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6bf0d24d46e7607e1483cb9f4ebad1a3fbadbc65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:22:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4af55ed97e524d75b26e36bb995199d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382f746e081148db81a5d5167c708f8d : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cf096344b07b80524c3888e44e0b895465598a74","revisions":{"5d12aecc0750a5c39e9a38d0721d6a04692a275a":{"kind":"REWORK","_number":1,"created":"2023-09-12 15:30:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/93/894793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/894793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/894793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/894793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/894793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/894793/1"}}},"commit":{"parents":[{"commit":"de424f5991b6fff9d0436aa29f138a3375d0947a","subject":"Revert \"[OVN][Trunk] Set the subports correct host during live migration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/de424f5991b6fff9d0436aa29f138a3375d0947a"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-01 11:43:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-09 11:18:42.000000000","tz":0},"subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","message":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\n\nThis reverts commit 955e621167987626806128d259a29d4e62acf102.\n\nReason for revert: the port binding handling done in this patch is\nincorrect and leads to issues during the cold migration process with\ntrunk ports in ML2/OVN.\n\nChange-Id: Ifc2d37e2042fad43dd838821953defd99a5f8665\nCloses-Bug: #2033887\n(cherry picked from commit 6ffad1f31fbb95247d6e5911298439ae61ef4ac0)\n(cherry picked from commit 5438714f84b791e93587c6ddc3e77f583232a4ad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d12aecc0750a5c39e9a38d0721d6a04692a275a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d12aecc0750a5c39e9a38d0721d6a04692a275a"}]},"branch":"refs/heads/stable/yoga"},"dcb18c88a1595b5fab2338965d3465ad835d2e1a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-15 13:00:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/93/894793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/894793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/894793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/894793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/894793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/894793/2"}}},"commit":{"parents":[{"commit":"f5523b2977bb68478c70bdcbb46c3abaa76924cb","subject":"Merge \"[neutron-api] remove leader_only for sb connection\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f5523b2977bb68478c70bdcbb46c3abaa76924cb"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-01 11:43:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-15 13:00:35.000000000","tz":0},"subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","message":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\n\nThis reverts commit 955e621167987626806128d259a29d4e62acf102.\n\nReason for revert: the port binding handling done in this patch is\nincorrect and leads to issues during the cold migration process with\ntrunk ports in ML2/OVN.\n\nChange-Id: Ifc2d37e2042fad43dd838821953defd99a5f8665\nCloses-Bug: #2033887\n(cherry picked from commit 6ffad1f31fbb95247d6e5911298439ae61ef4ac0)\n(cherry picked from commit 5438714f84b791e93587c6ddc3e77f583232a4ad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcb18c88a1595b5fab2338965d3465ad835d2e1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dcb18c88a1595b5fab2338965d3465ad835d2e1a"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"962be5a145c391e33b8b7b996980f7092721e428":{"kind":"REWORK","_number":3,"created":"2023-09-20 09:06:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/93/894793/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/894793/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/894793/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/894793/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/894793/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/894793/3"}}},"commit":{"parents":[{"commit":"f5523b2977bb68478c70bdcbb46c3abaa76924cb","subject":"Merge \"[neutron-api] remove leader_only for sb connection\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f5523b2977bb68478c70bdcbb46c3abaa76924cb"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-01 11:43:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-20 09:06:40.000000000","tz":0},"subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","message":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\n\nThis reverts commit 955e621167987626806128d259a29d4e62acf102.\n\nReason for revert: the port binding handling done in this patch is\nincorrect and leads to issues during the cold migration process with\ntrunk ports in ML2/OVN.\n\nChange-Id: Ifc2d37e2042fad43dd838821953defd99a5f8665\nCloses-Bug: #2033887\n(cherry picked from commit 6ffad1f31fbb95247d6e5911298439ae61ef4ac0)\n(cherry picked from commit 5438714f84b791e93587c6ddc3e77f583232a4ad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/962be5a145c391e33b8b7b996980f7092721e428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/962be5a145c391e33b8b7b996980f7092721e428"}]},"branch":"refs/heads/stable/yoga"},"cf096344b07b80524c3888e44e0b895465598a74":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2023-09-27 14:39:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/93/894793/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/93/894793/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/894793/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/894793/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/93/894793/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/93/894793/4"}}},"commit":{"parents":[{"commit":"a4efc2dd80f1628e794e761d2d3e9e6c27e992b5","subject":"[PostgreSQL] Subnet entity with ServiceType grouped by both tables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4efc2dd80f1628e794e761d2d3e9e6c27e992b5"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-01 11:43:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-27 14:39:09.000000000","tz":0},"subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","message":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\n\nThis reverts commit 955e621167987626806128d259a29d4e62acf102.\n\nReason for revert: the port binding handling done in this patch is\nincorrect and leads to issues during the cold migration process with\ntrunk ports in ML2/OVN.\n\nChange-Id: Ifc2d37e2042fad43dd838821953defd99a5f8665\nCloses-Bug: #2033887\n(cherry picked from commit 1b034f8d62091724a6fb639f34c2f8936329b501)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf096344b07b80524c3888e44e0b895465598a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf096344b07b80524c3888e44e0b895465598a74"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
