)]}'
{"id":"openstack%2Fneutron~895155","triplet_id":"openstack%2Fneutron~master~I1ffd5878489176f33146688de03367101ede2305","project":"openstack/neutron","branch":"master","topic":"caracal_db_branch","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-19 15:54:11.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-09-15 22:21:58.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-09-19 20:24:07.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-09-15 14:53:46.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I1ffd5878489176f33146688de03367101ede2305","subject":"Open the 2024.1 (Caracal) DB branch","status":"MERGED","created":"2023-09-14 10:07:42.000000000","updated":"2023-09-19 20:25:40.000000000","submitted":"2023-09-19 20:24:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895155-caracal_db_branch","meta_rev_id":"442d155856ba7930f94eeb1c49a53fce95b57d1b","_number":895155,"virtual_id_number":895155,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-09-14 10:42:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-19 20:24:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-18 04:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2023-09-15 22:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-09-15 14:53:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"date":"2023-09-19 15:54:11.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-09-15 22:21:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-14 10:42:12.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-14 12:18:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-14 12:36:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-15 00:23:57.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-09-15 14:24:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-09-15 14:24:15.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-09-18 04:36:10.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"383e754f12177c961907a4f9d0ff57c78c6a283a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-14 10:07:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e97375cb99efc2530e3bf1d1846821fb97f5bedf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-14 10:07:49.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"51189ed25ae19fc083ef3410e237a15789fefb60","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-14 10:07:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"3bd0c5af779ca37191d2e1973c2cabe70873ac20","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-14 10:42:12.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/55/895155/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7aaddf451879c8c0d28640112796d81e4fb20c4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 12:18:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac8dcfd251d04eb1b0be607e5e40bc9f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3fb2490ca254620a9b62e0b99d9e3de : SUCCESS in 52m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fff66ba9ce4b45c5a47b67ae1090226c : SUCCESS in 1h 06m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5c38bfe26414ddabefe530754181e4e : SUCCESS in 42m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b5fcb939769243348f72fea20f1cdef9 : SUCCESS in 1h 03m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"acfa356ff4172cd0b27c149f140a61c2b7843026","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 12:36:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cce44f5090ef48dcb4f048b455ac8fbf\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f0c44fddc1e34b9a91dd8b83906ff7b4 : SUCCESS in 1h 23m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b8e7806450f74eb596f124f854057aef : SUCCESS in 2h 12m 33s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/31bf72a6453b4ac1add6457009a96e45 : SUCCESS in 1h 34m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c53a482f0adf4c9ea2e118a16410424e : SUCCESS in 2h 14m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0142b5415f2d42b7b398329ed557edcf : SUCCESS in 36m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/247e4f4b6b3c4736906c4bb2c5475b53 : SUCCESS in 39m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c1fd29400540908d9149e67104f2c4 : SUCCESS in 11m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcd605bb62ae4a2ab641e38042543713 : SUCCESS in 27m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/298d8d7e26d543b2a21328297411ca99 : SUCCESS in 27m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38de88eab9c4448a8061be91dd4e3372 : SUCCESS in 24m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f7b0dc375084be98abc8973ee972f01 : SUCCESS in 27m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203c71ecf5df4767b2fa04ef2cbd19ab : SUCCESS in 12m 49s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/bcd4a4bfca934b01b0b06651bcd7540b : SUCCESS in 26m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f137d753164043f9ae20e8372318b910 : SUCCESS in 47m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/be39a2007db04ef58bce94dc3e03e2cb : SUCCESS in 2h 19m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/267e0d9e7a844ad19077b5a943857301 : SUCCESS in 57m 06s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/46cfe72cd0c34041b4d713983ef5b41b : SUCCESS in 56m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eeaa13c6eb614d29ac26e05aaaf43946 : SUCCESS in 1h 17m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6e7c6a419c1d40fcabdcceed69f2f137 : SUCCESS in 1h 19m 36s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/396b06ee5e2c40cea5b1eae92a3ce6ef : SUCCESS in 1h 00m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6da82e273e904120b7539aebd1a4ca0a : SUCCESS in 2h 02m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6a9e188f15fb485fabe3cf110c3462f0 : SUCCESS in 1h 55m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b305437c8d2b47bfb553517f13a8c4ae : SUCCESS in 1h 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ac920cb94a0fd612dc37f7f0ac15976b51eddd26","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-15 14:23:50.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"e87bdd426f24681235a0153e9551113aa19de263","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-15 14:23:58.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"929e16d1d284784c7bdfc4b0bf545b04d9e70b59","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-15 14:53:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9421f903af17c982221d1db89961165ef8b41e22","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-09-15 22:21:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3672a94ead54fc0c50766782e6f2b947884dba8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 22:22:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b077120a83c3e4502bbd7fb6d5997cb3cd3d23c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-16 01:18:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8722b830a6854e2eb60a472d63c58d47\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4b95803ed0894030b1775b9a8031870f : SUCCESS in 1h 54m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f837191799144ae89998771411a057d9 : SUCCESS in 2h 06m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1582390554d34988bb94f27c57112486 : FAILURE in 2h 05m 20s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5d8b0edfd93c426aae5307750cefbfb1 : SUCCESS in 1h 56m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79250a3e042b4aa9b0f4a6662f74f130 : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/211be86671254c6aa8c1dff46042c485 : SUCCESS in 20m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/faa84810663a4d05bba1d603790fe9df : SUCCESS in 18m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53945df2e5ec4c27ab98fac63c45e692 : SUCCESS in 21m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c3166ae8af4bb39570e51d60ac5ccf : SUCCESS in 11m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/16a09d1782b74295926c15dd78c77f4c : SUCCESS in 1h 00m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/679e55e34dd04abd9003b0c40f20c22c : SUCCESS in 2h 50m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ae8ad77ae8b647458a19d0dcc96c6f20 : SUCCESS in 1h 23m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b3ccdc8f165b4a55ae6728bb0dec3837 : SUCCESS in 57m 50s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e0d77889d11f4eb58ba1884a4262eb39 : SUCCESS in 1h 02m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/938227735f0e43a8b360d9f5219c428a : SUCCESS in 1h 00m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/9819c01b72fa43e8af22fabd63f71780 : SUCCESS in 1h 17m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6714b10164952e7c146a38a50e4598650f9e3918","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-18 04:36:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9738e7dda5ed101ddd87c5949198073abcc206dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 05:24:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/688ff2e2b9bd418eaa1125695fe74e82\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9650e25f95014848a0696dc45e42bc8b : SUCCESS in 40m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27916b9f3ceb4f7fb37f0fc8379bdfd8 : SUCCESS in 46m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f83874b450848968dada2a7f74c0812 : SUCCESS in 46m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/08ca67a49363425bac2a0d5640c7fbd5 : SUCCESS in 42m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e0ee124b59811f2c669fb848be26cbb82e700d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 07:40:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fa5bc9856934cb18a07f355dec4e492\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/29b1a81b6a2a4f43ba74b9c2a3e9dfab : SUCCESS in 1h 16m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e3447d57971c4476a5b6c191cec18d60 : SUCCESS in 1h 40m 55s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d79edab513ce4befa261352c01d4f33e : SUCCESS in 1h 23m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/804c171eea454ab1a97e6859fc6d1a4c : SUCCESS in 1h 39m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5a488a27e3684d02bfe84eb6bf3e5c27 : SUCCESS in 35m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b07fef52ebc74a8caf73b4f66ba53f0e : SUCCESS in 27m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3aa4964aa0e4a97a195bf8eb58e2221 : SUCCESS in 10m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07babe73f48649509ff03f6fd24eece1 : SUCCESS in 19m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a55e1c8f1e844f69db2bd390d888bb9 : SUCCESS in 21m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91805c68487d4c84bf1b3267da106297 : SUCCESS in 24m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7397565f07654eb48fa9a5bb0f0e2d66 : SUCCESS in 19m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec3f1afd0f342b698979f920ad6ae88 : SUCCESS in 11m 46s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9d9809e3dc764c8282700704a45edc7d : SUCCESS in 27m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f086b76c1c8e444296ed6939ed02139a : SUCCESS in 57m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f34a03f493949368d2864010fb9caab : TIMED_OUT in 3h 02m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4dd4901b563e4a69a0db3e35747e8111 : SUCCESS in 44m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9e98e103da884443a125d793a332093e : SUCCESS in 58m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d5d37bb47ff64480bc81081523d158a9 : SUCCESS in 1h 27m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6cf166b6a6dd4a6e9b6db43951e2b660 : SUCCESS in 1h 25m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/8bf3d09db85e40a7917452b794b0a2ed : SUCCESS in 53m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3cdcc6a176704b79b576fbf130e93376 : SUCCESS in 1h 20m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/32703e6d0c954564b8cb7101e9b32711 : SUCCESS in 2h 01m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/2cb3ac3671a74120b49a6e4ab9f534fa : SUCCESS in 1h 17m 16s","accounts_in_message":[],"_revision_number":1},{"id":"65bad0ae7ee060aef6f1e83dedba030d489f0a34","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-19 09:48:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"381ddd21e841e46d43bc7203b16ed0d92b553841","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 10:44:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5452a273b55d4532a8c326abba6e3a3d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae9d781dcea4496c8e9a62995803aa97 : SUCCESS in 43m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13e012d521934140b72a8e901e827762 : SUCCESS in 42m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e68f225fea2046d58a86a935997b4440 : SUCCESS in 48m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1a19b2bc43a24d119dcd8cf5e68baa71 : SUCCESS in 37m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9523b25871d7fa24d260f076ffa4e1b3d71100ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 12:30:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/776bdb53427d43f584d7316941f94e1b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d931f37084864266aacb0f2f9bd13630 : SUCCESS in 1h 37m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ab946184fbe14f47ba6f5d1d77dd989f : SUCCESS in 2h 02m 22s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1632d8de38cb42dd9b9f16d52ad0a706 : FAILURE in 1h 37m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dd29cb3a90404f4c84f5beb915993ca3 : SUCCESS in 1h 33m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1c8f76ba01474c81bf8f29eeb3d8a183 : SUCCESS in 40m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb083f9fee5e45178cb661ebcb5dfa3b : SUCCESS in 34m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a99585180ed7418db000d8b7f70863ed : SUCCESS in 10m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d84619dbe1c84ce7a881a46c13d08544 : SUCCESS in 21m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bac9552b68b4a99ad798bb2d525bf9a : SUCCESS in 27m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4cdcbd8566f94cb6b17aa2559bbec9fb : SUCCESS in 24m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e5c02b97b8f4cc5862ed3a0f02be838 : SUCCESS in 20m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f5794775f64f20afb8afc266cc01e7 : SUCCESS in 14m 39s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0856ae41b0084840bb5ba56ceb4ab102 : SUCCESS in 29m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2035badf84594fca90437d050c9267e8 : SUCCESS in 1h 02m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c89110b2ec4a4844a51919bf23670113 : SUCCESS in 2h 33m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d70603937b544839aa330f5370b212a : SUCCESS in 55m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/49690609dead44ceb5fa9154f3c79d43 : SUCCESS in 49m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5b24d47ed779412a8e2eed5b7ea786db : SUCCESS in 1h 02m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/452fbdf9031048acbe43ebcb5f9d33f9 : SUCCESS in 1h 21m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f414b10495bb47b982631bfb980ffcf6 : SUCCESS in 47m 56s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fbda039d5def4cd0ae0b29b09181f24e : SUCCESS in 1h 17m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aaaa4bf97080479d861f9819c1a19895 : SUCCESS in 1h 43m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a68a8dd25f554e05bdbd3e9fcc85db21 : SUCCESS in 1h 04m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0ffad0abd17534d8a95667ca381e97e812b8cf4a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-19 13:05:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36cf90381ce4794ae7e07be03f7595507abf7daf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 14:24:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d595a38a319491cb1f2a4cd8c1f1c35\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43f16a63810347748131979ddf391f82 : SUCCESS in 41m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/618728b733a0441cadc098039e45fd42 : SUCCESS in 52m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7152c451d495489eb9c84252df32204f : SUCCESS in 43m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/794d1399cd7f49eb8406342f375453af : SUCCESS in 50m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0acc004c96b71c91491e1c475a7c909f2db4e802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 15:38:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a2426aa3b4445959b9cfcce9cba2eea\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0bb0c6d4f5ce4f5b99e664f2d6b892de : SUCCESS in 1h 48m 58s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d575b0d96a53401abea4024b11920ef1 : SUCCESS in 1h 26m 10s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b93b0f90f8a94d1086c54d056ab55d9f : SUCCESS in 1h 51m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/ba41ac1e4a044889aceec17c04859897 : SUCCESS in 1h 34m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f61b964286554c3a95daa72a31f30f50 : SUCCESS in 45m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a936330e26b413b860e7405971e792c : SUCCESS in 47m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dfeac25a99b44009d2a24a8229c93c2 : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/096f847acf8d4024a20acdfbd47daee9 : SUCCESS in 24m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/863932c2e3644c24a106a3f8d009eaba : SUCCESS in 28m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2354a398c3a4d1898c9e327844d7549 : SUCCESS in 25m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ff881664c8e41b8aa680d4fc594d945 : SUCCESS in 24m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b051e377de64cbfad6ee7f640c6bf89 : SUCCESS in 12m 46s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/ee4d8274a7d947819b311a690de4a476 : SUCCESS in 30m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/65180525e6c74c1f9b39784e8c16350f : SUCCESS in 1h 00m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/138fb742e9604dd8956a117637c7fa2a : SUCCESS in 2h 27m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0bdc8e859425487f8c5da51bc1393d73 : SUCCESS in 58m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dfd7880134ca4ee78ae5f2e9dee99bd7 : SUCCESS in 1h 06m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ac1457f6604e4e62b5d398fa3faf069f : FAILURE in 1h 09m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e63a2a5bdf8f43d982db09b1b20bec28 : SUCCESS in 1h 22m 25s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c29eadac94f14f06b70b508479f20988 : SUCCESS in 1h 07m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b36765750a484306915c5ba006c1f3cd : SUCCESS in 1h 52m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4b1ef1bc6ffe4e9993a8d2e3bbb07126 : SUCCESS in 1h 52m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/80e0e45629404665ab0e77b79f2a251f : SUCCESS in 1h 21m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6a6a8a0cadfa3cf1f4981047349896a0fb44d7d6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-19 15:54:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4771b96259db9451451b7bbf642d165c13daa16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 17:33:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72c2cd3a0fa24fa7b456f326a62d37b6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03b8e304d716494896973731bd6a07b1 : SUCCESS in 45m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd8ff530a95f451db3f2bb54ed8b1582 : SUCCESS in 45m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/06857757164b4a06a2c4bd76acca5a71 : SUCCESS in 40m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/161c64b0e0e943caa64eecaeff013a4b : SUCCESS in 38m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b35727c8d81b644fa36f122c7a79b557fd69fe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 18:09:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/177e4c39f05047ef87792d5ad4a3f568\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/83e47adfff3f45feb5bfdc5d5c90035b : SUCCESS in 1h 27m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/57b243f89c36432b8e824da1f7e90567 : SUCCESS in 2h 01m 02s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/88bee1d5575b43d389fabfd96f81e02e : SUCCESS in 1h 48m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1c48a6dc74284341ab692a126bc3b4da : SUCCESS in 1h 54m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f10da638d9c84fc0af1eb6059c1d0ec0 : SUCCESS in 45m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/904dd505646f4b879c4e6480ddb87867 : SUCCESS in 29m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cef57630ed1440298eb7ebe289cce563 : SUCCESS in 9m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ad213e5596f4764b2f367f0b3227641 : SUCCESS in 24m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2867114c79fa4e51a1ca1e46f1852ece : SUCCESS in 28m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3c3a7ef96244032a60ec1da5a95d99f : SUCCESS in 27m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6ea317c6a9845cfa216ea8dd798a0a2 : SUCCESS in 25m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b667a57592a94fd3987c25781011e554 : SUCCESS in 12m 16s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/6e40fcfe22ce40f29c7feec5f710c5ac : SUCCESS in 26m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff88bb6798eb4ddb84f3abefd142c7d1 : SUCCESS in 53m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/37d981b211d846909aa357d298d366f6 : SUCCESS in 1h 56m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b12dec2f6e194c6c894b2a5ce7876d4c : SUCCESS in 44m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8010a772a39b4daeb4f2045d758d67cd : SUCCESS in 43m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ac8ccf27283e4ebbad15ac6939e050d6 : SUCCESS in 1h 17m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/29f3efb415124950bb346a24b3264e1d : SUCCESS in 1h 15m 34s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/df69662f231a4f8daeb6173be11f40f2 : SUCCESS in 47m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/075dc28eb5834711a7f512b0fc34b40e : SUCCESS in 2h 08m 59s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b2149004ed604b5eb1a274068e38fdbe : SUCCESS in 1h 28m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a59e6d80d34d45c49449e9c634691869 : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4bcd42ea4d29bc15786d37a1f4e0f60d38daf616","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 18:09:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b51ae808cbc30c213034c6b634f9201fe4c52e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 20:24:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30ef2c2ec180497da511f7dc8cc53deb\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e4c2fc0d7d50461185b536870639b216 : SUCCESS in 1h 23m 24s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eedbc3df567141a08dd0c59c48245390 : SUCCESS in 1h 54m 29s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eadfe6c0c8744540a773047f2386b757 : SUCCESS in 2h 03m 13s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4f213c4aa46a41d7861ffbef0f0c106a : SUCCESS in 1h 28m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec20db9accc84622844ab9206b2856ac : SUCCESS in 11m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a46336681e69445e89e61c815d1d1f8b : SUCCESS in 23m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d79a6fbe5a74fdf9353c47d5f4cda6b : SUCCESS in 21m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe3791aff5c7462e87485f423c9f7c04 : SUCCESS in 29m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0bcca239ae346f0b2bf01375df1996b : SUCCESS in 11m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b56414cb959441e49340cce7966680ae : SUCCESS in 46m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/636da84106114e419435278e3778adb5 : SUCCESS in 2h 13m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d9e3cb41cf7f4b50b52020b62c49146d : SUCCESS in 1h 26m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3222d5612f6c4988a1d6afa44bc32351 : SUCCESS in 1h 02m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7ddbc8e49d9e41188f3b42585ee6fe01 : SUCCESS in 44m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/7cbac807504e40f4a9cd9caf017fd475 : SUCCESS in 53m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6c650014bc4c45a49688151e0e1c6736 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":1},{"id":"986c4812cf07067208b757c42c81887d99d67275","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 20:24:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"442d155856ba7930f94eeb1c49a53fce95b57d1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 20:25:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7ae610569eb4001823aed1029f5c9fc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c8179422c34ac0a4b48f09c8599a22 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e0ba8bcdaeb1978cc11d8072a345637a6e44b38","revisions":{"7e0ba8bcdaeb1978cc11d8072a345637a6e44b38":{"kind":"REWORK","_number":1,"created":"2023-09-14 10:07:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/55/895155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/55/895155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/895155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/895155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/55/895155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/55/895155/1"}}},"commit":{"parents":[{"commit":"741f504c7badb3f6799977bd1e98ca5a9b7d1f53","subject":"Merge \"Prevent internal IP change for floating IP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/741f504c7badb3f6799977bd1e98ca5a9b7d1f53"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-09 22:13:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-09 22:13:40.000000000","tz":0},"subject":"Open the 2024.1 (Caracal) DB branch","message":"Open the 2024.1 (Caracal) DB branch\n\nAccording to the TC resolution [1] and the release identification\ndocument [2], the release number will be used as primary identifier in\nthe development cycle. Release name will be used in marketing.\n\n[1]https://governance.openstack.org/tc/resolutions/20220524-release-identification-process.html\n[2]https://governance.openstack.org/tc/reference/release-naming.html\n\nChange-Id: I1ffd5878489176f33146688de03367101ede2305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e0ba8bcdaeb1978cc11d8072a345637a6e44b38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7e0ba8bcdaeb1978cc11d8072a345637a6e44b38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
