)]}'
{"id":"openstack%2Fneutron~895402","triplet_id":"openstack%2Fneutron~master~I66fa3c541db54c5f57404681f30b141fe8dad84a","project":"openstack/neutron","branch":"master","topic":"bug/2036118","attention_set":{},"removed_from_attention_set":{"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-11-03 16:33:05.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2023-11-08 02:16:03.000000000","reason":"Change was submitted"},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2023-11-08 02:16:03.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-11-08 02:16:03.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-11-07 23:45:11.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-11-08 02:16:03.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-11-08 02:16:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66fa3c541db54c5f57404681f30b141fe8dad84a","subject":"ovn: Spawn metadata resources on additional chassis event","status":"MERGED","created":"2023-09-15 14:14:55.000000000","updated":"2023-11-08 02:19:11.000000000","submitted":"2023-11-08 02:16:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":56,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"895402-bug/2036118","meta_rev_id":"eea72aabebcb95090ac02be6517951da68d8d67b","_number":895402,"virtual_id_number":895402,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-11-02 21:08:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-08 02:16:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-03 16:33:05.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-11-07 23:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-03 15:50:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-11-07 23:45:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-15 14:17:22.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"state":"REVIEWER"},{"updated":"2023-09-15 14:56:49.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-15 17:30:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-15 18:59:15.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-09-19 12:44:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-09-19 23:13:20.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"CC"},{"updated":"2023-09-29 14:15:29.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2023-10-03 21:39:57.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-25 10:20:04.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2023-10-30 01:25:56.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"40eb6b9258016928b89ebd63658139bf648fd9c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-15 14:14:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd6722902dbeb0e35483ef827bc88c11575e990c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-15 14:15:25.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d71280259c9838d7be487a325dfaf037ce13e17f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-15 14:56:49.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/02/895402/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"0360821622de5a2796dbd1a94bbc3cd83d9ebd67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 17:30:48.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2347a7eba5c54edb99fc5c110811b931\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8f8b65661bcb4efe9a1ccd630972b7f5 : SUCCESS in 1h 34m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f857188604e243b0a084855e327723c3 : SUCCESS in 33m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52f657b370bf4592b2b4f597995c2d6f : SUCCESS in 37m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d62dfc40b5a84581b0ba58a6f6516fc6 : FAILURE in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d956cca65794435bb0f7ea38ee43d7ac : SUCCESS in 27m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fedfc2c749a5458c95e2e64a4c852a72 : SUCCESS in 20m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d49f3a56dcdb4136991d22994d2c0733 : SUCCESS in 23m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4825da5594674363806fa038c1f3e8f2 : SUCCESS in 17m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a378f19d864fa09640bc268421126f : SUCCESS in 11m 03s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/651a6a28edca4f74845c441ede89e625 : SUCCESS in 30m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d4970e3948e445c9a1dcf5b2e1bbd535 : SUCCESS in 1h 03m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/86f2ef5e14f44b9a88d50b1dd54034c4 : TIMED_OUT in 3h 02m 17s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/25d317a7e1fe4aceadc81a784ad08c04 : SUCCESS in 1h 06m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8a52a7e46ac0475d87ca2dc87e7e8fc2 : SUCCESS in 2h 11m 29s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d8ec360ae3704ba6b2ad17f61661bc10 : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":1},{"id":"720108dbf8721bde7e0f57c76570a23cfaa9c7ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-15 18:23:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f151f36f36c42de8ac5217f964976d6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc3af3c8430a4e69a34e3786b3bf4b74 : SUCCESS in 42m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c18fffbd1ad44901b4667734799cf60c : SUCCESS in 1h 12m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aff57545219745729bb59a5871a4d662 : SUCCESS in 42m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0fd05cd720064f0ba16f23546397aee7 : SUCCESS in 42m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60cb0c9f60e542cbe1e4cfd08992722c3d411beb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-18 22:33:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"56ba2c91212fc8bdce21ce2bd09b8a35fb593203","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-18 23:06:36.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/02/895402/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"1c67ca63daa7bb69e3a512709c9a7fc3d970637c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 23:32:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8516abd9470f4843a49f74741f3064c5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/39456e5382d3435f998a5e2bdfe15224 : SUCCESS in 56m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3120140c61f7482996b5384d7552bbf5 : SUCCESS in 56m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a778e64b21b94aa5a2299cbf4c969109 : SUCCESS in 56m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d8ced8f362dd4bd9b9debe99b7c84f03 : SUCCESS in 51m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84f12539a418b8329dceee3dedb71a1ada8355be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 00:38:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ddf5118b1a24717b12069e021445e74\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/85abcba273444ddaa88d9be24f1d156d : FAILURE in 1h 28m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a38dd430b8024448a5c34fbf163b4040 : SUCCESS in 33m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b575bb000b8c465c8c7e32e44e8fde8e : SUCCESS in 27m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70512bc621394007b5977e7d04f985e4 : SUCCESS in 9m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80ca9e10059d490da3381561eab82f15 : SUCCESS in 25m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd3ae30eb4764515a67dd9a6dc540b47 : SUCCESS in 21m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4164c56415b470ea35e2a328f2ca580 : SUCCESS in 20m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ed9d704e64840b68dc1f564b0cb4f94 : SUCCESS in 22m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da18fbba49442e8bbace726cbbc7a0a : SUCCESS in 12m 06s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/777a476b5f4d42f383ca82a9d76287ee : SUCCESS in 29m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f4dd99ddd1b49efa7ed497996c8e8d7 : SUCCESS in 47m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ab1391cfe84c4dab8259ad32da58f567 : SUCCESS in 1h 53m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/acbbe3e6bc04436e826a962947999976 : SUCCESS in 43m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d60ae2c1c43d49d995637a46e4e6e395 : SUCCESS in 1h 59m 03s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/aa56893612b742a4a893336373e95ae3 : SUCCESS in 54m 21s","accounts_in_message":[],"_revision_number":2},{"id":"045795eb772b6426541570b752a0f7f9b2235d3b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-19 12:44:14.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c34eb408dd9df1b853b3bb0eee7ad0bdb4634ed5","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-19 13:16:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e258899e47ff635aaef5bb640c6026c3fa17ec39","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-19 13:23:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"29a27e69cd32e4b7266584c5024d1ce8eea29ab5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 13:56:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95fc6e0cd1e540808090bc20dd34af29\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ce651df09c64dfaa11f18ac7c35df6b : SUCCESS in 41m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e97f89495fd54e3588721098a6704ea1 : SUCCESS in 40m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7aa65adc1e754769a950b9b89842d71a : SUCCESS in 42m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c99b3b1517624483852e07e73d154f22 : SUCCESS in 53m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab209be253234ba9608901893149a9974ade8ae2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 14:45:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a23d6d9f33b648aabacad42f62be2ae9\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1788076d63184a37a3f79232c781d92b : SUCCESS in 1h 17m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/734ae433bbeb459d9dd2a422d71e4179 : SUCCESS in 31m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16b076557b2c4dc3858797d11c31d911 : SUCCESS in 37m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8a455f174b544d9bf399be39d5271ae : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2de1079d6924a36875a55a87a3bd203 : SUCCESS in 23m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53a2b919daaf4098a8f6b86d303a2fa6 : SUCCESS in 28m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4673ce6a24be43b3a781d050dfcc1b99 : SUCCESS in 17m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bddd9bbc2aaa4177a5e50cd3b670441a : SUCCESS in 19m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9553349f86340ee921851b75eb7d586 : SUCCESS in 12m 15s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/de8bb524e35641398a87aa56679d0ece : SUCCESS in 28m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1df40f1b9a784226952dabe2972aac9c : SUCCESS in 1h 03m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/103752f222c2452eb032c3b033cd88f7 : SUCCESS in 1h 52m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/27269d95e2584f03a312e600228b6d03 : SUCCESS in 48m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fd2dbcd00e6c4d9c8cabdc9d2f8e9c55 : SUCCESS in 1h 56m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/bd38d44e61e4490383f950757b89604a : SUCCESS in 1h 21m 49s","accounts_in_message":[],"_revision_number":2},{"id":"5d6f67902e749316e2842e2c4184bcafd2204347","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-19 21:12:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2abe98b84dce64340460a3a8eab70d5ff5eaebb7","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-09-19 23:13:20.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"588e16501ad89d75725e405ba5426d35e6915229","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-20 01:21:12.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b0c8aaea6160f21f0697a0ae2a7b3a83674c00dd","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-09-20 12:40:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a695d5148d8b442a01c06dae2bdf4842c101d4d4","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-09-20 12:41:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae681a97f37b9b5febae91f1c334613ebaf60b43","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-09-20 12:42:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4c3f2935b2dc4d9e78f91d3020db9047560e01c3","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-20 13:46:13.000000000","message":"Patch Set 2: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7e11acbdfc7ffa111aeb2a3b50d98054d140c938","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-09-20 13:48:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2cc192ca1c0f97fc0398dc643b004ae8ff9073ca","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-20 14:47:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d95aeb7b96e81e3aba2fd8bd9ff92468a179d7cb","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-22 18:01:39.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"2f1e19c52e1fc41a70b7b7631e1494e110a78e24","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-22 18:01:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"934f94133330894b9948e7b0eb0ee85a14599751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-22 18:19:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c2c4eb4147bb0f862d61bfb22cd7bcc9ae46e0ae","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-22 19:05:13.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/02/895402/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"c072161d6e635744f11a948e410901f7fea58670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 20:14:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0d16251e80b42a0b68685c1f903318a\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bda7fdaa3f154e8abab64115644f65b2 : SUCCESS in 1h 52m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ea62e67c1dd84e5db56023c1dfce6985 : SUCCESS in 30m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49af3b4edc384352944363377c2124d6 : SUCCESS in 39m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/308c10de224247edbfd5dc9313292079 : SUCCESS in 12m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/498b2f5cc43041c9ae0e7b3113590cd2 : SUCCESS in 26m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/840db3db91ae4c9895b948b726782987 : SUCCESS in 26m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd1a9b8c7e394f3d855d11542787d66e : SUCCESS in 25m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e71994a55c1e4492841a0b4591e51173 : SUCCESS in 25m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f5297000aa491195ea2af3633605b5 : SUCCESS in 11m 54s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9afa9276aa094847950b9cfa8a5f7e39 : SUCCESS in 27m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/96909ce241474efeb5dabaafa6f7eea3 : SUCCESS in 1h 08m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/74820f66bdf84dbcbbd34793d2023a7d : SUCCESS in 1h 54m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3540924d1e5f49ff951964a032f2b9ff : SUCCESS in 42m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/593ff9b09cce4cc5b7315018f13e8381 : SUCCESS in 1h 19m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a6c5f085e8f64145a38a710be3496e4f : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":3},{"id":"80bafc323f86da8bf381ad9b1a133c5f98c525bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 22:17:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e544d996bacf426d8fcc8476f66aae8a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56294f20885741c1acc38702fbc4f3aa : SUCCESS in 45m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52637032182545a18afcedc397ccfc9e : SUCCESS in 41m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/035d291d8ee5422fbef7395aa8c1faa6 : SUCCESS in 45m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8bbf022dafaf40c88f156eb86951f196 : SUCCESS in 36m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1dee43ba495abfd6de7ecd15954d0ea8bf431340","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-09-28 19:57:19.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a4879502bbb8d5fdaacef97e355977f12d83fb93","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-09-29 14:12:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2d47a4aa8983bca4e09bd61eb70762ccf3967127","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-09-29 14:15:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a1bb7f7b8242b56955756ec34f46a11ce4cbdc81","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-09-29 14:15:29.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"fc9f980539b9cbd0853f0f412ea4a7da2c659636","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-09-29 14:18:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ee4325c27e5b83c85e9fc540b0b19a67e39d28d","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-29 15:20:16.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"081c8b064ea395c231e3986d36243aace15709ea","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-29 15:20:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b46043941f224aa3222731998bbacf7c1b8eddfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-29 17:00:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"6a2bb385f86fd83abff3f384fb09f3a1d67be666","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-29 17:31:29.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/02/895402/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"6fcf5aeb77e6b51a3eb60440ce3ae9eeda39f85c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-29 17:31:58.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"926250b065185eb39d918ace0299245c86a02eb8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-29 18:07:30.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/02/895402/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"3b15c666c967de4e598c142df33ba145e158331e","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-29 18:14:54.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b7d3c73262167d5fbb5ed59e2027def2bff60902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 19:11:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05eec990874843d3b4dab0ed2544100a\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6bfa276d0edf4373b037cf6a6a53612e : SUCCESS in 1h 28m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/72d9a61c13ca4ce583dff1b8d68fa0d1 : SUCCESS in 30m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c28c57fc4aca46848674d57913635de0 : SUCCESS in 30m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ac3959982e64f988ff45221562c4f1a : SUCCESS in 10m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a777f40029b04404b866654d59d4b14b : SUCCESS in 20m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66e2931e75c54e22a141fa6b0617b79a : SUCCESS in 29m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/598c5e890af744539494b4096f05b6fb : SUCCESS in 25m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8cf9cfa90d644af79bd421069cfbf9eb : SUCCESS in 23m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16cfae3a47e84fa6bd992d2998dd1e51 : SUCCESS in 9m 15s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/fce6f5b2575b405f9308bf80ff4d35ef : SUCCESS in 31m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/db4a9e82e7db4eb8bfa2b20cdce97ca5 : SUCCESS in 1h 01m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5799b28eecaa448e89a339349dfd93f6 : FAILURE in 1h 38m 07s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d686deec68fc445dbc6399f97bd9b585 : SUCCESS in 46m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dc7a71c8d4034612baaadc5dadaeaccc : SUCCESS in 1h 18m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/88418a0e0fca497ca087a3e8a6d4b3a1 : SUCCESS in 1h 17m 42s","accounts_in_message":[],"_revision_number":5},{"id":"e2673ae14f6148d00838e5d330889301fe38c191","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 19:36:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaff86a4fe8642c6860e51e7b06b8520\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/75ef08b224f74e54934db900c7513436 : SUCCESS in 54m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c7ee43e274d34550a466e7fbc9dc3776 : SUCCESS in 51m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/84147965ac814beea2145603616ef697 : SUCCESS in 42m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/adc76747d0d346ab84fef2601a08acc2 : SUCCESS in 45m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9178de9418c0d4de382fe978153c515174961f98","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-03 17:26:40.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"dc407fd9ce6f82c3cf8ef69cd614bed4ca89926a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-03 19:25:22.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"185a35385b7ea23f6053f81d82ecdcf9741876a0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-03 20:24:15.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/02/895402/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"98a47be473d487e4a5e746723aa26062071ff0cf","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-03 20:38:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c098c2bf8d7305d29063ad148106b8e9f6009989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-03 21:39:29.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"2ccfedce3e7ae09fd3d7787fac3e8ad84ccf11c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 21:39:50.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 895402,7","accounts_in_message":[],"_revision_number":7},{"id":"08847f389f4dec2a697a6e66315b23a684994d4f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-10-03 21:39:57.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/neutron for 895402,7","accounts_in_message":[],"_revision_number":7},{"id":"d7ac04cb7b51e13ede6c29469a73b395f84a8a13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 21:39:58.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 895402,7","accounts_in_message":[],"_revision_number":7},{"id":"a67a7afbef4dc2dac03322fd54f7006895e5ddd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-03 21:41:36.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"865b4c05bca74631a364908488501b88d2007daa","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-03 22:03:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5ed4fb9bb06562b9b969183d1350491752fd81de","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-03 22:11:39.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/02/895402/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"69d68649336527d768ec84bdd5f0679af3d0b9b7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-03 22:44:48.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/02/895402/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"67ce7fe2bb4bece17777a0a8791e0a57392ce7a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 23:52:04.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/407ecd53c6c842ea859c701bd1b575c1\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c650afa73fa94473a85a8672eade8d12 : SUCCESS in 1h 37m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/76b34a881ce44d2dba9b0593f8ded9b2 : SUCCESS in 43m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a648c1c7c9c45398d364cdd72699e80 : SUCCESS in 35m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c74f878eca644619de0f511c7b841fa : SUCCESS in 10m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f43d99a4fe654b84b29e1fc80b660f3c : SUCCESS in 25m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e843b558b754b36bebaaeb2a79812bc : SUCCESS in 23m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f07b7d87f04f4bce90269b5f723fe81a : SUCCESS in 26m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28264b6789784220b9a39ad1e2d76a04 : SUCCESS in 24m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1358d93351c242559f2308d831a6fe5c : SUCCESS in 12m 20s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/25941845913c46448612c0947a0f5730 : SUCCESS in 30m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/03e8c53473524c479620c484c30a5d45 : SUCCESS in 42m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8607bd1750254a0facfa48b524de848e : SUCCESS in 2h 08m 15s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/32039da075a24673a82c84bce4d57267 : SUCCESS in 59m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4acd0ad9869c4faeabbc211fa6e5d046 : SUCCESS in 1h 57m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0767c570e32b49aba4d248bf2ca2e983 : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":8},{"id":"49ab38765a314b8c6845b00920cb7b2ab296f656","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 00:09:39.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44dfbd76b7ba4abd8a17e6c2153a7055\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1df9e89f12aa4b69a088e49a3d24b24c : SUCCESS in 49m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0209b320cc7d4c19b6ac7fdf5a1845a4 : SUCCESS in 41m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fa63ba4fbf604648ba9809e863a98b97 : SUCCESS in 40m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c84b1502edfc4f5aa424bcf9bb2db3b5 : SUCCESS in 38m 32s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"87ed383796c4dd8dd5bf85d97244929e5d10a4c7","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-04 13:51:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"359475128e93e57eb319ed70c2a01ddcde92f992","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-04 14:02:40.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8b7112926c4076b17046e7abf7fbc8ce0044c5bf","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-04 20:46:28.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8462b0e11bba2c5acf8ea2e54cbec00acde1385f","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-04 20:48:00.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"79288b592a0e7ad3324058bc77f8e1dcbda16594","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-06 15:56:34.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"61ad8b4797dc2b4103a96e514b6d15dc5d4d84e7","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-10-06 20:21:50.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"6d68e9f22703e0b2647008755d49ee707d8089c9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-18 15:08:07.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"7bd894b2d035f15197dbe9f04b6616b09cd6bcbc","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-25 10:20:04.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"514c94436d2b22fc762cd5912a993cc475232c21","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-25 10:25:55.000000000","message":"Patch Set 8: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c8dd2ca28fa996c9566281ea331bb4f1b4df2a2e","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-26 13:36:25.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d6a721626389b01d6fd8674c8a3c2258a62e33bc","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-26 13:44:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"430d4f544bf6671ea75bf45c0d321c46826987e9","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-27 15:02:26.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a4102d17441159f744918c700280ce0a434174d0","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-10-30 01:25:56.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"821d674c48e0fb52fd72e46a239f4f1585eff367","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-31 11:00:20.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"c651ac935ef91ece32033f64be93f3be248ad450","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 11:01:23.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"d37fc3a32d54cb674c7c8a05e0c4b40bd72bd439","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 13:08:31.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c18f989a9434c5f9523d5d60f538bb5\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a318dc063f6c48c5a154cbd172cc2ddb : SUCCESS in 2h 05m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62cd9047cc034f2d939aa464693b8ab5 : SUCCESS in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6e191f6f2874f3ba637aef9a3b2689a : SUCCESS in 23m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/62739edbef55424096a79c6fefca3e63 : SUCCESS in 21m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65cd95b2d054c8eb795dbfbd891acc1 : SUCCESS in 9m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/60ba6c18398b487ab5cf9ff23fe785dd : FAILURE in 42m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb41d3f25be948ac9742e32491005af5 : SUCCESS in 2h 01m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/44f708a3943947faa97f7552ebeae11c : SUCCESS in 53m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c285d224b2514de8a137bbb51794c829 : SUCCESS in 1h 06m 58s","accounts_in_message":[],"_revision_number":8},{"id":"a4d278cebb852bd3b87b4e3e77c2f09db35d699a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-31 13:20:00.000000000","message":"Patch Set 8: Code-Review-1 -Workflow\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"30dc2a88a318a52cfd06764c03a662784337b844","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-11-02 20:35:36.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1, Verified-2\n","accounts_in_message":[],"_revision_number":9},{"id":"b20a264d73c644a966bb08305a4c0a6b080f1b2f","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-11-02 20:36:11.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"093bcbb5fd8542cfb42b60c37b35e85dea16067b","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-11-02 20:38:40.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2668e8c18fd3a1dbf48c79ff3b4a296d9311d721","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-11-02 21:08:19.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/November-2023/Arista_Tempest_Testrefs/changes/02/895402/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"d920696f30e291c5a48d8ff8c049bf6b44e4cbbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 21:33:08.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e255ea62bb44f15ac8b741f04cf4b45\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef83f286f89d4c01865807b5cb11e4ad : SUCCESS in 50m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c234406be62c41f9a6ab8da1cddb0a34 : SUCCESS in 50m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fa24b608137f4e08ae7fb791b23d9085 : SUCCESS in 44m 09s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4f46dda8b2ee07e79ec511519460a7574acb1595","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 23:14:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c726e5942dad4b099ed14c40f581aab9\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/822d0ff1275d44ea89b7e5cd44261a25 : SUCCESS in 1h 45m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4424415d308b43c7be625eb1482b6c15 : SUCCESS in 36m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75134288e9174fa491fc13c529b2a018 : SUCCESS in 38m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08925ce6078048458230337ca3d7f56f : SUCCESS in 11m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e02852d0d6b4556a4ef4d7ffb3493cb : SUCCESS in 30m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de9623c9fb9945928566de77c53f5afb : SUCCESS in 25m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07f9a93e4a4424cbc2b614331211d00 : SUCCESS in 13m 42s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d5f91f3b628243409e02346b685e3b4f : SUCCESS in 29m 08s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/3345397eb641455fb88ad14a7e4ba8c0 : SUCCESS in 1h 08m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a8e1f1fa69a483da3aa360282a80051 : SUCCESS in 1h 00m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5bbf237034e04d33b1d183b89d731259 : SUCCESS in 2h 32m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/91b1937d82044cf5a31f8490b20946f9 : SUCCESS in 1h 05m 25s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6380d9bef5b74dd396a36e49a7add77a : SUCCESS in 2h 02m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/859c2d191d8c4a2db359c6bd35910716 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":9},{"id":"26c60695c70c8307b8fea9af002009ac774cefac","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-11-03 15:50:29.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"57198c667c7164fdb9de6f06b0df05ac83910d4a","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-11-03 16:05:55.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"19e61bb59c79b5918e097b1255dfc5b324938d20","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-11-03 16:33:05.000000000","message":"Patch Set 9: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b1c6d1029d3bd1e93ee16a130637898b601a5ed9","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-11-07 23:45:11.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"11666c01723b5c60cfa27c576aff2d88e952b704","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 23:45:47.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"e4c6158b990df2810c796a9251a2fb3bc132174c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 02:16:03.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3da5c9e1a43e49efb885070bff531c49\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c056ce5e378e46fa9736ce356b92c633 : SUCCESS in 1h 54m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/018c2715037b40d6a50a2f5d8671a8c3 : SUCCESS in 11m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f97c902d39c4e64996471c87548f36e : SUCCESS in 27m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9fee55e855274da1926b0d3852a3ace1 : SUCCESS in 27m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/951b590be03747ef96d0a9d53a503b7e : SUCCESS in 11m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/032dd056dfa0484a886f73644bb2ff5b : SUCCESS in 51m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a118e38fbb844685924cfbe70c477143 : SUCCESS in 2h 17m 38s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/10817a545a1447f5a3c3a7975426e171 : SUCCESS in 1h 01m 19s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f040a0a9250846b0ae8f0b9bad4b45bf : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":9},{"id":"73d72d4a07d00d86f2b4271f904d2311066418d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 02:16:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"eea72aabebcb95090ac02be6517951da68d8d67b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 02:19:11.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf5a4d27bead4c81a6031553129add17\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b45284e7d9042b9bc135a6d4f82b346 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3ec7f96cbb0571c401406781562160db75b3b9f5","revisions":{"d398a734f00cb4fe0c4f5a3356c819274bc3f612":{"kind":"REWORK","_number":1,"created":"2023-09-15 14:14:55.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/02/895402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/895402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/895402/1"}}},"commit":{"parents":[{"commit":"741f504c7badb3f6799977bd1e98ca5a9b7d1f53","subject":"Merge \"Prevent internal IP change for floating IP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/741f504c7badb3f6799977bd1e98ca5a9b7d1f53"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-08-26 06:07:10.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-15 14:14:39.000000000","tz":0},"subject":"ovn: Spawn metadata resources on additional chassis event","message":"ovn: Spawn metadata resources on additional chassis event\n\nSince the additional_chassis column [1] is used during the live\nmigration in order to optimize the VM network timeout, we can also use\nthis to minimize downtime for the metadata service. The patch adds\nadditional match to when new Port Binding is introduced to the chassis\nassociated with a node. This triggers the provisioning of metadata\nresources a lot earlier in the process, while the VM is still in paused\nstate.\n\n[1] https://opendev.org/openstack/neutron/commit/35fade3b5fbe898885338cbe74843e903d4e8a87\n\nCloses-bug: #2036118\nChange-Id: I66fa3c541db54c5f57404681f30b141fe8dad84a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d398a734f00cb4fe0c4f5a3356c819274bc3f612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d398a734f00cb4fe0c4f5a3356c819274bc3f612"}]},"branch":"refs/heads/master"},"208020746b6482940849a5edf638756c329ac2f5":{"kind":"REWORK","_number":2,"created":"2023-09-18 22:33:36.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/02/895402/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/895402/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/895402/2"}}},"commit":{"parents":[{"commit":"741f504c7badb3f6799977bd1e98ca5a9b7d1f53","subject":"Merge \"Prevent internal IP change for floating IP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/741f504c7badb3f6799977bd1e98ca5a9b7d1f53"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-08-26 06:07:10.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-16 07:30:18.000000000","tz":0},"subject":"ovn: Spawn metadata resources on additional chassis event","message":"ovn: Spawn metadata resources on additional chassis event\n\nSince the additional_chassis column [1] is used during the live\nmigration in order to optimize the VM network timeout, we can also use\nthis to minimize downtime for the metadata service. The patch adds\nadditional match to when new Port Binding is introduced to the chassis\nassociated with a node. This triggers the provisioning of metadata\nresources a lot earlier in the process, while the VM is still in paused\nstate.\n\n[1] https://opendev.org/openstack/neutron/commit/35fade3b5fbe898885338cbe74843e903d4e8a87\n\nCloses-bug: #2036118\nChange-Id: I66fa3c541db54c5f57404681f30b141fe8dad84a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/208020746b6482940849a5edf638756c329ac2f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/208020746b6482940849a5edf638756c329ac2f5"}]},"branch":"refs/heads/master"},"e6de3164112af91e6a010974bd8f688918eb7622":{"kind":"REWORK","_number":3,"created":"2023-09-22 18:19:08.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/02/895402/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/895402/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/895402/3"}}},"commit":{"parents":[{"commit":"0b805b89f8ad06469170408fd7416f5a0d8a4ae0","subject":"ovn-metadata: Refactor events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b805b89f8ad06469170408fd7416f5a0d8a4ae0"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-21 21:06:41.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-22 18:05:35.000000000","tz":0},"subject":"ovn: Spawn metadata resources on additional chassis event","message":"ovn: Spawn metadata resources on additional chassis event\n\nSince the additional_chassis column [1] is used during the live\nmigration in order to optimize the VM network timeout, we can also use\nthis to minimize downtime for the metadata service. The patch adds\nadditional match to when new Port Binding is introduced to the chassis\nassociated with a node. This triggers the provisioning of metadata\nresources a lot earlier in the process, while the VM is still in paused\nstate.\n\n[1]\nhttps://opendev.org/openstack/neutron/commit/35fade3b5fbe898885338cbe74843e903d4e8a87\n\nCloses-bug: #2036118\nChange-Id: I66fa3c541db54c5f57404681f30b141fe8dad84a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6de3164112af91e6a010974bd8f688918eb7622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e6de3164112af91e6a010974bd8f688918eb7622"}]},"branch":"refs/heads/master"},"b0be48a523fdd003d8c5528f269d7a67e345ffc4":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-09-29 17:00:11.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/02/895402/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/895402/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/895402/4"}}},"commit":{"parents":[{"commit":"5f4e3ff0667751072b2b618a28e6c2dedcc809be","subject":"ovn-metadata: Refactor events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f4e3ff0667751072b2b618a28e6c2dedcc809be"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-21 21:06:41.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-29 17:00:11.000000000","tz":0},"subject":"ovn: Spawn metadata resources on additional chassis event","message":"ovn: Spawn metadata resources on additional chassis event\n\nSince the additional_chassis column [1] is used during the live\nmigration in order to optimize the VM network timeout, we can also use\nthis to minimize downtime for the metadata service. The patch adds\nadditional match to when new Port Binding is introduced to the chassis\nassociated with a node. This triggers the provisioning of metadata\nresources a lot earlier in the process, while the VM is still in paused\nstate.\n\n[1]\nhttps://opendev.org/openstack/neutron/commit/35fade3b5fbe898885338cbe74843e903d4e8a87\n\nCloses-bug: #2036118\nChange-Id: I66fa3c541db54c5f57404681f30b141fe8dad84a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0be48a523fdd003d8c5528f269d7a67e345ffc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b0be48a523fdd003d8c5528f269d7a67e345ffc4"}]},"branch":"refs/heads/master","description":"Rebase"},"5e630ef5fffbd5270d3969da460451579a3d29f5":{"kind":"REWORK","_number":5,"created":"2023-09-29 17:31:58.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/02/895402/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/895402/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/895402/5"}}},"commit":{"parents":[{"commit":"5f4e3ff0667751072b2b618a28e6c2dedcc809be","subject":"ovn-metadata: Refactor events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f4e3ff0667751072b2b618a28e6c2dedcc809be"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-21 21:06:41.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-29 17:29:35.000000000","tz":0},"subject":"ovn: Spawn metadata resources on additional chassis event","message":"ovn: Spawn metadata resources on additional chassis event\n\nSince the additional_chassis column [1] is used during the live\nmigration in order to optimize the VM network timeout, we can also use\nthis to minimize downtime for the metadata service. The patch adds\nadditional match to when new Port Binding is introduced to the chassis\nassociated with a node. This triggers the provisioning of metadata\nresources a lot earlier in the process, while the VM is still in paused\nstate.\n\n[1]\nhttps://opendev.org/openstack/neutron/commit/35fade3b5fbe898885338cbe74843e903d4e8a87\n\nCloses-bug: #2036118\nChange-Id: I66fa3c541db54c5f57404681f30b141fe8dad84a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e630ef5fffbd5270d3969da460451579a3d29f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5e630ef5fffbd5270d3969da460451579a3d29f5"}]},"branch":"refs/heads/master"},"14f367db3e62c4c75321a2a1792bc02b732c9402":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-10-03 19:25:22.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/02/895402/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/895402/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/895402/6"}}},"commit":{"parents":[{"commit":"881dad49aa245573b9fe37cc1d712d95fd412fce","subject":"ovn-metadata: Refactor events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/881dad49aa245573b9fe37cc1d712d95fd412fce"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-21 21:06:41.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-10-03 19:25:22.000000000","tz":0},"subject":"ovn: Spawn metadata resources on additional chassis event","message":"ovn: Spawn metadata resources on additional chassis event\n\nSince the additional_chassis column [1] is used during the live\nmigration in order to optimize the VM network timeout, we can also use\nthis to minimize downtime for the metadata service. The patch adds\nadditional match to when new Port Binding is introduced to the chassis\nassociated with a node. This triggers the provisioning of metadata\nresources a lot earlier in the process, while the VM is still in paused\nstate.\n\n[1]\nhttps://opendev.org/openstack/neutron/commit/35fade3b5fbe898885338cbe74843e903d4e8a87\n\nCloses-bug: #2036118\nChange-Id: I66fa3c541db54c5f57404681f30b141fe8dad84a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14f367db3e62c4c75321a2a1792bc02b732c9402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/14f367db3e62c4c75321a2a1792bc02b732c9402"}]},"branch":"refs/heads/master","description":"Rebase"},"960d4dd6edfa885a93e2b63873cabdf51067c5d2":{"kind":"REWORK","_number":7,"created":"2023-10-03 21:39:29.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/02/895402/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/895402/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/895402/7"}}},"commit":{"parents":[{"commit":"881dad49aa245573b9fe37cc1d712d95fd412fce","subject":"ovn-metadata: Refactor events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/881dad49aa245573b9fe37cc1d712d95fd412fce"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-21 21:06:41.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-30 12:18:35.000000000","tz":0},"subject":"ovn: Spawn metadata resources on additional chassis event","message":"ovn: Spawn metadata resources on additional chassis event\n\nSince the additional_chassis column [1] is used during the live\nmigration in order to optimize the VM network timeout, we can also use\nthis to minimize downtime for the metadata service. The patch adds\nadditional match to when new Port Binding is introduced to the chassis\nassociated with a node. This triggers the provisioning of metadata\nresources a lot earlier in the process, while the VM is still in paused\nstate.\n\n[1]\nhttps://opendev.org/openstack/neutron/commit/35fade3b5fbe898885338cbe74843e903d4e8a87\n\nCloses-bug: #2036118\nChange-Id: I66fa3c541db54c5f57404681f30b141fe8dad84a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/960d4dd6edfa885a93e2b63873cabdf51067c5d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/960d4dd6edfa885a93e2b63873cabdf51067c5d2"}]},"branch":"refs/heads/master"},"1ed129aa78f03f1596b86929ca5baa656466d692":{"kind":"REWORK","_number":8,"created":"2023-10-03 21:41:36.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/02/895402/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/895402/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/895402/8"}}},"commit":{"parents":[{"commit":"6801589510242affc78497660d34377603774074","subject":"ovn-metadata: Refactor events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6801589510242affc78497660d34377603774074"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-21 21:06:41.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-30 12:23:41.000000000","tz":0},"subject":"ovn: Spawn metadata resources on additional chassis event","message":"ovn: Spawn metadata resources on additional chassis event\n\nSince the additional_chassis column [1] is used during the live\nmigration in order to optimize the VM network timeout, we can also use\nthis to minimize downtime for the metadata service. The patch adds\nadditional match to when new Port Binding is introduced to the chassis\nassociated with a node. This triggers the provisioning of metadata\nresources a lot earlier in the process, while the VM is still in paused\nstate.\n\n[1]\nhttps://opendev.org/openstack/neutron/commit/35fade3b5fbe898885338cbe74843e903d4e8a87\n\nCloses-bug: #2036118\nChange-Id: I66fa3c541db54c5f57404681f30b141fe8dad84a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ed129aa78f03f1596b86929ca5baa656466d692"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ed129aa78f03f1596b86929ca5baa656466d692"}]},"branch":"refs/heads/master"},"3ec7f96cbb0571c401406781562160db75b3b9f5":{"kind":"REWORK","_number":9,"created":"2023-11-02 20:35:36.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/02/895402/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/02/895402/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/02/895402/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/02/895402/9"}}},"commit":{"parents":[{"commit":"1daa0dd5bfa06f44cc49662aa97d3c7a7f4d36e4","subject":"Merge \"ovn-metadata: Refactor events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1daa0dd5bfa06f44cc49662aa97d3c7a7f4d36e4"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-21 21:06:41.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-10-14 14:08:29.000000000","tz":0},"subject":"ovn: Spawn metadata resources on additional chassis event","message":"ovn: Spawn metadata resources on additional chassis event\n\nSince the additional_chassis column [1] is used during the live\nmigration in order to optimize the VM network timeout, we can also use\nthis to minimize downtime for the metadata service. The patch adds\nadditional match to when new Port Binding is introduced to the chassis\nassociated with a node. This triggers the provisioning of metadata\nresources a lot earlier in the process, while the VM is still in paused\nstate.\n\n[1]\nhttps://opendev.org/openstack/neutron/commit/35fade3b5fbe898885338cbe74843e903d4e8a87\n\nCloses-bug: #2036118\nChange-Id: I66fa3c541db54c5f57404681f30b141fe8dad84a\nSigned-off-by: Jakub Libosvar \u003clibosvar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ec7f96cbb0571c401406781562160db75b3b9f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ec7f96cbb0571c401406781562160db75b3b9f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
