)]}'
{"id":"openstack%2Fneutron~895583","triplet_id":"openstack%2Fneutron~stable%2F2023.1~Ib465649acab679602085239947c3b70b746eb561","project":"openstack/neutron","branch":"stable/2023.1","topic":"bug/2035578","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-09-18 11:34:01.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}},"hashtags":[],"change_id":"Ib465649acab679602085239947c3b70b746eb561","subject":"[DNM] Check devstack-tobiko-neutron","status":"ABANDONED","created":"2023-09-18 05:25:45.000000000","updated":"2023-12-04 05:01:38.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"983990645efc09f5a52c6d660d228cce1d4a34d1","_number":895583,"virtual_id_number":895583,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-18 12:13:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-18 07:26:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09e157a9d60c77a0436ac471d5e40f57e1733ae0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-18 05:25:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"880eff318510aa12bd22396b73ccfc9284b5eb80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 07:26:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ccc43b1301f46b7a1260cd7c7772b71\n\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/e7643d4f85144860b253ecac990ac8cc : RETRY_LIMIT in 28m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0876b3a0935a5f71a4900a8e5338acad79960303","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-18 11:29:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b544f38d624d2b66cbde406d3d58fd284ba4a97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 11:29:44.000000000","message":"Patch Set 1:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"b7d1236539e4640bfd3114de3c96686812f92c18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 11:29:57.000000000","message":"Patch Set 1:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"1a3a8c2a5fd774e46dd9eac1182f910e12bfd7bd","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-18 11:34:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7d5d1f694e3b99dc1946299a9b43df8a260a73a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 12:13:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e19756307e7490cba26f8da127313d8\n\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/91413baadba444d38f04d0843a768a32 : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":1},{"id":"75e8701a222b4d991a0aac6384a2edb702dd7374","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-18 16:04:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe7c0a4de1e25436d4e605b7701c15398d9e9ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 16:44:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d01ca54d288147d4a7002e960904575f\n\n- devstack-tobiko-neutron https://zuul.opendev.org/t/openstack/build/0d10bed9887446bd9bcc98637aebf072 : SUCCESS in 37m 55s","accounts_in_message":[],"_revision_number":1},{"id":"983990645efc09f5a52c6d660d228cce1d4a34d1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-12-04 05:01:38.000000000","message":"Abandoned\n\njust a test, fix merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"792db052e4725a3e36e22770fe1b92e7b663477e","revisions":{"792db052e4725a3e36e22770fe1b92e7b663477e":{"kind":"REWORK","_number":1,"created":"2023-09-18 05:25:45.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/83/895583/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/83/895583/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/895583/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/895583/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/83/895583/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/83/895583/1"}}},"commit":{"parents":[{"commit":"b92fc1ad6c87c02537d790e63dd9c8a391682ae0","subject":"Use HasStandardAttributes as parent class for Tags DB model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b92fc1ad6c87c02537d790e63dd9c8a391682ae0"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-18 05:20:31.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-18 05:20:35.000000000","tz":330},"subject":"[DNM] Check devstack-tobiko-neutron","message":"[DNM] Check devstack-tobiko-neutron\n\nDepends-On: https://review.opendev.org/c/x/devstack-plugin-tobiko/+/895582\nChange-Id: Ib465649acab679602085239947c3b70b746eb561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/792db052e4725a3e36e22770fe1b92e7b663477e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/792db052e4725a3e36e22770fe1b92e7b663477e"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
