)]}'
{"id":"openstack%2Fneutron~895668","triplet_id":"openstack%2Fneutron~stable%2Fyoga~Ic54bd6f699b91b53f22c93c972d8def6c99d0eb7","project":"openstack/neutron","branch":"stable/yoga","topic":"bug-2003022-stable/yoga","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-19 09:48:05.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-09-19 10:44:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic54bd6f699b91b53f22c93c972d8def6c99d0eb7","subject":"Add 3 secs to wait for keepalived state change","status":"MERGED","created":"2023-09-18 19:13:05.000000000","updated":"2023-09-19 10:45:42.000000000","submitted":"2023-09-19 10:44:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895668-bug-2003022-stable/yoga","cherry_pick_of_change":870679,"cherry_pick_of_patch_set":1,"meta_rev_id":"1df8372a325e9a3a8656d864476ec869597b3eeb","_number":895668,"virtual_id_number":895668,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-19 10:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-09-19 09:48:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-09-18 20:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-09-19 09:48:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-18 19:13:05.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"CC"},{"updated":"2023-09-18 19:16:11.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-09-18 20:02:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-18 20:33:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"3bec417fe8b7a51566398963b855ec6e83c4cf4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-18 19:13:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6463922349261ae15e0977886c47af58d572f6e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-18 19:16:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"759acafc9d0bb50ce397706cac5fffbb0b77f044","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 20:02:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09903f5e233f4d659800110884a10c42\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95c0ce6ad1ca4f6f834bb9a80b7e945d : SUCCESS in 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80cdadff481e447ca296d478c069334d : SUCCESS in 9m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9864df217e7047c0b7b57a92923a6da1 : SUCCESS in 46m 02s","accounts_in_message":[],"_revision_number":2},{"id":"556786a552743beab0f176f2ed705079c931e3c3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-18 20:33:32.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"49200930d85e3e2d051673d4bce8415805e88cb1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-19 09:48:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5bae630fb4b5ad924b06ce2731ceeba246626a34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 09:48:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"161dab76d78d875bace7b6bda8faf25f36e6e885","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 10:44:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/645b283920644cf9904ad6b0cdecbbd0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e30c36870e514a879073be9d8814481a : SUCCESS in 16m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db95fbc1f8714c18a3a2b476db8c8fda : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cacaaaec7ca24973bac07b039d609109 : SUCCESS in 43m 10s","accounts_in_message":[],"_revision_number":2},{"id":"94c411c8e8b8da372279f304e678fadbcc52c930","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 10:44:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1df8372a325e9a3a8656d864476ec869597b3eeb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 10:45:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a82cb6a386db4987ab17fb04e5a1f85f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0552021abd0740f8a712e8f2f9527c2d : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"edf33d2217f73b8775c76f3cac2fc75ce00632a1","revisions":{"1433cd784f79f86ab076f93834b78c0bdce47cf8":{"kind":"REWORK","_number":1,"created":"2023-09-18 19:13:05.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/68/895668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/895668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/895668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/895668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/895668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/895668/1"}}},"commit":{"parents":[{"commit":"ef31c17cb865eb9f591714d692c5ff5851076b2f","subject":"Use HasStandardAttributes as parent class for Tags DB model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef31c17cb865eb9f591714d692c5ff5851076b2f"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2023-01-16 22:41:20.000000000","tz":-360},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-09-18 19:13:05.000000000","tz":0},"subject":"Add 3 secs to wait for keepalived state change","message":"Add 3 secs to wait for keepalived state change\n\nWe add 3 seconds to the keepalived state change to avoid timeout\nfailures in test_read_queue_change_state.\n\nChange-Id: Ic54bd6f699b91b53f22c93c972d8def6c99d0eb7\nCloses-Bug: #2003022\nRelated-Bug: #2035382\n(cherry picked from commit f28e4165993f0c226bd5af54ad6d8c6bb48bff69)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1433cd784f79f86ab076f93834b78c0bdce47cf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1433cd784f79f86ab076f93834b78c0bdce47cf8"}]},"branch":"refs/heads/stable/yoga"},"edf33d2217f73b8775c76f3cac2fc75ce00632a1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-18 19:16:01.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/68/895668/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/68/895668/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/895668/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/895668/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/68/895668/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/68/895668/2"}}},"commit":{"parents":[{"commit":"ef31c17cb865eb9f591714d692c5ff5851076b2f","subject":"Use HasStandardAttributes as parent class for Tags DB model","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef31c17cb865eb9f591714d692c5ff5851076b2f"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2023-01-16 22:41:20.000000000","tz":-360},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-09-18 19:16:01.000000000","tz":0},"subject":"Add 3 secs to wait for keepalived state change","message":"Add 3 secs to wait for keepalived state change\n\nWe add 3 seconds to the keepalived state change to avoid timeout\nfailures in test_read_queue_change_state.\n\nChange-Id: Ic54bd6f699b91b53f22c93c972d8def6c99d0eb7\nCloses-Bug: #2035382\nRelated-Bug: #2003022\n(cherry picked from commit f28e4165993f0c226bd5af54ad6d8c6bb48bff69)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/edf33d2217f73b8775c76f3cac2fc75ce00632a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/edf33d2217f73b8775c76f3cac2fc75ce00632a1"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
