)]}'
{"id":"openstack%2Fneutron~895899","triplet_id":"openstack%2Fneutron~stable%2F2023.2~Ifc2d37e2042fad43dd838821953defd99a5f8665","project":"openstack/neutron","branch":"stable/2023.2","topic":"bug/2033887","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-28 10:00:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifc2d37e2042fad43dd838821953defd99a5f8665","subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","status":"MERGED","created":"2023-09-20 09:05:42.000000000","updated":"2023-09-28 10:02:25.000000000","submitted":"2023-09-28 10:00:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895899-bug/2033887","meta_rev_id":"2b8860b3c6ac7a6d8c45875ea31a4e04414f5ff7","_number":895899,"virtual_id_number":895899,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-28 10:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-09-28 09:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-09-27 14:38:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-09-28 00:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-09-28 09:32:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-09-28 00:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-20 09:46:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 08:33:39.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-09-28 00:11:46.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-09-28 09:32:15.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"e3e6fe879e945bcc828847cc0e02b249411f6696","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-20 09:05:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"253ae60bfbe870e8ea5eda633f79cce34567bbcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 09:46:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e35a5accecb44d78a158f45e0f4e3f7d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06ac5867c14c49318198ac61dee9b1dd : SUCCESS in 38m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df4e397257dd4126b3f971a0f1d1c97b : SUCCESS in 9m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cd0c3ef393d448db0be3c78ff5de4fb : SUCCESS in 27m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/833ba635c84e4e87856b181b1c5011f9 : SUCCESS in 22m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/013907d98b67408c8522da43235f1ece : SUCCESS in 27m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0adc5be6eee4f27a365ff5a2661c245 : SUCCESS in 22m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6587509e9d8b464486cf0734eb1dfd93 : SUCCESS in 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e565d59a3d2345d1c20cc39e544be777c94c4fce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 10:56:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d814c98e9b74fcc856def3eb4d6c29a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81260d9256eb4b90bfec31d0591b52c5 : SUCCESS in 1h 10m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/912ac25525454f13a2babf636c2fa441 : SUCCESS in 44m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b7a876c3cb434b3f8e3f2282099785ee : SUCCESS in 1h 19m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1712c741ad3c4465bb1bfba5625c042a : SUCCESS in 37m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb000108aa38bb7f4110520babf525c2164a6f70","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-21 08:33:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c6fc95487291ced175fb3ad149bf5235fb297ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-27 14:38:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3ff9a89a6a48028da9fa9c26b85b8d3b65ec9fc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 15:24:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76c9c6ef5c045f99c7d5d3f165f6fef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc591e4edc2a406f8bff2949fdd83164 : SUCCESS in 38m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eded76ef62e94a17a3b0217dcb9d0714 : SUCCESS in 8m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/733010b62d8c4169a24d20f35b96e644 : SUCCESS in 28m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fba36eeeac5472ea7253db6571dc3b2 : SUCCESS in 24m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd9a93a3d1ca435ab5060fec384969b3 : SUCCESS in 30m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be2a5566259d44feafad96af47df9cb5 : SUCCESS in 18m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05536ba88fb74af4b6782bccad0aa614 : SUCCESS in 13m 39s","accounts_in_message":[],"_revision_number":2},{"id":"75a6ed25c9641217c0890bd39e93af8fdba73a9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 17:02:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fdf36f7131c4ddba5bfe231cb78361f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f06fe3aa62d46abaa0308358ec16dc3 : SUCCESS in 48m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52b98db5868348068bd58338b1de13f4 : SUCCESS in 51m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3042e7654de04b3d9a5167f4002abf59 : SUCCESS in 41m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b602e81f5d8842c5a21071fe1e6d2e69 : SUCCESS in 40m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"23096a3f127b86e694466f89e98b68553d3d3c19","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-28 00:11:46.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a89f886811b6f2fea6df586210538f95ea8f8ae0","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-09-28 09:32:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9eed5ae8df7a69315f17b49ab14902c58281b8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 09:32:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"442d5a42f0e0c2c043e809d54f107ce2f2beceab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 10:00:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a461ca57851412a81221f53ca288505\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdcf4a14ceca495496b60a611b19992c : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4251e5632a4c4def969e65a676015f2e : SUCCESS in 20m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67f1e2da07a14942894a7c9d5b60c2b9 : SUCCESS in 26m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd80e8fc348a4bcbac392d3403373cf1 : SUCCESS in 26m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a13c8cfefd4b599deb90e9f43a9c7a : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":2},{"id":"e44cadec871c63aa2b80884da06d30fc70b7478b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 10:00:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2b8860b3c6ac7a6d8c45875ea31a4e04414f5ff7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 10:02:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63cd9c073ed94ea7a02e5dee7b65518c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c728ca4ae12e4f8ea6b378f0e9b4c2a2 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d28bcf0fe1648cfdded5e3bc38200b6d1315f42f","revisions":{"c8e2015014b33710e79755275679b1504076867c":{"kind":"REWORK","_number":1,"created":"2023-09-20 09:05:42.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/895899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/895899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/895899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/895899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/895899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/895899/1"}}},"commit":{"parents":[{"commit":"05aed84c2a4c6cbc4a088a9aa901eee38712ac65","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/05aed84c2a4c6cbc4a088a9aa901eee38712ac65"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-01 11:43:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 10:41:02.000000000","tz":0},"subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","message":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\n\nThis reverts commit 955e621167987626806128d259a29d4e62acf102.\n\nReason for revert: the port binding handling done in this patch is\nincorrect and leads to issues during the cold migration process with\ntrunk ports in ML2/OVN.\n\nChange-Id: Ifc2d37e2042fad43dd838821953defd99a5f8665\nCloses-Bug: #2033887\n(cherry picked from commit 1b034f8d62091724a6fb639f34c2f8936329b501)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8e2015014b33710e79755275679b1504076867c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c8e2015014b33710e79755275679b1504076867c"}]},"branch":"refs/heads/stable/2023.2"},"d28bcf0fe1648cfdded5e3bc38200b6d1315f42f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-27 14:38:20.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/99/895899/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/895899/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/895899/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/895899/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/895899/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/895899/2"}}},"commit":{"parents":[{"commit":"f1638bb6d130b5150bcdae669d2ce43bd6b9d688","subject":"Use safer methods to get security groups on security group logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1638bb6d130b5150bcdae669d2ce43bd6b9d688"}]}],"author":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-01 11:43:24.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-27 14:38:05.000000000","tz":0},"subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","message":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\n\nThis reverts commit 955e621167987626806128d259a29d4e62acf102.\n\nReason for revert: the port binding handling done in this patch is\nincorrect and leads to issues during the cold migration process with\ntrunk ports in ML2/OVN.\n\nChange-Id: Ifc2d37e2042fad43dd838821953defd99a5f8665\nCloses-Bug: #2033887\n(cherry picked from commit 1b034f8d62091724a6fb639f34c2f8936329b501)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d28bcf0fe1648cfdded5e3bc38200b6d1315f42f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d28bcf0fe1648cfdded5e3bc38200b6d1315f42f"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
