)]}'
{"id":"openstack%2Fneutron~895946","triplet_id":"openstack%2Fneutron~master~I2aca9a522bda2d69962369748b70fa9270fbe245","project":"openstack/neutron","branch":"master","topic":"bug/2036607","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-12-05 11:37:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2aca9a522bda2d69962369748b70fa9270fbe245","subject":"Make ``OVNMechanismDriver.post_fork_initialize`` callback cancellable","status":"MERGED","created":"2023-09-20 16:10:14.000000000","updated":"2023-12-05 11:39:22.000000000","submitted":"2023-12-05 11:37:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895946-bug/2036607","meta_rev_id":"58372363a2b04880859246c737e2d2266c667014","_number":895946,"virtual_id_number":895946,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-12-04 07:34:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-05 11:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-05 08:29:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2023-12-04 14:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-05 08:29:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-20 16:43:50.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-20 17:31:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-20 17:35:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-22 01:57:25.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-12-04 14:56:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-12-05 08:29:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"8489c7b020b66fa0d27d701cefbc9018cc15b721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-20 16:10:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c9d921ff126f1af04dae065479bc9e004abd0d0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-20 16:43:50.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/46/895946/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"7f55847ba05436352c5a77454560ce556592678e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 17:31:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6c3df90b13444bda9b76027f78e93fd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/838a81d25aee41ad9ae99d03d727becd : FAILURE in 52m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/522e01500fbe4a42a72134bf8abebbce : FAILURE in 50m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8b9ec298003d42eabc72b5badeb1139b : FAILURE in 52m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4311419bc5e140979dcbc73f67b632e0 : FAILURE in 46m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d33f625697db46d570065b3ea091d31d41e92369","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 17:35:09.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b887971d2f3b4d96a12dcecf34568c90\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/808c5498eb304babba528fd809a7024a : FAILURE in 17m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0424965cfd204fa9b835a0099657aaee : FAILURE in 33m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf1ee5d509a74b858064b7918804920d : FAILURE in 9m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de6bac0c14ec40fb997814dc927fc105 : FAILURE in 20m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b0ef55dc3f24708b9ae8d8c5f131742 : FAILURE in 23m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54c711df555245769bfa2a5dd8037771 : FAILURE in 22m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5663b85da52640ac9a627777716a89a5 : FAILURE in 20m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca7eece24ce44b18a081d770f17c3d83 : SUCCESS in 11m 01s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/c5e6308375124981abe9b8e416d18287 : FAILURE in 26m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/11e321eea7214c548b4b7f24cd161338 : FAILURE in 52m 26s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/a67c099fd2314e108f96757f74af59bf : FAILURE in 14m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/16c45b4704114e2aa2fc70249d6216ac : FAILURE in 21m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/952f57a058dd4cecbcacdf5910584e0d : FAILURE in 14m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1aa004ad652f0263146bc0bc0342115b9c2a2582","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-21 08:32:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"37681cea214fba469eef2ecacd97462b68dc0a89","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-21 09:04:41.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/46/895946/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"016917e47ef725b3daa57a4f71ee9dbf386e1118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 09:15:24.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d81fab054389488c9da1cf345940f730\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6b08eba8133d47e3878c1fb2b36b3f5e : FAILURE in 17m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e071af72ef74fa191e7ae413e14a486 : FAILURE in 29m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b8ce22cca20469992880b3a8b0fb638 : FAILURE in 7m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d40244457f15458189792304e71eb5e2 : FAILURE in 18m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a0e488ee80a40e5ae4d387e1123945c : FAILURE in 20m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1022a05c8844fa9aea7898a5db5d617 : FAILURE in 18m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e7f454455724bc3b5954ce58048f1b8 : FAILURE in 19m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f12ece730e7425fab3ff7d1e5192ee5 : SUCCESS in 8m 48s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/481937f4ea214c6ea8e168ecfc2f17de : FAILURE in 24m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a012d92dbc4149a1a4595de7e10b25ee : FAILURE in 41m 03s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f14e6c5264f340fe91d3bba17d482eb0 : FAILURE in 17m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e5fb228091bc485badfac70d565cbb81 : FAILURE in 22m 39s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4e22a9347d114098b63902787f1abe22 : FAILURE in 28m 44s","accounts_in_message":[],"_revision_number":2},{"id":"b1f238a48cd8cd582ff7a4b553adb9bd2ba260cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 09:34:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50de1e9b4f60426da31fb87120f957ef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2473a0da53f414fb601ab0d5d073952 : FAILURE in 59m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da5ae2e046394d75a611ec4eb4321433 : FAILURE in 59m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b919be82a4704ac699a641234b19945b : FAILURE in 59m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8e6fba8c1e5a4680aa4a0cbccd64dc56 : FAILURE in 53m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1f03d766a761967dd9f1574d152a3e882dc2b893","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-12-04 07:01:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c8f41b3f272f0f7381bd81529d9b7cfea14231b9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-12-04 07:34:41.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/December-2023/Arista_Tempest_Testrefs/changes/46/895946/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"024e056a95f087a3bf3d2e6ea081185726e6c195","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 07:49:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32a2637ebf6a4f88b79cebaab85b2931\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11327645f25d4259bfa4d211e4658ebf : SUCCESS in 44m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/00ef2396ea7541c7a6fa27cbdc06ddbd : SUCCESS in 43m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3e1188355897439ab3fb896d0ea1d0a6 : SUCCESS in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"263630f087b627ef0471dd5304a52bbb80efd198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 09:05:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bccd88a487f4a1386de8a4d9e7228f6\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4aca77c4ae4240cb99c6d74b4fa546b6 : SUCCESS in 1h 31m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/605fc396ea534310ba67b7fc01005c5f : SUCCESS in 41m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/910bc8b560754f8ea0597f270f65c935 : SUCCESS in 11m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77feee030c424822b1dcd7785b23bbc6 : SUCCESS in 24m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86003e1ec4f2451eaef1c045b85ec26e : SUCCESS in 27m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790793435128454694aa0e5683c1ca70 : SUCCESS in 10m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e7aa939e17d04bce846a060964c1d5c2 : SUCCESS in 2m 42s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d0422fd3d7f24ef6aa9e6a3d9fdeed62 : SUCCESS in 28m 51s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a2c1ccab477a42e2aff0476b38a6728e : SUCCESS in 1h 04m 45s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/a27b4b87c4404e3e8c63e0307e9c157c : SUCCESS in 1h 14m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/88820074a16b4c0aa09fb685abdf64ec : SUCCESS in 1h 00m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/16e064e9d2e34532b3ce31b2e1ccadc9 : SUCCESS in 1h 57m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee7a84bd69b644d5af4591fac5919ad2 : SUCCESS in 1h 00m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/de9f28881bfe41e99ebd50e9ae4268ed : SUCCESS in 46m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f1afd853fdd54a2c9120275d7558a065 : SUCCESS in 1h 40m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fb5d266facfb4e26a0ad84cdd50eafaa : SUCCESS in 1h 05m 43s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/ef4ae4bce0394337b2eef0a865f3ce1f : SUCCESS in 49m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cec40605a7c945f599defd174a4cbeff : SUCCESS in 1h 15m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ecd46e8c23f24a48a4db042a1e98624e : SUCCESS in 1h 21m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ee3f9954e437423b8970eeb067c7cc45 : SUCCESS in 1h 00m 21s","accounts_in_message":[],"_revision_number":3},{"id":"dbdcc11ea575a9e717d99595e52d564e9965889c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-04 14:56:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0f974c28c5a6d74f29a1b10f67b8728567b60e59","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-12-05 08:29:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a2db0cb38a0d28375ef6e3875f69534ebbacc512","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 08:29:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6b12f627824e8e9e2f2e563284b31bfde9f4c695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 11:37:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e82c473a53744ad9acec80244b5795cf\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/812f0133607442469715c621b2f202eb : SUCCESS in 2h 16m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ce2605e1a814922958e92ce8e05c92a : SUCCESS in 12m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c599f79c759343fbbed0706be2004bd2 : SUCCESS in 26m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86e7da52d77c4ae1bdd07d7e50049547 : SUCCESS in 26m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d36d4cf922d427eb34c458430da6ea5 : SUCCESS in 11m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d68fed5c2c6f45a88542de5fadc5b9a1 : SUCCESS in 3m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e6f12e46c64f46f9864298d632169b99 : SUCCESS in 58m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b466516b31045d3b3c91e76dba0553b : SUCCESS in 2h 37m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/196ad08dc21747ae8e8275d7412a11c9 : SUCCESS in 1h 14m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a278241419774176b1aa6813ab4825bb : SUCCESS in 59m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6cc747b57c454aaa92a1a77786b43a83 : SUCCESS in 1h 02m 58s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/1ab73eb89fa74d14aad50152f4d9dd35 : SUCCESS in 1h 03m 47s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/cb7e406c48594cf19a10dc468bfe18d8 : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":3},{"id":"ad450ed099aa9d97ab66bba26c40510ee80f40de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 11:37:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"58372363a2b04880859246c737e2d2266c667014","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 11:39:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80cb13158b564cb4b774e0e01e8ed30f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2670228a86664396a912e9f8e23abd15 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bb1114c8b1b99834f2cb782dae796fa778ec2319","revisions":{"80dd55f43f4b52e816796f797c8a8f8ac476ab13":{"kind":"REWORK","_number":1,"created":"2023-09-20 16:10:14.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/46/895946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/895946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/895946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/895946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/895946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/895946/1"}}},"commit":{"parents":[{"commit":"dbe4ba910b3236ff3ac42e33dcb4cc067b1f9177","subject":"Merge \"Open the 2024.1 (Caracal) DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbe4ba910b3236ff3ac42e33dcb4cc067b1f9177"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 17:26:56.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 17:44:51.000000000","tz":0},"subject":"Make ``OVNMechanismDriver.post_fork_initialize`` callback abortable","message":"Make ``OVNMechanismDriver.post_fork_initialize`` callback abortable\n\nIf the callback method ``OVNMechanismDriver.post_fork_initialize``\nfails, the callback manager must raise an exception and finish the\nprocess. If that happens in a Neutron worker, the\n``oslo_service.service.Services`` class will restart the process.\n\nNOTE: it is required to bump the n-lib version once the dependant\n      patch is merged and released.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/895940\n\nPartial-Bug: #2036607\nChange-Id: I2aca9a522bda2d69962369748b70fa9270fbe245\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80dd55f43f4b52e816796f797c8a8f8ac476ab13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80dd55f43f4b52e816796f797c8a8f8ac476ab13"}]},"branch":"refs/heads/master"},"64959e8cdc451f84f8906cafb35190b24bf5ad05":{"kind":"REWORK","_number":2,"created":"2023-09-21 08:32:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/46/895946/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/895946/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/895946/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/895946/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/895946/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/895946/2"}}},"commit":{"parents":[{"commit":"dbe4ba910b3236ff3ac42e33dcb4cc067b1f9177","subject":"Merge \"Open the 2024.1 (Caracal) DB branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dbe4ba910b3236ff3ac42e33dcb4cc067b1f9177"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 17:26:56.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 19:11:47.000000000","tz":0},"subject":"Make ``OVNMechanismDriver.post_fork_initialize`` callback cancellable","message":"Make ``OVNMechanismDriver.post_fork_initialize`` callback cancellable\n\nIf the callback method ``OVNMechanismDriver.post_fork_initialize``\nfails, the callback manager must raise an exception and finish the\nprocess. If that happens in a Neutron worker, the\n``oslo_service.service.Services`` class will restart the process.\n\nNOTE: it is required to bump the n-lib version once the dependant\n      patch is merged and released.\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-lib/+/895940\n\nPartial-Bug: #2036607\nChange-Id: I2aca9a522bda2d69962369748b70fa9270fbe245\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64959e8cdc451f84f8906cafb35190b24bf5ad05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/64959e8cdc451f84f8906cafb35190b24bf5ad05"}]},"branch":"refs/heads/master"},"bb1114c8b1b99834f2cb782dae796fa778ec2319":{"kind":"REWORK","_number":3,"created":"2023-12-04 07:01:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/46/895946/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/895946/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/895946/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/895946/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/895946/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/895946/3"}}},"commit":{"parents":[{"commit":"119c3ba85c99fbcd864f2cb5cae7e06a2fa48d73","subject":"Merge \"Allow neutron managed ports to bypass PREROUTING chain\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/119c3ba85c99fbcd864f2cb5cae7e06a2fa48d73"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-11 17:26:56.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-12-04 07:00:08.000000000","tz":0},"subject":"Make ``OVNMechanismDriver.post_fork_initialize`` callback cancellable","message":"Make ``OVNMechanismDriver.post_fork_initialize`` callback cancellable\n\nIf the callback method ``OVNMechanismDriver.post_fork_initialize``\nfails, the callback manager must raise an exception and finish the\nprocess. If that happens in a Neutron worker, the\n``oslo_service.service.Services`` class will restart the process.\n\nThe neutron-lib version is bumped to 3.9.0. It contains [1], needed\nfor this patch.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/895940\n\nPartial-Bug: #2036607\nChange-Id: I2aca9a522bda2d69962369748b70fa9270fbe245\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb1114c8b1b99834f2cb782dae796fa778ec2319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bb1114c8b1b99834f2cb782dae796fa778ec2319"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
