)]}'
{"id":"openstack%2Fneutron~896273","triplet_id":"openstack%2Fneutron~master~I8d5f5392fb7a6ab9b20e9222c143f4e67c925cae","project":"openstack/neutron","branch":"master","topic":"bug/2019960","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-10-09 21:27:54.000000000","reason":"Change was submitted"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2023-10-02 10:10:41.000000000","reason":"\u003cGERRIT_ACCOUNT_6773\u003e replied on the change","reason_account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2023-10-09 18:55:21.000000000","reason":"\u003cGERRIT_ACCOUNT_4694\u003e replied on the change","reason_account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2023-10-09 21:27:54.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-10-09 21:27:54.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-10-09 21:27:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8d5f5392fb7a6ab9b20e9222c143f4e67c925cae","subject":"Remove the publish patch in SG rule BEFORE_DELETE and BEFORE_CREATE","status":"MERGED","created":"2023-09-22 16:10:21.000000000","updated":"2023-10-09 21:29:09.000000000","submitted":"2023-10-09 21:27:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896273-bug/2019960","meta_rev_id":"5c6c9dff83f88f6d9612de9962cc943b00e0957c","_number":896273,"virtual_id_number":896273,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-09 21:27:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-09 14:11:43.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":2,"date":"2023-10-09 18:55:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2023-10-09 14:11:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2023-10-09 18:55:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-22 17:56:08.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-29 15:35:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2023-09-29 15:35:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-09-29 15:35:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-09-29 15:35:49.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2023-09-29 15:35:49.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-09-29 16:31:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5291d92690528a3acb4393e5342d8a1e23980d6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-22 16:10:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9024fcab27d6c052b783aad1ca7a93846e14cf53","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-22 17:56:08.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/73/896273/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9bda6d9e783245902cb7657fb4b5e706dede7578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-29 13:40:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"822db331de1500198e82d89a1e9f1e4045823a01","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-29 15:13:24.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/73/896273/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"f807c94da971855bec42be28223288a955216fcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 16:31:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae67b253b16f4bfcab00eff2b14db07c\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3a9865e1b9124af1868cb4ef64e73be1 : FAILURE in 1h 28m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/446e9fd758f74cb9a392a5c092d1e59e : FAILURE in 1h 53m 39s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/db55d79f0bf848859cc719461ea95685 : TIMED_OUT in 2h 49m 20s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/81955a50ccec440ab4945a08e7dc9075 : SUCCESS in 1h 44m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/094ae16a367045d7952ae2045f43e12d : SUCCESS in 33m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8faf0b30d1e04d199ef0ea966f3172a3 : SUCCESS in 37m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c895bae5b53463eb56f7004e1f4059b : SUCCESS in 14m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/919ad54c1d654473abeebc01dadebccf : SUCCESS in 29m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f2c2d9ec1d14c90b636cc76c1eb240e : SUCCESS in 24m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4550511e6d3434a9a4de4a63de61ed6 : SUCCESS in 31m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0aa231122fed4744bd279ca3af68e0b0 : SUCCESS in 23m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b97afad9b1624a4f99c0918cdc4a1a73 : SUCCESS in 12m 26s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/9be47b7bf5c9479abbd5ddfa78d6f5ef : SUCCESS in 30m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/58aaf52f0af341aba51252886061dd33 : FAILURE in 56m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ad50bdb217a4d71bb066e4ac87ff5f1 : SUCCESS in 2h 25m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a42ad1985f64204b4d6654e3e15fcbe : SUCCESS in 59m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a43f8da667b74d29b80172c95adaa7c5 : SUCCESS in 1h 05m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fee9475be6574b51a1288cc29c6fdce6 : SUCCESS in 1h 21m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1663f867a4be4434a62d6186d61912f4 : SUCCESS in 1h 27m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/375450bbc9a54bce95695fff463db97c : SUCCESS in 40m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8f03b0767d1d4152b094d4462242a6fe : SUCCESS in 1h 55m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/27cd6a151b514429b8e5c585ed82150e : SUCCESS in 1h 22m 25s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d0a66ea416e94fe3aa41b0aa502201ad : SUCCESS in 1h 16m 13s","accounts_in_message":[],"_revision_number":2},{"id":"3224f26a1eaa38149113797eb65de6264d7e4c70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 18:01:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b897fc61399541e8867363d2f9991153\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61e454bdaad64a9faaf334bb71bbbd14 : SUCCESS in 43m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e972d2f39866461190eb26bc4b3aca6e : SUCCESS in 53m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/65cfb4a97de94e5683918a15190e6de1 : SUCCESS in 53m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d7984fe09ddb4ddfbdd3c99638630de6 : SUCCESS in 40m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"757d0d4b60219decaf291217ff5b6837dd6fdc64","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2023-10-02 10:10:41.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"372dbb14ad4f16c3758e563e04bce86130884cb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 11:02:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/138afdd3256840c789dbc64c098c7b51\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62a45530c8f44c90ba18ca69a1182cab : SUCCESS in 46m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73e4b563522d403186fa8d432210bed7 : SUCCESS in 47m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e331dbd0a53f416d988e1cb4f0fec631 : SUCCESS in 49m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5e0ff200ec304d8ab7bfac27d8ea7ef7 : SUCCESS in 35m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4b036edb68f20981aeea4806464aa05cdd08d2d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 12:50:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da45de7e064241f4805d56fbc172dd3a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9ac4c4d6536e4c63b3dea6d39bc48146 : SUCCESS in 1h 50m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dad6b32c21e34fa08db8ff24a467acc6 : SUCCESS in 1h 52m 30s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a100f49358074d7cb914b510cb9f334e : FAILURE in 2h 10m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/94d0e19c45224d56b9b2dd780c969f4c : SUCCESS in 1h 44m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b5edb5001e1b4f3c9f7d7830f50bc1f5 : SUCCESS in 43m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25d68994395c4dd181902c711b4de005 : SUCCESS in 40m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e335cf2883f4a2b9b0b595e6dab2c03 : SUCCESS in 10m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8af30f6664e64190a96f7e1b785c3d1e : SUCCESS in 24m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/324fd75edf8a42b884451e3b681b78db : SUCCESS in 26m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/980728d2facc492f9d0593c9904fd12f : SUCCESS in 33m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63b8146ea00e4211b1acfe25ad468d11 : SUCCESS in 27m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5da5675e66640fc9665718b6af85408 : SUCCESS in 11m 02s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/8afa163205f347ad9d1100a2894de382 : SUCCESS in 29m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/176c36c54e4d481a95c5cf2894edf973 : SUCCESS in 48m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/29e774c515a6448fb7ed806e26bdc94b : SUCCESS in 2h 32m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7206a48821f44ffc82c4c5222c56de77 : SUCCESS in 55m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c7e0ed90f5da421090792755454e1370 : SUCCESS in 1h 10m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dd9caabeab07419ab63b646950d33e40 : SUCCESS in 1h 16m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/421a7877ad3440b188e8ee01dab547c0 : SUCCESS in 1h 23m 39s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/75c2568fb2124876a7c0a6ff112a0e1c : SUCCESS in 1h 06m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8e4adce3580048eabfbce7f23457e94e : SUCCESS in 1h 30m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dd3524de7c1a438c82953b73684f45e1 : SUCCESS in 2h 08m 10s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/872a4e498cfb4417b2a187ac35a8e451 : SUCCESS in 1h 05m 03s","accounts_in_message":[],"_revision_number":2},{"id":"449f610ef69fd0c96a955d83810f4fea08f3b772","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-10-03 07:18:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e33677bd3f720efdcd33e21548097a5c4ad156a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-03 07:50:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dbc066f4e00a57050cb4ba27e2f17cb8a1f5130e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 08:55:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e33092033dd4fc39e01cf8c56962c46\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/90bb8e9384664bd78b64eb200bf0cba3 : SUCCESS in 1h 00m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b63b4200854412997f22f2552b22a2c : SUCCESS in 1h 01m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8e3705598a084fe88edf227b0f704d2f : SUCCESS in 1h 02m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bc9cf2dbaa864658b631b3293fd34784 : SUCCESS in 57m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"77579c03885c05e3d818918401db54998459b5a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 09:41:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa31581d1d1f4b50afdaf7233d2bae62\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/58801a209a604a4b8672ae9f41232da9 : SUCCESS in 2h 05m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/409a90de920b4b6ab9ab2bd88425fdf3 : FAILURE in 1h 53m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ed0130d74e6e4c8da10d92158000e70a : FAILURE in 1h 34m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c431524a302a48ef9f5a798784852228 : SUCCESS in 1h 35m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4f37f56ee49d4b17bcb7af8a29b902b6 : SUCCESS in 31m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e544104abbf401981d7d6dab6e15e52 : SUCCESS in 42m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513e097fc28945439fa80d714c938d72 : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11959512ec8f4e4292b65d81bb23ba74 : SUCCESS in 26m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f35aa4da94e410b9670854c224d47da : SUCCESS in 22m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/864d5e8e4bc34eb99083a55a751a6925 : SUCCESS in 22m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a26046394a7a46cf92f37aa7d00ececf : SUCCESS in 24m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8962c7e10b594821b7f8a92472339e32 : SUCCESS in 13m 18s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f5e05b4b2bdf48b587d8cc0e099a9376 : SUCCESS in 31m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/21515f821a744a41ae29a886ae27a4ca : SUCCESS in 49m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a7bfa3f9e1eb49d189effb506ac6158e : SUCCESS in 2h 20m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/672fda8f63834bea855bb0018a95c524 : SUCCESS in 44m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6a9601bbeace4b46adc94308b995704c : SUCCESS in 59m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f5014cc8e98b446b8a194ca365dbe24c : SUCCESS in 1h 28m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c73e4426221248a5b04783c2fde38426 : SUCCESS in 1h 25m 59s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/502db9d604444458b34a4825b9e09d2d : SUCCESS in 47m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bcf071a6958148c5bd7e6e5c22349576 : SUCCESS in 1h 27m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d05625a7464a4ae98e737783288029b2 : SUCCESS in 1h 34m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/66e5230dc1e844c986be90dd17bb5241 : SUCCESS in 1h 11m 35s","accounts_in_message":[],"_revision_number":2},{"id":"2cd6597498feb6d1525319ddbab3966f85f207d0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-04 16:28:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04ad635d61ea1804c082240adb6e8a5f21fe6a09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 18:53:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5441dd0b0b814587941028f0b622a3fc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dc2b8599966d48809141709a3425a7c1 : SUCCESS in 1h 41m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a3f6fb5ea7814fd8aca7e951c3c0f6de : SUCCESS in 2h 08m 42s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5b83b7a0e7ae488e92e0fa6a056df756 : SUCCESS in 1h 33m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/89189b5dbc4840f4ac8c2433ddf756fb : SUCCESS in 1h 42m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0e69bb2461f4407090ca5aa34839701c : SUCCESS in 43m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/702f86a95b954680ada249dac589dadb : SUCCESS in 43m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4acf80478d2848af85db4e1f5e2c846c : SUCCESS in 11m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c1c801b8eda472e8a241b1143dfcc34 : SUCCESS in 26m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27c3a9225e924617b53c05ecb488affa : SUCCESS in 23m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4367ab175f3649e485d993ee215eed32 : SUCCESS in 24m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60be13c1a95a4bfea2b9df977fd24b63 : SUCCESS in 24m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab724a151c64608b80ca9d32d6c3f95 : SUCCESS in 12m 51s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a285a97460b84317af85594bc53fb3a8 : SUCCESS in 37m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c8593e2fbbf4735b2f50bbfdca387ca : SUCCESS in 1h 07m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e9ee187f2e449e498e54baa2db7b8ff : SUCCESS in 2h 13m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/90489524db9341d1b0826f6e6cd702f7 : SUCCESS in 53m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c6c2f2b0b5b34dc6ba72d78b217172f1 : SUCCESS in 58m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9adc9c392d214c21b66fdb5a529b71ff : SUCCESS in 1h 35m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/90e81eb839e34614951ea3531a5465c5 : SUCCESS in 1h 09m 57s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/52e666cfa9474eb589f08ed5305ecefb : SUCCESS in 1h 06m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6e8a7f7434094ccc9632da9d774cf262 : SUCCESS in 1h 29m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e3cdbb0955a14a49b7ed9019e22b8a2d : SUCCESS in 1h 51m 22s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c5108f6501a545eeb69f35121e0db946 : SUCCESS in 1h 23m 13s","accounts_in_message":[],"_revision_number":2},{"id":"a4638f6a74a4f35bfecb9321d17d03368465992f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 23:10:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fcf7b27e49041829d562361dec957bd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8e224e8d527497da845832447b2e3f2 : SUCCESS in 44m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/078bcc7291fa4485a86f2e1389362b60 : SUCCESS in 1h 08m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b8f9b7364594be1adf6e0353b32dd4c : SUCCESS in 1h 07m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/916224ede8534951b97e78f891a6c8c4 : SUCCESS in 39m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21bcfba6b62709a83bb5d76e7e3f450eb1722575","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-09 14:10:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c74bfaccba70e38ed6aeebb7d46e0398c3ed492","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-09 14:11:43.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d88e4be8df99711c406c3a67e115bd50e422503a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 15:34:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c99abfc7ca054910bdb299ca898a07b8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/590849c9d2714a568806d45ba5b322b0 : SUCCESS in 51m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05b1aab7f4c94e9ab4c3ed37ff413154 : SUCCESS in 53m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/df3eef1e212741fcb747321ad888c103 : SUCCESS in 49m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/be1e81395e57408f9f9e057164f67b99 : SUCCESS in 43m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"87431b72f2568cde300c2f0edebddd1f8428bb58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 16:43:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ae36d0009c41a7b0361fa4c871b30e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5d5d96bf463d43f699d6e7c4a42c799e : SUCCESS in 1h 18m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7b989804a58b4ab482f659c17d05a361 : SUCCESS in 2h 21m 08s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2ff25bf6326a4569835be8ece5513825 : SUCCESS in 1h 37m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/47468ce2c7a3487bbb6ce66d7cadd9b8 : SUCCESS in 2h 04m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ac139aa232964df9897d0732e68fec40 : SUCCESS in 34m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b6b827779274acba6b193126bcbaba9 : SUCCESS in 35m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b8c31cadeae484b8c96d4a9a9681f6f : SUCCESS in 10m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b62a22cd42a34b638c9fce660c557e1b : SUCCESS in 25m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e008bb5634d42fe8491f90ef95e48bd : SUCCESS in 22m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83dcc2f6651e481e912e9bd6452459a9 : SUCCESS in 23m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/895e7ee0f09e4c409ff6c908a7884484 : SUCCESS in 18m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb173b5ed0ca4150b30ad20025a66c5b : SUCCESS in 10m 57s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7034505ec7664c03b4cdf1713020ceb7 : SUCCESS in 22m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ed317769f30a43abb9758efbf83daa86 : SUCCESS in 1h 04m 37s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/af609f8c651b4e08bcbf71b74abaf4fd : SUCCESS in 2h 24m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b4085c2e67f94b6d977b359573c8c30e : SUCCESS in 56m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e91723d559884f66b7ac47e61622f106 : SUCCESS in 47m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a74537410b614d7b8c75ee49679e348b : SUCCESS in 1h 38m 49s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ddc0ee9a12244124b631126c2b8da251 : SUCCESS in 1h 29m 54s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9f6f3851c0c14f10a61cf94596923f13 : SUCCESS in 58m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/00b5f7210b42413c8254574f4e851920 : SUCCESS in 1h 26m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b84c735ff7214c699d5b80f2f6e67641 : SUCCESS in 1h 39m 58s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/509c2bb817864f6c83acbe725f2de878 : SUCCESS in 1h 15m 21s","accounts_in_message":[],"_revision_number":3},{"id":"931028beefbca5cefabd414aead3301d6eafd96a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2023-10-09 18:55:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8b84209f93c419bf4fc74822e0c5d0c76249fcf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 18:55:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cdf9c0280f57b78fa2a840aeb70860561fa3af96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 21:27:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c73fe22ebef4b198043420ac3bd1344\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ffe2f8d2474e4b81ad988f809168e487 : SUCCESS in 1h 39m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b4b2eed7715e48a2bff25630fe996f14 : SUCCESS in 1h 38m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4cb5d37cc9e4ea1a7e6b25c455c7c13 : SUCCESS in 1h 57m 18s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/09e4e7ad27874b8db5d1945c63c495f7 : SUCCESS in 1h 27m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a93c37e0aab14fecb078491e3ad6f6ce : SUCCESS in 12m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bf5fd005b3a4c3d8f2025836b4a139b : SUCCESS in 26m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f3d0c981d8c43dcac9fe51c616a95d8 : SUCCESS in 24m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff38b54435224fb19500c2100fc4d790 : SUCCESS in 25m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642705a094aa4f489039e8e355178c6b : SUCCESS in 10m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f4454a1f380a4ba5a702966b2c1d35ee : SUCCESS in 57m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2886f3c0d1ce42e1af890b0fec527059 : SUCCESS in 2h 26m 58s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dc8a7bea063c413a94f4946e89606364 : SUCCESS in 1h 32m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7b1f7968ff3c46a4b4621f58811ac6d4 : SUCCESS in 1h 02m 43s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/06b82244fd614c018260d4c2bfbc1559 : SUCCESS in 54m 24s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f5eb4ca48b5f49f38d24890baf910ba0 : SUCCESS in 58m 48s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5f166961488345eeb95c7a028fdbdb6a : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":3},{"id":"05e2ac9d5306f89d411e1c2f932f2afba0277041","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 21:27:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5c6c9dff83f88f6d9612de9962cc943b00e0957c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 21:29:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f453741a27de47b8ac8a607c1a2284df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3104fd20171049e3aadd9cf355d9103c : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"78027da56ccb25d19ac2c3bc1c174acb2150e6a5","revisions":{"22f32e450d4440d0f5d70f952d4ca95042e57b88":{"kind":"REWORK","_number":1,"created":"2023-09-22 16:10:21.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/896273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/896273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/896273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/896273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/896273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/896273/1"}}},"commit":{"parents":[{"commit":"b5ac9ae6d6920214e20a72e495a7e5c955ba0d66","subject":"Add a new extension \"security-groups-rules-belongs-to-default-sg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b5ac9ae6d6920214e20a72e495a7e5c955ba0d66"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-12 08:57:45.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-12 10:11:19.000000000","tz":0},"subject":"Remove the publish patch in SG rule BEFORE_DELETE and BEFORE_CREATE","message":"Remove the publish patch in SG rule BEFORE_DELETE and BEFORE_CREATE\n\nThe method ``delete_security_group_rule`` is publishing the\nBEFORE_DELETE event before starting the security group rule deletion.\nThis event is published using a wrap method called\n``SecurityGroupDbMixin._registry_publish``. This method is capturing\nany ``CallbackFailure`` exception and raising a\n``SecurityGroupRuleInUse`` one. That makes no sense because:\n* We are hidding the real cause of the callback failure.\n* The BEFORE_DELETE is not checking that the security group rule is\n  being used (NOTE 1).\n* If any new implementation makes this check, the corresponding callback\n  should return explicitly this exception.\n\nThe method ``_create_security_group_rule`` is publishing the\nBEFORE_CREATE event before starting the security group rule creation.\nThe same argument applies here: the callback manager should return the\nexception raise by the callback method (NOTE 2).\n\nIn a follow-up patch, this events will be captured to check the\npermissions related to the user creating or deleting the security group\nrule. In case of error, it will be needed to raise a ``NotAuthorized``\nderived exception, instead of a ``InUse`` one.\n\nNOTE 1: this is the current use of BEFORE_DELETE event in the\nOpenStack repository:\n* [2] Omni project had no activity for the last 4 years.\n* [3] networking-arista: the method ``run_cmds_on_all_switches``, that\n  calls ``run_openstack_sg_cmds``, returns its own exceptions.\n* [4] networking-opencontrail: same justification.\n* [5] The ML2/OVN mechanism driver, that will raise an exception if the\n  OVN ACL deletion doesn\u0027t succeed.\n\nNOTE 2: this is the current use of BEFORE_DELETE event in the\nOpenStack repository:\n* [2] Omni project had no activity for the last 4 years.\n\n[1]https://codesearch.openstack.org/?q\u003d%5C.SECURITY_GROUP_RULE\u0026i\u003dnope\u0026literal\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d\n[2]https://opendev.org/x/omni/src/branch/master/neutron/neutron/plugins/ml2/drivers/aws/callbacks.py\n[3]https://opendev.org/x/networking-arista/src/branch/master/networking_arista/ml2/security_groups/arista_security_groups.py\n[4]https://opendev.org/x/networking-opencontrail/src/branch/master/networking_opencontrail/ml2/opencontrail_sg_callback.py\n[5]https://opendev.org/openstack/neutron/src/branch/master/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n\nPartial-Bug: #2019960\n\nChange-Id: I8d5f5392fb7a6ab9b20e9222c143f4e67c925cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22f32e450d4440d0f5d70f952d4ca95042e57b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/22f32e450d4440d0f5d70f952d4ca95042e57b88"}]},"branch":"refs/heads/master"},"84124e8df1295fea7b2425996fcbd357f6441cac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-29 13:40:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/896273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/896273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/896273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/896273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/896273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/896273/2"}}},"commit":{"parents":[{"commit":"696b7cbec6c881b51401e70405843853bede651e","subject":"Add a new extension \"security-groups-rules-belongs-to-default-sg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/696b7cbec6c881b51401e70405843853bede651e"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-12 08:57:45.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-09-29 13:40:27.000000000","tz":0},"subject":"Remove the publish patch in SG rule BEFORE_DELETE and BEFORE_CREATE","message":"Remove the publish patch in SG rule BEFORE_DELETE and BEFORE_CREATE\n\nThe method ``delete_security_group_rule`` is publishing the\nBEFORE_DELETE event before starting the security group rule deletion.\nThis event is published using a wrap method called\n``SecurityGroupDbMixin._registry_publish``. This method is capturing\nany ``CallbackFailure`` exception and raising a\n``SecurityGroupRuleInUse`` one. That makes no sense because:\n* We are hidding the real cause of the callback failure.\n* The BEFORE_DELETE is not checking that the security group rule is\n  being used (NOTE 1).\n* If any new implementation makes this check, the corresponding callback\n  should return explicitly this exception.\n\nThe method ``_create_security_group_rule`` is publishing the\nBEFORE_CREATE event before starting the security group rule creation.\nThe same argument applies here: the callback manager should return the\nexception raise by the callback method (NOTE 2).\n\nIn a follow-up patch, this events will be captured to check the\npermissions related to the user creating or deleting the security group\nrule. In case of error, it will be needed to raise a ``NotAuthorized``\nderived exception, instead of a ``InUse`` one.\n\nNOTE 1: this is the current use of BEFORE_DELETE event in the\nOpenStack repository:\n* [2] Omni project had no activity for the last 4 years.\n* [3] networking-arista: the method ``run_cmds_on_all_switches``, that\n  calls ``run_openstack_sg_cmds``, returns its own exceptions.\n* [4] networking-opencontrail: same justification.\n* [5] The ML2/OVN mechanism driver, that will raise an exception if the\n  OVN ACL deletion doesn\u0027t succeed.\n\nNOTE 2: this is the current use of BEFORE_DELETE event in the\nOpenStack repository:\n* [2] Omni project had no activity for the last 4 years.\n\n[1]https://codesearch.openstack.org/?q\u003d%5C.SECURITY_GROUP_RULE\u0026i\u003dnope\u0026literal\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d\n[2]https://opendev.org/x/omni/src/branch/master/neutron/neutron/plugins/ml2/drivers/aws/callbacks.py\n[3]https://opendev.org/x/networking-arista/src/branch/master/networking_arista/ml2/security_groups/arista_security_groups.py\n[4]https://opendev.org/x/networking-opencontrail/src/branch/master/networking_opencontrail/ml2/opencontrail_sg_callback.py\n[5]https://opendev.org/openstack/neutron/src/branch/master/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n\nPartial-Bug: #2019960\n\nChange-Id: I8d5f5392fb7a6ab9b20e9222c143f4e67c925cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84124e8df1295fea7b2425996fcbd357f6441cac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/84124e8df1295fea7b2425996fcbd357f6441cac"}]},"branch":"refs/heads/master","description":"Rebase"},"78027da56ccb25d19ac2c3bc1c174acb2150e6a5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-10-09 14:11:43.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/73/896273/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/73/896273/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/896273/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/896273/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/73/896273/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/73/896273/3"}}},"commit":{"parents":[{"commit":"e066cab875bef07308dc91a163ae03f82006d97f","subject":"Add a new extension \"security-groups-rules-belongs-to-default-sg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e066cab875bef07308dc91a163ae03f82006d97f"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-12 08:57:45.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2023-10-09 14:11:43.000000000","tz":0},"subject":"Remove the publish patch in SG rule BEFORE_DELETE and BEFORE_CREATE","message":"Remove the publish patch in SG rule BEFORE_DELETE and BEFORE_CREATE\n\nThe method ``delete_security_group_rule`` is publishing the\nBEFORE_DELETE event before starting the security group rule deletion.\nThis event is published using a wrap method called\n``SecurityGroupDbMixin._registry_publish``. This method is capturing\nany ``CallbackFailure`` exception and raising a\n``SecurityGroupRuleInUse`` one. That makes no sense because:\n* We are hidding the real cause of the callback failure.\n* The BEFORE_DELETE is not checking that the security group rule is\n  being used (NOTE 1).\n* If any new implementation makes this check, the corresponding callback\n  should return explicitly this exception.\n\nThe method ``_create_security_group_rule`` is publishing the\nBEFORE_CREATE event before starting the security group rule creation.\nThe same argument applies here: the callback manager should return the\nexception raise by the callback method (NOTE 2).\n\nIn a follow-up patch, this events will be captured to check the\npermissions related to the user creating or deleting the security group\nrule. In case of error, it will be needed to raise a ``NotAuthorized``\nderived exception, instead of a ``InUse`` one.\n\nNOTE 1: this is the current use of BEFORE_DELETE event in the\nOpenStack repository:\n* [2] Omni project had no activity for the last 4 years.\n* [3] networking-arista: the method ``run_cmds_on_all_switches``, that\n  calls ``run_openstack_sg_cmds``, returns its own exceptions.\n* [4] networking-opencontrail: same justification.\n* [5] The ML2/OVN mechanism driver, that will raise an exception if the\n  OVN ACL deletion doesn\u0027t succeed.\n\nNOTE 2: this is the current use of BEFORE_DELETE event in the\nOpenStack repository:\n* [2] Omni project had no activity for the last 4 years.\n\n[1]https://codesearch.openstack.org/?q\u003d%5C.SECURITY_GROUP_RULE\u0026i\u003dnope\u0026literal\u003dnope\u0026files\u003d\u0026excludeFiles\u003d\u0026repos\u003d\n[2]https://opendev.org/x/omni/src/branch/master/neutron/neutron/plugins/ml2/drivers/aws/callbacks.py\n[3]https://opendev.org/x/networking-arista/src/branch/master/networking_arista/ml2/security_groups/arista_security_groups.py\n[4]https://opendev.org/x/networking-opencontrail/src/branch/master/networking_opencontrail/ml2/opencontrail_sg_callback.py\n[5]https://opendev.org/openstack/neutron/src/branch/master/neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n\nPartial-Bug: #2019960\n\nChange-Id: I8d5f5392fb7a6ab9b20e9222c143f4e67c925cae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78027da56ccb25d19ac2c3bc1c174acb2150e6a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/78027da56ccb25d19ac2c3bc1c174acb2150e6a5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
