)]}'
{"id":"openstack%2Fneutron~896299","triplet_id":"openstack%2Fneutron~master~Ie9598e6801b093b80d736b313941695c3ed52c02","project":"openstack/neutron","branch":"master","topic":"use_utils_ovn_port","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-12-01 17:39:38.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2023-12-02 20:05:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie9598e6801b093b80d736b313941695c3ed52c02","subject":"Use is_ovn_metadata_port from the ovn utils","status":"MERGED","created":"2023-09-22 20:42:26.000000000","updated":"2023-12-02 20:06:36.000000000","submitted":"2023-12-02 20:05:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896299-use_utils_ovn_port","meta_rev_id":"c7013a2e76abf05bfdce0a0d98e13d98a17ab4d7","_number":896299,"virtual_id_number":896299,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-09-22 21:17:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-02 20:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-03 11:59:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-10-30 01:36:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-12-01 17:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-12-01 17:39:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-22 20:43:35.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-09-22 21:17:52.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-22 23:10:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-03 11:59:58.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-30 01:36:56.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"3040c40341f2ad34ed661f43b280e0fc5c6586d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-22 20:42:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5ea47feb97bebac324b8e2721d1410f79d5ffac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-22 21:17:52.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/99/896299/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"23ca43cab2df3d76b781a107b8fc20853a19d8c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 23:10:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cba0f13daaa64477a21c5422f85c6451\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/61cc665cbeb540a6af1bc24291dbe320 : FAILURE in 1h 24m 05s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2a0c1c5bdc454a168a483d4c25bfbf40 : FAILURE in 1h 50m 06s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b29ffdd90ff74ccab837657ea8f950d3 : FAILURE in 1h 18m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/99d49da592034275a07c0a84e2f4133d : SUCCESS in 2h 02m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5645bb6eeecb41c590f47766b83a4e03 : SUCCESS in 28m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03d644381544496aac7759df21fa324b : SUCCESS in 9m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f7761fbb2ee481bbd321b5e12acb209 : SUCCESS in 22m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4d5625d8ee9429181889a53c52a27ea : SUCCESS in 20m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cfc080917754fb49bff322b2fb90b14 : SUCCESS in 25m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/776210e8cd954ac7aad254a309bc7922 : SUCCESS in 19m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c380393522c4643971788795cce763d : SUCCESS in 11m 55s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/0fd9919e30a34bc39416df61a11500b4 : SUCCESS in 22m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d52555cc339646d98e6afda90dea5b9b : SUCCESS in 57m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f960e31c5c7048f1a14cf6c2af3ef2e0 : SUCCESS in 2h 22m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4b0d7b23030474188420bae76b83809 : SUCCESS in 49m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7fec57dd95a241a8b858d6b2a654204c : SUCCESS in 56m 42s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/003733aa198243df8035dba203b79ebe : SUCCESS in 1h 26m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f854d9e898b743af88ffd35124335d3b : SUCCESS in 1h 31m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/f31b9e25c89241779f87e4478599f2d0 : SUCCESS in 1h 05m 07s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/438a453a4cc345b5b7674b661eafd2dc : SUCCESS in 2h 02m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8f7150123bf744808de92425ecb1ba79 : SUCCESS in 1h 40m 57s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/33697ac7196b466191b1af13ddbc463c : SUCCESS in 1h 18m 41s","accounts_in_message":[],"_revision_number":1},{"id":"cecc46039f8b2c7502ba8432c0c74965466a08a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 23:17:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/934db123a58245e29cef2520fb1776ec\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a73e6450e8864567b3b3122582786c09 : SUCCESS in 47m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18bd2848938f470897d08dba349cae88 : SUCCESS in 48m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f159433415e04f12b2779d795765634a : SUCCESS in 43m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b859a944e2d74b2596bc9e627368f5ce : SUCCESS in 37m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4f9ccd7951d73e62a1866670777af437bbda0d5","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2023-09-26 13:52:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a16e2149c1786350aee76854cece3f08debfbd51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 15:06:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7d6872f352c43f79dd8ac9a06313ed2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a219261dfb146e89ff0c5f8c72eba70 : SUCCESS in 40m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43d9d892551f432c8376ec3422966faa : SUCCESS in 46m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/65b61d32c8ad4babbf188b35f98b1cef : SUCCESS in 1h 03m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6cd83e1f355f4f108ab253230e8a0a5a : SUCCESS in 40m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dddcc9c3766e5d5d3be811615c0b95190102840b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 16:44:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/68e4da7e39be4d0cae225809d9d76601\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/96107639ab03426ca3485be5dd4e8043 : SUCCESS in 2h 00m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ae9838a3dafe45129fc8286c946408d2 : TIMED_OUT in 2h 49m 30s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/df2f01b9b2694266ab2dbcb5b76d4496 : FAILURE in 2h 10m 46s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f8a27203050542598424d0550d1a1f8c : FAILURE in 1h 32m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41cb74fe98b44fc88a00561336bce5e4 : SUCCESS in 36m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9c4d0ec081041a9a8eedd56a38fe92c : SUCCESS in 9m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9204d7d9500d4e8db248a818a1ddac6d : SUCCESS in 25m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bbb78b808f2429096910f3089b97c8f : SUCCESS in 26m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03d07e2424324bb8abe38549486fe7d6 : SUCCESS in 28m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae0d72fb2601487682f3661d82e6113d : SUCCESS in 25m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/305a0a675aec42d6bc05e61f0653afee : SUCCESS in 11m 44s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/d7788500631447f689df902e4c66b799 : SUCCESS in 24m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/01db44dbef1a49708713513ce560b155 : SUCCESS in 52m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/babfda3a596541798aa28d616691002c : SUCCESS in 2h 02m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/129cb357c0134c6fb0664b627df314b3 : SUCCESS in 1h 01m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4a155351553746f1a992948d267f6558 : SUCCESS in 1h 01m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b69d807dbdfc48ecb7200c18d0cd2e5d : FAILURE in 1h 21m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4fbf2a2dd77b4c26a5335aad80e7503a : SUCCESS in 1h 18m 12s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/c743262924c847ba835d67f1af3b1b34 : SUCCESS in 51m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/eba3c778702c4ac590869ef1dc0dfba8 : SUCCESS in 1h 36m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e35b0880d1f540fca601c620687c5fa5 : SUCCESS in 2h 04m 45s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/970760efd63c4bc98e03df137c39feb5 : SUCCESS in 1h 35m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e2b6644f4aa43b5090cff76870a66bf2f0d11865","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-03 11:59:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a2f3a229dc004759889e9e6351e3293cc5adf37","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-03 12:00:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb3c5f01f0f727883029581ad14aa31c9e72cfd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 14:00:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29a51282fa70466a98e509b70ca29bf8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46dc610f21274690a1ada02d24ce45bb : SUCCESS in 40m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e962526bc2348c48514e110e11e2fbd : SUCCESS in 50m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d12ca6b0a1e44b7cadb9132511aa9a2e : SUCCESS in 44m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d021f4255c2546539dd75bfbf1809d1a : SUCCESS in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fe0ebd86db0ba49543b57e45684cc23052acc4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 14:20:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6e27734205d4048b7a245f0745b2fea\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f1699b9bc1a74f33819f0f11eddae554 : SUCCESS in 1h 48m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/99659544ddec40bba343c03a0871144f : SUCCESS in 2h 03m 58s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6fd7e36d985a4f9ca57e7c469714e312 : SUCCESS in 1h 46m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c64e3a0b537c4301b1e6484db20462f9 : SUCCESS in 1h 59m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6652f4ead14941299438d61a6c26a6bd : SUCCESS in 40m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01955d099cbe4773b82a91aca8f7ce52 : SUCCESS in 11m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/011cb3062576488bb5b0c67de01ec5e1 : SUCCESS in 27m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6916ce1dd69b47c2a43cdfd639cbb8ac : SUCCESS in 26m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c2da58aed2944fba5364df9c85bedd8 : SUCCESS in 26m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6790b5d421fe45939dba9ef54fc9b6bb : SUCCESS in 25m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c055c266f5e4f079aec9ec659095bc7 : SUCCESS in 12m 20s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/1095afc7b00d4efdaae321c3dc9260b5 : SUCCESS in 30m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/548a9a54929e49649076b1b874223b90 : SUCCESS in 1h 03m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b03cf0ccb5d47e284bb66dadb5514de : SUCCESS in 2h 17m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f03ebcffec434df89997814830aed47f : SUCCESS in 48m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2f115a0cd91945ae87a3198529732dba : SUCCESS in 52m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/eb8afcea67c743e981170ce525e795fd : SUCCESS in 1h 29m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/57570395d40d45679b5b72ce2359530b : SUCCESS in 1h 25m 21s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e9ca0126a2f246d693112610a1cb5acf : SUCCESS in 51m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3f34db2809054fef863b11aa35d6795c : SUCCESS in 1h 35m 47s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1bd265f04124447d80b3c14fcefe66e7 : SUCCESS in 1h 34m 42s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3938d25e5b4d4738b131ad5189f48a3c : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":1},{"id":"15b52e15d4b8ad67e2ba1cc11ae0ae243c8ff6b5","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-10-30 01:36:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ad981bb903badacdc7cf820666c20bbfe6fe09a3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-01 17:39:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"19e0967fff8576451450aa9007ba9f678c1e8b2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 17:40:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3097baa6b167f068ed29d7b74597f11f1759ad6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 19:53:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d425d1a8c664a23a09b8d6133d2613e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5bdaec059c654ca7901dba683a79fe66 : SUCCESS in 1h 43m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/97cebece16874509b229ab02e40c4228 : SUCCESS in 1h 57m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/edabcac03e9e4b198731148c4bbcb037 : SUCCESS in 2h 11m 14s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/f33f77ead61a4445aa41e472d1d7c10b : SUCCESS in 1h 59m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcca4a2f116c410eb57ec6733c9c9792 : SUCCESS in 9m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82cd502ad9404662b7a396d4f725a626 : SUCCESS in 23m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5b11e32e99c443bb893e31e1ed4396b : SUCCESS in 24m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b32def468f043348ba30c77bc278957 : SUCCESS in 11m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9897e98e773c416db1f599ec7a8bac1e : FAILURE in 46m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5c7ac610aaf45158b342771bfed6875 : FAILURE in 1h 44m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/73b57f68be914a3d823b0efa669df28d : SUCCESS in 1h 19m 36s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f954e2d2785432890527c5343903631 : SUCCESS in 50m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d9131eb0dab54cd7ac4e9d66c8dd66f1 : SUCCESS in 44m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/3f9b224250df456187b6171e32bedf40 : SUCCESS in 46m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b7765a63594e48e6814f5173473cd6a2 : SUCCESS in 1h 08m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1bb86a8959798dd03ba19ac5405b79c42a550069","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-01 21:08:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"194c6736c622a70ce6ad4cc50bc8c84566aa1003","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 21:27:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b41294bfc396419d93b31ccc1d50f753\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa6453a1c0bb4647acadb8a3160b9078 : SUCCESS in 18m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/506259b555d14706b3a6264c97bb3b91 : SUCCESS in 19m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bbdd21457016437fbe62fe8107fe2f5b : SUCCESS in 17m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1890387aede8a8a930e66e0eb1ba74042cb37ca8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 00:11:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c38891ffd734216ba78d4ced371575a\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/498bb81837ff424a993e79b15af0b073 : SUCCESS in 1h 58m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c13252d0e2144abbb9b2428dec994eff : SUCCESS in 2h 15m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/0a437d7ac9b54667aae4e8049d7b3333 : SUCCESS in 1h 45m 33s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/748cfc3532824d7b9812901b3046cde8 : SUCCESS in 1h 55m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ae7660b15714ca89143841a81f6de50 : SUCCESS in 31m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3d269ec0c194f01b4a5091689773860 : SUCCESS in 9m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1dba5b399dd4eeca5f7a202878041c4 : SUCCESS in 24m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5cb4e9bffc54513b78842e940bb6ad9 : SUCCESS in 26m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ea9e4918e94c83877f8b8d04d09961 : SUCCESS in 11m 33s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/572d88a331ac415c8ca1289d8cda1ced : SUCCESS in 31m 24s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/2cc1d6bc3e1a4f52a1566590da1b4bab : SUCCESS in 1h 04m 48s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/8ad2a0aadf5a42c7a13b6c0e2d8b6081 : SUCCESS in 59m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bacdafa010714bb083ff8c8a36b9b83e : SUCCESS in 1h 03m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b313680796e34c728545015a7144f6ca : TIMED_OUT in 3h 02m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f81f3cc625694a97948357ea1430f0da : SUCCESS in 1h 06m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c2bb5a5e0eff45e0b98e21f38eb7d262 : SUCCESS in 1h 05m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5c4d1d0b0c104a5892481d2975e7b47c : SUCCESS in 1h 18m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eb9b34164f5c4c1aaa5173c738e9507b : SUCCESS in 1h 22m 27s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/256e8de3f4fd42d1aa2f39e9b8b24eb6 : SUCCESS in 52m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7389e2accd2c43c68d9f128b8a44a467 : SUCCESS in 2h 00m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/e2e37a0c60114fa58b5894683a36bcb7 : SUCCESS in 2h 06m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/7646415d49a548d4a6cbe5baa2111305 : SUCCESS in 1h 19m 32s","accounts_in_message":[],"_revision_number":1},{"id":"66df14f55557bd98829883ad77a0faa425861fc5","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-12-02 15:57:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7924d3bbb872f28435b076fff86dffd44b5f350c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 16:48:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b5b5e82889042f2a937be44704124b8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/114d9a395a85456e817ff13ac58ef885 : SUCCESS in 48m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ffe09dc4adc34a6c96dcaec13084c5b8 : SUCCESS in 49m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cec10fda46e84547845f7e25517ca261 : SUCCESS in 43m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0c410147147b6d63f43c7f18c5591ee7737b04e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 17:58:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6f79a5e05e34605966393c5749d62a8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/600519f0aaf0481b86292c37a458b1a5 : SUCCESS in 1h 28m 26s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6f6eeee755cb43829fa4469c99f8fa8a : SUCCESS in 1h 52m 04s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/b23a75bafb3c4777810630489caa7701 : SUCCESS in 1h 07m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/153c4640436549a18df983333728020e : SUCCESS in 1h 33m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/864de297b8704e98ba6fb2b11c676404 : SUCCESS in 35m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add9ea4b948b42a59f48c512e75c9697 : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bd54f73c68c45e19156e08dcd4f91e6 : SUCCESS in 26m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25e50b130da1405c953ad3bb4b4ad3e2 : SUCCESS in 23m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c56c2352c74ff39304f8384f09843e : SUCCESS in 9m 12s\n- openstack-tox-py311-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/869ae8c508a147f2bf58cb641da6e949 : SUCCESS in 28m 39s\n- neutron-ovs-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/48a2c3a2d6064befa75616ff1c726b8b : SUCCESS in 41m 30s\n- neutron-ovn-grenade-multinode-skip-level https://zuul.opendev.org/t/openstack/build/69095c4a47c74f3890a4480be0c4e839 : SUCCESS in 1h 07m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c6d66f15e204cb7aac1ad7ff6b860b6 : SUCCESS in 1h 01m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fb9856bc8d24d55b498c683cc4d3c05 : SUCCESS in 1h 55m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b1e99e6c08d4126946dc75e00b762e5 : SUCCESS in 52m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dc8e118a05df401bafb455cb40d123e8 : SUCCESS in 58m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6fa0e04280a947de9f25c08bb6a8f67d : SUCCESS in 1h 24m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dac8123f75c5408d8edc08dac9bbdd4b : SUCCESS in 1h 28m 46s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/e054e90c85604376bd9bbf424bd5c7e8 : SUCCESS in 1h 22m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2f881e740df9459faa0ca599460921b6 : SUCCESS in 1h 55m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c07a4e6c2aff496ca63573d07cb0afd0 : SUCCESS in 1h 40m 04s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f6ffe304751e4d1c965763cfa74dfd32 : SUCCESS in 1h 26m 25s","accounts_in_message":[],"_revision_number":1},{"id":"378a3137171011b9630c307645620bc3af1bf08f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 17:59:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65bfc33eccd9e1f57dfe31772bfd8d6c72b368b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 20:05:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15e258cd395741a792c6f2744f7661de\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aed24f9c835c4529a6fa8c9c47385075 : SUCCESS in 1h 20m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cfc2e952022247c09615ccdefa86f497 : SUCCESS in 1h 26m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/28954d5faa3f4a179d4a18f27edbe2a0 : SUCCESS in 1h 39m 07s\n- neutron-tempest-plugin-openvswitch-enforce-scope-old-defaults https://zuul.opendev.org/t/openstack/build/212437aa03c240a1889e3772decf0bdf : SUCCESS in 1h 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f4c6d5283e24b428d4f61b3b75ff003 : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a901c461991649d090487ef437ec87a0 : SUCCESS in 31m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/457d4d1ff47e4b2b82a1c30802a0e0d7 : SUCCESS in 17m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a7dd663ce6e4dcc912160b63b2896bf : SUCCESS in 11m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c2829c0ae6874c0d81a2f2fb72b4c00c : SUCCESS in 47m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8dfd38637f3f424c88186dd2509259b1 : SUCCESS in 2h 00m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c6bf27882b2c468bbfea14ef3f5ea04a : SUCCESS in 1h 29m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7afee95ff134a4e9664b511354eb1a0 : SUCCESS in 38m 24s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/248882161ff5417eaf87b403bf8207b8 : SUCCESS in 1h 03m 53s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/07ac6707510d4add98157ac89ad10035 : SUCCESS in 1h 05m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4d1fc8ab5c1840509d0784429867422b : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":1},{"id":"afca0b15c4a171e04afc737d9cda1c2f3b8a1ba9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 20:05:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c7013a2e76abf05bfdce0a0d98e13d98a17ab4d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 20:06:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7104276e5b6d474fb1aaa7418ae9437a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f1f586525e449b916d54fba383daaf : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c571cedbb1901dc96f33e50572a2a333abc4c27","revisions":{"4c571cedbb1901dc96f33e50572a2a333abc4c27":{"kind":"REWORK","_number":1,"created":"2023-09-22 20:42:26.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/99/896299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/896299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/896299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/896299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/896299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/896299/1"}}},"commit":{"parents":[{"commit":"82f2a21d1c9e27999d3fd7006a7ecf961039a370","subject":"[DHCP agent] Add route to OVN metadata port if exists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82f2a21d1c9e27999d3fd7006a7ecf961039a370"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-09-22 20:26:37.000000000","tz":-240},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2023-09-22 20:26:37.000000000","tz":-240},"subject":"Use is_ovn_metadata_port from the ovn utils","message":"Use is_ovn_metadata_port from the ovn utils\n\nUpdate the code to use the new is_ovn_metadata_port from utils.\nUpdate the test such that mock port objects are subscriptable. This\nis required for the `is_ovn_metadata_port`. I have utilized already\nexisting `Dictable` class. However, I have removed `__class__` as I\ndont think this correct intention of this class. It was introduced\nhere [1] feel free to disagree.\n\n[1]https://review.opendev.org/c/openstack/neutron/+/205181\n\nChange-Id: Ie9598e6801b093b80d736b313941695c3ed52c02\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c571cedbb1901dc96f33e50572a2a333abc4c27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c571cedbb1901dc96f33e50572a2a333abc4c27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
