)]}'
{"id":"openstack%2Fneutron~896504","triplet_id":"openstack%2Fneutron~master~If32c876c6c17c8786b90e4515af463d5a0e9c25a","project":"openstack/neutron","branch":"master","topic":"bug/2037239","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-10-04 10:10:33.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}},"hashtags":[],"change_id":"If32c876c6c17c8786b90e4515af463d5a0e9c25a","subject":"[DNM] Test oslo.log pin","status":"ABANDONED","created":"2023-09-26 10:25:36.000000000","updated":"2023-11-09 04:44:15.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4fd81b414ea432cde6a2f768c53fb5df35e18543","_number":896504,"virtual_id_number":896504,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-10-04 10:01:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-26 12:28:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"79dbdcd895a842f5732093c575aa82a538dae5b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-26 10:25:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42a6ce4185792ab93587b3fe88c03c4faf6df64b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 12:28:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57290400ddc44b28a4d60944040d7301\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5a412dafbeab4e61a67f7dc0d6cca04c : FAILURE in 1h 27m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a208fb5b59c64f26bec109af086ff487 : FAILURE in 1h 59m 56s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e1d5b4de44674973abc3f52e16a6c0b8 : SUCCESS in 1h 27m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5efc1948a89849bf96c5beb8533194ad : SUCCESS in 1h 25m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5dcaf832ddc5434f97b4834c18193ec9 : SUCCESS in 47m 16s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/bfabcc769641417889e5517ba0c85f9e : FAILURE in 1h 44m 42s","accounts_in_message":[],"_revision_number":1},{"id":"eefd042f644dfe1f8fa4ae624a06322845fb2736","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-26 13:02:24.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d548664dc65a60523edb48139009764ad6799b42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-26 14:47:50.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c6909da343fd905cbcbf164bab9ddf60dcfb90b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 16:52:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b7c699fa4a3443eb853dc93d75be521\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/49774e5d06b14b42815417db5bf27798 : SUCCESS in 1h 47m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e984bc80aa07494c92668b6b83217b85 : SUCCESS in 1h 50m 11s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ed89217cd732408296c2e2c4a65a0c5e : SUCCESS in 1h 33m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/733b47f68c68446690ee7f337c071d90 : SUCCESS in 1h 59m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d574770108ff4531b9048aaa8b2c65b3 : SUCCESS in 39m 29s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6c086eaf82f14422a7e24c41e4178d11 : FAILURE in 2h 02m 28s","accounts_in_message":[],"_revision_number":3},{"id":"9b4726c456908e77e2b5b4e6feaa4bf37214dd32","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-27 08:12:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aeeca9d1f24fb7a1ebc853050734c425cc97be45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 10:26:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ddcde25985d40fdaf467bd6a5bcebf1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b627045e78384aa79a8ccd05c9975909 : SUCCESS in 1h 39m 15s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/85d310daa4c643d2b1dde8df01221491 : FAILURE in 1h 54m 35s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fdc99fbb9d7e484497169c8e865648e7 : FAILURE in 1h 55m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2d9117ae721d48088c8088572b505ff4 : SUCCESS in 1h 43m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7081ff5d849a417ca9bc216fe3a99d8c : SUCCESS in 32m 48s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/a402e05b4b03447b8aeec1132bbe97dd : FAILURE in 2h 11m 37s","accounts_in_message":[],"_revision_number":3},{"id":"c3f27d4c4d5822398edaccaf9688fb1399b75f98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-29 06:44:02.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ec33cc6cc4cb6e41697cacc4e598ea21d4584ed5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-29 08:56:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2a324a965ae4f57cc19b5f3cff04ed8248aa460e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:30:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9ecee67263046b2a56eefd426cad8e2\n\n- neutron-tempest-plugin-linuxbridge-1 https://zuul.opendev.org/t/openstack/build/af2de09a90784566827052a8ced82020 : POST_FAILURE in 2h 25m 56s\n- neutron-tempest-plugin-linuxbridge-2 https://zuul.opendev.org/t/openstack/build/f162f4c508ab466fac4fba877d681677 : POST_FAILURE in 2h 32m 05s","accounts_in_message":[],"_revision_number":5},{"id":"434e034df57cf2ac8fbf394530d1660caafae2b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-04 07:28:33.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"16bbbf42686cec9bf18cd3afb35c7cc3db5b1114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 10:01:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3206e3c74a4403cb48bae3434a0f00c\n\n- neutron-tempest-plugin-linuxbridge-1 https://zuul.opendev.org/t/openstack/build/b56541df33f04eeba51ab1e116c96f0f : SUCCESS in 2h 03m 51s\n- neutron-tempest-plugin-linuxbridge-2 https://zuul.opendev.org/t/openstack/build/ad606b8ebcf443209303f17b5d62ff47 : SUCCESS in 1h 36m 14s\n- neutron-tempest-plugin-linuxbridge-3 https://zuul.opendev.org/t/openstack/build/85fc19b16dd7420386e436aecc4517cd : SUCCESS in 2h 23m 56s\n- neutron-tempest-plugin-linuxbridge-4 https://zuul.opendev.org/t/openstack/build/1f81d59405a04fe493bb4c0daa68d15f : SUCCESS in 2h 07m 30s\n- neutron-tempest-plugin-linuxbridge-5 https://zuul.opendev.org/t/openstack/build/033398a1a33a444e99c273c881814452 : SUCCESS in 2h 09m 00s","accounts_in_message":[],"_revision_number":6},{"id":"6187415b1184d1fdc0bc9e85bd8a3ebc89371c14","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-10-04 10:10:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e95868d6902e806c59c3cce5b487cec5e1421db6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 12:21:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bd8c3c280c7486dbb29ac8e20308d82\n\n- neutron-tempest-plugin-linuxbridge-1 https://zuul.opendev.org/t/openstack/build/3d6d7cf67eeb4d1281f2d4e8d4d35938 : SUCCESS in 2h 09m 02s\n- neutron-tempest-plugin-linuxbridge-2 https://zuul.opendev.org/t/openstack/build/b637f5dee8b84efea3e992a31bad7b81 : SUCCESS in 1h 59m 35s\n- neutron-tempest-plugin-linuxbridge-3 https://zuul.opendev.org/t/openstack/build/f62bcb7864134a35a4c618b40fa596f2 : SUCCESS in 2h 06m 00s\n- neutron-tempest-plugin-linuxbridge-4 https://zuul.opendev.org/t/openstack/build/8af297e314ad4041ac288e90d978fa60 : SUCCESS in 1h 46m 05s\n- neutron-tempest-plugin-linuxbridge-5 https://zuul.opendev.org/t/openstack/build/9ff453ebf0c141659b9d39a33290dd35 : SUCCESS in 2h 01m 29s","accounts_in_message":[],"_revision_number":6},{"id":"4fd81b414ea432cde6a2f768c53fb5df35e18543","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-11-09 04:44:15.000000000","message":"Abandoned\n\njust for test","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b156c8a6c4ebe7b45904b66fa754a1380ef85630","revisions":{"7a13e34e2406d87891b204442293762a3d474bb5":{"kind":"REWORK","_number":1,"created":"2023-09-26 10:25:36.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/04/896504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/896504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/896504/1"}}},"commit":{"parents":[{"commit":"0ceecd328276cd446c351f0e6f425424f961e3dd","subject":"Merge \"Fix pointless-string-statement pylint warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ceecd328276cd446c351f0e6f425424f961e3dd"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-26 10:25:02.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-26 10:25:05.000000000","tz":330},"subject":"[DNM] Test os-vif revert","message":"[DNM] Test os-vif revert\n\nDepends-On: https://review.opendev.org/c/openstack/os-vif/+/896440\nChange-Id: If32c876c6c17c8786b90e4515af463d5a0e9c25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a13e34e2406d87891b204442293762a3d474bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a13e34e2406d87891b204442293762a3d474bb5"}]},"branch":"refs/heads/master"},"ef8b836079f25a60009c1d641f9b4e65ae98069a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-26 13:02:24.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/04/896504/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/896504/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/896504/2"}}},"commit":{"parents":[{"commit":"0ceecd328276cd446c351f0e6f425424f961e3dd","subject":"Merge \"Fix pointless-string-statement pylint warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ceecd328276cd446c351f0e6f425424f961e3dd"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-26 10:25:02.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-26 13:01:55.000000000","tz":330},"subject":"[DNM] Test os-vif revert","message":"[DNM] Test os-vif revert\n\nDepends-On: https://review.opendev.org/c/openstack/os-vif/+/896440\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/896514\nChange-Id: If32c876c6c17c8786b90e4515af463d5a0e9c25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef8b836079f25a60009c1d641f9b4e65ae98069a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ef8b836079f25a60009c1d641f9b4e65ae98069a"}]},"branch":"refs/heads/master"},"66bfc14ca3dabfb071c4ae7dd43809711880074e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-09-26 14:47:50.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/04/896504/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/896504/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/896504/3"}}},"commit":{"parents":[{"commit":"0ceecd328276cd446c351f0e6f425424f961e3dd","subject":"Merge \"Fix pointless-string-statement pylint warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ceecd328276cd446c351f0e6f425424f961e3dd"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-26 10:25:02.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-26 14:47:26.000000000","tz":330},"subject":"[DNM] Test os-vif revert","message":"[DNM] Test os-vif revert\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/896549\nDepends-On: https://review.opendev.org/c/openstack/os-vif/+/896440\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/896514\nChange-Id: If32c876c6c17c8786b90e4515af463d5a0e9c25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66bfc14ca3dabfb071c4ae7dd43809711880074e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/66bfc14ca3dabfb071c4ae7dd43809711880074e"}]},"branch":"refs/heads/master"},"af879da39e34ab53bd187bb9d1e4a87180955e5a":{"kind":"REWORK","_number":4,"created":"2023-09-29 06:44:02.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/04/896504/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/896504/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/896504/4"}}},"commit":{"parents":[{"commit":"0ceecd328276cd446c351f0e6f425424f961e3dd","subject":"Merge \"Fix pointless-string-statement pylint warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ceecd328276cd446c351f0e6f425424f961e3dd"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-26 10:25:02.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-29 06:42:51.000000000","tz":330},"subject":"[DNM] Test os-vif revert","message":"[DNM] Test os-vif revert\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/896549\nDepends-On: https://review.opendev.org/c/openstack/os-vif/+/896440\nChange-Id: If32c876c6c17c8786b90e4515af463d5a0e9c25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af879da39e34ab53bd187bb9d1e4a87180955e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af879da39e34ab53bd187bb9d1e4a87180955e5a"}]},"branch":"refs/heads/master"},"bc2827528c786f877ff027b5122601ff68f9cb1c":{"kind":"REWORK","_number":5,"created":"2023-09-29 08:56:28.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/04/896504/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/896504/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/896504/5"}}},"commit":{"parents":[{"commit":"0ceecd328276cd446c351f0e6f425424f961e3dd","subject":"Merge \"Fix pointless-string-statement pylint warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ceecd328276cd446c351f0e6f425424f961e3dd"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-26 10:25:02.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-29 08:56:02.000000000","tz":330},"subject":"[DNM] Test os-vif revert","message":"[DNM] Test os-vif revert\n\nDepends-On: https://review.opendev.org/c/openstack/neutron-tempest-plugin/+/896549\nDepends-On: https://review.opendev.org/c/openstack/os-vif/+/896440\nChange-Id: If32c876c6c17c8786b90e4515af463d5a0e9c25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc2827528c786f877ff027b5122601ff68f9cb1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc2827528c786f877ff027b5122601ff68f9cb1c"}]},"branch":"refs/heads/master"},"b156c8a6c4ebe7b45904b66fa754a1380ef85630":{"kind":"REWORK","_number":6,"created":"2023-10-04 07:28:33.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/04/896504/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/04/896504/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/04/896504/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/04/896504/6"}}},"commit":{"parents":[{"commit":"0ceecd328276cd446c351f0e6f425424f961e3dd","subject":"Merge \"Fix pointless-string-statement pylint warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ceecd328276cd446c351f0e6f425424f961e3dd"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-09-26 10:25:02.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-10-04 07:27:46.000000000","tz":330},"subject":"[DNM] Test oslo.log pin","message":"[DNM] Test oslo.log pin\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/897301\nChange-Id: If32c876c6c17c8786b90e4515af463d5a0e9c25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b156c8a6c4ebe7b45904b66fa754a1380ef85630"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b156c8a6c4ebe7b45904b66fa754a1380ef85630"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
