)]}'
{"id":"openstack%2Fneutron~896522","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I18e65d79e56fe5995076eb9166da23fc14c92fc5","project":"openstack/neutron","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-09-26 16:00:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I18e65d79e56fe5995076eb9166da23fc14c92fc5","subject":"[PostgreSQL] Subnet entity with ServiceType grouped by both tables","status":"MERGED","created":"2023-09-26 11:54:11.000000000","updated":"2023-09-26 16:02:19.000000000","submitted":"2023-09-26 16:00:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896522","cherry_pick_of_change":896511,"cherry_pick_of_patch_set":1,"meta_rev_id":"50f21eff35dde4e42a48a9aaccb9ffdc8c898677","_number":896522,"virtual_id_number":896522,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-26 16:00:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-09-26 13:18:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2023-09-26 13:21:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-09-26 13:21:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-26 11:54:11.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2023-09-26 12:33:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-26 13:18:34.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-09-26 13:21:54.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-09-26 13:53:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33e80cc5b9e90248994fbc379d89bd5c41700a01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-09-26 11:54:11.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"3e644db3e124486b506dd1dba431b1a44ccb575c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 12:33:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d92cb53b2d7b488c852902de640aab96\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e792ff02069a44698c41f53ae7150fd9 : SUCCESS in 37m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0260e8d64834989b86c669603922b6e : SUCCESS in 36m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"109eaa9207cec32d613bc83a75100c751e04e3fc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-26 13:18:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0eb2cdae0f1a169041e0c51c19eb1debad37ba22","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-09-26 13:21:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e7be59de898ce21142598319539dffe799d8e62a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 13:53:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3bb8e6715c646d9abbfd9c105b7e994\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/d6265a21e7ae4082af9aa20779c5cb03 : SUCCESS in 47m 53s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/13495247b9b8434f8ef0cda4edc7fb1d : SUCCESS in 1h 13m 12s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/474520144c9d4912bead4cfe8698b708 : SUCCESS in 54m 40s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/89c5f9751a6349bf852c3e7108721c4a : SUCCESS in 1h 15m 56s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/1dbe169e99f842ab9de557441d61715e : SUCCESS in 1h 13m 49s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/955bbf4633c6466ea9f34859fdbf9fb0 : SUCCESS in 31m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f53080ad1524400d9a5a6cb9be920b24 : SUCCESS in 27m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7da8bed11af4c76887281760dc24777 : SUCCESS in 10m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7316079e81e14ad9b6216c239876268e : SUCCESS in 19m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec18f37f05a647fca2e0f052faeef499 : SUCCESS in 24m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dbdc26161ee4bffbb9b392a588e83d2 : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bd3f518e3ec749a79fd1a987ba30a2e3 : SUCCESS in 44m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1eab8f1ebd414274bed956abbea760f7 : SUCCESS in 1h 57m 04s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8ba14ed7e34d4d6da57c94798bae92b3 : SUCCESS in 51m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e81f669791d7411783552eeebbc35c56 : SUCCESS in 41m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e76a76d5b814490d84c26546ccd5a286 : SUCCESS in 59m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/53028456b4ef41f5b2eb23bbcad05b12 : SUCCESS in 1h 14m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f5f366ff9bf04806bc2ebcea0f17a5ef : SUCCESS in 1h 01m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f258530d708043b4b93b6b625328bfb8 : SUCCESS in 1h 06m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aafc417df6624014b79303ff69375f54 : SUCCESS in 1h 04m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e16b194910409dc97fbb795bf95b10169f54dd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 13:53:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b0f10688bb96278c59cf4126decff6969ef8b04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 16:00:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10d2c6e2837840ee9a7cf5fb5fa36897\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/f60d294dc6c24790af16cc2074a27f88 : SUCCESS in 44m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e36d2c41fa1c4110b0c7742939a87f62 : SUCCESS in 11m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/452fa46b10c9413b8492789bc3e13ca2 : SUCCESS in 25m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b143737a82de4fb68047b009e3810a04 : SUCCESS in 25m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4270c992c1964be5b3c5b950286d3a66 : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6e966a34124450a8f2d572177dd8468 : SUCCESS in 48m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/805a9f79d57349aab6edeea82669d797 : SUCCESS in 1h 27m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/86395e6f1e8f48cdac105473bc35a58a : SUCCESS in 1h 17m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a494836d17d546908965a432f3601389 : SUCCESS in 1h 03m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e19d570b72d241bda0a58ba82b568506 : SUCCESS in 1h 04m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a5f549b1f384e2a8938b4e7548837e4 : SUCCESS in 40m 27s","accounts_in_message":[],"_revision_number":1},{"id":"77a90ff528a8ed4dfa3c85af4ddb4e0e7109ffb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 16:00:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"50f21eff35dde4e42a48a9aaccb9ffdc8c898677","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 16:02:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2838ad92f534accbdd5b454194ca4a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf43113347e045dfb3ce25ecffbd6c1a : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4efc2dd80f1628e794e761d2d3e9e6c27e992b5","revisions":{"a4efc2dd80f1628e794e761d2d3e9e6c27e992b5":{"kind":"REWORK","_number":1,"created":"2023-09-26 11:54:11.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/22/896522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/22/896522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/896522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/896522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/22/896522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/22/896522/1"}}},"commit":{"parents":[{"commit":"edf33d2217f73b8775c76f3cac2fc75ce00632a1","subject":"Add 3 secs to wait for keepalived state change","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/edf33d2217f73b8775c76f3cac2fc75ce00632a1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-07-18 02:38:47.000000000","tz":0},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2023-09-26 11:54:11.000000000","tz":0},"subject":"[PostgreSQL] Subnet entity with ServiceType grouped by both tables","message":"[PostgreSQL] Subnet entity with ServiceType grouped by both tables\n\nThe SQL clause \"GROUP BY\" in PostgreSQL requires the presence of all\nfields provided in the \"SELECT\" clause; this is not happening in\nMariaDB not MySQL.\n\nSince [1] (and the related patch [2] that are part of the same bug\nfix), when a resource with RBAC registers is selected, the \"GROUP BY\"\nclause is added to reduce the number of returned registers. The\nfiled used is \"id\", present in all the RBAC controlled resources\n(\u0027network\u0027, \u0027qospolicy\u0027, \u0027securitygroup\u0027, \u0027addressscope\u0027,\n\u0027subnetpool\u0027, \u0027addressgroup\u0027). That is in opposition to what was\nstated in the first paragraph \"requires the presence of all\nfields provided in the \"SELECT\" clause\". However it is possible\nto group only by a primary key [3].\n\nIn [4] a prior change was introduced that modifies the \"Subnet\" entity\nto include a child table \"SubnetServiceType\". That introduces two new\nfields (\"SubnetServiceType\" fields) that are not present in the\n\"GROUP BY\" clause, causing the reported error in the LP bug.\n\nThis patch is adding these two fields (that form the composite primary\nkey of \"SubnetServiceType\" table) to the \"GROUP BY\" clause, when\nneeded; if the query is executed by an administrator, the RBAC query\nwon\u0027t be performed (an administrator has always permissions) and the\n\"GROUP BY\" clause won\u0027t be issued.\n\n[1]https://review.opendev.org/c/openstack/neutron-lib/+/884878\n[2]https://review.opendev.org/c/openstack/neutron/+/884877\n[3]https://learnsql.com/blog/must-appear-in-group-by-clause/\n[4]https://review.opendev.org/c/openstack/neutron/+/744512\n\nCloses-Bug: #2028003\nChange-Id: I18e65d79e56fe5995076eb9166da23fc14c92fc5\n(cherry picked from commit c831771053c29521f7e6e3a17997201f695925ec)\n(cherry picked from commit 96267a25821f1fa2117bbc364012da7d1de8bf0a)\nConflicts:\n\tneutron/objects/subnet.py\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4efc2dd80f1628e794e761d2d3e9e6c27e992b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a4efc2dd80f1628e794e761d2d3e9e6c27e992b5"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
