)]}'
{"id":"openstack%2Fneutron~896544","triplet_id":"openstack%2Fneutron~master~I9771bd4f76a9ec073afeeb80a787832102446cd6","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-10-11 00:17:02.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-10-11 01:59:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9771bd4f76a9ec073afeeb80a787832102446cd6","subject":"Add has_lock_periodic decorator for OVN Maintenance","status":"MERGED","created":"2023-09-26 14:19:18.000000000","updated":"2023-10-11 02:00:45.000000000","submitted":"2023-10-11 01:59:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"896544","meta_rev_id":"1bb336af03484b866a2a01fb464bfb213cc91cac","_number":896544,"virtual_id_number":896544,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2023-09-26 17:02:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 01:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"recommended":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-09 21:52:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":1,"date":"2023-10-09 07:39:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-10-11 00:17:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-26 14:46:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-10-11 00:17:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-26 14:46:08.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2023-09-26 14:55:34.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-26 16:06:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-26 16:47:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-26 18:56:54.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"},{"updated":"2023-10-09 07:39:36.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-09 21:52:51.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2023-10-11 00:17:02.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"cc20627ff16326aad53e89f6ab15ec8ec4255658","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-26 14:19:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0455f298383dcbf5866dafb1234f094e815ae31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-26 14:27:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c04c25336bba1237582029d67c23b70a91973350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-26 14:40:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"47a2789e5b3ece24553f4423344490966720d04c","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-26 14:46:08.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6298dabeb7bb0d69180d7ce11da3672c422db76e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-26 14:55:34.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/44/896544/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"bbc2062994de66dce5c6754da09c9d4a1c429c54","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-26 15:59:39.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/44/896544/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7cf7b41bf2c2b44fd89f69854b5185b64f96ec42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 16:06:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3170193b3cc34d2d9efd90144051fba3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb2eca3757c84c269893171b97a6f9ce : SUCCESS in 40m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4e935910d7840d99a752f3eaa7f69c0 : SUCCESS in 46m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89f7853151ea469ead73099dd90331ac : SUCCESS in 52m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5672e3f2762c45bea2954d2776ae926e : SUCCESS in 36m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e39894352f760156dde91e59e40a95e1f937acba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 16:47:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a5861d83e3d428d87084cd46ed3ca84\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7ca45ff6279941779bee0f1e57f6faa2 : SUCCESS in 1h 20m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5abe3f4ad284ca4b79cd9beeac00317 : SUCCESS in 37m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90da97c4378749fb872a8330302e6871 : SUCCESS in 11m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/000723db74eb4923af59ae68d7214f31 : SUCCESS in 27m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c846bc480d04e44b09102d93b7d7e30 : SUCCESS in 26m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad1706f4701147a8a1506a94e93d032a : SUCCESS in 27m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a7e5de4ca6f9409db2f6351c16a07349 : SUCCESS in 23m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee3f7d834c6d4f39b3419ca03c15e818 : SUCCESS in 11m 24s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/34ab44cbf69a4909b6612558f2092c80 : SUCCESS in 29m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3741f4c383d6443a8636878a3ccbaed4 : FAILURE in 47m 31s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/40c927f12eba44768964fc456f5f8e43 : SUCCESS in 56m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1679b54688794f8ea32bd1759a26e274 : SUCCESS in 2h 01m 08s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d722fdeb286344bc926f22ac617f648d : SUCCESS in 1h 09m 07s","accounts_in_message":[],"_revision_number":3},{"id":"496283bf5316c700b38559cf7835eabf1e7e4184","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-26 17:02:00.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/44/896544/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"855eeac1d5e4421513a9576c83fd77c126838d8b","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-26 18:56:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"84fb9a91444da2a68c4ba6b26b7de05f19d15aff","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-26 19:21:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5cc7e527c2a64210732a5a877f4875170708bf90","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-26 19:31:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ee1d0229fd27144a54d55aa6790548fa6ed15f3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 21:30:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/632b7d14d8e64fa6a3cf851abefd68f3\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/de1413f1fee84b44adc77c748b0e2b3c : SUCCESS in 1h 51m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51377263d9d6433abf2c3012ef1809de : SUCCESS in 36m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3061cd09bba94398986192d6c3d617a4 : SUCCESS in 10m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/896d9a7c8e654eb2937422d97b5789f7 : SUCCESS in 23m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0037564f028247dea59ced0966e6aa0a : SUCCESS in 24m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56f9c1cd913a48d88739a98616944714 : SUCCESS in 21m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d23c9151f9924458b77661c4c83ad467 : SUCCESS in 19m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9689f28375949a4a6fcff9fcdaecfff : SUCCESS in 10m 35s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/b107b24e379c4c9b99b61be212603bb7 : SUCCESS in 25m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7a06ea64dddc4a50ad24900ea1c835fa : FAILURE in 42m 30s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/d99cc58d05bb4bd38eabb614035a70d0 : SUCCESS in 54m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b5a405248c944bdc865d1ba4b19b8731 : SUCCESS in 1h 21m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/f7678b45f7254f2095e36ddd97dbbb4d : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":3},{"id":"149ad2381dba2c5e47ea932cf18735085a716006","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 21:57:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/444cec4e54ab4d859512acae84d1a03e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13666864a3684688a65aa53916821643 : SUCCESS in 55m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e20a0cf3804464c9a21073755207ecc : SUCCESS in 51m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/684ec06c52b04a1daa10b187361853ce : SUCCESS in 44m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c3a4adc46eed4868924eab1b15080d5d : SUCCESS in 38m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"51053bc6000279f21e95d5c5f570e0d82b3902bc","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2023-09-27 13:46:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4d97d1ce9d6aa689cfa91cb29bd20421ca38253d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 16:06:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1618a6b84fee4a0b9847d9f14ec3445e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8af0fa25facc423c88662e4c73749624 : SUCCESS in 55m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f91af00ac7ff4218bf7d3e2682e26046 : SUCCESS in 57m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/35848818f6564481a4f3495688fc6516 : SUCCESS in 55m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eb6c108ba492462a99b0fa2597ecaca3 : SUCCESS in 36m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fd1628df922f139d7a654084340014679d650696","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 16:11:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/793899560df1490aabca42bcdd5d7560\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/dbf6d41bad56456786aa080c2f490e93 : SUCCESS in 2h 22m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d8e905c331e4345857b43a046527c4a : SUCCESS in 40m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2708dc5d392e48c9bdca5246edd82920 : SUCCESS in 11m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b77a6783b2df4a778af0b4afa1215c3e : SUCCESS in 25m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a6302710a1448f3a48e40f15389e923 : SUCCESS in 25m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eede5d30be474e0187d025bd625d1fa7 : SUCCESS in 25m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/861bdc52927c4bed85d4f80da60c2e9b : SUCCESS in 20m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b6c5707d1845fc943cd76713ecd4bc : SUCCESS in 13m 24s\n- openstack-tox-py310-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/f455d42b2e7346d68e1364d78fb314f9 : SUCCESS in 27m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/eae1e188e6a64563835aea588a36ea74 : SUCCESS in 51m 19s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/9522b780d89e4ffcb9959c3c99ad39f6 : SUCCESS in 1h 10m 44s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4f6a8f6fbab64d08b1fed7af9a423974 : SUCCESS in 1h 20m 53s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/5f1ea7bf5ba54c9e96a766e31c6d8fd6 : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":3},{"id":"858f16b70cbe220fa09f5ef7e1f64ec06906db24","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-10-09 07:39:36.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"af702a18b42fbbe72d98ffe98ef3fdc22c41dae8","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2023-10-09 21:52:51.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f180db6338fe0deab5b11f12e55656fb9473c127","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-11 00:17:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e5f3a0b15b3fdc35bb855856544585f7f2c23f64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 00:17:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"afe24af2610b396bcffc49d2850eb89760974242","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 01:59:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51daa28bc3a9410eb188e249b49c796a\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bbc7f62cb33f4f8798efd24093aeed99 : SUCCESS in 1h 40m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d4149b02b914c55ab39fcd3e9e80aab : SUCCESS in 8m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e772e3c40c834cd4a50a0d74bed9d6d6 : SUCCESS in 19m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d023e5d086e43bfa2503de2f18921a4 : SUCCESS in 27m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fdc1ce30412c4fb48ad2b2000299d375 : SUCCESS in 18m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af6d7677344b492abb2e3fb9d506fe5e : SUCCESS in 11m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/07c41f9e804f49aaa05f4b6531c89711 : SUCCESS in 46m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/292601cebaa64b919c139e2836031e30 : SUCCESS in 57m 49s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0af886426858496b9080346b93ae56a2 : SUCCESS in 52m 24s","accounts_in_message":[],"_revision_number":3},{"id":"7582a035d8a29132f16a52d15793f3de7dd39412","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 01:59:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1bb336af03484b866a2a01fb464bfb213cc91cac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 02:00:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fc1a86e63364fba96311a04446fe34f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34c5141204b44b05813e20a6f1e4d335 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1d7e99bc0f10509166b240ad503e649fef4ebab3","revisions":{"902dd5e6ccbf1c9160b075305df9a6a22fbd0bb4":{"kind":"REWORK","_number":1,"created":"2023-09-26 14:19:18.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/44/896544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/896544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/896544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/896544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/896544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/896544/1"}}},"commit":{"parents":[{"commit":"0ceecd328276cd446c351f0e6f425424f961e3dd","subject":"Merge \"Fix pointless-string-statement pylint warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ceecd328276cd446c351f0e6f425424f961e3dd"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-26 14:15:51.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-26 14:15:51.000000000","tz":-300},"subject":"Add has_lock_periodic decorator for OVN Maintenance","message":"Add has_lock_periodic decorator for OVN Maintenance\n\nAlmost all of the periodic methods in ml2/OVN\u0027s maintenance code\nexit early if the IDL lock is not held. This adds a decorator\nthat wraps the futurist periodic that exits if the lock isn\u0027t held.\n\nThere are still a few methods that exit raising NeverAgain if\ncertain features aren\u0027t available, and those have been left alone\nfor now.\n\nChange-Id: I9771bd4f76a9ec073afeeb80a787832102446cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/902dd5e6ccbf1c9160b075305df9a6a22fbd0bb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/902dd5e6ccbf1c9160b075305df9a6a22fbd0bb4"}]},"branch":"refs/heads/master"},"19d3449086a2d204e542537b301106711c47f9dd":{"kind":"REWORK","_number":2,"created":"2023-09-26 14:27:24.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/44/896544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/896544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/896544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/896544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/896544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/896544/2"}}},"commit":{"parents":[{"commit":"0ceecd328276cd446c351f0e6f425424f961e3dd","subject":"Merge \"Fix pointless-string-statement pylint warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ceecd328276cd446c351f0e6f425424f961e3dd"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-26 14:15:51.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-26 14:27:17.000000000","tz":-300},"subject":"Add has_lock_periodic decorator for OVN Maintenance","message":"Add has_lock_periodic decorator for OVN Maintenance\n\nAlmost all of the periodic methods in ml2/OVN\u0027s maintenance code\nexit early if the IDL lock is not held. This adds a decorator\nthat wraps the futurist periodic that exits if the lock isn\u0027t held.\n\nThere are still a few methods that exit raising NeverAgain if\ncertain features aren\u0027t available, and those have been left alone\nfor now.\n\nChange-Id: I9771bd4f76a9ec073afeeb80a787832102446cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19d3449086a2d204e542537b301106711c47f9dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/19d3449086a2d204e542537b301106711c47f9dd"}]},"branch":"refs/heads/master"},"1d7e99bc0f10509166b240ad503e649fef4ebab3":{"kind":"REWORK","_number":3,"created":"2023-09-26 14:40:41.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/44/896544/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/896544/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/896544/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/896544/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/896544/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/896544/3"}}},"commit":{"parents":[{"commit":"0ceecd328276cd446c351f0e6f425424f961e3dd","subject":"Merge \"Fix pointless-string-statement pylint warning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ceecd328276cd446c351f0e6f425424f961e3dd"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-26 14:15:51.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2023-09-26 14:38:36.000000000","tz":-300},"subject":"Add has_lock_periodic decorator for OVN Maintenance","message":"Add has_lock_periodic decorator for OVN Maintenance\n\nAlmost all of the periodic methods in ml2/OVN\u0027s maintenance code\nexit early if the IDL lock is not held. This adds a decorator\nthat wraps the futurist periodic that exits if the lock isn\u0027t held.\n\nThere are still a few methods that exit raising NeverAgain if\ncertain features aren\u0027t available, and those have been left alone\nfor now.\n\nChange-Id: I9771bd4f76a9ec073afeeb80a787832102446cd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d7e99bc0f10509166b240ad503e649fef4ebab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1d7e99bc0f10509166b240ad503e649fef4ebab3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
