)]}'
{"id":"openstack%2Fneutron~896728","triplet_id":"openstack%2Fneutron~master~Ic2074bef53ec6c1d311a18203f3f0612b7e95a6b","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-10-13 18:17:16.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-10-13 17:27:46.000000000","reason":"\u003cGERRIT_ACCOUNT_1131\u003e replied on the change","reason_account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-10-13 18:17:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic2074bef53ec6c1d311a18203f3f0612b7e95a6b","subject":"functional: Enable test_fip_connection_for_address_scope test","status":"MERGED","created":"2023-09-27 20:54:02.000000000","updated":"2023-10-13 18:19:02.000000000","submitted":"2023-10-13 18:17:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896728","meta_rev_id":"cfd54fdc27c6fab16f28e5a2bd2116a8d07b4d0e","_number":896728,"virtual_id_number":896728,"owner":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-10-04 21:14:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-13 18:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-10-13 11:26:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2023-10-13 17:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2023-10-13 17:27:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-27 21:25:35.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-27 21:45:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-28 01:04:55.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2023-09-28 07:10:52.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2023-10-13 11:26:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"c7b69572497fa373640ac98c0e52b3e09f91f86c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-09-27 20:54:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ae3fc0cc036d6741340800d9d9d97dc26626f8b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-09-27 21:25:35.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2023/Arista_Tempest_Testrefs/changes/28/896728/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"33ad86c23e80a40d95b93474f503a825e26feac4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 21:45:44.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0739dddeb6b4002a89bcbf54c3fddd8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6155d2d7d26244edbc4a12ef2e381a52 : FAILURE in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66eac83ec13847baaf931ae2b952062f : SUCCESS in 12m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fc1ef8c9bd1e456493ead7447676bf30 : SUCCESS in 50m 46s","accounts_in_message":[],"_revision_number":1},{"id":"53699b731b5f8db40dc51f2a02cb9af1a2422ee9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-28 07:10:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"92891a8ee9890c72382b2cceb84ab896923c85c5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-09-28 07:11:20.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d738fdb2c3b7318fb94c900dbcfd12636e112d63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-04 18:58:51.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1, Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"47aa4756177e04c22872796e0b019d2e383af2d0","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2023-10-04 18:59:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0cf3a2a72cc17b57aa73c08ccd04c20df1e6675a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 20:00:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1925353c04f4fc38e431c5fe0b23e71\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae59f3f00851414583d5a458afbb26a2 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c009694164433ab036d04c4fdd6d79 : SUCCESS in 12m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5b2238b8bc924c48b11e2940e7c58249 : FAILURE in 49m 26s","accounts_in_message":[],"_revision_number":2},{"id":"0898dabf0594dfe65f995e904ca341fd01ba77bf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2023-10-04 21:14:28.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/October-2023/Arista_Tempest_Testrefs/changes/28/896728/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"dd6de8059cc4932c99caaef32f01fe2575f4b715","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-10-13 11:26:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c4597b2f5805ed2e60b2d008b4c84531c2487da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 12:14:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5d23e0f9c2f4363bb6919c78913eaba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da03bffbf03b415f9734bfe519a25166 : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87cfdcb082434fb096ddba70a36b5073 : SUCCESS in 10m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e0dcdfaaa5494a5ba2ea52327d77782d : SUCCESS in 46m 04s","accounts_in_message":[],"_revision_number":2},{"id":"f61c70921ff1593f7b39a0937e1b0f9e4026cb8a","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-10-13 17:27:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ac373d61a0626a80e2a44593f06b1baa332afe1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 17:28:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ad7784200ca4a6e42a299b84e40d51a296256dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 18:17:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2776cc3a5ec441e8bffcbf1cbb65dfc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36679039d00d435db4d35fd9127c74f4 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a979d9abb9e541868b40793aa2d0746d : SUCCESS in 10m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0f86288c783b4b3ba545090f1acc5a97 : SUCCESS in 47m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c75c20f99de8a7fbd573f3d8908aa723c001385f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 18:17:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cfd54fdc27c6fab16f28e5a2bd2116a8d07b4d0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 18:19:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6afab88cddc40399d25a28190d565ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc1b0204f7e346f0814610c090945b03 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e274d676295c864bdc36e459790a9082e468d12","revisions":{"d94891bd8153b7cab74b9d7fa82a83b65d8cec9a":{"kind":"REWORK","_number":1,"created":"2023-09-27 20:54:02.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/28/896728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/896728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/896728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/896728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/896728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/896728/1"}}},"commit":{"parents":[{"commit":"61ac046cf7ed81a7502ddfdfb6fa785f4184554f","subject":"Merge \"contributor docs: Architectural overview for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61ac046cf7ed81a7502ddfdfb6fa785f4184554f"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-27 20:52:23.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-27 20:53:25.000000000","tz":0},"subject":"functional: Enable test_fip_connection_for_address_scope test","message":"functional: Enable test_fip_connection_for_address_scope test\n\nThe test was marked as skipped because of a known issue. The issue was\nfixed 6 years ago.\n\nTrivialFix\n\nChange-Id: Ic2074bef53ec6c1d311a18203f3f0612b7e95a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d94891bd8153b7cab74b9d7fa82a83b65d8cec9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d94891bd8153b7cab74b9d7fa82a83b65d8cec9a"}]},"branch":"refs/heads/master"},"1e274d676295c864bdc36e459790a9082e468d12":{"kind":"REWORK","_number":2,"created":"2023-10-04 18:58:51.000000000","uploader":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"ref":"refs/changes/28/896728/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/28/896728/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/896728/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/896728/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/28/896728/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/28/896728/2"}}},"commit":{"parents":[{"commit":"61ac046cf7ed81a7502ddfdfb6fa785f4184554f","subject":"Merge \"contributor docs: Architectural overview for metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61ac046cf7ed81a7502ddfdfb6fa785f4184554f"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-09-27 20:52:23.000000000","tz":0},"committer":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2023-10-04 18:58:51.000000000","tz":0},"subject":"functional: Enable test_fip_connection_for_address_scope test","message":"functional: Enable test_fip_connection_for_address_scope test\n\nThe test was marked as skipped because of a known issue. The issue was\nfixed 6 years ago.\n\nTrivialFix\n\nChange-Id: Ic2074bef53ec6c1d311a18203f3f0612b7e95a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e274d676295c864bdc36e459790a9082e468d12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e274d676295c864bdc36e459790a9082e468d12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
