)]}'
{"id":"openstack%2Fneutron~896799","triplet_id":"openstack%2Fneutron~master~I803a34c5b55530f889ef6938d64f64ad2133b89d","project":"openstack/neutron","branch":"master","topic":"os-vif-pin","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-09-28 16:05:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I803a34c5b55530f889ef6938d64f64ad2133b89d","subject":"DNM: Pin os-vif to \u003c3.2.0","status":"ABANDONED","created":"2023-09-28 14:26:30.000000000","updated":"2023-09-28 16:05:09.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6b23abc7fe7766c0bf6935d7790d4344efd93005","_number":896799,"virtual_id_number":896799,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-28 14:55:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 14:26:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-28 14:27:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-28 14:27:13.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"9e96ac3f5b49d01ac15a1ee1675246dc6df66a2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-28 14:26:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bd2f2771b9af2f6b1adac809ef3bc543c47e3bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 14:26:55.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/neutron on branch master.  The error was:\n  \n    found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 21, column 31","accounts_in_message":[],"_revision_number":1},{"id":"84354244d454dc99228a16d28368371eda05afba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 14:27:04.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  \n  Zuul encountered a syntax error while parsing its configuration in the\n  repo openstack/neutron on branch master.  The error was:\n  \n    found undefined alias\n    in \"openstack/neutron/zuul.d/project.yaml@master\", line 21, column 31","accounts_in_message":[],"_revision_number":1},{"id":"569d4f020522d834c9c6cf76bf38aa0c25c2d7e6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-09-28 14:27:13.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/neutron on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/neutron/zuul.d/project.yaml@master\", line 21, column 31","accounts_in_message":[],"_revision_number":1},{"id":"e84be191e43a09a77626a15e2ef78af778c11e2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-28 14:36:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ca3572ef2aef16b27b3ba55625dddecadd518389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 14:55:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d28c2b3f86584d638c346c068840520e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/7cfc0db3362f4845af7687485a40cca9 : FAILURE in 13m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dbc208a091c1453b99a93ad228e6d7cd : FAILURE in 14m 21s\n- neutron-tempest-plugin-openvswitch-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ca18cd5c7a8e499e89f31afb2af65703 : FAILURE in 13m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3dafe7c2e47b44feaed13662ad8fafc5 : FAILURE in 14m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/79cb93d5ea114c3ab449a47738eb7c90 : FAILURE in 13m 57s\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f9aea19644b243e2966a6c6a9256314f : FAILURE in 13m 51s","accounts_in_message":[],"_revision_number":2},{"id":"6b23abc7fe7766c0bf6935d7790d4344efd93005","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2023-09-28 16:05:09.000000000","message":"Abandoned\n\ndoesn\u0027t work","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"799810f007064c370e4fd95fe29c06c47402ec64","revisions":{"4ca09641b66486bcaa6840c9368b780b8df63e9f":{"kind":"REWORK","_number":1,"created":"2023-09-28 14:26:30.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/99/896799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/896799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/896799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/896799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/896799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/896799/1"}}},"commit":{"parents":[{"commit":"5ba85194285d984c34381ae9a6b576de998216ed","subject":"Merge \"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ba85194285d984c34381ae9a6b576de998216ed"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-09-28 14:25:54.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-09-28 14:25:54.000000000","tz":-240},"subject":"DNM: Pin os-vif to \u003c3.2.0","message":"DNM: Pin os-vif to \u003c3.2.0\n\nLet\u0027s see if this helps with bug #2037239\n\nChange-Id: I803a34c5b55530f889ef6938d64f64ad2133b89d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ca09641b66486bcaa6840c9368b780b8df63e9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4ca09641b66486bcaa6840c9368b780b8df63e9f"}]},"branch":"refs/heads/master"},"799810f007064c370e4fd95fe29c06c47402ec64":{"kind":"REWORK","_number":2,"created":"2023-09-28 14:36:52.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/99/896799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/99/896799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/896799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/896799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/99/896799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/99/896799/2"}}},"commit":{"parents":[{"commit":"5ba85194285d984c34381ae9a6b576de998216ed","subject":"Merge \"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5ba85194285d984c34381ae9a6b576de998216ed"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-09-28 14:25:54.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2023-09-28 14:36:38.000000000","tz":-240},"subject":"DNM: Pin os-vif to \u003c3.2.0","message":"DNM: Pin os-vif to \u003c3.2.0\n\nLet\u0027s see if this helps with bug #2037239\n\nChange-Id: I803a34c5b55530f889ef6938d64f64ad2133b89d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/799810f007064c370e4fd95fe29c06c47402ec64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/799810f007064c370e4fd95fe29c06c47402ec64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
