)]}'
{"id":"openstack%2Fneutron~896803","triplet_id":"openstack%2Fneutron~stable%2F2023.2~Ide3204748274cbab9731310c2e2bd8bd8c9a53ff","project":"openstack/neutron","branch":"stable/2023.2","topic":"bug/2024160","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-29 15:40:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide3204748274cbab9731310c2e2bd8bd8c9a53ff","subject":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent","status":"MERGED","created":"2023-09-28 14:36:44.000000000","updated":"2023-09-29 15:44:58.000000000","submitted":"2023-09-29 15:40:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896803-bug/2024160","meta_rev_id":"6a8ee52464d55240054b5dfe1016f56e156ef429","_number":896803,"virtual_id_number":896803,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-29 15:40:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2023-09-29 11:34:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2023-09-29 09:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2023-09-29 11:34:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 14:36:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"CC"},{"updated":"2023-09-28 16:50:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-28 17:12:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-29 09:52:39.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2023-09-29 11:33:13.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2023-09-29 11:34:10.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"57cea2cc9b91ba6df5f995b29c40ba401e9242b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-28 14:36:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1f4d5e2ae10902080c9292a920016f23f5e8b94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 16:50:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e310b30f39546019f98cb000dc39e27\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70ce9b0a170041c3ae79d9c70b6a57fd : SUCCESS in 39m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9229a769c32a44c583a0b76866653853 : SUCCESS in 43m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a9f22a41fdbc44998542d1580b5c2ec8 : SUCCESS in 41m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8b2302a0a5f9431489e09be8304fbbca : SUCCESS in 40m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b158c2e6a7e2ebd8d7f1f91c78b4e4760d4a6676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 17:12:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aada2690afb7448fa8d00d2b1bdda693\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a67608fb7e1246baa5b004abd4f09b38 : SUCCESS in 2h 15m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53fd84b2c6d2480eb0297041b795be78 : SUCCESS in 34m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2436343fc4e4dbcabab6ac6d335438e : SUCCESS in 9m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75d518e197ab4957a34b9e6a60d5a634 : SUCCESS in 31m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af060cc29ebb48aca12dfb9c96b3f09d : SUCCESS in 26m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4abcaf8b989e440ea180fbaf1a6118cb : SUCCESS in 28m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5de03ff0662d466698b74e3111823d07 : SUCCESS in 26m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fe0759a6a0a48f0b0d0a6f6792f6e95 : SUCCESS in 11m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/89c4023be1a44f61b03a6fb112017901 : FAILURE in 57m 45s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/b134e973480e40dba07de0c41dabb2d3 : SUCCESS in 56m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2eefef247d4d400b9ae25ceb2757065b : SUCCESS in 2h 27m 15s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/925a3567e0354008a1370529fa345651 : SUCCESS in 1h 19m 33s","accounts_in_message":[],"_revision_number":1},{"id":"515299e410bba826110d7e526de5959b202c71fc","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-29 09:01:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d2e9d2bb212d35351335dc00590468cc5da7d55","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-09-29 09:52:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"491711dd1d18c614750dfe95c258ab749788e5b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:24:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc67bafedbba4a32adc5a9a37b80c39e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a88a6f92ac142889629e752d068f352 : TIMED_OUT in 1h 21m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70dabbf56fb648f3b2f73d30c51c7289 : SUCCESS in 1h 21m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c128bbd9e0c419fb4119c963392cbed : SUCCESS in 1h 18m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0e8d03f104074fcaabea0d1e7e9058e4 : SUCCESS in 1h 07m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c8f3b1a0951e96d4c7e5097d6199aa93d826ce8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:56:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8df2695fcee04509bcf9e9f759534cd6\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/24a2472a59f748f793feb665851cbcab : SUCCESS in 1h 53m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e972f639d32845ec88767d3bde73facc : SUCCESS in 39m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d81bf66c8c34937a393bab5143c81df : SUCCESS in 11m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f11eb57164ce40debbc493412c84a9c1 : SUCCESS in 24m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a36e80fd1186415987f170261373da5d : SUCCESS in 22m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fc2fb3daa4e44c38d44dbbe0a92753f : SUCCESS in 26m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9af1042fec37485f88a293dc0077212c : SUCCESS in 27m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b486cbc09644a66a8c624c54e5e8b8c : SUCCESS in 10m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2a14aca9de064c7e837811c10c736130 : FAILURE in 57m 29s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/98925f17636444fc8abb5fb92559432c : SUCCESS in 1h 00m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fb04cc1bc00c46349127d9e1f57d39f9 : SUCCESS in 1h 29m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/3e6ad497c8e74b6f9ec42651be5de6ef : SUCCESS in 1h 00m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4f95356779bd52e5c4cde02cc5ee693cb93cc5d8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-09-29 11:33:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29ef5ff7e988303ad35dc63f2fc95f035d00541c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-09-29 11:34:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b124c035d8a44160d698510b14b7c891a8917fee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 13:32:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01114289df7f4cfb96d9d8a8ffec0db6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec664fe3c98942b9b8dae45a7c31bb79 : SUCCESS in 57m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/454c44d756204edeb16a626e5b67c452 : SUCCESS in 1h 01m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/436206d404824949b191fab6417f48e2 : SUCCESS in 59m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/544ac4817ac34843a74f56e70809f55a : SUCCESS in 38m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ee3bef8c8d2e0a07495c93e4bd5f411274ee89d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 13:38:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c321393b4ec34667beffb48c995ae396\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bf66a338dc534d5db973ae820e1358a4 : SUCCESS in 2h 02m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f182327324e4020ba6ece13a02c3237 : SUCCESS in 33m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27b5cdcf099439a910a0bf51a0d71e2 : SUCCESS in 12m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08bea978681f467c95f34260fc71ca78 : SUCCESS in 21m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25b64ecac1534e98bea0196224bda287 : SUCCESS in 23m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0279d400bb7441b4ade710676467a5b8 : SUCCESS in 19m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4686c8855e2540f5974f7cf84317dfb1 : SUCCESS in 24m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c17d7157e143ea8bd42f26375efcad : SUCCESS in 10m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/103d378a20b447c7bb271189c98ba860 : SUCCESS in 48m 55s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/fa70a2c6a3c64d3d8477df56a6ba2924 : SUCCESS in 51m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/76f6df4702f242758ac02308d5b46363 : SUCCESS in 1h 33m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ad6f50f5525d4802b0c01a09eb72c86c : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e3bae86bc8b1ff9fd64fc9f09850e1893789ef4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 13:38:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62ceb71a81afdd06b1c81db26664b1b2e2d3d7b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 15:40:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc99dee4782d4e9fa161435124b94b49\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/66abfb054f9e4b6996ec21401e01de21 : SUCCESS in 1h 52m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67bcd368e38a4071a884c3d5903b6584 : SUCCESS in 11m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/268ebf7578854348a90b5c2d2b57ee11 : SUCCESS in 27m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d065f73069a8497685058c29d6c3b232 : SUCCESS in 28m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/849deaa1d58e4233af2d6e6b9e64a424 : SUCCESS in 25m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8a7285ae1e5411f9c24310bf130e480 : SUCCESS in 13m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b77b0e3c4b34cc68e663d7ce76524ed : SUCCESS in 1h 00m 42s\n- neutron-ovn-tempest-ipv6-only-ovs-release https://zuul.opendev.org/t/openstack/build/4ca95bed7e9e41b897abafa1201d4f3e : SUCCESS in 50m 01s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1c4c35f8baae4200aaed9aa1077d343d : SUCCESS in 1h 41m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ec0440d6f1df3b29d0dc25eb5fc3b0df9f9a9555","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 15:40:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a8ee52464d55240054b5dfe1016f56e156ef429","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 15:44:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/924b1630b7af4a53b252bd42604de047\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed9bd2fc5f7a4a9a9bb653e4e0de9073 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04e9b063ad526778ba80438fc851e03fa16e5085","revisions":{"04e9b063ad526778ba80438fc851e03fa16e5085":{"kind":"REWORK","_number":1,"created":"2023-09-28 14:36:44.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/03/896803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/896803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/896803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/896803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/896803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/896803/1"}}},"commit":{"parents":[{"commit":"d28bcf0fe1648cfdded5e3bc38200b6d1315f42f","subject":"Revert \"[OVN][Trunk] Add port binding info on subport when parent is bound\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d28bcf0fe1648cfdded5e3bc38200b6d1315f42f"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2023-09-26 20:02:55.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2023-09-28 14:36:28.000000000","tz":0},"subject":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent","message":"[OVN] Match LSP_TYPE_VIRTUAL in PortBindingUpdateVirtualPortsEvent\n\nWith newer versions of OVN, port bindings are removed and added\nfor both virtual ports and container ports (aka trunk ports in\nNeutron world)[1][2].\n\nWith this, Neutron recently landed a change in order to handle\nvirtual ports for the update/delete steps[3], however, this has\nresulted in unexpected behaviour where the binding information\nis wiped for trunk ports, then subsequently removed on the\nnext maintenance run.\n\nWe should make sure that we only have row deletes for\nvirtual ports specifically rather than all ports.\n\n[1]: https://github.com/ovn-org/ovn/commit/ec933537f9b04b35ef4b79fb2b4743f9095da209\n[2]: https://github.com/ovn-org/ovn/commit/28d40c0a024227c9c845117087b7594723ff0ec7\n[3]: https://review.opendev.org/c/openstack/neutron/+/883681\n\nRelated-Bug: #2024160\nChange-Id: Ide3204748274cbab9731310c2e2bd8bd8c9a53ff\n(cherry picked from commit 1160aaa4716904fdba06a8dcfbc1d5a2ce419e72)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04e9b063ad526778ba80438fc851e03fa16e5085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04e9b063ad526778ba80438fc851e03fa16e5085"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
